)]}'
{"id":"openstack%2Fswift~852602","triplet_id":"openstack%2Fswift~master~Id3b29107a5b3331c7078fb5e63f08e286a7371f9","project":"openstack/swift","branch":"master","topic":"better-invalid-path-errors","attention_set":{},"removed_from_attention_set":{"26274":{"account":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"last_update":"2022-09-08 10:02:54.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-09-08 10:02:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id3b29107a5b3331c7078fb5e63f08e286a7371f9","subject":"s3api: return InvalidURI (400) instead of 500 and traceback","status":"ABANDONED","created":"2022-08-09 15:55:09.000000000","updated":"2022-09-08 10:02:54.000000000","total_comment_count":13,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"f03019eeea42dd00ad1dc48b4aee504822b1f5e7","_number":852602,"virtual_id_number":852602,"owner":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-09 17:09:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":1,"date":"2022-08-09 19:12:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-09-07 20:38:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 17:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 19:12:42.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-08-15 00:17:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-09-07 20:38:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"e595adb3f642e2181e64944e193510e620a791c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"date":"2022-08-09 15:55:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43c75ed40e9c3872441a5fd80e1c385bef6d8154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 17:09:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55e6aaccc5ec4771b1c471f49c7f023e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600d89e8de68447f96a6dd055c81a658 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e572dacd7a584062bc050e41b282a630 : SUCCESS in 1h 07m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d089ef045494e1b90eb70178e5b143f : SUCCESS in 52m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2e7ca6d1bf345959a4e9f35f8a51b4e : SUCCESS in 56m 21s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fa3fb1f047684e3c9d4148b23434c442 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ac00eeaacc054bdbb8d039fabda0ee48 : SUCCESS in 22m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/66368ff61f6b425d86568454dbde8d4d : SUCCESS in 29m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d17b2054bb640248ffd77f91b8edddd : FAILURE in 4m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8615104a8ad949f49933e23431df5ed7 : FAILURE in 4m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17cd7ae3667b4593a1a30e3e71826211 : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/36d8c6424b67477393b44ba97113e0be : SUCCESS in 15m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/418cdbb64cd74777903a6ecef4d9a7a1 : SUCCESS in 17m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af4319dc2cf74265864fbe00fff92c50 : SUCCESS in 23m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd7f96701cc447d7970592245cef9cf5 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7aa3d550aab74fb3acb2e457d4bc7d29 : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2f51de8d4bd04e9d944c1d8685728b64 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8621af8a191b47c1b2f8bbe13706e9ea : SUCCESS in 22m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/283ae959557a4ee887c4275c3757eba7 : SUCCESS in 18m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e2570f1ccc3649c8a110212608838eb4 : SUCCESS in 8m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11202dc38df24bca9fcbb73e2a08df07 : SUCCESS in 21m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/93b932208ab84c9b82814f050dfab197 : SUCCESS in 13m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0359cdb39994472983d4ed67cf83bc54 : SUCCESS in 1h 08m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/91349689558c4433a1b5f88cb5673b58 : SUCCESS in 58m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3c7253c23944e3c91ec8fe074529766 : SUCCESS in 48m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f349c4755737403dacde166c2089ea90 : SUCCESS in 49m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d23facfa7d184c6297603bf652bde160 : SUCCESS in 18m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3fc340f11e2453184a625f01e7bd73c : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23cca8948219429992ffd477936e2307 : SUCCESS in 33m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94dbbacea2cb44f29cde4a14c1ee71f4 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":1},{"id":"76d9b698c1d897a917771e50cb93f52472004586","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 17:13:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72665bc486d24e4aacd11f06d8efca90\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62cd8c071e28407b8ba010397108a331 : SUCCESS in 31m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/948e0553e91f48eabce70e2371a4e066 : SUCCESS in 1h 17m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/297ece30ac9e4f45855173875431603b : SUCCESS in 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4e0d388a2454ab1abf8a8cf248f4c4f : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a5d5f04f79954fe9e6be23354771ad764e25a8ed","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-08-09 19:12:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2710ec936d6ecef2cfddec78509674ddffc7c761","author":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"date":"2022-08-10 10:04:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cc26dfc93802c23fd451c9242c0cd12e0287927","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-08-10 15:55:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3de939a77ffffebed48306997b2dac074d1dc32","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-15 00:17:11.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1b5acd4af6827915c9a84f1a96a945d57b3ad5f7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-07 20:38:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f03019eeea42dd00ad1dc48b4aee504822b1f5e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"date":"2022-09-08 10:02:54.000000000","message":"Abandoned\n\nWe have a better solution to the problem.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5be3a3155c4da2c5a74dead79f90974450275f7a","revisions":{"5be3a3155c4da2c5a74dead79f90974450275f7a":{"kind":"REWORK","_number":1,"created":"2022-08-09 15:55:09.000000000","uploader":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"ref":"refs/changes/02/852602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/852602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/852602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/852602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/852602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/852602/1"}}},"commit":{"parents":[{"commit":"0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e","subject":"Merge \"ring-builder: Better format large weights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e"}]}],"author":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2022-08-09 15:21:40.000000000","tz":120},"committer":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2022-08-09 15:39:47.000000000","tz":120},"subject":"s3api: return InvalidURI (400) instead of 500 and traceback","message":"s3api: return InvalidURI (400) instead of 500 and traceback\n\nWe have seen some broken handcrafted client application sending a\ncomplete URL instead of the object path, and things did not go well:\ntraceback and error 500. Now just tell them they are doing it wrong.\n\nChange-Id: Id3b29107a5b3331c7078fb5e63f08e286a7371f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5be3a3155c4da2c5a74dead79f90974450275f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5be3a3155c4da2c5a74dead79f90974450275f7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
