)]}'
{"id":"openstack%2Fswift~853116","triplet_id":"openstack%2Fswift~master~I0128123c61eeabeca609b049746ffce1ce76a978","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-08-15 16:22:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0128123c61eeabeca609b049746ffce1ce76a978","subject":"remove unicode prefix from code","status":"MERGED","created":"2022-08-15 04:46:41.000000000","updated":"2022-08-15 16:23:24.000000000","submitted":"2022-08-15 16:22:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853116","meta_rev_id":"0ab963f13d34e12d4567144a5dc7c6425ead31b1","_number":853116,"virtual_id_number":853116,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-15 16:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-08-15 15:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-15 16:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 04:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 15:54:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-08-15 16:10:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"737d75ad66f1ea74b2f416612b49d0e04950c0b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 04:46:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"540c07e7fc0bf7e5bbda0d90f9827674c74f26fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 04:56:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb313a1c26a945039b908f1d44c27f12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9468c0294d9a4a8e8b00f2938febc636 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ae117dec2944fab88c2bb49ea35da7a : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d835a4171b0d7ff243e369cef4551e44f9de1193","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-08-15 15:54:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0f50724f04f7d865ee59afebfe149df13dd619a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-15 16:10:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5be36b07a33a2bfb912d503b5a53d3620ef087ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:10:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9c7132ab636127017ea68527ccf4e9b2daefac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:22:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfb3f026c754ab89ff51fcf59a2da75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb5696f674b4cdf931ba25f7ccf79aa : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3325a97dd78343b38e77abceb06cb99c : SUCCESS in 4m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/666fd599bbd941ba848c633b8fbee464 : SUCCESS in 9m 59s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0a3a05840fd4fc1b00e0a4af5fb9a72 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"28c2e5d4f56db4c5e9beec078f0e664e51760386","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:22:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ab963f13d34e12d4567144a5dc7c6425ead31b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:23:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a58b38dcb62401fbc70de1fe0e25964\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d5f846b0fb4c1b93c29e8d0b7b8205 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f096d8907c44f14812deb13405f79a0 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/014d38bc08a14cb0bc28bb19ecf4fa4b : FAILURE in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c0e8c59f008f4b34b4f5e8e1f98aa9a5 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd99514e600dce801f4af3bde1e78cc0bc83576e","revisions":{"dd99514e600dce801f4af3bde1e78cc0bc83576e":{"kind":"REWORK","_number":1,"created":"2022-08-15 04:46:41.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/16/853116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/853116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/853116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/853116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/853116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/853116/1"}}},"commit":{"parents":[{"commit":"9aa740f4f5344ae62464b513d33638d951fcbb25","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9aa740f4f5344ae62464b513d33638d951fcbb25"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 04:46:00.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 04:46:07.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I0128123c61eeabeca609b049746ffce1ce76a978\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd99514e600dce801f4af3bde1e78cc0bc83576e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd99514e600dce801f4af3bde1e78cc0bc83576e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
