)]}'
{"id":"openstack%2Fswift~853590","triplet_id":"openstack%2Fswift~master~I53153c4fde043e964e1daa7bbf2089e0471dede2","project":"openstack/swift","branch":"master","topic":"drop-py2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-05 17:43:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I53153c4fde043e964e1daa7bbf2089e0471dede2","subject":"Drop py2 support","status":"MERGED","created":"2022-08-18 06:09:54.000000000","updated":"2025-01-13 17:18:06.000000000","submitted":"2025-01-01 07:20:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"853590-drop-py2","meta_rev_id":"b91a061aac3c95a5681da91ca144f2e75eb85261","_number":853590,"virtual_id_number":853590,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-01 07:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-01 06:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-01 06:18:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 06:42:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 07:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 23:30:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-18 17:52:23.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"9a7100408d644c62414c50986841c41d78e92cd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-18 06:09:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6a22322b698a2a29d2be63ea1322e6cdb1b3a2f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-18 06:10:17.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2f2ddf70ee89652490fe10f0a31d68269ec81f3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 06:42:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/032cac7d60ab47ddabd8aa8bbe5cb0f5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36be9d1621c64cc38243a457b367fe99 : SUCCESS in 26m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c3813779e4b4881957c5facff8d51cd : SUCCESS in 26m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/085195b903f74654a745a50a4b5a41e2 : FAILURE in 3m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3dfe0be2fefd4eb3a36987e5807e0225 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0d3c1eff79c51fa6a62392d2c357a70a7c7dfd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 07:21:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d6ef339694f48d3a3b2a80ce0108da1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5bbd2dee394344e49c336a7177c7e0bd : SUCCESS in 2m 31s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3702f319dcfd4f6b910a51cebc984f9c : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caecd2a1db37484dac2d96ee7bac63ef : SUCCESS in 8m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d586d5381da45099e8d8a498fc942ef : FAILURE in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e9e31e380fc44d6a5cc182e2f8099b6 : SUCCESS in 5m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/97d4458093894125bf3784abbc7a826a : SUCCESS in 1h 04m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29eb94724f7c4bba9449facf5f08b2a1 : SUCCESS in 1h 00m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8ff5951703d4ff38974545e7539097e : SUCCESS in 56m 06s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c074095b8779428d864fcc2d1db2c44f : FAILURE in 6m 36s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c8eae28cc7c34ecb974202c0afafb696 : FAILURE in 6m 46s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b5daa1390adc4ccb92d97a0ea8a7c6ee : FAILURE in 6m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/755fbd3d1f9b41759e659022397c028b : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a4d31fc70be4014a759fee9c7d9810f : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/339c768368c9478aaacf0b72aa34c060 : SUCCESS in 15m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/627a4432dcdc4264805697460ee25051 : FAILURE in 3m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/daa04acb54b0491395c69a972c27cdc6 : FAILURE in 3m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/56948ae101b841bfa683dc7e12fe6e00 : FAILURE in 3m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/538108cd818344dfa0d3389372784317 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/116444f5d3244f6c888c12ba9900d198 : SUCCESS in 8m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/735e6eb3a62745b8af9b404ca3d71f27 : FAILURE in 24m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e54d66d0b0b476889376b80fb5fd113 : FAILURE in 24m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/739be0941663469e9da2b1efbe682c67 : RETRY_LIMIT in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964f162283f24f5c94df5409c49331de : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eaf39e5d8a0e49839fb404e4bd415336 : SUCCESS in 22m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e969aa9afc24faeab19eec5ccac8a85 : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fa361af03e08494f45fd1dcaf074c8f03d749e6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-18 16:35:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e518106af8ae0eb7799da487f580e82d7134a283","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:37:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4363a2f4a8c74180baed02b9d43f11a5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48000626b62545e88b9875aa9d7f392e : SUCCESS in 59m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/509f4903d8884f78b6353cc33c6cdbc9 : SUCCESS in 29m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bc71b0fe3cf4e4b99b2716c8912d28a : SUCCESS in 59m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b47f630acce42e4b0d601e0faafa8f5 : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e4bd90991d683ca18ea4606df51227f59af62719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:43:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de3a4d87fae145e2962da87f83c033ac\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f060381e51ec4794ac408c06a8129790 : SUCCESS in 2m 47s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1bc33a3716e34eec9b8298fba17aa775 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4ef457cd9d478e935112b0bbaf5e1a : SUCCESS in 8m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e67c7a6f80dc401fb918e540f7695cbb : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71192a5dbb4747beaf65f0e541b15d91 : SUCCESS in 5m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/37c6dbdfab2844e5bd992c813a974408 : SUCCESS in 49m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/204b78d3fea8473cab309db975b21b25 : SUCCESS in 53m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cc02c09ec494e8187a3e5af45f34d5b : SUCCESS in 42m 41s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eba21447339c448184297b03db2518ae : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/17420c6887064c098c53d7960b436046 : SUCCESS in 21m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/451e29531dc340569d33cf0fc4e1d837 : SUCCESS in 16m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94d784376ffe4c9181ccdd38008b5b19 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3fdfb27dafc145b3836c056344be9b94 : SUCCESS in 15m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3633d8c933934dd8ad52923a9304fa08 : SUCCESS in 15m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e94a05145bd94528a45e393fa03e4274 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dfc4f2196e5c4cb0bdd2aefb507b2394 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6d98c1c0259248839dee2b60737e1cf8 : SUCCESS in 16m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/649d99377f5f4a5c842c89df2b78a42d : SUCCESS in 24m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f46ea6cda3f34e25bad6ad45c5fd78d5 : SUCCESS in 8m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0753b484777477d9646ef2d8d2a8ddf : SUCCESS in 46m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9f4d2986f8b47dba660edc25c1c22d4 : SUCCESS in 54m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb6c8e14a9ac454ab9f9ac59549e72cf : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad77d30d14774b2fbc9eab01799ed2e3 : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f2ff42b6548478bbbddd559d4216177 : SUCCESS in 29m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c22bccac67b44357a6b88927842dc1d5 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d66175c8b906ee52c39d313d9435225785f4152b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-18 17:44:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"097ff7859f1a0134c0851d2a14b0c4e04ffcd05e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 18:34:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/701ca5dccdb1493384ea7ab612ace8c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/488e835d3fda41208c7ca3a46c4cf2e7 : SUCCESS in 2m 29s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b0c23efed4004aaf9f811d8de45cb09d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8001a3765ae241e3b4696bb74fc45ccf : SUCCESS in 8m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a4108465aee43cf82bc6ab51fbd9fc3 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74eefa66a0a5431ca233748bd0594a4c : SUCCESS in 5m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e8972ccd64e34a349fea7173a92834eb : SUCCESS in 48m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd17e1464a4e4ac593035642a7ff0876 : SUCCESS in 37m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5270495186242dd98f5ae5e2da23130 : SUCCESS in 43m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/056b69f17cae45b7a943b22c4d0b22b0 : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b43a4b239e864ad6a0fd3f699d49b253 : SUCCESS in 23m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/47f480b01f644ff28551a29f72b05de4 : SUCCESS in 22m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2765005d24eb4e07a4f19e4809bfa930 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ac1f21e8d3248c1867caca98681a96e : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f4847ff6826c40caa4c2a707237639db : SUCCESS in 16m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/16b44f26a0604ab59d3d1bd5e8f11565 : SUCCESS in 20m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/880f3144dcd54968895c1255834a7b9c : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/852bbedb946447e3b8461453b5f2cb02 : SUCCESS in 19m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6d64a74556e4349971a7d4aa5a8366b : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa1779b8f11a454cb09c3875fb06cc40 : SUCCESS in 8m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e46305511ef4f1dbbc38e94417c3455 : SUCCESS in 40m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8187a633cc148be91b1f984f875f9bb : SUCCESS in 37m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/378ea78d4bab47a5b698a12b6dae5a71 : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f30568e8b6b4e12b23e1c23ce550909 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c772cd775324d099ded6330c90ddd60 : SUCCESS in 23m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/396633c78c164c5a929375ba69ff2eb4 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b5128320cd7198f5e5f5dc1a1f4f54fb00646019","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 18:49:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/437954f900874695bf3b6f7cd204f700\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f3c0e6849644f12a9bd6468d6955e92 : SUCCESS in 53m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c80f42718c943b2bbd222104b3f4c7a : SUCCESS in 27m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/65dfe0ec857f48a3a5ecef4f904a0d77 : SUCCESS in 51m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce3129fe40a54473a26e83ec181ada5e : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b22744ab06df6734ec8afbf64c2b57f12b31276f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-18 19:25:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0540beb475657db9881a00b1a27d27e035ef6cd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 20:15:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79aa2a69eaf64b34af4bf565d6fe8845\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0cdba94f34c41b3b6242d27778e8012 : SUCCESS in 28m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed32abc82d4a4a4fac6648574e23c1b0 : SUCCESS in 39m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bec95e5fe27c45ef844e911feb9729bf : SUCCESS in 47m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5feb598e9a3418ea9e999204045d048 : SUCCESS in 28m 25s","accounts_in_message":[],"_revision_number":4},{"id":"3e5ef088d11b558ac511f54afd0133ad5156a9d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 20:44:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2c2553e34a54cc5b1a5b66e8b84b00f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/75e8a819c9fe4fdba237bd7dd6da63a4 : SUCCESS in 2m 25s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/df36f047d8c244099010ffe66e05ed1c : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c5056b28e6412cac67e701248cbfb2 : SUCCESS in 9m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dce9af28ba234df4b31f7fd759a39063 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b09afe46be94febbaebd8b4b0bd1083 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3ce88a1d92994b9f817b13fa7b4be345 : SUCCESS in 1h 04m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb2f83cb341e40d193bd100df0729231 : SUCCESS in 57m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41f4d9060c024a84a12516a0b267a428 : SUCCESS in 59m 37s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6a839016662f4b199864f54be09837cb : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d3bad0a341274c8691c8898c4d7dbf80 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/788849d7e19643b8bd178a9c03df68f1 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee6bb9fd5f3f46ff84450179bc6f5578 : SUCCESS in 9m 54s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6603fc8065d54702ab173468e1edb44e : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d449132175842b5910633dac7d3dda7 : SUCCESS in 16m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a389be95eae348d0987d4905f9f64dc4 : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a091580b4ee4c2c975f48c9eb7902c6 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5de4483a7e05404caca98dff74a5cfc0 : SUCCESS in 17m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5e70815cd224a2fb3dd4f35a35ea9a4 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef1abbfe1128457a811fe9829051de60 : SUCCESS in 9m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/320d84de860f4fd39ed4eacdb0f7f52e : SUCCESS in 47m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b8bd34d53c54df5b7e39942e1c59207 : SUCCESS in 45m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5d00921895247349e7b027649ab651a : FAILURE in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa1ecf51da34dd9a19f191580b568ca : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ecffb3c464a4b5992f0b8d63159c876 : SUCCESS in 33m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d504f29c9bf46dbb62a3a8754573e41 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":4},{"id":"fc5fc967b811248f335cd5f2588ac122883688d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-18 23:34:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae3869be758045f053cdd4d03199380981742a4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 00:42:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e901fe1b33b43478e8d8a56563fd841\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6100e649f6d54ceaa8b34952c5137f15 : SUCCESS in 31m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40267b8f44254b358c261f11e1e1c57f : SUCCESS in 59m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a0b3d09b0b642688546ad0869af59b7 : SUCCESS in 32m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/88474adb48244583959290a2c7cf4f5e : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":5},{"id":"0c5c3a81bee261deea70c820174dbad7a4a05bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 00:49:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3267541d6f74f52848d75fd986d6ae6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f3a71a1a6ed45eab1db15e47ad4405a : SUCCESS in 2m 31s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/65fa3400a83a4b3c8d6de98b68a90997 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91737bcf614d4666bdc5c6e8d257cdbd : SUCCESS in 9m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e84a4e2b1d143c3a75dbafb4bf2e3f7 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3395ae51b6f40e588f4abc7e4d0627d : SUCCESS in 5m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/fbc5b0104ed74dc2becaa12a7838e14a : SUCCESS in 1h 08m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28a567a38d0f4e29a7a00a952bb4eaca : SUCCESS in 1h 02m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5acec6a4f21644a5aa6fe1852b84e9f4 : SUCCESS in 1h 00m 15s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bcc170a5f22e4a1693ae0bea89cce82f : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5efaa65fb57541ef85791b8c0473d477 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/716453f12dd64708ba50ba56947e7874 : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72a85d4c48654bf3bd6b051f3d87d77e : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aaa3d839fe114d788ca479dce0849f10 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc33269df45f4b3698baab8f2c1cdc30 : SUCCESS in 16m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e4c3da7156d94c819111ea1e00ab94ae : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9158ae6d9516471b8f4eab2b86483f2a : SUCCESS in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2293e7f6e50f4d7d978a26aa9949866e : SUCCESS in 17m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b80475c2d16e40798733bb599d6f932d : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac316863e8e44439b79331e38a335482 : SUCCESS in 8m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/771224ed7c264842af96fdf1fd1677f1 : SUCCESS in 48m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54b5ff3aa59d4db1b1a16aa1b0e9e1f8 : SUCCESS in 44m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39e0075bbbd847deb02c41f5e469ef42 : TIMED_OUT in 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32433bee2db747b09dfac214355b7f4a : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01d6ebc98e1b455b9f856f508696725e : SUCCESS in 28m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8174e05fa23742208df3b60616a69217 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":5},{"id":"605fff2f6b8ef6944a863921bf6ef8f3cc05b80a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-22 23:13:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f4b7eed4af4395dec54bf47b59d53f1c33b4541e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-22 23:13:54.000000000","message":"Patch Set 6: Code-Review-2","accounts_in_message":[],"_revision_number":6},{"id":"7bd3ac7b34ba6790ed6a827548d39f5b8dd315ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 23:52:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe603896117495a8c66dff9ec1b6766\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9106f9d31d58413fa38eaa90071393cf : SUCCESS in 31m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbf71ecc4192448eb5bde4b46e139887 : SUCCESS in 32m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/738eef04493d4f569f18b149318acba2 : SUCCESS in 32m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dece32136c9947999eff260d19def1a6 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":6},{"id":"63095456fce9c043bab42b582601f99f67fe1843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:23:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b79bb97a10e9469ea0f508b0e9d210e4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f12532e139c47b28baffa6b72638907 : SUCCESS in 2m 36s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7fa3d594ddb04be397a2c84768e22a96 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d828de54b38c458390a0da0f0ec95be3 : SUCCESS in 10m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0dfe28b937a04d40bb1ee108a019b303 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0b884ac6a344732a2a1e8a168be7b1c : SUCCESS in 6m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2437ec1784ce44ba963f09ab1f8d0c08 : SUCCESS in 1h 07m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3fe32e92f19749ac9ada981595fb7eca : SUCCESS in 53m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/748bb39cb6934eeb82b23e5f51fd5c76 : SUCCESS in 51m 46s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c096dc34684e4ab1a82e36bc9f2d0bb5 : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/00704bca3c474fcc9fccc51f2f81fb7c : SUCCESS in 24m 08s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/be809551862d4417ad43d4047b5978da : SUCCESS in 23m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e916104538f451eb3173505c8a139f8 : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/091182e41d994c20a313c9aabac80184 : SUCCESS in 17m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/391ad1616ab24066a04bb4c24d39f201 : SUCCESS in 15m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7814a5a374e74bf69785eb8cebf6774d : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c8c6b3c0a1d344ad862070ddcfc821ea : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b391dbff58b497d8dabbd941b902e6c : SUCCESS in 17m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/30567b0e18c642e4a273dd6872ffbb62 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae8e03363ffe4b5f8c742e9e8c132384 : SUCCESS in 9m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e63a22d257b4999995fd5247346f9b0 : SUCCESS in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/615413115553416380f6bfdf9be3f7af : SUCCESS in 46m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f44866abe7124c919360ee299605e041 : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e574a0ca3765427a967b15b14bb1af23 : SUCCESS in 7m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/593d05d541304ff8a6761407dadb658a : SUCCESS in 25m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caee8f11f17e4ead90ef03d470984ba0 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":6},{"id":"3575b1f6bb13d0122ad2bcaed6c9f86d940369b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-21 23:29:51.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"679dc7cd5806fcac448f90c1f940d5c03b1e93ab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-21 23:30:09.000000000","message":"Patch Set 7:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/swift/.zuul.yaml@master\", line 133, column 21","accounts_in_message":[],"_revision_number":7},{"id":"fe6ead9db73a3b53dce0fd832fdc30b87a7393b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 23:30:13.000000000","message":"Patch Set 7: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/swift on branch master.  The error was:\n  \n    mapping values are not allowed in this context\n    in \"openstack/swift/.zuul.yaml@master\", line 133, column 21","accounts_in_message":[],"_revision_number":7},{"id":"cadefece07c79f62d789de2f3ed21646539d5986","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 23:30:20.000000000","message":"Patch Set 7:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/swift on branch master.  The error was:\n  \n    mapping values are not allowed in this context\n    in \"openstack/swift/.zuul.yaml@master\", line 133, column 21","accounts_in_message":[],"_revision_number":7},{"id":"15c61175bd76cf122ac6711c00746326a40692a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-22 17:07:12.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"752d59fc422986be5e62ee7fd2b7b7d285f65e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:07:34.000000000","message":"Patch Set 8: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/swift on branch master.  The error was:\n  \n    The key \"nodeset\" appears more than once; duplicate keys are not\n    permitted.\n  \n  The error appears in the following stanza:\n  \n    name: swift-tox-lower-constraints\n        parent: openstack-tox-lower-constraints\n        # This seems defensible for a l-c job\n        nodeset: ubuntu-bionic\n        nodeset: ubuntu-jammy\n        vars:\n          bindep_profile: test py36\n          python_version: 3.6\n          tox_environment:\n            TMPDIR: \u0027{{ ansible_env.HOME }}/xfstmp\u0027\n    \n    # Image building jobs\n  \n    in \"openstack/swift/.zuul.yaml@master\", line 458, column 5","accounts_in_message":[],"_revision_number":8},{"id":"3351c8b00d8b20a904444cd7ef705deeb2a7d282","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-22 17:07:34.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  The key \"nodeset\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: swift-tox-lower-constraints\n      parent: openstack-tox-lower-constraints\n      # This seems defensible for a l-c job\n      nodeset: ubuntu-bionic\n      nodeset: ubuntu-jammy\n      vars:\n        bindep_profile: test py36\n        python_version: 3.6\n        tox_environment:\n          TMPDIR: \u0027{{ ansible_env.HOME }}/xfstmp\u0027\n  \n  # Image building jobs\n\n  in \"openstack/swift/.zuul.yaml@master\", line 458, column 5","accounts_in_message":[],"_revision_number":8},{"id":"f510cd7e1d46ce3031f82f6143ff8138c9827a34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:07:40.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/swift on branch master.  The error was:\n  \n    The key \"nodeset\" appears more than once; duplicate keys are not\n    permitted.\n  \n  The error appears in the following stanza:\n  \n    name: swift-tox-lower-constraints\n        parent: openstack-tox-lower-constraints\n        # This seems defensible for a l-c job\n        nodeset: ubuntu-bionic\n        nodeset: ubuntu-jammy\n        vars:\n          bindep_profile: test py36\n          python_version: 3.6\n          tox_environment:\n            TMPDIR: \u0027{{ ansible_env.HOME }}/xfstmp\u0027\n    \n    # Image building jobs\n  \n    in \"openstack/swift/.zuul.yaml@master\", line 458, column 5","accounts_in_message":[],"_revision_number":8},{"id":"07478c83473d2634101fcd4b69836f73be7063a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-22 17:26:49.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"dbf57befb08ef259c71fc3ef333e91fa90c3ca89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-22 17:32:36.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"213cbe326e0d5eb2f9afb71186bb8af32d2c6a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 18:49:34.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dba61c735efb4f9d8fa066cd085cf50a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/133c79a8cc8c4dba8714a8961a839bc1 : SUCCESS in 2m 07s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b38f247d47ac47008315180a1f7a1d87 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ec75e21a44495ea49a1d1b3b5bf7fe : SUCCESS in 9m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/742b29db5bf14779a4ff485a16ab6d03 : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b61275d649764785a5d5ccd652f8169a : SUCCESS in 4m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0a0fe2242be5412d84d9d9da72b3b2df : SUCCESS in 49m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/71aded51def145d3b95ea6794adb25a8 : SUCCESS in 49m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4d9dd4234b84276a8cfba65b29188f8 : SUCCESS in 46m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6fdd4249f674272ba1d9ed200f129ba : SUCCESS in 24m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0f3629821944b619ad90e7504a309fd : SUCCESS in 25m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42004fa723ee46a0a9da548b563eed2e : SUCCESS in 29m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f69528111ba4f1392b5391620026bc7 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/68ed0810ab524af1b3f0d22ae80857bd : SUCCESS in 23m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fbd27b14898940bb8c09c099949f5de4 : SUCCESS in 23m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/529835b94c794d518c44a85a49a4fda7 : SUCCESS in 22m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cd5382a78c714e798f7a4e74e7de1b89 : SUCCESS in 25m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a89a1912051a4740865a8dae5a2e02f2 : SUCCESS in 24m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc2b61537f524f88b764da1f75f2ed29 : SUCCESS in 25m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e87c6d242ea146c79da2bd7474cd8448 : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81dfb58ae3294f3890da329704e15c6e : SUCCESS in 7m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2e7481db552484a9cdeee7b58792aa9 : SUCCESS in 16m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9c6c7e1689845dc9e7e521d49e0bd46 : SUCCESS in 7m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5c55ba9466743a7ac0df34f1f0e478c : SUCCESS in 1h 01m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9c94fbacc064f979df6eded5866bdb9 : SUCCESS in 54m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0601bc56fa664ad983d9e7447a0ae253 : SUCCESS in 52m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8218f2f344304a748d3f50fa5e29f72f : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdcd32ef7ad4820814c4d5efb33a61b : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc8b005c5034477e8f85276ad6c6c952 : SUCCESS in 21m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b081a3794164915b7a179de651e3247 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":10},{"id":"426d62c158f00a350b03664b75607a580ebbb851","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:34:18.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e167663eb97e4f3cb8d50282d74d7aaa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c15ef50cd1eb4142afd4b0f58d45d85f : SUCCESS in 1h 07m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74c82c5f87634e8dab10ad62fa7f1315 : SUCCESS in 1h 06m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/452203409093428289533df74df56c98 : SUCCESS in 1h 11m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/78aef5dfe7c045b599ad205050ece3a7 : TIMED_OUT in 1h 31m 04s","accounts_in_message":[],"_revision_number":10},{"id":"a963afe43a8861fa11253be14f165b9af92bd528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-25 00:05:30.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"1239596c2eb39080dc49aaee7083cbc6b859982d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:13:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16db1aa267cd4c30b7c20aad61ae73b1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b461ab2cfe9e42dc8cea6e395583b48d : SUCCESS in 14m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3813caf1a7047eea44b4177cf84fffb : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/be51de47dfe6457d8e1709d018999546 : SUCCESS in 1h 04m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/66f7f927f5cd4f8fb846bc0c925bdb3f : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":11},{"id":"f4662f0341b1e8ac6bdf7425f2dc0426974bcae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:25:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00055a828dd34bfbb4d6fc4a4e88a8e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa9f41d30f28450f86b79606325ebbde : SUCCESS in 1m 55s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7189d4d5566f4422b2d07ce9d6c5cdd2 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eac7713ddfa4d869b91de92e5e5ad22 : SUCCESS in 8m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90813ea743e340a897c405f990ce5876 : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec8de067e3bd44a4871c60c397412ec4 : SUCCESS in 5m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/d6424cdf318548529d0f2ba6a7f70956 : SUCCESS in 1h 10m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6b88c16b655409fada4dadf8dc64e01 : SUCCESS in 1h 04m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dce67aa1c1ca4bbbaf276aa82f659400 : SUCCESS in 51m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9319ebad07044e38affa57eb82c01aea : SUCCESS in 34m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ab793f827d9415ea881c81bc313e808 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7e5f3752ebb4bdea46ae689028b0360 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5e353341aa34a179bd7b3b495762623 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/618faa682e2b44ce9d12f63c7d30d574 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b53d8407b3e4e108160c37825c53a83 : SUCCESS in 17m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb3b1f43a45f4cf5bd03e675b013bac7 : SUCCESS in 16m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b6f57dfb5c754b8d820c8bd9c3c9c8f6 : SUCCESS in 16m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aa9b53395bac47cab40e81716b64b886 : SUCCESS in 19m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/20650e134b174263adf8054b922b0f30 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7370838a3f9d4bb380212ae35844ae5e : SUCCESS in 15m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/04ec51e0f1594d018a7a6e11a8371f26 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a03dc811455d4ce0aca39ce14ab65325 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c65800e6d45d421cbe678bd630571281 : SUCCESS in 12m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33a1000836fa4692a3f12ef1b28d14b9 : SUCCESS in 8m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/231f03a9240844dfb2096db236f4341b : SUCCESS in 52m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbd840c684e8498e9b411c14fac26a65 : SUCCESS in 58m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4723ece41a44862b001c2ef3b6ecc75 : SUCCESS in 40m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7856999eb58466eac62319aa521caaf : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7547d9565b3a4e6a85dadf1e480dfcdc : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2b54731a253487ba82e3b85b034afc1 : SUCCESS in 20m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b7aaf8785d414386f5dd6563767cd8 : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":11},{"id":"10908c3737e8daa5f707f4c9b2f1a8171818cb29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 18:36:16.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"587e753ad2bfc1dc54abe1a2ccb744cda45a7190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 18:36:41.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The key \"name\" appears more than once; duplicate keys are not permitted.\n\n  name: swift-tox-func-ec-py38\n      parent: swift-tox-func-py38\n      description: |\n        Run functional tests for swift under cPython version 3.8.\n  \n        Uses tox with the ``func-ec-py3`` environment.\n        It sets TMPDIR to an XFS mount point created via\n        tools/test-setup.sh.\n      vars:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 219","accounts_in_message":[],"_revision_number":12},{"id":"84525e2a487bd749fd407537b04ec43c474630ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 18:36:46.000000000","message":"Patch Set 12:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The key \"name\" appears more than once; duplicate keys are not permitted.\n\n  name: swift-tox-func-ec-py38\n      parent: swift-tox-func-py38\n      description: |\n        Run functional tests for swift under cPython version 3.8.\n  \n        Uses tox with the ``func-ec-py3`` environment.\n        It sets TMPDIR to an XFS mount point created via\n        tools/test-setup.sh.\n      vars:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 219","accounts_in_message":[],"_revision_number":12},{"id":"ed4cc43823ca1afbad623684d675895b17f49a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 18:42:55.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"afe3a591a42bbdfb730ac87deb7ada36d222fe4f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 18:47:34.000000000","message":"Patch Set 13: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e80f396ad9ca7a34897a5bcabcb34410c13a7ac1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 19:58:17.000000000","message":"Patch Set 13:\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b21565ef4c6b67b04fccf2a304a38a784d08f6e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:08:17.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e893f7f5c90428795420048e17b6021\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb96bf5031474507b88c018cd7eb5b4e : SUCCESS in 17m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f019313a539946d095ccd8939c1df31e : SUCCESS in 1h 20m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1bc9d74c756439f98b2f73739808823 : SUCCESS in 23m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ab2dd03fdce414795b4f2131b2038d4 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":13},{"id":"ed9ac41ececd0c5148aff0b12ad7782f7ebb21f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:24:24.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d3fce6a20314049891dbd3ff21362e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbf662cc073640a49232dada1c2add26 : SUCCESS in 2m 10s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44a07d87318d4206b698d5b25c43c1c7 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d21eb5b09b45d8bb8c8a85bd971eba : SUCCESS in 8m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c14c645aba444a42bab3c39f818e0c1f : SUCCESS in 2m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44ca3d48c31646f0bd51f48005bddbb4 : SUCCESS in 5m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/1f232d79821549e8b680dafc9c829014 : SUCCESS in 50m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a45a8f33818e4101aa16c4f2f490940c : SUCCESS in 54m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2142d0c6d9d0417bad25af19c9e2ad48 : SUCCESS in 48m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c40d444eb9e47288d333596434a62f6 : SUCCESS in 23m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25a3038b93a54b29bd5028076361c06f : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94823705956c46019e2c25094237e123 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4bafa9a8fc84ac3b0227ddac378995a : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c36a106f0b474d8504df0361886def : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/803dc20783884cd59650cc706c9a6162 : SUCCESS in 18m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/96f9f00936c24a66b67ebb12a6a1288a : SUCCESS in 17m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/78d007e236af4ab4ae5490d40c87ab0a : SUCCESS in 23m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ae22a87cd38c429e8bca1d1a86c1ce06 : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb7e38b00eac42239b42b868b76bcdb8 : SUCCESS in 18m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/49f0519546b2451d8b863670ef637bce : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/91d5076eeeca40378eb979bd7980328d : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a605887a7934994b0e92ccae0efccc4 : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/609702e63bc0442a8c7e5b32ff9c53b9 : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cab730726e96466abb1819b4c23c8e2a : SUCCESS in 59m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/594c4108e1434b7096181ff76fddec9d : SUCCESS in 57m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53549d3ac13147a1812b9750dfd1b177 : SUCCESS in 59m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2302a13534384dd082c7cec40555cb4d : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa403487bdc743a9983d1dddec495d1e : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e61155d8ba14405b0c195eb36984c00 : FAILURE in 1h 11m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f02fcc3c29304feaa3a590dd25c60250 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":13},{"id":"30ad1845a58232e142c06756419e0b96dc7f3d3d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:25:54.000000000","message":"Patch Set 13:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ca5da48e4d24b99b95f837e1ad77bd1\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/364baccbb7d947678ef6e5b750786ed1 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a271ac833a84c739397a9a92b79fe35 : SUCCESS in 18m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7a279dc74fcb429b8cea27f2ea1f984a : SUCCESS in 20m 33s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f13dadb7ed14908b49eba64ed738da5 : SUCCESS in 20m 13s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/73e4cde15ba647b29e0d73d57365d16c : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd43a5e3026f458d91678e80129f9f00 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c06f5e491f64874a2d3c329252eb0cb : SUCCESS in 20m 27s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/e9e55caee52a43be9cb485dab6aef107 : SUCCESS in 20m 36s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/29417ac6b11144df8b1b360df80dfcd5 : SUCCESS in 24m 09s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/1102f76b36414d59a1a9d0f2203a5b08 : SUCCESS in 20m 53s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/847294e861db4e7383268a5fd2818219 : SUCCESS in 20m 57s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/b8cb4d1b137d438c9f11e2b056083f02 : RETRY_LIMIT in 9m 42s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/7669c826f60d47529461af79c7254d28 : SUCCESS in 22m 12s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/7c3a2f59483c40b18ac4f65346aa9363 : SUCCESS in 26m 45s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/8db446a2d88243aa819206475025c52a : SUCCESS in 17m 22s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/384b5c850faf4622a12ba79d46f144e1 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":13},{"id":"054e3b5dc0ca8e2a3c3ba7aef5b308afdc891810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-21 22:25:59.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"920850ae22cdf9b321a45aebb04319c108f41843","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 23:41:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbcc41b242be45b9b657233a076c203f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9daa716773834051bb85ab1527a1ca65 : SUCCESS in 18m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89632171dc1b4ca49af4d3b696e24cb2 : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/70d3e783bae944019560a9103067e1d8 : SUCCESS in 1h 13m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d69a2c566cef4a96bf0931f1dcd5ca81 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":14},{"id":"687bbbbbfb204600b6dd58cc80e4d6159b8a676c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 23:42:14.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddc9d93ed6484a20b9522e5300fefe31\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dcef0d6c31ce4b44a8a1fe3ba7cd6194 : SUCCESS in 1m 40s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/563912e5d43a467d93283e857bf96cb7 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102d9e98ec3341328b92691972b6e4e2 : SUCCESS in 9m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91eebd4289e54dff805d704bf96c94ac : SUCCESS in 2m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9af30f5573c040839078435869dbac08 : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/139ea2c506e943fbbea56e545cb4b1e1 : SUCCESS in 48m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1a2739584e14b61a80c5b8cb9e597b2 : SUCCESS in 46m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff770a70f2454e7aa3b44a38f1f036a8 : FAILURE in 1h 05m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4edd28f3dcb42e481c3b6d2302fb872 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46ffccbfc668456489dd5c6f75c9c5fa : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94b46b3460b848ae90809d7f7de1cadf : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58a99f0a175344799665db7b90f0c996 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99ec20575d314c6189fa086ba1d910a2 : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/10c1d3a518d54ee68d9c6d1ddd01ff6f : SUCCESS in 16m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2d439eb86dee4299bf25edaf47f48d61 : SUCCESS in 18m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c435b42f373f48999cc618af4dba3416 : SUCCESS in 23m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2a3e644db3734f65801b5e05f17d3c43 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fa843cfb83a944228eeaaff507ca914d : SUCCESS in 17m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b190cfa843c4a5d9dce32dc70608704 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0abe85606125461398bd713c085ace49 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66974c8059134db38f262f0304e1eca4 : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f0f6601c191408bad4b48f89ea5fa4d : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/394c4de7199447339253719483820013 : SUCCESS in 48m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b885d4de080411089d4472d60a8c794 : SUCCESS in 38m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce4da6671c454fec9866983d6b85e2c5 : SUCCESS in 36m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b516607aa7054a42b5d7cb4076fede94 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ab6fae3c95456a992808c91393bc80 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83b50bd43c0d4a769441a20f83c01218 : SUCCESS in 17m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e71df9337444adead1a946164b6ffd4 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":14},{"id":"5e6a604a2c6325cd2ab991f5b05a08c010c1163f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-15 22:42:20.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"694016e6e28ac2deb1345cf58cbc2ecca19c2e85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:07:05.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ee9bfb71cd84e8b965c5f1e633e017b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc5b6d48991e4efc8bf6a580a0f49fea : SUCCESS in 22m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/deb90748664042ec91f2233c5e7edc39 : SUCCESS in 20m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1c7c58fbc934ff4acaae2a97a97cd91 : SUCCESS in 23m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2627ea8fdbb54809bc08e28b31e09846 : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":15},{"id":"41d44942442b0204c4ced94b8444f4891555eb58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:49:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/816c47478ffe4b1282ff0ed0c92df7d9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81154351cb6a425b9ab1d175e206633f : SUCCESS in 1m 56s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/238e7c666aff442db67d3df49729a083 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471e9ed4cac943d89174c4327ea114de : SUCCESS in 8m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc17d2796ef24608825568b53ab001e2 : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77d4c7cf34b946cdb2ecd7e256a736e7 : SUCCESS in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/5f34670daa1e46fcabf05716dbe38d95 : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a0060f3419141cda21f89088b324071 : SUCCESS in 43m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36dd07683b9048d887fdb4138031a593 : SUCCESS in 1h 00m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b73e0e60b81347478b4924fff29e6999 : SUCCESS in 22m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4732c5cb4ec04cec8cf9c713638d6527 : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3dfecdf752a433296fde06eff61d1f5 : SUCCESS in 21m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/980e2f2ebb0e4d9d8829087a460e5c7a : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/36ce104677844f8e86b8a7dffa8ff15d : SUCCESS in 18m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cc184385b9b641dcbf6b9f6fc9325e3d : SUCCESS in 21m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4bbd0beba527492d9f08ee30a9572a68 : SUCCESS in 20m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68a51eb6176348768132f3d5c279fc3d : SUCCESS in 25m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fc9fbfb48ffc4ca19a3c39842ad2579f : SUCCESS in 21m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d4e092d0db244256b655b6d42dbcb818 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e9375721eaaa46418562761c2bf86c2b : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f3399e156eae409b88bae7a09d2c6bf4 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/869c39ad91664b4a9addb119d0cc419f : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c31da532c426477c8310a26c647ed935 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56181f279b1e484f8bb1e43a98eb4a83 : SUCCESS in 55m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a282248bf984cfda74ff527ce07c954 : SUCCESS in 53m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26e86ac8211749ef8693c6bc2862a6bc : SUCCESS in 53m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a76532367424b759c42bd01d67081ea : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216c7be729cc4a8f9028520d7c5bd59c : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6da61f57c224de9bfbc1ca5adc3cba8 : SUCCESS in 17m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f6a467287b3447c867a963091279de9 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":15},{"id":"d3627fb24977d05583256753b99c94550a9421e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 20:06:30.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"d5f38f7a96f8eba1bbe4caf96ec5e86826e8fd0a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 20:06:50.000000000","message":"Patch Set 16: Code-Review-2\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"2b1aa76c486da83b47aae4742b243864e81e6ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:24:46.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5221fe4cde40b9960a17f9aa916165\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b63a0d600d24060bb00294bd5afa9ee : SUCCESS in 1m 43s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6027437def6146a8b3cbe7697d88df35 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c19cdf3a73445159c54617920f7f833 : SUCCESS in 8m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1e5a80a695a447a975e31b0a67d6d37 : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc2da099ebf44d6984b4fd057917a8cf : SUCCESS in 5m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2dd2aab009a04562869f88b9479f88c2 : SUCCESS in 57m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3dc29e9295474a4ca7a188730c361619 : SUCCESS in 53m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/011bc803f17a48f9809943cd19d1516d : SUCCESS in 53m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc1a448b40f24355afc5960e2969d123 : SUCCESS in 23m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e52639679eaf46ada053621d7043afe8 : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f10f9ad29181494b863b572f4ffeabfe : SUCCESS in 24m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8191260bea9f447699b46b284979b1bf : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/750cbe36d8f949bcb5733b0ab2d0b02c : SUCCESS in 16m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6006cbc8f82d43d5ae08b6ae523607c8 : SUCCESS in 17m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7bb3f15c61da44a0904388bd0f59f0c5 : SUCCESS in 18m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01f3689355b94372b7c9152803a81279 : SUCCESS in 24m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e90bf7e3785451a8ad784d019ae5c22 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/83fd6baed6f84ea5a0e50cdd063dbcc4 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/73db4a6b5a8b4e6a85c162317a95bb82 : SUCCESS in 19m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2febb2787b344d58f8ab728075e3238 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d99cbaceb36b4e61bee6e795e2919023 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0f6fd701ef744b3fbb2a28c6469b63dc : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2218a5cbb9664d3a93fc4d753c1c28b3 : SUCCESS in 54m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28de8e88e0854278a786932c1b8e9631 : SUCCESS in 48m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b94f6a1387e3466da9b167df73b2b4f6 : SUCCESS in 54m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0666786e0de04c229e59308ba89a870e : SUCCESS in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5d5b101c8f480d8cb36fe41d7bb322 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/66ccfb65c1744e23a09c8bdc7e2aa3e0 : SUCCESS in 16m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c9f114db241486ca478861de1c8c262 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":16},{"id":"0f6c9989926d96622bd6c31bd2eaab558424e4f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 00:44:13.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48530b6a07c94fc1b7bf690f746ec739\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cae81f474c6464f8ea0fdc6b6920ec7 : TIMED_OUT in 1h 31m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdd4a4d6c34a494cab060135cb56b870 : TIMED_OUT in 1h 30m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4f7e29c11b54e9bbbd547db7a2afede : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3952be01bc7342989827281ca2d94f8c : TIMED_OUT in 1h 30m 48s","accounts_in_message":[],"_revision_number":16},{"id":"05024902d34ec0722f8993397c5800bae2b5f2fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-28 20:19:21.000000000","message":"Patch Set 17: Patch Set 16 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":17},{"id":"95603fee12f25a0a6854bbd02d3313c36bd23327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:30:27.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ec4594cfcf491cb9ce8bccafd43029\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6cd235455ee644ef97cbc1c7b84fbe76 : SUCCESS in 1m 59s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aef5b78a65024a6c925bb85b8d6b3ac0 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e0b205214140d4bdcbc06a0b86081f : SUCCESS in 8m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c017e64d7b2542ab9987d33b26e06ce7 : SUCCESS in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a24d459cc21407b9dcac38decc47fb6 : SUCCESS in 6m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/8cadbc47ca3c4c8bb83689112793d6d5 : SUCCESS in 1h 04m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2915d9f267cc4cccb06fa0e58911339c : SUCCESS in 53m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/733cbabb8f0f40f7b15d38840b8a5bba : SUCCESS in 55m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21d89bd6add94c15b9115dd74aa599d2 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c531e607e07a4ca08fa5199163284d4a : SUCCESS in 30m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/915c65eb33c243d08dfb29d8d1abf588 : SUCCESS in 25m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b70e719808c745a7970bbcac9e3af75f : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88455741a8fc4de3bce85fe281906f09 : SUCCESS in 19m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc0da1aef1a74432b9ce39f2665d2aa5 : SUCCESS in 20m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f70724fafd594a93a3e2bdb28bcae672 : SUCCESS in 27m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/86a610a5607a4a81ad2235bf2852f3c8 : SUCCESS in 30m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4f1f4961d8c7485ca874a5764932ccf4 : SUCCESS in 22m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/afea79f240fb412c9b29459f298029c6 : SUCCESS in 25m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a39b599f036e4305975c9a14813e5b4b : SUCCESS in 22m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e95383e211174746b698249ac063a0a7 : SUCCESS in 8m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d37468f3ef9418096cecd889b569a45 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/68ecb90fc7464c1cbe5ab06b25b9156d : SUCCESS in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ac003f68580467391fe00b27f3ab992 : SUCCESS in 57m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1c7068c2dcf4262926080a9b21cc72e : SUCCESS in 42m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0dfa01401b3e40388aed7411f3a59c46 : SUCCESS in 53m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d0a4696819d4cde8c859776eacc1498 : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8c059168b841168227f36fa460ce97 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ddb42b9a890e4414b000502c7d14792a : SUCCESS in 24m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/586cf714c4e14f798f78ac5ecc8337da : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":17},{"id":"1f228a6bc0b66d29aa74cb9972dce757febdb9e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:56:44.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16861a535924435cb761734d15d4e288\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80b7c595260b45c98d7c84d651cb76cd : TIMED_OUT in 1h 30m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fbc159169a14b8292d5d7a8bb62cb90 : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e8b48475d2946aab30e90970e37f942 : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/21c8fd12d6a2408991ea0485f1d50141 : TIMED_OUT in 1h 31m 05s","accounts_in_message":[],"_revision_number":17},{"id":"09feece9e981eaef13a5bb38bcf707a3b43a1d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-04 00:09:48.000000000","message":"Patch Set 18: Patch Set 17 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":18},{"id":"b53541d967f80bc61fa2bc1023167e984581fdd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 01:22:34.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca238b6917c048118f5fdf66b8fcc133\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8960ade8ec749ee8c1117c805c5a9bd : SUCCESS in 2m 11s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1e4d47067e6479188775b25dd1640cf : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed8f23ab61b483ab3ebabe285c54730 : SUCCESS in 7m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0eddf9fc0de04309a4bbe00e92d14af0 : SUCCESS in 2m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7386657e46264f5e8811155eb41db99e : SUCCESS in 4m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/135a8034e663448993f239b924089871 : SUCCESS in 1h 06m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a4d2787dd81442184bbbbccf688bef6 : SUCCESS in 51m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19737b62fe57421197633c8a5f542e75 : SUCCESS in 59m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6845e5b88e047e99c65c92464af9323 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36666a04423a49efa0684a2f8a8579ba : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df698839593c4e3aa814d2be61a5e411 : SUCCESS in 23m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c60f4ea49d6f4a68930c66127ec2ce41 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c565d0844784693871e3a4ee7b4db60 : SUCCESS in 19m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ecfba23d546e445dba8138f16a6a460b : SUCCESS in 22m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/eac61da7fd6c4addb72fc9fc74d1e33f : SUCCESS in 27m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1d5e4577140646e992f9b118ea25471b : SUCCESS in 28m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/850e86f4d0a64a91a3c6838d938e82ca : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6f98cce9966e48c8afe549c0e7fd6f37 : SUCCESS in 19m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fbf7b507201f403793e8465d6386f4df : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ebd70b783ea4edb9035b342a5400b1e : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57b6e845f99b4e66b9324b49c00a4f37 : SUCCESS in 13m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/988a7c524f874516b0a329e6e9235661 : SUCCESS in 6m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e726971354f94b4693dc3396d5fd4b48 : SUCCESS in 1h 03m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/484632f7e2694499b39afe2b657e507a : SUCCESS in 48m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b00d8bcff3bb4507b536096fc3acbc35 : SUCCESS in 39m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f8ac9fdd108471b8173a7b066c9fa7b : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517763e179a9404cbe322e8fad6b5600 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc6540960ee942ec91122859afcc0854 : SUCCESS in 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38c413a5f3e642c081f49d4b2be69cf2 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":18},{"id":"24c073c59e55ce2a7d9f0270e69479a064c0cea8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 01:42:35.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f29c95e1caf41d6a9cfe4b7fbe058dd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3bcd930549b4e0c8b935884893c6c18 : SUCCESS in 1h 09m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28cf126b7f6b4761832e6d5ccde59f3f : SUCCESS in 1h 13m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b14b14b949343b09144e46325f3fa42 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8c113f1ee4a40768ca79637ee2017d7 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":18},{"id":"f7336e07d6cd1d58799db6d9a361684e415d41f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-18 19:16:00.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"6e984e469ea96bf6c2cdbbf0fe9dfb88f7516e01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 20:16:02.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/724b0c39559047ecb1218091650e7505\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/29acaf9d25e4421e86c9ef7b9ad4c5e9 : SUCCESS in 2m 15s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0b3162ee96724dfda6cb9fe26260da66 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f821b4d8e85471a86bb067a1685b043 : SUCCESS in 8m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d7a18c19ee64c9a9362c5322f84f0f5 : SUCCESS in 2m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3556e2120d0494b9db1ff742b0bd1a9 : SUCCESS in 4m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/841640e7260743fe9ab45c572c49fdad : SUCCESS in 45m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14603e74e8684429a39bb8045cb6e178 : SUCCESS in 40m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18956f7c58c14c578e3d29f4bf5dce3c : SUCCESS in 39m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6af6a0a52ea2401da665fddbbb919d74 : FAILURE in 37m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/141c6a6eca314e469c0301be4cfe23da : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9a5fcc5815c49efac5339c3cda97853 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fca3bcf7a7f449581f2c2b982de573c : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f016e99e95c5427db0f82735743bb713 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50c35878442d4df19fcb8693293e1d86 : SUCCESS in 19m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/51f3ca8e8ab446edb76bd38d0baf14e6 : SUCCESS in 18m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/48a2de56b2a64cfeb7ed6dd5aa107e11 : SUCCESS in 21m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c4ab6cda518d4993871dcf72b1a8068c : SUCCESS in 28m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2411e645ee5e4408a58301243f74996b : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b947998d4ef4af485cd16e2873c22f8 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/efd4e5485870451184415e3f4dfaaa3d : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fab01bbae6e14afb912a781ddb2f90f7 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5888e9d66dd940baa8cef1c213b50386 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/256b64d7d5c74e4c9cee62180e2d1269 : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e7919a66d504e4e9e68c77f13a9d79f : SUCCESS in 45m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d16a9590617f445499333e2b0de29939 : SUCCESS in 52m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5a6370720f0f48039cfb7d02d510ca44 : SUCCESS in 38m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad5f954cbfdf4c4b9696b8bac48afc3f : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb661caf3c90426e81d144ec725a07a2 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5bb574bed0534771827e7486c1b8dac2 : SUCCESS in 17m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be848515074f4cbdab6ff7e641ecd07f : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":19},{"id":"48d15bbfb51e747b2c7533389f56aa6831534b69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 20:32:14.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d3976c2942450eaa995ea4fab4700d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/329e80c9743242ee9ad7f3014aa564b8 : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":19},{"id":"857eb299da2fb2ca13664ef3904d1851a64d8f22","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-25 18:19:58.000000000","message":"Patch Set 19: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8d3c73103aa91528d6e46f2fb5df18bed43c875b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:41:04.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/593896df49534797b62677722807c562\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cce8790e24c749b483335a139612ee5b : SUCCESS in 1m 28s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/60e8c046ecfa425287e7128574707298 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cece5a5b71842b190a7ec1d0429c450 : SUCCESS in 5m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b03e0bd3d1743599191d4dcce3ae849 : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3c0a051d7a14da89091a1d29b2c9be9 : SUCCESS in 4m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/a88c5c81e8324d6b973bb2cdfc23a4b4 : SUCCESS in 1h 07m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80947b934f3747e485bde5ab9280e8c4 : SUCCESS in 33m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95248d97277646af8b7ca51537622b6c : SUCCESS in 54m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f6e87a90f1a46e0934d14688b01e721 : FAILURE in 34m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bec345ab53d4ab1bc3e4fcc37a2c723 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28760d158bb5431b9acc9c2c817c9301 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/632fcc8d447d49f5be2d24d29686f0a1 : SUCCESS in 21m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/974e93ebce2a41849d0a35f166f22b36 : POST_FAILURE in 34m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70e78b147e414685b10908291ed350b1 : SUCCESS in 19m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/93c64b7435874261a0675f046ee5a290 : SUCCESS in 19m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c76c06dddf0d4e68850d967003341f42 : SUCCESS in 14m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49370a3456ce43d995ef5a879bdee255 : SUCCESS in 20m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/30dd4a4a48e64fc8a0829ca28cbafb3c : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c92f84efaf7f47e899e4d60fedb269ff : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8cdc9e99b9194e8281e4827611ed6f63 : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c3a2e57917034d089097e7a35317e755 : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc6b6d97752149198edac43a21556075 : SUCCESS in 12m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7f64ee28c6ea45979baab6e42461f62a : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d5fda1c0fad45e9bfb63f508d19b141 : SUCCESS in 33m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/366a115043b64685b30d174623ac51bc : SUCCESS in 29m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6ed34a5bc2c47dc893a4b970eb41e07 : SUCCESS in 30m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c307af7742204107a9050448b230c3eb : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f02dbe7a84f64b4dbaaa26f4cde86619 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fab1a653daaa4314ad3c3bcafd2a9112 : SUCCESS in 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38ff722180544fc4a273b49337f2c2dc : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":19},{"id":"049f78a82d92334b624a27df0d1ed80b2862357c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:48:37.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc83c0029364d89a0f73893697a09e1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a64fa6d64fe45fc9e28f3fe8b4036aa : SUCCESS in 1h 26m 31s","accounts_in_message":[],"_revision_number":19},{"id":"775ce5ca3aed66dd0cee0038e7da1e2c21178533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-05 05:10:57.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"94bae797db0e2537231f93d72855a52f9c696ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 06:18:56.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc1df09cc85148818caab159246c5c97\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c5c2b9987b7843c385f79ab44cd93222 : SUCCESS in 1m 43s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fb31223fb4db4c5981cf15e7819acc66 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ed9b1a9595490da158e91de7770b53 : SUCCESS in 7m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f63072a695264d7fb5f58a3c00a30dea : SUCCESS in 1m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe04cbc783c472192dbfd419491a758 : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/4ca2baa7baf644939f02c923d352bad6 : SUCCESS in 58m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6633b6e89fa48f8943feb6f96cb2c6d : SUCCESS in 39m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/22fc28f967774c2d8d15d2c45e8d1b6b : SUCCESS in 39m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d356c4e03ea148a49b12737f78ee67e5 : SUCCESS in 1h 01m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13d258d2848043959ea0af4edc6efde3 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/161ca1e149c14ba6967bec48ad66b3cf : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c59111165a54d919fd435a0c08f4370 : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b786dce24964a779acdd95a64694df8 : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d14d06e8b81b4943910ef7d664dc71b3 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d95cccaf33649d58b0c2b665fc00e9c : SUCCESS in 20m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8da4fdeb32a84ff3bbd309805199171e : SUCCESS in 28m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc1f7b4e92df4c2fbe42cf82838aad0e : SUCCESS in 9m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bddc97b6ca37404e81e5810d021bb3ce : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4c6a6ddf7c4d427bbff3545394acb149 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/173fd87f20b84d58bb273d715160077d : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a8fbf89c9ca4994bf30c83bd2594d31 : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eed88896a0aa473cbffa1a93a8917524 : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6acc87062844db09f5a807382c319e1 : SUCCESS in 48m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b724d828a77e46d6bff4ad26e2d38443 : SUCCESS in 53m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bfba50f6107f45e2bfcc7501cccaa41c : SUCCESS in 36m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3a674585cdf4c25b1c0a736fc258e0e : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6bcf73d5224cb2a8bdb50f9799c959 : SUCCESS in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2935c91e33745b8996baec6f76f8875 : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e722d3abeb847ad8b40c4d860d4f9ec : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":20},{"id":"057c36e1ef0ce42cd3d328a804ed304d139bcff1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 06:48:28.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18d84844b334ade8e152b6fb1cf0bf0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/907f9bc4b3a946df8188bb9af1f4ce61 : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":20},{"id":"0462000ac0ceeeb7c9cd8bd45754d990a5065ee0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-05 17:43:59.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2b1d61a5c2158c068201f1fedd1d538786512222","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 19:55:30.000000000","message":"Patch Set 20:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f36c38d7ee6b45e5a236a22abace600a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de09ba77d81047c2a82f3de061103610 : SUCCESS in 1h 12m 17s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/186ee40e6a8a4cf5be9dc5622f397737 : SUCCESS in 1h 39m 54s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4f8a528ebee42cfb6772647297941bd : SUCCESS in 1h 23m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2025fcdde5544f7ca9f6d263ccc4a83b : SUCCESS in 16m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a61f586261cb4e6fbc392db10ecebbd8 : SUCCESS in 20m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5946aae839e34896b98f5b0eb1872bc0 : SUCCESS in 22m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b44efaded8bb4a8cb91910138a7fd0e8 : SUCCESS in 10m 27s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/01e356df6bed402c9465e6586d7b3d01 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f67c3cb3bbd642249af2fb81029762d9 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbd4a53f5cdb40528abf742a42efca1b : SUCCESS in 16m 33s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/f6982e4b962843d7827293fb555c8c3e : SUCCESS in 17m 57s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/623d7adfdf1445c68887cd45560a5766 : SUCCESS in 24m 15s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/89e70c877bd043d6ac39e87976e9fae6 : SUCCESS in 25m 32s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/7234b7f82647461fa59a8a58b33e3189 : SUCCESS in 18m 46s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/c1018b2c99ec4a4ba8af674762571f6f : SUCCESS in 16m 03s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/27a541939d1f42ca94f9f465d40b242a : SUCCESS in 10m 50s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/25e3e764cc404fc9bacc64d045e10332 : SUCCESS in 23m 09s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/159a320c54ef4793959e6f63ec4c63b3 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":20},{"id":"47f443a69e7f167ce8a2b65e24fa32a96bf0502d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 18:39:34.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"3df48bc38a93e287721fbd4784152917fc159f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 19:38:14.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1eff6e564a8454e8c6c1aa914796e99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d5b4e4a30794ab5b425e6a25e5ff16c : SUCCESS in 2m 01s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0ce78428b252455c8272752da8670b25 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa0619fe57c49a0b14e4f0ed6521919 : SUCCESS in 8m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff2360fcfc1b406d87fd8e35b1c6322e : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6a290d2f0b454698eac2441e8d7495 : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/5f3edade8da64e019d8af35f5b87d29f : SUCCESS in 56m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/358bd6eee6144c5f959d0de1c272cbe6 : SUCCESS in 41m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f809a8a645c44a8b799d1a5c3908fa3 : SUCCESS in 50m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f69027c885e34d2e98d175eba566ad48 : SUCCESS in 42m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87c2149add7d4041987528c4a480acea : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbdbf4eaecef40d58b91f8b669979032 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e439e6cd25824150a9782a245166edb0 : SUCCESS in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52ad85b8328341ee90ac5467713eafba : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06bba0d72c8d43a7a70e5f01a447616d : SUCCESS in 13m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39fc85d18ace4c22bc8d3ef772c820e5 : SUCCESS in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24d5c423f0664f61ba41e787b31cc3de : SUCCESS in 30m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/78afbf7dc9074eb198d44efa1de83f24 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/93ca3175ecee47669211a076a16db46b : SUCCESS in 16m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8f48ab3258564703a0bb268c643f0d1d : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/892f42be40814c1f8264c4f1d00f036f : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c36a2a2bcefa4171a61afaf71b64a858 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07c46d55b53e4cb6b2a3f92dcd22a574 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3868fdc093134efe8a45be712cb5f429 : SUCCESS in 49m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13d80d772752498d88aeeb95e6835864 : SUCCESS in 35m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9160c95d09e44e3f924c8cdea806d538 : SUCCESS in 48m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfa1140f082f430686b5ae991b136075 : SUCCESS in 21m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649331c39cf64913851fcea15563d0d7 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4e5fc93b0ae495eada0dfa714a8eb5b : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77609f9ecbd9447cb823a293fe68668b : SUCCESS in 41m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a495f799a0748d89872350a62ef39da : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":21},{"id":"013baf74d144aeded8ad114c9f1311f25a69ddb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 20:57:44.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87cba533f4ea495d8052e833f356ac32\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07b746ac1c664de0a1080ed67238eab4 : SUCCESS in 1h 49m 24s","accounts_in_message":[],"_revision_number":21},{"id":"412b9ddaada27fb76e56227837b7e520a2c52d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 18:50:15.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"234275b0ebff716c4d173c6e0a96fc800bec3531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 20:12:22.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4594a5125fcc4c7bbcb074c909ef990a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/971e0a71fa2849009a3785b9019030c7 : SUCCESS in 2m 58s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2aafb9c07e9a4c49b8f668c20a4f1088 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2881b3b538541b79487910ae20dd19f : SUCCESS in 8m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/38c1c82fe02c42349c5a74cfee5d7940 : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44ed041a377d4bcabd5010efa600da1c : SUCCESS in 5m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/10e4bdd93c964eb7bafec3cea2468383 : SUCCESS in 59m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84336a988a8d4a60a7330c8c0358a883 : SUCCESS in 54m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/642f7f9f9b5240c3b7a41e65d4ceceb7 : SUCCESS in 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a4ffd6ce79e4fdca989595a2bf893e9 : SUCCESS in 1h 05m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a00d6b55b74a47008d971437f748a8db : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ee19c248bae47659cd51ffd10085959 : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c6f397736534c81b3315a34290a7c05 : SUCCESS in 22m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be85f898a1004b49b26b46e73da0d194 : SUCCESS in 7m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e8aeb44d5e46049fab1449e7f5e273 : SUCCESS in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6fd0b6149de24b4b96dc09b2d1201d88 : SUCCESS in 14m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/245694f8311b4860b85dc9d36dd0f789 : SUCCESS in 17m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/29325f434c154e4fb1f06cd1c4d29c1d : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b2faaee4c0fa40dc824f8f8e9157a81a : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3cbed3f9ce984a65b948889474f4fa30 : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a3405e542ea442e9b628d5ef9f6a646 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62abb38659184497bba367b176bd786f : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59924937bd864a30b047ae644ec7a8d6 : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6db9cc6c1cdd4b5383bc1a276b8148f3 : SUCCESS in 48m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51946a8b13d44dce8850980d68d6b036 : SUCCESS in 41m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9e45bad713740fa804351585593f646 : SUCCESS in 42m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d192650b574f4567b78d159a8b104f47 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944f33bef63b407c9e361cd7178d046e : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca23ae77a0ef4d56b7285f99d5bde5ff : SUCCESS in 19m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/326dff4608f648e392c8ed6bf5886827 : SUCCESS in 50m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98931213c47b4813b321c2967b4a6c66 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":22},{"id":"ed9a9017d09953cb1127950f760853c8a05f58cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 20:34:46.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c270ef9784d4489f9b9e176c33f58d98\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ede12fdb32f46af8f456842b31225e2 : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":22},{"id":"e689cef3edddc6f86a24b30ac4d4b06e05eddbc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-20 17:11:30.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"e84838f39ecff6403e6734704dfdbab6a25539a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 18:29:39.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2f95ec94cf44950834c46276a0ef8db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a2f5d17913b41088054b78780bb73fa : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":23},{"id":"bf9513e4e44b35a6169505956b928424e924297d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 18:39:02.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d00b51134764ef99a8dac444ed9f0fd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18402a0a24a348e88b27b28b34806239 : SUCCESS in 4m 19s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b41187c9e5443a7b5d17dbba45fbe34 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63dec3063cf342f58d3a3a4b2932fafb : SUCCESS in 10m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6144ec9b05bb43c49344bc91a5898770 : SUCCESS in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3de5cea8b7ac48ca9b1fce9fedd6bde1 : SUCCESS in 6m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/fec8d52c2d574c31bdd74bbd0dfaeb08 : SUCCESS in 1h 01m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38aae93f19fa4fb6b3f24730bcf6a06c : SUCCESS in 1h 20m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a24fb3fb1cdf4760b28fb4435bd0b254 : SUCCESS in 54m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/478fe8cc79e34ace859fa87bbd7a4428 : SUCCESS in 1h 03m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa3321d964fb493e8a3a80c2e737fa55 : SUCCESS in 13m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d4fee66f3d349559b6d3a0e0ec7a825 : SUCCESS in 26m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb633823937646d2be47b233ac4e6335 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3d03735cc7e4311b1ea6b2b0705915b : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8848ba3b472e42b4964042120e28091d : SUCCESS in 20m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1413e7e5056e480e9a2d85b24c6dad71 : SUCCESS in 21m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/29e668d0374f4fb18337efe126fc0194 : SUCCESS in 27m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c39bab4769a41a2b910f9b2d7d0839d : SUCCESS in 11m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4fa30f9c58e48679cd0f50f2e1d3720 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/28c18efe5d1e4465a69d0a01ba068602 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f39e6a04a5c34753b152a0a572e56a39 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78385ad914594a299574d23a081b5591 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2b9db12f1d04f7f80d070c3fd688dad : SUCCESS in 9m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36e4e46331964c5dab78db8bb3e8bf33 : SUCCESS in 51m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/043533e957694ec190b034a4434f5f46 : SUCCESS in 27m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49d89a232353488996d6ef0960a61d61 : SUCCESS in 49m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eb4ba5140a442ff86683d0fdc604cf1 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da6f31fab7f42399194310357f4ac18 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4eb509fb3804b618228a40951185a5b : SUCCESS in 17m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db488c845aaa44ba8714a6139a931235 : SUCCESS in 44m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3db26c411264617993780ee226b8d91 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":23},{"id":"4655b478ea7e199446d426e1c0b1848841248456","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-01 06:18:48.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"f451505a8a2a2bf2ce1e32509aa1a88741ee0a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-01 06:19:29.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"2699fa68cb0d44bfcd5dd0142b45915ad96fcd7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-01 07:20:04.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0201a6b851c4bcfa20fa639718fb869\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/882f6971658a42c2a48df91c628784f4 : SUCCESS in 2m 18s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/749d89be5cff4215878334f55797ca70 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bcaa41df4c4cecbb2b27a4a2c77b76 : SUCCESS in 5m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a02dccaf15bb4b768a911cbddf35a57a : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69019f2d67744a4d9a2d703f40c5d327 : SUCCESS in 4m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/c075e1480328449685db24be498caf41 : SUCCESS in 39m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/75894bedab26497a887fe8508270efd1 : SUCCESS in 34m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4fc10a32c85f4cf890610b6b01c84529 : SUCCESS in 54m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d989e97eaaac407b92114972b215b640 : SUCCESS in 44m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/39195102edcf42e28aeb705e7cf7ae69 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/273c93e1cd084e9e9334a8ec32bb5c06 : SUCCESS in 8m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a601b6d826364ce2b709d7e8878ce717 : SUCCESS in 21m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c80b184f6a434b96941478944d626a12 : SUCCESS in 27m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3041371b7c95453abdb84ec3a633d588 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9738614268fe433e9cb35b3a0e34ac37 : SUCCESS in 9m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e78ca64f04e54bc0a18c51808d19e4b6 : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b82da408217745659115998358c677f5 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0343ae963eac4d7cbf787ebf9c4266ca : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9157a7aa3fb6476697898f62db154a85 : SUCCESS in 35m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a6a3e8c713841c0a15f854c17db4400 : SUCCESS in 55m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c8e749c4ba474f6a85c76f53fba2aa39 : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/021e45605c884c06935c1c7fa2eaa466 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea83209325174606b7a7f9b20d05f78b : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/472a179607d84f43b612aca669d2d85c : SUCCESS in 20m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d931dc48d5734097baaa1c6d043ddb9a : SUCCESS in 33m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3f6312240614a0080d57a52e80a2406 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":23},{"id":"37122f8371e24c0839a9e4ea8c5dfa68f7162936","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-01 07:20:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":23},{"id":"24e9e171398eca79868438d1290c3f35f616c252","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-01 07:22:01.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628ec92a90c6406bb662dd4aa707d880\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/05e9d79eee7643ab9e12444cdba4e554 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184fbe87e1cc4d7e83537a3755845ebd : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61d5e128c23547d89ff58a19e670f86b : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4d44489d351d4e9a918d628fb9772eea : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"b91a061aac3c95a5681da91ca144f2e75eb85261","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-13 17:18:06.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"7367907c58933944aaed52f0659be6203791a047","revisions":{"d090c296696064d689e9e393aef666dd33175cba":{"kind":"REWORK","_number":1,"created":"2022-08-18 06:09:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/1"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum pythong version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d090c296696064d689e9e393aef666dd33175cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d090c296696064d689e9e393aef666dd33175cba"}]},"branch":"refs/heads/master"},"97294ac52fc636271d515d085ef8744c4bb0115a":{"kind":"REWORK","_number":2,"created":"2022-08-18 16:35:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/2"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 16:35:19.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum pythong version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97294ac52fc636271d515d085ef8744c4bb0115a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97294ac52fc636271d515d085ef8744c4bb0115a"}]},"branch":"refs/heads/master"},"a7b9e096433fd35b19324589b769d755ad979671":{"kind":"REWORK","_number":3,"created":"2022-08-18 17:44:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/3"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 17:43:53.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum pythong version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b9e096433fd35b19324589b769d755ad979671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b9e096433fd35b19324589b769d755ad979671"}]},"branch":"refs/heads/master"},"f87542932b0a19cb29f5f63151a77fb990158ff9":{"kind":"REWORK","_number":4,"created":"2022-08-18 19:25:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/4"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 19:25:09.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum pythong version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f87542932b0a19cb29f5f63151a77fb990158ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f87542932b0a19cb29f5f63151a77fb990158ff9"}]},"branch":"refs/heads/master"},"6c247fe312415a65b97b39cc124f695501aaf839":{"kind":"REWORK","_number":5,"created":"2022-08-18 23:34:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/5"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 23:34:28.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum pythong version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c247fe312415a65b97b39cc124f695501aaf839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c247fe312415a65b97b39cc124f695501aaf839"}]},"branch":"refs/heads/master"},"ed3134993f77c183f9530c8e54e71bc5615961f6":{"kind":"REWORK","_number":6,"created":"2022-08-22 23:13:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/6"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-22 23:13:23.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum pythong version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed3134993f77c183f9530c8e54e71bc5615961f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed3134993f77c183f9530c8e54e71bc5615961f6"}]},"branch":"refs/heads/master"},"7b08d76945e28f200c9e894b0d9fb8f1d46a26b3":{"kind":"REWORK","_number":7,"created":"2023-09-21 23:29:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/7"}}},"commit":{"parents":[{"commit":"fe30715d973581386a6838cc87311ad09314d1c4","subject":"Merge \"CI: Stop collecting coverage reports for py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe30715d973581386a6838cc87311ad09314d1c4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-21 23:27:25.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b08d76945e28f200c9e894b0d9fb8f1d46a26b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b08d76945e28f200c9e894b0d9fb8f1d46a26b3"}]},"branch":"refs/heads/master"},"e81eb7d66eadc3d7f509d810d4f591344c635b28":{"kind":"REWORK","_number":8,"created":"2023-09-22 17:07:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/8"}}},"commit":{"parents":[{"commit":"fe30715d973581386a6838cc87311ad09314d1c4","subject":"Merge \"CI: Stop collecting coverage reports for py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe30715d973581386a6838cc87311ad09314d1c4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-22 17:07:01.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e81eb7d66eadc3d7f509d810d4f591344c635b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e81eb7d66eadc3d7f509d810d4f591344c635b28"}]},"branch":"refs/heads/master"},"976493785e9a736db354874c1ada416c42fe29de":{"kind":"REWORK","_number":9,"created":"2023-09-22 17:26:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/9"}}},"commit":{"parents":[{"commit":"fe30715d973581386a6838cc87311ad09314d1c4","subject":"Merge \"CI: Stop collecting coverage reports for py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe30715d973581386a6838cc87311ad09314d1c4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-22 17:26:46.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/976493785e9a736db354874c1ada416c42fe29de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/976493785e9a736db354874c1ada416c42fe29de"}]},"branch":"refs/heads/master"},"e8a7d2821ffc535d8b0ad6840cc439c31c3db755":{"kind":"REWORK","_number":10,"created":"2023-09-22 17:32:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/10"}}},"commit":{"parents":[{"commit":"fe30715d973581386a6838cc87311ad09314d1c4","subject":"Merge \"CI: Stop collecting coverage reports for py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe30715d973581386a6838cc87311ad09314d1c4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-22 17:32:33.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8a7d2821ffc535d8b0ad6840cc439c31c3db755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8a7d2821ffc535d8b0ad6840cc439c31c3db755"}]},"branch":"refs/heads/master"},"ad24bb8bfcc054c5dbfc7e1ea5f230d02ad1bacb":{"kind":"REWORK","_number":11,"created":"2024-01-25 00:05:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/11"}}},"commit":{"parents":[{"commit":"1d9ab5d0a8372e907ae0f1029e743b7a97554aaa","subject":"Merge \"Only try to install py2 dev libraries for py2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d9ab5d0a8372e907ae0f1029e743b7a97554aaa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-25 00:04:15.000000000","tz":0},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad24bb8bfcc054c5dbfc7e1ea5f230d02ad1bacb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad24bb8bfcc054c5dbfc7e1ea5f230d02ad1bacb"}]},"branch":"refs/heads/master"},"b11ebbe99159270ed10ea077a01bfcb977b3ee84":{"kind":"REWORK","_number":12,"created":"2024-05-10 18:36:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/12"}}},"commit":{"parents":[{"commit":"47cde0b28b53531927cd8e52f11d55fb3d97a8b3","subject":"Merge \"Test under py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47cde0b28b53531927cd8e52f11d55fb3d97a8b3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-10 17:59:00.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b11ebbe99159270ed10ea077a01bfcb977b3ee84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b11ebbe99159270ed10ea077a01bfcb977b3ee84"}]},"branch":"refs/heads/master"},"2fe58e8f017182c4bdf0c36b6b0e4fa4253dfe31":{"kind":"REWORK","_number":13,"created":"2024-05-10 18:42:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/13"}}},"commit":{"parents":[{"commit":"47cde0b28b53531927cd8e52f11d55fb3d97a8b3","subject":"Merge \"Test under py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47cde0b28b53531927cd8e52f11d55fb3d97a8b3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-10 18:42:48.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fe58e8f017182c4bdf0c36b6b0e4fa4253dfe31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fe58e8f017182c4bdf0c36b6b0e4fa4253dfe31"}]},"branch":"refs/heads/master"},"be282c9bde62dc7d472549a0a126c381c03304d4":{"kind":"REWORK","_number":14,"created":"2024-06-21 22:25:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/14"}}},"commit":{"parents":[{"commit":"68adc5bff358cde91544c63f5ae21277cba3cc29","subject":"CI: up-rev a few py2 constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68adc5bff358cde91544c63f5ae21277cba3cc29"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 22:23:55.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be282c9bde62dc7d472549a0a126c381c03304d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be282c9bde62dc7d472549a0a126c381c03304d4"}]},"branch":"refs/heads/master"},"1f0d08c6bc77e0d277e71489961889007412d36c":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-07-15 22:42:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/15"}}},"commit":{"parents":[{"commit":"e8affa7db5817e50cd82130a8f4c250725b80884","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8affa7db5817e50cd82130a8f4c250725b80884"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-15 22:41:27.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build py3-only wheels.\n* Specify minimum python version in package metadata.\n* Build py3-only docker images.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0d08c6bc77e0d277e71489961889007412d36c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0d08c6bc77e0d277e71489961889007412d36c"}]},"branch":"refs/heads/master"},"ff16f2046ea75cec2ce53b7f8a72a31994a04633":{"kind":"REWORK","_number":16,"created":"2024-08-27 20:06:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/16"}}},"commit":{"parents":[{"commit":"92eebe24c6837ef36302c801547c2de2e2726b2a","subject":"Improve test coverage for proxy object DELETE and POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92eebe24c6837ef36302c801547c2de2e2726b2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-27 20:05:11.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff16f2046ea75cec2ce53b7f8a72a31994a04633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff16f2046ea75cec2ce53b7f8a72a31994a04633"}]},"branch":"refs/heads/master"},"9cea1f18b92f8cfcc5f048510ecfe8a00267c66f":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-08-28 20:19:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/90/853590/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/17"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 20:19:21.000000000","tz":0},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cea1f18b92f8cfcc5f048510ecfe8a00267c66f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cea1f18b92f8cfcc5f048510ecfe8a00267c66f"}]},"branch":"refs/heads/master","description":"Rebase"},"c3c0b9c3f29e092c388923b83114f73fe073339d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-09-04 00:09:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/90/853590/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/18"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-04 00:09:48.000000000","tz":0},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c0b9c3f29e092c388923b83114f73fe073339d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c0b9c3f29e092c388923b83114f73fe073339d"}]},"branch":"refs/heads/master","description":"Rebase"},"99fc659a5876addd6ca67da971b56f3fdfb1325c":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-10-18 19:16:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/19"}}},"commit":{"parents":[{"commit":"8ded39bccd9192b570e24cf039a3a4c8e3ccc216","subject":"Merge \"logging: Add UPDATE to valid http methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded39bccd9192b570e24cf039a3a4c8e3ccc216"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-18 19:05:04.000000000","tz":-420},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99fc659a5876addd6ca67da971b56f3fdfb1325c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99fc659a5876addd6ca67da971b56f3fdfb1325c"}]},"branch":"refs/heads/master"},"8855a3ac785f73d0821685c4508a2b80f72c0a3c":{"kind":"REWORK","_number":20,"created":"2024-11-05 05:10:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/20"}}},"commit":{"parents":[{"commit":"97c5deae87e4120f09df38382ffeb6d913eb065c","subject":"Merge \"slo: add more functional tests for copying\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97c5deae87e4120f09df38382ffeb6d913eb065c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-05 05:09:05.000000000","tz":-480},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8855a3ac785f73d0821685c4508a2b80f72c0a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8855a3ac785f73d0821685c4508a2b80f72c0a3c"}]},"branch":"refs/heads/master"},"d7a29b24c0644c86ff91d63622ff4104f7dd8abe":{"kind":"REWORK","_number":21,"created":"2024-12-02 18:39:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/21"}}},"commit":{"parents":[{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-02 18:37:49.000000000","tz":-480},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a29b24c0644c86ff91d63622ff4104f7dd8abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a29b24c0644c86ff91d63622ff4104f7dd8abe"}]},"branch":"refs/heads/master"},"5de2ef99542526426d6c816ed7ea1a816939479f":{"kind":"REWORK","_number":22,"created":"2024-12-10 18:50:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/22"}}},"commit":{"parents":[{"commit":"f9a3f142ab16594c6c43b14b258ee417a80eecbe","subject":"Merge \"Make OldestAsyncPendingTracker timestamp float\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9a3f142ab16594c6c43b14b258ee417a80eecbe"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-10 18:49:00.000000000","tz":-480},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de2ef99542526426d6c816ed7ea1a816939479f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de2ef99542526426d6c816ed7ea1a816939479f"}]},"branch":"refs/heads/master"},"7367907c58933944aaed52f0659be6203791a047":{"kind":"REWORK","_number":23,"created":"2024-12-20 17:11:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/853590/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/853590/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/853590/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/853590/23"}}},"commit":{"parents":[{"commit":"394463000731ab90fc9365130a2ec7738b783ab7","subject":"Merge \"CI: Clean up deps for various doc builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/394463000731ab90fc9365130a2ec7738b783ab7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-18 06:09:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-20 17:11:14.000000000","tz":-480},"subject":"Drop py2 support","message":"Drop py2 support\n\n* Remove py2 gate jobs.\n* Build non-universal, py3-only wheels.\n* Specify minimum python version in package metadata.\n* Clean up requirements/constraints/bindep (a little, anyway).\n\nChange-Id: I53153c4fde043e964e1daa7bbf2089e0471dede2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7367907c58933944aaed52f0659be6203791a047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7367907c58933944aaed52f0659be6203791a047"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
