)]}'
{"id":"openstack%2Fswift~854251","triplet_id":"openstack%2Fswift~master~Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c","project":"openstack/swift","branch":"master","topic":"jh","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-09-09 00:19:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2022-09-08 18:52:15.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-09-08 19:17:18.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-09-08 19:30:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c","subject":"ShardRange: add is_child_of method","status":"MERGED","created":"2022-08-23 18:22:31.000000000","updated":"2022-09-09 05:13:26.000000000","submitted":"2022-09-08 23:19:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":69,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"854251-jh","meta_rev_id":"dcd7ecc668745e25e7207a624f95ad1bc455c14b","_number":854251,"virtual_id_number":854251,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-08 23:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2022-09-08 10:57:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-09-08 19:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-08 19:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-08 19:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 19:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-23 19:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-25 05:30:48.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2022-08-25 05:38:07.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2022-08-25 20:41:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-08-26 23:29:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-08-30 11:29:11.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-09-01 17:14:27.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"2823d2d60ec8234738a70310aa29cae5436f16ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-23 18:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fe780955825907cd261f6f3e268145959851e99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 19:17:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03744d903d734b0dbb519f4e1f57afa7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79ece08157c04dfc9eaa82b37f0c8ba1 : SUCCESS in 41m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96b6f813daf14b4f95714d658deb5d5f : SUCCESS in 50m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4eaefd386a14cdc98e2eef8c44563ad : SUCCESS in 40m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/606664fbb13e4e69b2236c520f5f9503 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b375ec825a471c3ac5341fb1cfdf5e0b3a8aa210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 19:27:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fb9131a1573470e9bab46318ddd5fe2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b942bc815749fd983f55cfa9b1d431 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/bf8aa66af13444118e1d6cad692bf778 : SUCCESS in 45m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e94190319d844949a12b15f1990ef36a : SUCCESS in 47m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79c714f390004ff19eda2244d73377e3 : SUCCESS in 38m 51s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/009600f364b24c5a846fdad0c160ae50 : SUCCESS in 24m 34s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/55cc4a4f27154ffe9c54a506ddb148e9 : SUCCESS in 23m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6934e99c8b634a0c8e33ac6c9c77d24c : SUCCESS in 18m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26e1f29497c4402fb68504d210701590 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a36b6ccce9ab4d2da8115a113dd608bf : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bdb3ca2ab7f492d9cfd40ce480211df : FAILURE in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e4a51a3da384ed9907da437d55468d5 : SUCCESS in 14m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df54b06846c348cabc3ff9207f624b95 : SUCCESS in 19m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/008f3f2097664514914d4f7d4c7b4246 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad79d9cdbea44a0d928dee8c2c187f21 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6e7775d5d6f047c68335ae9ed5773bfc : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/caf47efc7bcc486185a74116cd248e4d : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a2488abfdc1465bb94462a20a1e5d2d : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/288d9159779b426aa5e8f04f8756d8d6 : SUCCESS in 20m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a18fd2917d445109c4cccdff0710b46 : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd5e7fb0002646da814ca50e1be433ac : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e9b8824481d04c1c9cb27dc25934f271 : SUCCESS in 9m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/24b05fc37d024b3fbfd22c84b9f84a48 : SUCCESS in 40m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/322c19254f9e42518a9d94b0e389f501 : SUCCESS in 1h 02m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/687fd7bb4d0f4b678f866435aae51a84 : SUCCESS in 46m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56575128957e4dbdaf7a7d02f2554d47 : SUCCESS in 39m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12d2253ab8a34626a7539f1b593affc6 : FAILURE in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c6d85f6b5548b48efd686edd692463 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/901297d5905e4abe9e1da94ad6ff748d : SUCCESS in 22m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d2433bff8384086b1a8b297a1cb845a : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c0ff879c6b895e4e40a4c990b634daeb18622d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-24 17:50:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"651b747f7a6f5a36526589eb144e32961abde903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:41:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0740145cc5514e0597cb8046b1d2e8e3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56268c2736304ea2a1d75aa447d32dbf : SUCCESS in 48m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b242660a81b54760b47129d29963ee4a : SUCCESS in 44m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9ee2bc8821741d3af7a39401fdbbaa3 : SUCCESS in 33m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ce41f7c02b64e8aa5862aeff2898af9 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":2},{"id":"105f0ff3de567c6c7a2550253469f5301ab4da33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:02:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41da4bad38e3443f9aa178195d9271b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc629bd2dd634ccd98b604c029a682d9 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/068b319d03814bd7a1b9f168293f78eb : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0179ea508034bf196c88dcfdee9087c : SUCCESS in 53m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bad3572b63a545d291f78a6f56284fde : SUCCESS in 54m 25s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e8e788a4cd32486b9ef201569c921352 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4fc9253b718d4e35ace0bb57c6ccfef8 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0e46a8f4be80478cb9a8e9195b45b546 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1661ba57a72486f8ab622f397375bef : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/722f43e83f2a489aae8b41d92859caa8 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69af3cc5f6dd45a8b3c263295cc82a73 : SUCCESS in 15m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/742e529ecfd145e9a73d246ab98a7119 : SUCCESS in 15m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/401009002e934fae8a96f4bcae235ee5 : SUCCESS in 23m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/894c848d27cb43a5b5a09045723550c4 : SUCCESS in 21m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4caf39f6e9fe42e6b2a375cb36e33868 : SUCCESS in 19m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ef0a505cbd24e428dcf871420f6e529 : SUCCESS in 21m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/24d75b7241d04cf289e00ea0552c5e35 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/846c444869564eacadb11c13b476fa0b : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/590cae0dcfb84113bfd526f9bde4f5b1 : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07e0dc830258441b87945c1f39afa1dc : SUCCESS in 10m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42c533562c09487a928ef55f146f4fa6 : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8db3e539f9604640a1c72f487ab48341 : SUCCESS in 10m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f2ec02f95c24d6682f847dfbce027be : SUCCESS in 55m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f26270dbb89743fdb8e1494b23ab6f15 : SUCCESS in 1h 07m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/777a2becb8914fafb563c38c3cd07a0e : SUCCESS in 47m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a5b4010f8cc4890be676f5de8592082 : SUCCESS in 46m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2309d347ea9242b9a21f17c91bbaf4d0 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe3126414614994bb4a569169c5a856 : SUCCESS in 8m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2053517814504c9fad5ed71956977336 : SUCCESS in 21m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/894336e3188b466d8cd3a330b0d12cf7 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f01331a93a5f4b5eb588648dc3910a8926442151","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-08-25 05:30:48.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2c36be02ed50d8d35ee5b765336454a3c7693e84","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-08-25 05:38:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"87a618fa3b84266d023ec0163a89efc4694673e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-25 09:24:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7adc72715af45a90cd1e9d4889442c66b663eca9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-25 09:24:43.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3010fb31e031515e242f2cd60cd0183fd93ea48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 10:31:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5508e28c599445e2af6546a1a23c1f4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6daa2407508140e492f433b27e70af94 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/9f42ec6a5dd54cfa811cb9424c549cab : SUCCESS in 1h 02m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bbe6ebc2ba9b4ae6bdb3ac01e19bdf89 : SUCCESS in 39m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c7ee3de72ae47489986dfa715e6c09b : SUCCESS in 53m 56s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a8811d8c4ec847cab24e10ed8bc76d96 : SUCCESS in 30m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/81196cd60ea54ad08887bac23e775f16 : SUCCESS in 28m 19s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6bb58a6915d54d079571d25aef7324de : SUCCESS in 29m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa8e445b24ef45d5825c4c95da6fb71c : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ef7c9e1787943cda65e34b4edd9e758 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9d8ea97987543e6a63e1de8e143d7d0 : SUCCESS in 17m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f0defbc466c54a62a3352b7cbbca5e0d : SUCCESS in 16m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd689f0e51b946609260634c23c0b787 : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9e7049f16c148b8a4e5103e9dc0abac : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45ca66ef01b9427aa3e85569fc8dd726 : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da407b02c51a46889caeb05979aff891 : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ad9dda057f3747ffbec4c8886f5af15f : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1c7f0040de1441d1be08b29294146003 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4614f7a8dd5a4a4e84bd40117eae6bd8 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b30f5dc374e6417ca0ed32441bbed74b : SUCCESS in 13m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fdbe27e213054291b767854566ecfe52 : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc58b201d1204963a030df1c41d1e168 : SUCCESS in 6m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fa64f0752fe340b9876f3d797ff1004d : SUCCESS in 49m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/64866484562347e081673b8e6a483291 : SUCCESS in 59m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d24cda73c8e7497c9b03c319d6aa5b87 : SUCCESS in 41m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0739034063004829a402811e532d45e1 : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6aea50de1b4471bbdbf924630f70284 : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33bdb1937c2455e87bb321f1d385978 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b3b685bf221456fa8699e55102d1475 : SUCCESS in 31m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/025bdb6b7e0f4a71a2e002c68ddf5dab : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":3},{"id":"de58c4d01bdd652ade63d670e74616c7a35e81dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:26:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed54b3507f634f0f94ec5daf81963fe4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bbdd84b4ac446c09e1e951a5d4e83bd : SUCCESS in 52m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95b586a107584031b3ddceab44940c19 : SUCCESS in 1h 16m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e91996826be44e68d5da19f7cdd689d : SUCCESS in 1h 17m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbdc57a339844c31929fe240e05ada75 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":3},{"id":"f9d0fd1ef7f39d2bca20bb81309496d6900d812f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-08-25 16:49:59.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ee329e446aba3c1f138fb4700f0062e627fac1f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-25 20:41:53.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6cab0f8c2b20deb1511e4518b42d526dcb97d1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-26 12:07:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5beb03cc2413c59265779c547a8994f0da86e75d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-26 12:08:14.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"aa55de6e0961327ab4146cde0635b89d5f3c74a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 13:24:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37607ceed32e4bab87f48f817bd40b99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64adc7083c4493f897e2dcb78bc0fed : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/834f22a69f0041daa384426d2ef3862d : SUCCESS in 58m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1516f14e43f84e17940e0520ec43b3ca : SUCCESS in 50m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48b2347d883240a09db63c270044c9f0 : SUCCESS in 55m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3b2f929bc02b4e12a313a864fe1d133c : SUCCESS in 26m 53s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bf7a6d0fce744a82805585be9feba619 : SUCCESS in 26m 10s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d86024f1c195492581f993e27fb1b602 : SUCCESS in 26m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d27d136fe386478e8e9ac2f3d242b78d : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/48a1fca8f34b437aa82d3e2a736dbbf9 : SUCCESS in 10m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59edb5f4a1c740aea573147c21adcd80 : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e3e053c096a84cd19df619e3d090a749 : SUCCESS in 12m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1a26866df6c343da8c559d266c0a9b6f : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/647c32c6dd6745adb40718abe0154313 : SUCCESS in 21m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/746bbc2b21714644842f88a8e9d15fd0 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea41cb437da34d3786222cb961772e1e : SUCCESS in 24m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/041f51f9a3d14b33b03770b206b6697a : SUCCESS in 12m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5f08df0dd29a4dc1b6b4d8e254e99e3c : SUCCESS in 19m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/822ef34d84ad4feabc5e2ee10573b301 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ef66665c04341f08c0bfc8371274ef0 : SUCCESS in 17m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6617bc540d444ca5bda2db999ad33c1a : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b272df52c4a74cb4a9cfd322ada886b4 : SUCCESS in 15m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0f73334da98e47bc8602ccfcb3cb1153 : FAILURE in 55m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/630f12d04cc3441b9895026af1a45592 : SUCCESS in 1h 09m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b195cde245ef4e09b5c2551e19629825 : SUCCESS in 39m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75f5a92940394f13a0fde6b3976f1c3d : SUCCESS in 57m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/080d6bba36c94007ae06d35225fd73f1 : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a8355bfeeb4090ab07d68bcc97b23f : SUCCESS in 11m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/360c86cb98b54dc1bf2c397d898cb4e8 : SUCCESS in 25m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cb7e8be64124ec19bc27b29a65d57c9 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":4},{"id":"edd790225248249bfd447895629b279813e9c563","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 13:45:21.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef376f4131364c85a32f1a7b5f548a74\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec4e85db5f9644359fa167d5dc684a55 : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a59da08e47c48b78c38cd2f2ebcc0bc : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b314e289260445c8a3c1a8f94663de5e : TIMED_OUT in 1h 30m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4efa1883d3344e5bfe1ed39b69e00c5 : TIMED_OUT in 1h 30m 47s","accounts_in_message":[],"_revision_number":4},{"id":"ee599ea94a1bac58d5b8ca4fc8621494c1237a8a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-08-26 23:18:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bcf03be7745baa37c0d67d757d0f601af6121843","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-26 23:29:29.000000000","message":"Patch Set 4: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"772f5d676f27fd64768a315da62e617fe0d473c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 00:28:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab9d1ef8a8ff4242ac61c94ffa0dda08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ea849ffca0442bb5c54aa6f21f4368 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/5a2e2e898c0446028f5c325c1946d1a2 : SUCCESS in 1h 02m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/81d41e4c3ed44a3fa63157eacad73621 : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48f51522af724ceb8e9c29ff5cfdf22a : SUCCESS in 38m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/472aa2764117467492fa15655fa2a858 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cc76ad3fc80f4a04954828674aa46605 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b04de2e62dbf4b1abf52ebe50240698f : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/062e7dfc4a074bd1ad816af72e3b27e8 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8e5a25a4ec804c6b80b1876cde5b4735 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8503e00e1e3447edbced14356369919b : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79d0e1b3a1ee403e95ad048fcd767cb0 : SUCCESS in 14m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55fb7c9942e94aaa90cf741de9b45c38 : SUCCESS in 14m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d58b3a95688149da99d560165bc047ab : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac2a698623cc41e9b13b295cd3b4b757 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc5dede86e8f4b0db4e7b734f3019c59 : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/88838f231e4049409d92120713fc76f4 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/075e65acae3c4812ba7aeb23d0409865 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/98a0590755f4439c86604765706d048d : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27bb33b451a14a35ab14ee831c03c9a8 : SUCCESS in 12m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d3f1680fb0d4d7f8a6d5f5b493568f2 : SUCCESS in 21m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fab8f1bb97904b72b7c8f5942a1fc5bf : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db3963fb57864642831dcbc21313ee69 : SUCCESS in 1h 06m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/11b84b060d584207a28ed3f500e9e935 : SUCCESS in 52m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab467bf6426c454490d24099df4db0de : SUCCESS in 47m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1de2701571e54d83be287327352ae79c : SUCCESS in 44m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eefaba23603d48ffa76fe963456f2974 : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83e0960c32d49c9a9fd7f3c468c4312 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/407f1388f512490987109caffe925179 : SUCCESS in 22m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4edf3285e7e647f78ca9f4106d442b16 : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":4},{"id":"bc60d687d0be941c20ec38f1a0ded6bf2a2e63dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 00:35:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c71ac1ea7a428c847070b27afb25fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba6c741fb24a48a2a2ebb4f9c4821153 : SUCCESS in 28m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96a76d9a868f40fa884695038ecd8033 : SUCCESS in 1h 14m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5abffb7deae4879b92f6b773b78faee : SUCCESS in 28m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/41870e3551bd43bca6638c72c6141ede : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":4},{"id":"363cc468ca0f61c2928442c8356c550b2352f281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-08-30 00:05:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8a26d3c5ef678e6483c30e0a9f31664fb1e60f76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 01:09:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19959e8fbb324d699b56793b159a056b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bfdcaab4e3c43e58b691f935135ba03 : SUCCESS in 30m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/903c7b35b0ca416cbdd8af0d184da87f : SUCCESS in 29m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/482a906ef3ea4295bc6b20370c64e11c : SUCCESS in 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/649de07bcebd4ee79ba93c1ad7c91a36 : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":5},{"id":"cc58227b4d206c2e4e8b70cdf51506dcbe3daa99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 01:19:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305947085ade4fdbb8575c4d887d6816\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e643a74c7a244d4a6121f4d42dc764d : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/03fb7b1b00704dedb6a5502f048ca41f : SUCCESS in 58m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/46659834a2e441dea51ebf61e86aa5a1 : SUCCESS in 53m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9bf8798f2194820b424da5a8659e549 : SUCCESS in 52m 55s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d9cd09faa5a640eea9505806df0dec17 : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/69d55963c4f1405e86406ab2d95d341e : SUCCESS in 28m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/847bc8e9a91e497bac1293ac8027bbf5 : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f50682771c044d2db7fdd1e5eff6c0a3 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/71750b46fa674550be2c84ce599d615f : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/afe59a09e53e49da8a2bf15e5829a8b1 : SUCCESS in 12m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26a55883206e4dfe8b04c4b0b7779498 : SUCCESS in 16m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2a89f3101514e27aa55b36f6a1c8eec : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e39dd8380dc845728997d104396e4be4 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc4a52e76fa04106abd040f20e475eb6 : SUCCESS in 21m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9fc67c1429b14d929f4a2d43ec313dc7 : SUCCESS in 30m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9cfee0e9de0c4e6489fbbb75adffa49d : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8b588edf4cdb4eb0a0112e5b7c7f7ca1 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/064418c595a448f08462f01f7f3c4286 : SUCCESS in 13m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59b01e3b568a4bb483a6b1ef47f0aeaf : SUCCESS in 8m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bc74e4a8a3941c18a9c4d2c32655f0a : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/88f141c817f9491db40c98278a113c3c : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7730a6307d944218fbbd10b705acfb4 : SUCCESS in 58m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/58bd1a8b79924e61b0b8eb1c0cc549e9 : SUCCESS in 59m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99926c84c8a3476fab476b52b6e0b64b : SUCCESS in 45m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30b35effd0b241a9a92fff30e145813c : SUCCESS in 44m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77be23145f914bbbaaf3c6d3a955bf63 : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f52e30abc1492caace00248dd1f70a : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04acd35b017549f19b816f43e3453031 : SUCCESS in 30m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ea13e53b8a9488e9722b55159733eba : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":5},{"id":"9259c54d0be6c0b2b5921373c3d0e5a6452e8ad8","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-08-30 11:29:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fd66fa97635afc48cfd055973447bd49cd8f9fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-01 15:56:37.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"1ad9962782f89625d3705c0910f039e75c07caee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 17:12:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb026b9cbef4d6588bb09b112673966\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04008cb869a48408c8104fc4f50eaff : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/318cadb8443c4059b4e3343b7905a808 : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1339fc0bbea049d587e5ee7fc443c4d9 : SUCCESS in 57m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66c0c630ac454425b53e34c9a5b54b80 : SUCCESS in 1h 10m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/028306f7b4f14248b160b52493b3fac5 : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d74f78bd6c4f4602a5a838815d2da589 : SUCCESS in 23m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9152348784964182b0bb93601ee5108f : SUCCESS in 23m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86b7cc15cde6411e9330508eb2f17bb3 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/01cefeb2e8034dd3827dd746c23de085 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/735b94dca9644afa839dbb8b66ffd97c : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/636fb38575114788b245f08148a04f89 : SUCCESS in 14m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/633bd6cb29bc4c72b16d4dea9abe758e : SUCCESS in 13m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e94498658d2f4b168f0b49a845154d3b : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06dc1405ed1942e9af91a0a4d7a4c38b : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0642b6ff495d4a82abe5483fe8ed2acb : SUCCESS in 17m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8efdcd5c92f9448fa141b307f26df8a3 : SUCCESS in 22m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ad8b477b40c641c8865f182b620742f4 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e82eb7a6bebc428f8f66dcaa8da9aa20 : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f24053cdd07473ba7a7c24823b3c1c4 : SUCCESS in 9m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6e4bb52881e4f94bedb4cecb9bfac56 : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a43b8be5923c4ec9ac5e0e77bcbe35c5 : SUCCESS in 12m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/696811c0d24a448bbf01c2cf9b7b7aa1 : SUCCESS in 57m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/93411e23ac7b451781e798d1fb9e06d1 : SUCCESS in 56m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e02e5a585dd4d3f93ed17e819418c3b : SUCCESS in 47m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c7b7e9684e5406594d2754f8a24ce34 : SUCCESS in 47m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e83e8b5664c34de4b6c830645eaa4483 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9df1715fec4a1bbed2986f440b12f0 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0019f13f5f2c494782ec4b533c3223dc : SUCCESS in 22m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de2441ad6ef04d95a5a57a531ec4f197 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":6},{"id":"d9c9389a60b6f3d86a228428a1f01278b1b6e4ec","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-01 17:14:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a000ddc2954c588e5e65aa55644ad70a39ec2d42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 17:41:46.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be567c9f57f4accafc7a69f23294dc8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1efd0ef6753141acb580f3abeb478835 : SUCCESS in 1h 11m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39b1d256d8cd4202b1d5f537550935b3 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c334909576e148c6969aed4d928f254f : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0bb550063be403aa602b8c52920205a : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":6},{"id":"c63d375a7dfb2db98b20ad1aee0d2d6235e2c7a2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-02 15:22:54.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e9736441054fe83bda7cad6c03be50f1f6d188c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-07 12:56:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"412583fd541a2d0ca57399f6bcffbc1b7ca22c32","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-07 12:58:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"79514229f4288ae522d73157587f650ab1d4b2f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 14:04:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e628261277654152b2b7f99f11de197b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59f358bd87434ad3a496aff10fa565ba : SUCCESS in 51m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/149f152e0ae949fe85060f13802edc39 : SUCCESS in 30m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/391247e0f8064762ae1d0c600c1308cc : SUCCESS in 1h 03m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e44020b69661472d9c4a16f06840da10 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":7},{"id":"51fea33dfc1d5679ec080165a800a1e55cc2900a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 14:05:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e81de058420043d197e0932fe9caa21e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac21d41929b464abd96af5e764e358d : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1c70e745d74b41d8a26e5ece081787f5 : SUCCESS in 53m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c129167e11548eb848b7a0bb31d3b6c : SUCCESS in 48m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b8a477b36bf4d77bda3cfbbb63ea3ed : SUCCESS in 1h 06m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8156773b62a846da8e0feac892b145d1 : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/652c34e670e9448f9748fbc379b6b734 : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ec014bc44a04d54928101db246da2ef : SUCCESS in 26m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6ea36704163457d9cefec3bce8695ec : SUCCESS in 9m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ee410c27814401084c0881cdd8e5e2d : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32ddc4cf8b9d471d88b4161f3d8fb2ab : SUCCESS in 19m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b01ae1d8f2942cdbc111f5e49f6f3ee : SUCCESS in 15m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e55f8a2eaac4e6cbae180b34d40be02 : SUCCESS in 20m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8f9f0d5d5854f488eb5181b46146437 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6546c8d223f546f5a07425d565d0e5e0 : SUCCESS in 21m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d3f92a49a573405f9c385092cb2b78de : SUCCESS in 23m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e73f6d616e0a486ab95a33f268855ee8 : SUCCESS in 19m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c9b46b5274ea43e390822b4cfb5b4c97 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/82a1c44b996e40dc80c3d4e4021c5628 : SUCCESS in 22m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d14ba8e9613d4bc4874714048eb0aa20 : SUCCESS in 7m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47bee683e8174253a92dbc36d196e789 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/085be98977eb4652be5acc1a1454235b : SUCCESS in 8m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f6a7681a8c7047cea0512b61a1214c9e : SUCCESS in 1h 01m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6219715855ae43bcbb4de7108968ce01 : SUCCESS in 58m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae2b815025cf436da1f4d42c79383497 : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f67c2fab3464128a1ae9ad8753cdd29 : SUCCESS in 51m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63779fd8045a4526a44de1678f4b70f1 : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f0a7b64cfec415fbc9ba905d2d36026 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5d5297185d64fc08356a726d377bc9a : SUCCESS in 28m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a92177b1e074353aeda18e6a5237893 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":7},{"id":"e66935314139b13eaa2ece89c29eb6416339c97d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-07 23:34:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"740c4ac090962b19b3a674d40d7dbf7a5aa74540","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-07 23:47:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3d6ff91f43b67d2dace48a30967ead643560fef0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 00:32:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b2dfaf649448caa95bc724af062c7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f35e976a8b5349a4896ef2597377d70d : SUCCESS in 36m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53effa8dd1814fd39242b1ab8b104ce6 : SUCCESS in 28m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7affca11bad47ed9e39eaf4e83cb3cb : SUCCESS in 28m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/035de372e1a2456ca9f8e83742a23321 : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":8},{"id":"772dc22c4008d038e5ea6a6c5193578e96aeec70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 00:44:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/199142edf1604c488dc82dd031b94745\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f8bd3d642f4be9aea69e6e2eda4adc : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/60b0048c6192424e83c41517e95eefc0 : SUCCESS in 56m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ebc092e72b04bd88e39bd43d1895dc7 : SUCCESS in 52m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/891a2528d7ea467f876eb5421b1a8951 : SUCCESS in 1h 07m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc79e8705b4c4daa8cd9e80eeccf565c : SUCCESS in 25m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c246b7ecf1c14375ba835c77df178e20 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8ef49b5cc8b4b35a40631a671033fcc : SUCCESS in 25m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0fc5867f450342d4ae47b23421019f98 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8037688cb2834101b3b0b5077b91f333 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1aa0ffc955c64c6aad6648a31ab47153 : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f64c5ec40734155a946a8fe3001287b : SUCCESS in 20m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a27f1eab053485d8b4cda39bf225088 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c10bf473cc444371b32f89d5e4b0992a : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/337adece7ee1412cb50d8b9105ab06f3 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ce59f0bc8f04dca93e70cb4d815b36c : SUCCESS in 17m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d5b57bfe68e4064a4f473e6e17e59f2 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e0146c16947e4bd793e54bb0885cdf37 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c751356d39d0478cb567130a12418d1f : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2579f63da4494ba79d896cfbfec772d0 : SUCCESS in 9m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4989b3f977a6447ea8a5bda7d689660c : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/48bc9c7820ac4d0ba5e7e6be699cba90 : SUCCESS in 9m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59c49a08e4044d679eddf2f584b90318 : SUCCESS in 51m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/89d0904a742a4f01a0f7efdee565bb52 : SUCCESS in 59m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf8ef862404c49569523c313eef10f7c : SUCCESS in 50m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91c0a6afad4f4cad8e47808566f15024 : SUCCESS in 45m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b49246f72d0479d88c1fc7fc99e88d5 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5eaed1983549c985f58fb50ea507af : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ef77ab75027460794796891a6efabc7 : SUCCESS in 20m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66ddbc331fc34d75895a58693e6f6900 : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":8},{"id":"252107a50a9f07ef5bd8a04fb4c281fac717dd12","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-08 09:54:31.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e743946f0cd7856e5e33babdf3a1d74ddb7842d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-08 10:57:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b208933215874dc9ea48e1fc2567d2b4b60b6b55","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-08 10:57:15.000000000","message":"Patch Set 9: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"99fccca1f3f39455c7544b44639f1b713ce46e94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:56:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35065d3c80504039a522f54e7d261431\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7eb2404afd6448998faa8ac5b185758 : SUCCESS in 55m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebf125e2221b4fae9bb3c28270f2164e : SUCCESS in 38m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e5202e72e03493dad3cca01c256cfa5 : SUCCESS in 40m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7523ed1ed98e4099a0dbf05e54948314 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":9},{"id":"e3223260f7e44b561a47e6bc9ed6b4505a057ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 12:20:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a30d9f83b512442780fc669e06ed8e4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aea3b170afc4e4c91b144aca26f2383 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/caa2ad903fa540fc819101d4bf139956 : SUCCESS in 1h 02m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15bb13e5a24442359fde938fb42273e5 : SUCCESS in 51m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7db6eab87a04d3d828e911fc75623e5 : SUCCESS in 1h 17m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f6c012dc54149d6a02938cde1acd6da : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dce871b4e55049d98b89dffe729fb9dc : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08a6c9ac15124334a0ca37ec81b4bac3 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2d34ab5ddd3403a9f2ebe248b3e41fa : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02592a822fc74489890b2a8d9773a0c5 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55b0637b248445fd979aab8ca76b840d : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb4f2a70602a4af29047668a371c6760 : SUCCESS in 17m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b96a077e9309478985ba3883bd21cfa0 : SUCCESS in 15m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0aa76e5e12fc446996da5b97983a5e89 : SUCCESS in 21m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cd06d97eb7a040599e3fe70e634db8b2 : SUCCESS in 22m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a93eef725b246ff9b2b120074deb5c5 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/922c77e453f2491ea9ba7db37c95d3d0 : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f28d25328f86405493f85078dcfff811 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80c82df6885246a6ac92eb5fa0d93eb4 : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be22da3648d24a9aa3bd98703b295fc8 : SUCCESS in 14m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a26c8d37901f4ef2a7d7c6987803de3f : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/013bc37bf9684782ae008921afcd320b : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/846e76d71bec4c7990f61659096bbd24 : SUCCESS in 1h 03m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4aec7dd754d4cb0aefd43ed872378c8 : SUCCESS in 1h 07m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4a85c23f788471c9651cca068cd6604 : SUCCESS in 53m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/770870ac33324a508f12cb999c27e4c8 : SUCCESS in 42m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95445838974b49c89e1644fdc8728ea0 : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2807c73120b486ea6d086c95bf2a4c2 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d0d095f8be74b378c34c10652906ad6 : SUCCESS in 26m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3f6d088b1c14a59ae84628b499e1415 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":9},{"id":"1cd87dd88c9e1cb2dcc37baf77374e056f5d41ba","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-08 18:52:15.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2cf705427506ab1248db291ac9481d5a756a99d0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-08 19:17:18.000000000","message":"Patch Set 9: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6caf78dc2f56c0b4a994570459e8bf49bcb2c125","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-08 19:30:41.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7706652688e1a154d3815603ba10245a39d70031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 19:31:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cd0b8c1e51fb6333410d40e7cfa988a89e89736a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 20:40:06.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61dc977f667b423db98d28945b68c6b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f04f99b6634409bbd5f0f9c818c8fc : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/3cdb282bbce94c70809ff7cfa2be11ba : FAILURE in 50m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2bf9d0df797942098f1431415c6305df : SUCCESS in 41m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9b91af969ec44e3bd4f1b081eb8107b : SUCCESS in 1h 08m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0e2cd51369a14864bc1aef0a14b48ee2 : SUCCESS in 9m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/66af3cb47991444b9dc8c9ed1dfa7bb7 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24e1840e48ca4f3dbf4ce080461513f1 : SUCCESS in 19m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a6fefb0dc9a43fab622948df5659ab4 : SUCCESS in 19m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4ae51cb2bfd64e3d973871ef8db6765e : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac919acb9e5647429cce9293910b1f5b : SUCCESS in 23m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08c184f1bc82469db27eb23cae998b37 : SUCCESS in 22m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d252f292f754412a48fd0845ed4bd71 : SUCCESS in 24m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/224b759c4dcd44ceb319ff0ebc7481dc : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/207d79fad2f34498aca35337ce8226a7 : SUCCESS in 23m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5998a41e5fc04daaa0e05b3f89e9c080 : SUCCESS in 20m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/753fc05b3081443e81672467736c7eb7 : SUCCESS in 8m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e69c7ce9e9cf46f5aa2537d52b3d2c1e : SUCCESS in 9m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b2b39c2cfd4f4f6fa8d16a1c7df61165 : SUCCESS in 43m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5108a0d6657a44a0b9df192baeda245b : SUCCESS in 1h 01m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a172c4992bc54f06b0445aae03202f58 : SUCCESS in 46m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eadab127eccb4a95a453dac076dca203 : SUCCESS in 51m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5e8b6e457cd4394a641679d0734aab8 : SUCCESS in 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e865bba304cb4195a9d78d782228ed63 : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/930f1bccabac45518be241b0d4bf4dd4 : SUCCESS in 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b928ac72efbe4cdd9dbadf67b21e3485 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":9},{"id":"33279374419fe2a550d58b475b87fa2282a5cef8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-08 20:53:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2000fc29e0a7012ff6b55363d8a8ab9687c96871","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 21:31:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31917a2ec49f475e8bf1fb3e444d35d6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46db6b3987654998a36dd95e6fcc30e4 : SUCCESS in 29m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f342905462f6406996dcd29383a74826 : SUCCESS in 28m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0dc815302be5428f8a5e432c6a9095d5 : SUCCESS in 29m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/00bf43d106f74ece8f4aec772a9eb52b : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":9},{"id":"ffa56168560b787acbda535f84b612d88e0bb8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 22:03:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da39e450a7145b5bb176a42ad674c3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201e74a072b145388de29cfb6765871e : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/95f0b4ead9cd4761b367a34dfeea1c33 : SUCCESS in 59m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4771f7b61870487eb68384a0bf5214bb : SUCCESS in 41m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdeab396fa3f49fd9aa094c6f1f3a6dd : SUCCESS in 1h 04m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f17ef5c32e22483fa0ab758770c0a13d : SUCCESS in 27m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d0d755b5000451f9958231e43b2a40f : SUCCESS in 24m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa51ffbf71ad415d9287499a9bec8482 : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/537f898b9cce46a3a6b1a5c0176cd9b8 : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91db5cb8d633407e87062c880dcbf827 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/753c0c7d6a0542f3b14474cfda99f7e3 : SUCCESS in 20m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aaec11a6adb74df88c224fce5a624e8d : SUCCESS in 18m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b5eaad3592c84efc8b0759a4059ed063 : SUCCESS in 20m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/938dfb1971a2450891adfcb383436162 : SUCCESS in 23m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab9b96102415402d91587b7d2e0af4c2 : SUCCESS in 21m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3938e2a261f5483e895a229678ced0a3 : SUCCESS in 23m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d9df817c58245e998ac593e6a47a515 : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/71396a7bbcc34628a02c4a35e5ab60a9 : SUCCESS in 21m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b17442788fc40d6a9de96f5ae776066 : SUCCESS in 21m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c83952b114a541d4ad41f22b1b09d1fb : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bc6651dab27429da6200a9c95ef13fe : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9dbad0ed53ff428083fa5c541688754f : SUCCESS in 8m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1fe8fc2c1935435d9f795b8e09782447 : SUCCESS in 1h 01m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c7a5137cb084a72b9b0e1fca4e86e0d : SUCCESS in 1h 08m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d61f629605d4498b255fe48be7da152 : SUCCESS in 44m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa02d90e9fe5474998429279d5a7c0a1 : SUCCESS in 44m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4213e7b97e984af7a28e3f596a0651bc : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102fab88e7e34a4da87202f7cef313c7 : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9179ee3439443b995bf1b7048d9088c : SUCCESS in 20m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61987da4b64342879204d448a9529865 : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":9},{"id":"30269b9fd61e3c8b41a7cc319853b5eb9c741095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 22:04:05.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"f3abac6c4c1fcb9759ba243659ee6f3c3a870c32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 23:19:42.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7cadb7f60b48be927f396c231f8baa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ba030ee1de44e491aced0ba12b9e99 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/3401152ac4574ad29860aa57b3112d95 : SUCCESS in 51m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e6073c60ee94ae091119ec9aa8fca73 : SUCCESS in 40m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/168ba283cc42420e99418272e706d08c : SUCCESS in 59m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fe3a8f46359f4477b9c2c4a4ce636b85 : SUCCESS in 9m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a75e750320894442a2c0c32b7207b96c : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/befc5f2c15bd479195f7bb755135d144 : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08d26cec25e5495e9667083e661840fc : SUCCESS in 14m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4d6d7dea789e471bb3e88db916cf21f8 : SUCCESS in 17m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5744717d514c4a30aa15ef8e2aad11ed : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c51a4b370df34aad83c680f17b024f61 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b92302f2f41f4786b3ab904ddb572228 : SUCCESS in 20m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7d9a965066984f83a2284855ddafc65d : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2fa1f1a193e402296481b45da9d3d9a : SUCCESS in 15m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/374b02f5ed0843adb510820f5382bc1c : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33818ef2489044488ba67ed0aa4c772a : SUCCESS in 8m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e987c55a7e61464cbeaf44961b794128 : SUCCESS in 9m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a6474cc636242b38d3aa1ea58bb6048 : SUCCESS in 58m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/acf81a1e2d5047d99fca355ec660b3d0 : SUCCESS in 1h 04m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5894107698c470aaa4e6d073c90624b : SUCCESS in 45m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e2ba8efe2bf4aab97e3b3c69e196376 : SUCCESS in 1h 10m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/276df788bbaa43a8a2396ea438970785 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b21465d3ba414d889bd40df9ce5413 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ab7d22056fd4add9d5f4f00ed741a3e : SUCCESS in 24m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/969d90a2124a4ac88fb2dc118c3aff3c : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":9},{"id":"fc6b0a16e26de3e95659de57b1ec27ee82c76586","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 23:19:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"2b0b56933d2d6fe5f0244deb2ef679e31500f813","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 23:20:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92119782d47640a087b0f5c136d48751\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab647e70e4f4652a80c7faa4467a416 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a874d5ad79f647f18a4bfa5f01453d8b : SUCCESS in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5fe6daa92584463a9d4903dbc0bbb995 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5b6e8245ca7a54a9b836977ca35f07c9d393958d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-09 00:19:30.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"dcd7ecc668745e25e7207a624f95ad1bc455c14b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-09 05:13:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"73bce6331221a7ac71f798f3868ba2fdeed55b9d","revisions":{"113fefd152661a646eff3bfa6677e348808789b7":{"kind":"REWORK","_number":1,"created":"2022-08-23 18:22:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/854251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/1"}}},"commit":{"parents":[{"commit":"bdd2b4b74800650a04fa9f6531098870d6d17ee6","subject":"Merge \"Add Jianjian to authors.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdd2b4b74800650a04fa9f6531098870d6d17ee6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"subject":"WIP: ShardRange: add is_child_of method","message":"WIP: ShardRange: add is_child_of method\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\nTODO: more tests\nTODO: deal with hard coded .shards_ account prefix\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/113fefd152661a646eff3bfa6677e348808789b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/113fefd152661a646eff3bfa6677e348808789b7"}]},"branch":"refs/heads/master"},"61348fb0385db889c355f20b3e030b005a4ec719":{"kind":"REWORK","_number":2,"created":"2022-08-24 17:50:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/854251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/2"}}},"commit":{"parents":[{"commit":"bdd2b4b74800650a04fa9f6531098870d6d17ee6","subject":"Merge \"Add Jianjian to authors.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdd2b4b74800650a04fa9f6531098870d6d17ee6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-24 17:45:00.000000000","tz":60},"subject":"WIP: ShardRange: add is_child_of method","message":"WIP: ShardRange: add is_child_of method\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\nTODO: more tests\nTODO: deal with hard coded .shards_ account prefix\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61348fb0385db889c355f20b3e030b005a4ec719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61348fb0385db889c355f20b3e030b005a4ec719"}]},"branch":"refs/heads/master"},"7822cfcf95776b4e75bf8fb3f3ec6e7e75948725":{"kind":"REWORK","_number":3,"created":"2022-08-25 09:24:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/854251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/3"}}},"commit":{"parents":[{"commit":"bdd2b4b74800650a04fa9f6531098870d6d17ee6","subject":"Merge \"Add Jianjian to authors.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdd2b4b74800650a04fa9f6531098870d6d17ee6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-25 09:22:30.000000000","tz":60},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names.\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7822cfcf95776b4e75bf8fb3f3ec6e7e75948725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7822cfcf95776b4e75bf8fb3f3ec6e7e75948725"}]},"branch":"refs/heads/master"},"4569d80ab7437260bdef64d346e13eb4d259ac88":{"kind":"REWORK","_number":4,"created":"2022-08-26 12:07:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/854251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/4"}}},"commit":{"parents":[{"commit":"bdd2b4b74800650a04fa9f6531098870d6d17ee6","subject":"Merge \"Add Jianjian to authors.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdd2b4b74800650a04fa9f6531098870d6d17ee6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-26 12:07:25.000000000","tz":60},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names.\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4569d80ab7437260bdef64d346e13eb4d259ac88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4569d80ab7437260bdef64d346e13eb4d259ac88"}]},"branch":"refs/heads/master"},"17cb6ef80ad7dd59d51d45791a3151e6ea1bda02":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-08-30 00:05:49.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/51/854251/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/5"}}},"commit":{"parents":[{"commit":"6504a5f59e84c3d56409836af522866c414142ca","subject":"Merge \"swift-manage-shard-ranges repair: ignore recent overlaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6504a5f59e84c3d56409836af522866c414142ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-08-29 19:09:18.000000000","tz":-420},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names.\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17cb6ef80ad7dd59d51d45791a3151e6ea1bda02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17cb6ef80ad7dd59d51d45791a3151e6ea1bda02"}]},"branch":"refs/heads/master"},"26b7bf2dd81f6976b1ff87da1a5e5beabb4c144a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-09-01 15:56:37.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/51/854251/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/6"}}},"commit":{"parents":[{"commit":"b05b27c0b63cf350702a094a4ca112e850926a91","subject":"Merge \"Add note about rsync_bwlimit suffixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b05b27c0b63cf350702a094a4ca112e850926a91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-01 15:56:33.000000000","tz":-420},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names.\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b7bf2dd81f6976b1ff87da1a5e5beabb4c144a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b7bf2dd81f6976b1ff87da1a5e5beabb4c144a"}]},"branch":"refs/heads/master"},"9f2367e06c7379aa3b83f59856a3aa13077a31f6":{"kind":"REWORK","_number":7,"created":"2022-09-07 12:56:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/854251/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/7"}}},"commit":{"parents":[{"commit":"b05b27c0b63cf350702a094a4ca112e850926a91","subject":"Merge \"Add note about rsync_bwlimit suffixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b05b27c0b63cf350702a094a4ca112e850926a91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-09-07 12:56:42.000000000","tz":60},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names.\n\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f2367e06c7379aa3b83f59856a3aa13077a31f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f2367e06c7379aa3b83f59856a3aa13077a31f6"}]},"branch":"refs/heads/master"},"5bcaaae5cc40f68ca43d225f97523f327f7c663e":{"kind":"REWORK","_number":8,"created":"2022-09-07 23:34:56.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/51/854251/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/8"}}},"commit":{"parents":[{"commit":"b05b27c0b63cf350702a094a4ca112e850926a91","subject":"Merge \"Add note about rsync_bwlimit suffixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b05b27c0b63cf350702a094a4ca112e850926a91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-07 23:34:33.000000000","tz":-420},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names. This new method is limited to\nuse only under the same account.\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bcaaae5cc40f68ca43d225f97523f327f7c663e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bcaaae5cc40f68ca43d225f97523f327f7c663e"}]},"branch":"refs/heads/master"},"73bce6331221a7ac71f798f3868ba2fdeed55b9d":{"kind":"REWORK","_number":9,"created":"2022-09-08 10:57:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/854251/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/854251/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/854251/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/854251/9"}}},"commit":{"parents":[{"commit":"b05b27c0b63cf350702a094a4ca112e850926a91","subject":"Merge \"Add note about rsync_bwlimit suffixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b05b27c0b63cf350702a094a4ca112e850926a91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-23 18:21:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-09-08 10:56:53.000000000","tz":60},"subject":"ShardRange: add is_child_of method","message":"ShardRange: add is_child_of method\n\nAdds an is_child_of method that infers the parent-child relationship\nof two shard ranges from their names. This new method is limited to\nuse only under the same account.\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Iac3a8ec5d8947989b64aa27f40caa3d8d1423a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73bce6331221a7ac71f798f3868ba2fdeed55b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73bce6331221a7ac71f798f3868ba2fdeed55b9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
