)]}'
{"id":"openstack%2Fswift~854420","triplet_id":"openstack%2Fswift~master~Ic854dc3c393d382a8acd20d89f56bff198a2ec5e","project":"openstack/swift","branch":"master","topic":"replace_setdaemon","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2022-08-24 21:22:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic854dc3c393d382a8acd20d89f56bff198a2ec5e","subject":"Use daemon property instead of setDaemon method","status":"MERGED","created":"2022-08-24 14:21:54.000000000","updated":"2022-08-24 21:24:28.000000000","submitted":"2022-08-24 21:22:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854420-replace_setdaemon","meta_rev_id":"63e70733b026eb62115e3f2c9e6b7a0410eea955","_number":854420,"virtual_id_number":854420,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 21:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2022-08-24 16:39:48.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 19:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-24 19:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 15:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 16:39:48.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2022-08-24 19:09:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9c9887b9f839856f6d337d52a55c65fe73053497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-08-24 14:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4adc9f499cca9e6603301de250a9d788f11a7060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:21:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf50550e70048da9b19f5fca6fe438f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c72169852b3450bbde3eab6fc49baf8 : SUCCESS in 8m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d92bbbf80dea40b3a9d2d3b1a23c7d8f : SUCCESS in 58m 12s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/81b327b917924b5886aed00c09212d46 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/412c4d1ab7eb422e92a782cf4934fda0 : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0e87810ad8144eafaffb2ad1a4065aaa : SUCCESS in 22m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b26965a992bf421db10594f05a704820 : SUCCESS in 10m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/19d870e6ddac42ab8d3efe9470f3ac77 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2cbcf05285ee4027b8a13b11c176d059 : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c729d8133af4bb8ba424d3baffa5560 : SUCCESS in 16m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef12d85ec9f4554a486a9982ebc8257 : SUCCESS in 15m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f8cf64f43044cdda91b5626b1863183 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7e47079304242fdb1161d682ed34877 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d6339572273432f9dd36d0bdcf4110a : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f4716fe47914f92baba7d7c7e0501f5 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/474e720ea35b4cf3a925071e4a49aefa : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ec3bb2d7398d4bafac20dae52b463552 : SUCCESS in 17m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14bdacfdb434404aae627a99401892f1 : SUCCESS in 47m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f52cfb6df944117aa968a873dd35b7a : SUCCESS in 47m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79d148bc4d5f4c15b464c76bb2d59942 : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f2ce1301754f5aa15435d1c68d12ea : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/172098371fff4bf2bafe5f3038cdd3f3 : SUCCESS in 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5241730213b3761168001851f33c94e3c7b531ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:50:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95ab2c9ede0945009f19b3a13ace7db4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7b31b9fdd1840c8b0293ef44167f484 : SUCCESS in 54m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98fc977abeea482484c7e5771f4b39ae : SUCCESS in 46m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5c84ac778284b29b4fa9654720a0427 : SUCCESS in 1h 00m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbb1ec34e3b34e35865eaebd87ab05de : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d8eaff0c60025a08131f2839d217bfe264abcae7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-08-24 16:39:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4a2c8d8a7ddbebd4612f4041f39d38582e69db7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-24 19:09:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eec9c24777b135c361c6dacb0ac3eb15eb7870da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:10:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"273faa4fce464ab66fd4127dbdd2fcb7130dcf46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:22:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e134a4d21248405d926c30161ea3403c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e384bbae29ea4c65ad69f989b3eca00c : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d36eac7a2734e84ae0ea6e5a7df63bd : SUCCESS in 54m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/085e1f988dc843bea94fad8cd4230ad3 : SUCCESS in 14m 09s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a48c35265d4d49ea988cae5a42679d3b : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9797d18c403d4ac4aa2d1632581d3148 : SUCCESS in 21m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af6ba2d059c944f89ea0d99d3fb8d867 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c8b2d3d777314430b7023dc6e20953f0 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/73375e8244c9454ca3110c9191897007 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a031ee0df1b94ebcbc28ba806f8936ab : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0aa8d91e95234196baf8eac369102de6 : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ebd111a5faaf451cacb494d08ef27206 : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b61c093da9c04dd7b3055bbd120dbdb2 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/01ad3d111332470e813f5bf21427627b : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eef8c4b7e8eb462dac8590d2de07175e : SUCCESS in 8m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb3c87c4c36d4ec19dc3207b55e54168 : SUCCESS in 14m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09ac0703771b478a883268708d379a55 : SUCCESS in 47m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89c2b55ea5be4216a3e9f235032cbaea : SUCCESS in 51m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de8bd5ac2e704d23a6b7f9c1e14931ba : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915489bc3e8543719ce2d8da607b24a9 : SUCCESS in 8m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/183c6515d6674752af62b97413a3ae2d : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6d86bfb2efb434ff50e6080e6bc944068c3c9725","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:22:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"63e70733b026eb62115e3f2c9e6b7a0410eea955","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16afced23f484965947c1b6b46dc42fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b028ee42c78f4d59a2a2683e94074e04 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c2756265bf1047158fee922c459e0dab : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6d26b12884ae44a988880b3ed47f9c95 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17664cfb98b2deb5552ad0cbc659b3c57aad5a73","revisions":{"17664cfb98b2deb5552ad0cbc659b3c57aad5a73":{"kind":"REWORK","_number":1,"created":"2022-08-24 14:21:54.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/20/854420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/854420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/854420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/854420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/854420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/854420/1"}}},"commit":{"parents":[{"commit":"6504a5f59e84c3d56409836af522866c414142ca","subject":"Merge \"swift-manage-shard-ranges repair: ignore recent overlaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6504a5f59e84c3d56409836af522866c414142ca"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-24 14:18:48.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-24 14:20:39.000000000","tz":540},"subject":"Use daemon property instead of setDaemon method","message":"Use daemon property instead of setDaemon method\n\nThe setDaemon method of the threading.Thread was deprecated\nin Python 3.10 (*).\nReplace the setDaemon method with the daemon property.\n\n*: https://docs.python.org/3.10/library/threading.html#threading.Thread.setDaemon\n\nChange-Id: Ic854dc3c393d382a8acd20d89f56bff198a2ec5e\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17664cfb98b2deb5552ad0cbc659b3c57aad5a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17664cfb98b2deb5552ad0cbc659b3c57aad5a73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
