)]}'
{"id":"openstack%2Fswift~855548","triplet_id":"openstack%2Fswift~master~I777f52706974ebc5172cf22a3bf18ca11495a185","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-10-21 05:54:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I777f52706974ebc5172cf22a3bf18ca11495a185","subject":"Ring v2/history follow-up","status":"ABANDONED","created":"2022-09-01 20:03:29.000000000","updated":"2022-10-21 05:54:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83a7b7c0bd56fa7d01959f4b1e35b8accfafc9e6","_number":855548,"virtual_id_number":855548,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-26 23:43:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-01 20:54:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-01 21:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e0cd552f5638140aee2b59c816210070c4d2777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-01 20:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b8462762ba473a432ae706e26d511eabbd2f07c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 20:54:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843e6d5625894f4ab58434121dd85e73\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a783613ceb254419954b040bcc91d64b : FAILURE in 34m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53b7482652a34ecbbb06af63afea699c : FAILURE in 48m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ef151c5ce69409b93dc9db47b7e7e97 : SUCCESS in 36m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/302926b008d548d587a6d327f0f1e76d : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"df24abad5f2035988b195b7e9eff29f998ee02c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 21:39:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d547ac7483844d0089c0ca046d0e2e34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec8434f3fd243f5af57ecb4d9a2d420 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/d8c022800f6a482e92720216301544d2 : SUCCESS in 54m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/13aa46a5001e40a3871dc7292f9450f9 : SUCCESS in 44m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fda89e263e78483cbe1ec167da31f875 : SUCCESS in 57m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7d25bc25f1247e795084a6bfeabce7c : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cdc8d392d04443cbd14ef0dbf029d53 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fce54fb082c142f085625fa95b08ce84 : SUCCESS in 19m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb5144ed3ceb4332ab16647a25b1e40b : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f25131f09d24c20a22ad00eb3df3e44 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/237136bc6a9b4f419b0c52350b7ec75b : FAILURE in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9035d9185e094adb9ca16bd21c7dec6f : FAILURE in 16m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4dacb72687540b99236ac73312d77e7 : FAILURE in 14m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/713e0619322d4be2b0892df9a4ee8474 : SUCCESS in 21m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/19db919b9fa44443b47c6a764774a271 : SUCCESS in 22m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be2fbb1ecb15404992719b3539334992 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7c2320ba03e4b6a84b7d037aa2122b9 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88112e4ca76944588f3cee644e3b01ba : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8c4e31a675df4a73a24d2151498b5b9a : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33ff403bd8834ef9a204ecd3939e6269 : SUCCESS in 7m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2b039d0aed7440cbd6e6d4cd54787b4 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/70ca5b4a9ccc4a84939b854f07588274 : SUCCESS in 8m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/197774fd5e8c423bb7b5b88dbbc61375 : SUCCESS in 59m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b156008fe1a648429f0eb80b3776cb9b : SUCCESS in 1h 06m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e46cc05d16774602af1b8affd814e6d5 : SUCCESS in 33m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8efa945f484d440c8d4f1f3b7eaf7841 : SUCCESS in 49m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c219dfedb574448a38da67fec912541 : FAILURE in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e167a1843d94f5680aa3c3cdb3a993b : FAILURE in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e49ea717a36d4a12abdeb6b391ceb2d7 : SUCCESS in 23m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f99d7b42ee542c6a89a4fa54f9b21bb : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4a64eb19a573177c968fd8f964757d7ab49b2423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-12 20:57:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7d89692e5f44cc7757abe3f4378297624c8ffc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 21:38:44.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d50a444f00774f469c6145debb40b7ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43c9330f8ded4c58a8a5859dde056269 : FAILURE in 31m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c208f9b832b40708d9971d452cdbab8 : FAILURE in 30m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/26d21c7f2ccd4f3b9e196b533c2405ba : SUCCESS in 31m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a29a2a4da4a478c8c348ea2afe3dfb3 : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":2},{"id":"97f0de4e168e09e9447371de48130c3f71e345b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 22:06:23.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c942dfb122649cd9f98e22335974ab5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e139503ea5849d88fd122acfadb59cb : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/b3be0fa470894ee8b95d5e268e9cb3d9 : SUCCESS in 43m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/550d7fd800d44da9ab21072ac39775b9 : SUCCESS in 39m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91b61d044a194f299708a28e35f4c0bb : SUCCESS in 1h 05m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11cef03283c94299940877807d2100e6 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e4c13fb3b90486183d2ccd775c4d8ce : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2dbb764e3c064511a4043a5a1396f2cf : SUCCESS in 15m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a948e237e0643ce8e16cac4facda877 : SUCCESS in 8m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0688a8b65112474d8523b359777e9290 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf5c8294faa449bb84b576c6deac8e59 : FAILURE in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1826a66fca5c4d7193d6d2cd33d401a8 : FAILURE in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f39e61ac615043fdbb9ec33649cdd0ec : FAILURE in 11m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b913b719fe9f435f98ddc552a28861ad : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66dced0b921a41f5bda3cd11c3f9a047 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f45687ab43934334b89c7ac4527888ec : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3a757f315679493e9e26c393da791aeb : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8af3ddf1a2374977961ee4bc4f4f9f68 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6b11929530244af818e1baaabc6db76 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/326811a9428b4d7787510dcf40db0920 : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c30789fe824643d2bcd65c6d52f5fdfc : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6a5ba6ac63494484a91f5ad63fcfeca8 : SUCCESS in 9m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7c050e5b63e46f5b48a8de4c22d9efc : SUCCESS in 1h 01m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a68c3c00247b45cb8c54c4e97fde215c : SUCCESS in 51m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e43dc0fc2f8b4561a4610a20a88af695 : SUCCESS in 1h 01m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39f8a7af12674de9815643d7174a262a : SUCCESS in 45m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9a16342218e49088a116ccdf5522860 : FAILURE in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47ee8f98b8e147c0a0052a94c44c20af : FAILURE in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16614c20792a457b85579632216cfe48 : SUCCESS in 24m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/212e97ee5c08440c8b1648683ef148c5 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e9d07f33b1f25f0f390488ad01bce6d9b159263d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-14 18:45:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e104c0b2bb24af1e699d539b29119607f8fbc7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 19:39:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d01f27fad244b0be5dbcccb87afd23\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8d20ab8bf314f88974b696d2d9e676e : SUCCESS in 51m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4da961b79d240f09bae0f550075f394 : SUCCESS in 35m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a65abd22832e4e4abdff440309fb91e5 : SUCCESS in 36m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d0d0e69c55e4d62a652bd729dc6a84c : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":3},{"id":"20d858d5c48256dfc481e263bda6a9d992c1e9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:02:23.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/985fb0a817c248999f173caf5d1c64cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb9c0f89eb446679e394b5edc3e2c8d : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/0c54c8d2a5e0437bab32b64d5160801f : SUCCESS in 54m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfe10da250f94dd49cb4b4bc4a30dbd1 : SUCCESS in 44m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/829040d54a9a4615b46eda9091267f7b : SUCCESS in 1h 14m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e950d4258604bca9d1c62360ca27554 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f38bef98ac0432d85ac0114c432ab1f : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14403a99386046ae8c89c6c09c49f130 : SUCCESS in 17m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f94a090a8ec8400d85a96feb080ce9eb : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bfbd2b0a34634cd7a4eb82540b84515b : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e49061762874bbe9be123f5879ddf44 : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d9c343ca78947b88988b63c32dca70c : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e752473c7a5e4d9f85705e996e1ee074 : SUCCESS in 14m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25af9f100d9b4ae0be9bb2181ea38e2b : SUCCESS in 21m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad0e92bb17d94039825246cbd4024f70 : SUCCESS in 39m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22bb4653bf6f4bdc9e5283a069ade36b : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2bb01298352b4713aa306db7c51225db : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1cdabe4fd915408f9c75bc2c913d107a : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f875510a3a1544e1af9b2ba4f074f4b6 : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b027590aaf834fa390f551430271498a : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c91425096cd4507884e29cb955f8bda : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7851a4d0dbbb45878e2abcbd4a50b9f4 : SUCCESS in 8m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f5e10801f27d4fd3aeb4f75b248ade91 : SUCCESS in 47m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6097b33f72a43c1be890c42475d33b9 : SUCCESS in 1h 06m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4df942023fae483d8da72cd188208b08 : SUCCESS in 45m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0bd9a63a2bbe49a8a0a33cd74cab57ef : SUCCESS in 37m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e4b3c1fda9e4b1fa50c77698b113857 : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3aaaeae2c2140c99c5e77fedd80a843 : FAILURE in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b1bdc5e6cdf41e18160cc8f4f261bc9 : SUCCESS in 25m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48ad9296b8124bc9b3b157bb27f01c16 : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":3},{"id":"3d0b6d0efb8277670002b101dfd729162acd0dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-14 20:47:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4c218ce3c3f8fe0d3f4f718d4d83236d8688abe9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 21:35:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a1b814972441a59295d298df5c82db\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf9d0bfb91fe4fa599467deff5dbb527 : SUCCESS in 33m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef032781dbf5406db82b6452f07e0f9f : SUCCESS in 34m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d015d0d87864878b7bb779902464a4b : SUCCESS in 34m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b3eb4bfecbd414a9e72a5dc5d5ca10e : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":4},{"id":"6afa06fb5b68248c16b62182627ec62c77ae3fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 22:09:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b18c625fb17644db831b0008b11b926e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e998748d2e49bc86fe7c694cfdd731 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4616b185d9214cd7a296e6249f0e1f56 : SUCCESS in 1h 05m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c3f7d18e54a4e7ebbeac5f24db04a5f : SUCCESS in 1h 00m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4043b31ef5944d8eaa6e00b89b70dac2 : SUCCESS in 1h 17m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c215a6829b7f487bb9bce1369941b7e2 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac87fbb4fa95487b9629242f8a367921 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50c01057b6264c5298854578592045ce : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a686c03f8d274538a90f88da18d68833 : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/103fb29913be4223b57016287091d912 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f75dd65d9e5543de9d876a462957acd1 : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b53d9bf86c64458a85fd44e69a5bf2f : SUCCESS in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31b318de4c204d9e83d31d4a67c6b43b : SUCCESS in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a33ed38708b547d0ad87737a8bf94938 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a73b38c8f5cc4165b6e12007a7f9d6df : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30b18d43211b4755be4b0a62fc91f456 : SUCCESS in 18m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6355f9caefb045408b0613ed830ff127 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8aa892f40e7473d8b45c82e0f6bb139 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a8876bde14274846af71064c88c6913b : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f607857b8ba846d2bcaf9638d574bcde : SUCCESS in 8m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c039e8b158243e7b663f0d927baecbf : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfabd5887e7743dc9e51f60f133aba38 : SUCCESS in 8m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cd82cebacc084c12983188473f8a8683 : SUCCESS in 53m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c274faab660646688032d4f9d2c0bb0c : SUCCESS in 56m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44be4f91ae8346e5ae849cf863b4cfbd : SUCCESS in 44m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed286da094114e37af8faec3299716cd : SUCCESS in 38m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6de6aeea22a4d83aa83bfccf5ad545f : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b83bf765de4af08909c7fea549f68c : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80eb6aa41d434502830c70800e32888d : SUCCESS in 21m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b74f1f1d1654b939ee53d7ae5a50acb : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":4},{"id":"83f125f03f1989da4e5c77044433c9423b947c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-26 22:39:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"3d07ebdfd4911620cc7aacfe862c70c71f3302ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 23:43:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9915b18bf1b432aa7f9a544b90d9b9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197edac32ae7441aa3b69be7401967eb : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/78fc05d5de784d669e9bdb5a59a97947 : SUCCESS in 46m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80c7bffedf294b6dbc96c80b9985c1c5 : SUCCESS in 36m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3da1fe62d92c4c418f2118846859d59f : SUCCESS in 41m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a152065474da4a9289164544ac6741f7 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1adfa3b2a3164c22a327d7aeb25cba95 : SUCCESS in 24m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f32ad73c3a7f424987b1ceb018c039cd : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96aa58f713bf4edc93be64280253d2c3 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/784385a30dfc4140818c6186061a1e22 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3de46bd89fef4d91b544a3f6b53d16ea : SUCCESS in 16m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ca79a62c7734a34b37ab6b98238eaa9 : SUCCESS in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1bcfd65596eb4fc79d4a77702f7fd354 : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cd1f65bbeb041f7b9fe2cd395df49ee : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4edd7bbe9a094c38bd8e50a6d6b8980e : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2250842ffacb40b1a8a4c765afeb7e34 : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7670d17fe53546fda9abdaab4926cb1a : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b2d1304fe014ddbada441dc34936013 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/779f45d7908842dd8c2c035d05c53950 : SUCCESS in 17m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/572b4d46be444800b3d9c57e94a9fc7e : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55cb0bd865004cfb858b92235af1ddd8 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fba417e09bb41268f03fef8c090d285 : SUCCESS in 8m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1f4311d0aba4bcbbca035b56e3933e2 : SUCCESS in 57m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a6773b821544e04ab00071c46c68872 : SUCCESS in 53m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/513a44c0cac749dca17910d96636b2ff : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cec8bfcc0f704173a9522f218cd566fc : SUCCESS in 34m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03f51376b018473480b05470cc7e80bf : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbcf4b473e544eea0418d957a9258ba : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba84c1d4a8c54394bc7be1ca21692103 : SUCCESS in 30m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9ed10107e54647b74ed2bc15c10ef8 : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":5},{"id":"a9c6c56da71397c6b9e779a72acd85f5cbe4bb89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 00:00:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e593e4fefff647afb0a99ac45f305a95\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6ef0454d55646e4ba1a39e6fb9f7841 : SUCCESS in 1h 05m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f5eda19b6b343aeae320350f2304484 : SUCCESS in 49m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2157c5edaf1a475db507a2a0d4aebb8b : SUCCESS in 1h 04m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7545b34f5e343d487713eb6b482cd30 : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":5},{"id":"83a7b7c0bd56fa7d01959f4b1e35b8accfafc9e6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-21 05:54:55.000000000","message":"Abandoned\n\nSplit up and squashed into the two parent patches","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3d86afa04d2cc69ef8510fe1c98880affcdedd08","revisions":{"e90cdbbd9cec939f965654d1cb2762bbfc330fb1":{"kind":"REWORK","_number":1,"created":"2022-09-01 20:03:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/855548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/855548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/855548/1"}}},"commit":{"parents":[{"commit":"ff8c15a1db21ed4571cb3fb90f50a316747d7b5c","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff8c15a1db21ed4571cb3fb90f50a316747d7b5c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-01 19:59:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-01 20:03:17.000000000","tz":-420},"subject":"Ring history follow-up","message":"Ring history follow-up\n\n- Only rotate history when we\u0027ve got new trackable history\n- Purge removed devices from history\n- Make history work when increasing part powers\n- Change get_last_part_node to get_last_part_nodes -- have it yield out\n  whatever entries it can find\n\nChange-Id: I777f52706974ebc5172cf22a3bf18ca11495a185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e90cdbbd9cec939f965654d1cb2762bbfc330fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e90cdbbd9cec939f965654d1cb2762bbfc330fb1"}]},"branch":"refs/heads/master"},"e33fcecab655584417da87df5c2fc732a57cce87":{"kind":"REWORK","_number":2,"created":"2022-09-12 20:57:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/855548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/855548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/855548/2"}}},"commit":{"parents":[{"commit":"ff8c15a1db21ed4571cb3fb90f50a316747d7b5c","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff8c15a1db21ed4571cb3fb90f50a316747d7b5c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-01 19:59:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:51:45.000000000","tz":-420},"subject":"Ring v2/history follow-up","message":"Ring v2/history follow-up\n\n- Only rotate history when we\u0027ve got new trackable history\n- Purge removed devices from history\n- Make history work when increasing part powers\n- Make history work when decreasing replica count (max 1, and if any\n  parts move for balance, they get priority)\n- Change get_last_part_node to get_last_part_nodes -- have it yield out\n  whatever entries it can find\n- Ensure builders without history can grow history\n- Rename FixedLengthReader to SectionReader\n- Add read_ring_table/write_ring_table helpers for v2 rings\n\nChange-Id: I777f52706974ebc5172cf22a3bf18ca11495a185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e33fcecab655584417da87df5c2fc732a57cce87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e33fcecab655584417da87df5c2fc732a57cce87"}]},"branch":"refs/heads/master"},"c004b791a1015a22eaed07af25d16a0711d408eb":{"kind":"REWORK","_number":3,"created":"2022-09-14 18:45:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/855548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/855548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/855548/3"}}},"commit":{"parents":[{"commit":"e8952a8d20aa143ce003be1cb47b16209dcef7ba","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8952a8d20aa143ce003be1cb47b16209dcef7ba"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-01 19:59:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-14 18:44:03.000000000","tz":-420},"subject":"Ring v2/history follow-up","message":"Ring v2/history follow-up\n\n- Only rotate history when we\u0027ve got new trackable history\n- Purge removed devices from history\n- Make history work when increasing part powers\n- Make history work when decreasing replica count (max 1, and if any\n  parts move for balance, they get priority)\n- Change get_last_part_node to get_last_part_nodes -- have it yield out\n  whatever entries it can find\n- Ensure builders without history can grow history\n- Rename FixedLengthReader to SectionReader\n- Add read_ring_table/write_ring_table helpers for v2 rings\n\nChange-Id: I777f52706974ebc5172cf22a3bf18ca11495a185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c004b791a1015a22eaed07af25d16a0711d408eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c004b791a1015a22eaed07af25d16a0711d408eb"}]},"branch":"refs/heads/master"},"bd58b05fa7eca41886400a910a213114b273aa5f":{"kind":"REWORK","_number":4,"created":"2022-09-14 20:47:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/855548/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/855548/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/855548/4"}}},"commit":{"parents":[{"commit":"e8952a8d20aa143ce003be1cb47b16209dcef7ba","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8952a8d20aa143ce003be1cb47b16209dcef7ba"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-01 19:59:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-14 20:47:03.000000000","tz":-420},"subject":"Ring v2/history follow-up","message":"Ring v2/history follow-up\n\n- Only rotate history when we\u0027ve got new trackable history\n- Purge removed devices from history\n- Make history work when increasing part powers\n- Make history work when decreasing replica count (max 1, and if any\n  parts move for balance, they get priority)\n- Change get_last_part_node to get_last_part_nodes -- have it yield out\n  whatever entries it can find\n- Ensure builders without history can grow history\n- Rename FixedLengthReader to SectionReader\n- Add read_ring_table/write_ring_table helpers for v2 rings\n\nChange-Id: I777f52706974ebc5172cf22a3bf18ca11495a185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd58b05fa7eca41886400a910a213114b273aa5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd58b05fa7eca41886400a910a213114b273aa5f"}]},"branch":"refs/heads/master"},"3d86afa04d2cc69ef8510fe1c98880affcdedd08":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-09-26 22:39:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/855548/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/855548/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/855548/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/855548/5"}}},"commit":{"parents":[{"commit":"b5f15a87f3bc6ca10da7ee5805bdf6f7cd31dd6a","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5f15a87f3bc6ca10da7ee5805bdf6f7cd31dd6a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-01 19:59:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 22:38:46.000000000","tz":-420},"subject":"Ring v2/history follow-up","message":"Ring v2/history follow-up\n\n- Only rotate history when we\u0027ve got new trackable history\n- Purge removed devices from history\n- Make history work when increasing part powers\n- Make history work when decreasing replica count (max 1, and if any\n  parts move for balance, they get priority)\n- Change get_last_part_node to get_last_part_nodes -- have it yield out\n  whatever entries it can find\n- Ensure builders without history can grow history\n- Rename FixedLengthReader to SectionReader\n- Add read_ring_table/write_ring_table helpers for v2 rings\n\nChange-Id: I777f52706974ebc5172cf22a3bf18ca11495a185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d86afa04d2cc69ef8510fe1c98880affcdedd08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d86afa04d2cc69ef8510fe1c98880affcdedd08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
