)]}'
{"id":"openstack%2Fswift~855670","triplet_id":"openstack%2Fswift~master~Ica0cd5277c93cb2005eb6e6443e51b3162196d0c","project":"openstack/swift","branch":"master","topic":"jh","hashtags":[],"change_id":"Ica0cd5277c93cb2005eb6e6443e51b3162196d0c","subject":"squash: fixup gap repair probe test","status":"ABANDONED","created":"2022-09-02 14:51:00.000000000","updated":"2022-09-05 16:02:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f433b29271baf724220e717b97153bf757ec4980","_number":855670,"virtual_id_number":855670,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-02 16:05:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-02 16:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6aaa0254228ba69cd44a104fac945b1c8c10878d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-02 14:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4612c6b89cb45e45fdd5b001204132476252e54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f798e554e9fc4a398e888ad48fe502bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a28433b87ce45308c443db9eda531ca : SUCCESS in 9m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79d664bf12ed4788b25776d367f534f5 : SUCCESS in 1h 09m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/668fe93b977c4f48b8bdb4c9d1eddb01 : SUCCESS in 55m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/06162b23414b43e49868acf2901c9d7c : SUCCESS in 1h 00m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0c7fee7acc48949cf59f12c03bba1e : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c272256fb2038dc8e9d62cccf20a233be8f67ab4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:38:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa41c3a6903f417eada6d92d06147611\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93e8c23c355a4b8380838f0de36025c5 : SUCCESS in 56m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dd65ce2323a498198116b33cd9d241e : SUCCESS in 59m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/284df389518b4054a34a3289ad1ba575 : TIMED_OUT in 1h 31m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5ad46addb8141a0a007917f2fbe7668 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f433b29271baf724220e717b97153bf757ec4980","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-05 16:02:49.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/855100","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16c24e9be07474c21678abf21ba68084c6017f21","revisions":{"16c24e9be07474c21678abf21ba68084c6017f21":{"kind":"REWORK","_number":1,"created":"2022-09-02 14:51:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/855670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/855670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/855670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/855670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/855670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/855670/1"}}},"commit":{"parents":[{"commit":"202e4189e95123e1b8c5d81628de9794ec02dbad","subject":"swift-manage-shard-ranges repair: check for parent-child overlaps.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/202e4189e95123e1b8c5d81628de9794ec02dbad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-09-02 14:39:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-09-02 14:39:36.000000000","tz":60},"subject":"squash: fixup gap repair probe test","message":"squash: fixup gap repair probe test\n\nThe probe test is no longer appropriate because we\u0027re no longer\nallowed to repair parent-child overlaps, so replace the test with a\nmanually created gap.\n\nChange-Id: Ica0cd5277c93cb2005eb6e6443e51b3162196d0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16c24e9be07474c21678abf21ba68084c6017f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16c24e9be07474c21678abf21ba68084c6017f21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
