)]}'
{"id":"openstack%2Fswift~856329","triplet_id":"openstack%2Fswift~master~Ic67d51ca79fd685219655654a470c1d5d52d6c09","project":"openstack/swift","branch":"master","topic":"s3api-absolute-url","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-03-02 04:56:28.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-09-07 20:34:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic67d51ca79fd685219655654a470c1d5d52d6c09","subject":"wip: test for absolute urls","status":"ABANDONED","created":"2022-09-07 20:11:44.000000000","updated":"2023-03-02 04:56:28.000000000","total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"301bbbdccfd0dd3373e923d38de15f9e5d0204c2","_number":856329,"virtual_id_number":856329,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-07 21:13:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2022-09-07 20:16:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2022-09-07 20:16:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 20:21:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-09-07 20:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-07 21:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4a56abf537764e09750efe4fbccae102a72e984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-07 20:11:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"151b63044e2a0e57cbca8038bcd050b4e7cf3546","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-07 20:16:28.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"89acbb535f908c4d1ea4094c65eed4746fb7b850","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-07 20:21:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6e8fd095ed210c80d7a60456a301df8bceaa4cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-07 20:34:09.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"680d889a9e911b665cd97219a0843cc3a19e393a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 20:47:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a271ef07803946529a1ee0cb10f8c898\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2aebf3244c543a7bec9e1fe4b422763 : SUCCESS in 29m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fc1c833fd464306aa0a1c47b53166cb : SUCCESS in 29m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/003d9a5bb8ee46799df2feab1a527cab : SUCCESS in 29m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a9d9b61c8af4a99bc86c94e07e1e977 : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5121b3908e393fa19ea511f56263473a1f620403","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-07 21:02:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f308d34114d66a57eb928b528da66ccd7be46ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 21:13:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e6c033a4e6a4df58d71dbfbc71a9f48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a70df3cc984bdeae759ee8eba0fc2a : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66575b0ab57442e789516c1361f74500 : SUCCESS in 59m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2aeb4b8a1ff24729bc64a9c59f98404f : SUCCESS in 13m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3717ebe4832545d3b5e3a433e38aa22c : SUCCESS in 9m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/48ffd51766114b46be918eaa8d4c4d7e : FAILURE in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347bcc931fb449a4bae3ba337edba344 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"301bbbdccfd0dd3373e923d38de15f9e5d0204c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-02 04:56:28.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/856327","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8528fe4b837e435ef7f27e15513b929522c8dd5f","revisions":{"8528fe4b837e435ef7f27e15513b929522c8dd5f":{"kind":"REWORK","_number":1,"created":"2022-09-07 20:11:44.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/29/856329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/856329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/856329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/856329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/856329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/856329/1"}}},"commit":{"parents":[{"commit":"e3aef0c2ee749dc2e7a72a0ecd63d6e85a9b270a","subject":"Merge \"Add validation method for metadata in ECDiskfile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3aef0c2ee749dc2e7a72a0ecd63d6e85a9b270a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-09-07 20:11:35.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-09-07 20:11:35.000000000","tz":-300},"subject":"wip: test for absolute urls","message":"wip: test for absolute urls\n\nChange-Id: Ic67d51ca79fd685219655654a470c1d5d52d6c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8528fe4b837e435ef7f27e15513b929522c8dd5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8528fe4b837e435ef7f27e15513b929522c8dd5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
