)]}'
{"id":"openstack%2Fswift~857233","triplet_id":"openstack%2Fswift~master~I19716aeb4a4bf7b464928616a3ccb129fff7a7f2","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-09-13 19:00:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I19716aeb4a4bf7b464928616a3ccb129fff7a7f2","subject":"Fix test.probe.brain CLI","status":"MERGED","created":"2022-09-12 20:54:48.000000000","updated":"2022-09-13 21:49:02.000000000","submitted":"2022-09-13 21:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"857233","meta_rev_id":"3995f02f60169aad7113c5c3666679a0f1730e44","_number":857233,"virtual_id_number":857233,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-13 21:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-09-13 09:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2022-09-13 09:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 21:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-12 22:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-13 09:28:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"94f10c7eb3085b78444f442c55451ae8df9499a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-12 20:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9081cd617681a090ba38f3520fc851af765fefaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 21:33:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22cb2ec6f66e4818a5d20bf62bc0bacc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f920d8ce28245479baaa1254310437e : SUCCESS in 33m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2775e68062dd4f2b986fdbade69d9cfe : SUCCESS in 35m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/95942049eaa744bb96478534cd0faea5 : SUCCESS in 36m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/35daf8d305ad4bcfac70cf2a85b5366c : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2a7d52dc7b9d7c1f3b59930394d648c2a24314f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 22:02:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04404a7249cd44468b832be43937e847\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74d34c36f2c4dfcbe5dfded5d3c743b : SUCCESS in 8m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02ab5e8d2614404da1eb878c51333351 : SUCCESS in 1h 05m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02802257e68f463b94be09e524224fa6 : SUCCESS in 40m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b1d0e89f1d64343881fd2e5032565a9 : FAILURE in 55m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cdd1824fd354e4d9ac622ed9b724632 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1f7ba273ed2ebbfe243e021df539ccdd3c725032","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-12 23:09:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"045475c94b65a1ddf6c212060c8d2ee0cc834f51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 23:58:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bacf9a9ea644473bfab464c067fe0d6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa3fac31985b460c92893ac22283eea5 : SUCCESS in 46m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/452222311a824a4887e41f059491650f : SUCCESS in 27m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/632baa1017fb4fada16c1d686a87aa52 : SUCCESS in 27m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e656e02df7b456697f1e28f650f1ee2 : SUCCESS in 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9f86ef55a073fe956d0252e85f051f697db8ba5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 00:28:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d555f5084c49c59a66b7f9325c9c55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3941b56b431e4729a7200087f0bb44b5 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1c5007bec5e4f46a21f83129e33ea9e : SUCCESS in 1h 17m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1671223a33e542b4a229ae72fc60f669 : SUCCESS in 56m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc0125e1496c4f9cab0ec16b2ab5316f : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3bb2ca806841998b03956d0d5fd2ec : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9d60c999e2d964e3e81e8e50800e219488e26777","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-13 09:28:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c32931f5ba27dd503e30ce94f6b155e5aeab992c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 09:29:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"452656df3883f2093e5abc919938428ca3c1f8e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 10:39:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e312572960fe4da887b88cccb14d94f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d37966cdfba4a3db7de7f533346d44b : SUCCESS in 9m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7123529bcd945a384788e48572c2df5 : POST_FAILURE in 1h 09m 14s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/20d8de8d53bb42cda7e8dc3f14804d5b : SUCCESS in 14m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56f39e420a524eb4a61b2abd0bd6aab1 : SUCCESS in 14m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99820528efca4ef6ad30bb861fb04364 : SUCCESS in 14m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce5953d9000049f2a8a3a12bd15ae4e3 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf4f6e5b242f4357a5441d7662e3fec0 : SUCCESS in 21m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c00371a9b57456da021c1768aaaf5cc : SUCCESS in 19m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47ee58926ac94b759143a9233bd7c64c : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b155057fd78f4a158fe792fe9e75708b : SUCCESS in 18m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4edb0445cb184f1ab28af856ae5cb031 : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a15b283b5bd948b78fea49aa18c5f41f : SUCCESS in 9m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83ef8b1fa6584d97bdc5f422112b5f0e : SUCCESS in 9m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7cbd89b7a394a4892f3527c9a4d263b : SUCCESS in 57m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/300b9cb762b346c9a373fb76bd597bd2 : SUCCESS in 49m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7691cba4cf4a48a289aa2dca18bf7d1b : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6a117afacf7d6c8fc95a5a0b58b98c17df8ee287","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-13 14:15:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"175c1f7f1b2b7a2384a3594ac4343d133ced0268","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 15:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2204144f54345bebe2c6e811f7e618b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68240ee73dc34408880eed3afee68f76 : SUCCESS in 31m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76ea9c3542a14e52b641daf1e2810455 : SUCCESS in 31m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d0b95101197461fb48e4fc8b2de8a54 : SUCCESS in 32m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5857876df6f49c19adbbc12308db5e5 : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f033a9f7a6cdbd42c8c6899ebcc9289420c86061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 15:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d7cfaec44949c38ee6770850d4c7dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c60d6b262d04d99b06193d6a7c70cbc : SUCCESS in 8m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79301860104044e196b4462a0a72cf7e : SUCCESS in 1h 19m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e563fa65d88493faf7403fd0088204b : SUCCESS in 49m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1386dce0496e4fd0ae6f73e766bb805b : SUCCESS in 1h 01m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b165bf907574c93a70b53c95cfe48df : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"357f1344727f6c200effa4e3961fffebb21494b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 15:40:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30950414fe87b056b65764587644d64d897087c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 17:06:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/605bc9dd0fa9469a816f9dc685b039ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ceda7da0fa14112a903163fcb062403 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f3c5dab5304491981e30e7546cfdd61 : FAILURE in 1h 26m 25s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae00d12b9f4549589906961f5b46c6e8 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab769bc194fc4d96bee0518a8217f6bb : SUCCESS in 17m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ec2c5f3e4a8445a09d42acec07efb0fc : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bba176a6ba5c4750928f581f27c9fc43 : SUCCESS in 22m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8de648ecfe96446aa00bc277c24cf5b7 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86746ffeb01d472085f5f100c9d5c677 : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/56df38cabbcd4c379f5a7efc31c572f5 : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c6c7fd643f80477da7e48700e522e4fa : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2dea9fd4c4be4f6b9a290ccefcbf0d72 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7bc2e54d61f84e0c812c2a62d2746ba5 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/009349a96a924e45aec12ca570bd10cb : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c5fdb82bf1e494298f42a7b9898dd00 : SUCCESS in 59m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b62f4b2584b844c084e66a008c444111 : SUCCESS in 1h 08m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428c76c336be42a5bc70687b172f1a47 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0dff4917bf352de9ec79a0ba9d9c000d58461c5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-13 19:00:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"590253acfa6bb253da680dcd423afc2acc8c6f2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 19:39:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7849e074fb72454fbf8eca47a93b7ba7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/719ddf266ff641e0a8d51a9e26ac39b4 : SUCCESS in 31m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89f08e5ed20b44bc8fbe1e1fc03efe2e : SUCCESS in 31m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a93c67db9f6459eb3b8a5334adec93e : SUCCESS in 31m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4368081378304c189e75c1ec0b8b7ecc : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3372d445d07b98a005319271604edc3d381b4133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 20:07:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4f004bfef0424a88856faf9963cb43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad4be05bf3649bbb61adbcec6efcfd4 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a641e25a7112442995af7b604b0887c3 : SUCCESS in 1h 06m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a38194e86b7e4cb2a7eb7c98347e168f : SUCCESS in 53m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0ea5bb92b1c4063b55b4723a9e0e9b3 : SUCCESS in 56m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a80f2f5203e43a0930d9b452f18a4cc : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"476bd736bf9951f328450c0f56ace7a40e13f1c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 20:08:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6d15dee16ae5932eb93ef6bfeab81e290f20fe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 21:47:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd1adc9c701489883e35709520a8027\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7542d63fa1b44f05aff358914a1a264e : SUCCESS in 8m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9812b950943749a9a54a5766dc209cca : SUCCESS in 1h 25m 34s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0143e1ec3324f2b8d732656c62e8344 : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4598241e855243fdbc06320bf41b777d : SUCCESS in 20m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed3f67267c2d4dd7864a42998f76156c : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53ebfcde50df4af192090cdbf7558df8 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7914163dc8dd4912ac612869eb091549 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f63460b2d277421ebf8f8a5a001c2e49 : SUCCESS in 18m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/038afe1648e14f6fad5df7279666733b : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/20b3685354ca4914b54b02fbf01315d4 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9956a212f74f4520ab09afd54275b984 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d07e529fd8414865bd2f53905f53cc7c : SUCCESS in 8m 10s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9484cda39ae04c098f6f1769a4c80d5f : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e0e40d2ac40f4e56901ca4cf5c006df1 : SUCCESS in 54m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0c9f632b8054056bc835aa6d37f67fa : SUCCESS in 58m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eecd9a42eea42828ce9c537f4b672a0 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2582e21bbd61d69836d696008be36cca264eabb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 21:47:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3995f02f60169aad7113c5c3666679a0f1730e44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 21:49:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ad0f9e60d8045b889afde1ac3d1429e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15108eac66e245a8991386b2dc042686 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/20a01ebde12f4f0ca6d45c9a9876b433 : FAILURE in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/dc763e0ac0b649bba5f0c8fef8e3fa42 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0860db1f60963b63f96146ad627fd55630da793d","revisions":{"0860db1f60963b63f96146ad627fd55630da793d":{"kind":"REWORK","_number":1,"created":"2022-09-12 20:54:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/857233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/857233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/857233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/857233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/857233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/857233/1"}}},"commit":{"parents":[{"commit":"5de15c3b6d3d4c2cc4b69c119e6ebff93d013104","subject":"Merge \"ShardName: add test and clarify docstrings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de15c3b6d3d4c2cc4b69c119e6ebff93d013104"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:54:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:54:20.000000000","tz":-420},"subject":"Fix test.probe.brain CLI","message":"Fix test.probe.brain CLI\n\nChange-Id: I19716aeb4a4bf7b464928616a3ccb129fff7a7f2\nRelated-Change: Ib918f10e95970b9f562b88e923c25608b826b83f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0860db1f60963b63f96146ad627fd55630da793d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0860db1f60963b63f96146ad627fd55630da793d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
