)]}'
{"id":"openstack%2Fswift~857234","triplet_id":"openstack%2Fswift~master~Iefea826ed3ae766163ba0bccb9ace521ada0ae0b","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-25 02:34:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iefea826ed3ae766163ba0bccb9ace521ada0ae0b","subject":"Add swift-ring-info utility","status":"ABANDONED","created":"2022-09-12 20:57:05.000000000","updated":"2026-04-20 22:24:22.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0603713514738b6858c2020835bb8a66eb5b6664","_number":857234,"virtual_id_number":857234,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-25 03:47:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 21:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-12 22:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b83cef9389ba8acd1b4779416902281f95fc8926","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-12 20:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55968af4fbf5efcb69911e1024f76be2c2f5f95c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 21:37:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebbe123bb51465ca0dcbf335b301e32\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ee110d5d12e4a6fb356eebb66ef3a6f : FAILURE in 30m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf02e4c6dba441c3ab87631a49251039 : FAILURE in 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c73538ae74640049fec3595dcd1e56c : SUCCESS in 30m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/973797decc784d7e962eb71d7f39d596 : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7bf52b4b9da98eae6aee8b2584e1e4db3d4d4bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 22:16:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7aeaae93d3a4c40be222f76db146778\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ce00c3366e548b29ef2d3e028548cd2 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5754c09bb74a4c6cac362088ebcb306f : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/58ad2c86271a40cd82f1843dd82d3ab8 : SUCCESS in 57m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2806f788423345a1affed86a57bee06d : SUCCESS in 51m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49b4419f35be414e901393e4f341645c : SUCCESS in 1h 17m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a00c38fd038457a8d3971a07db04e06 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3c66286defe495f8549bd9e94c64897 : SUCCESS in 24m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55192c61fe3146de89b320d7d5e66d3a : SUCCESS in 25m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7b750efc0e249d4b7c83eb97001efe0 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08fbcf4a99d84fa9934fd03917437099 : SUCCESS in 9m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d95f9246b4484335b6d0c90981180124 : FAILURE in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd3fa0a50cc844a89c3f9b725f772848 : FAILURE in 20m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/045d9f0389954b7c802c5906df9cbc8c : FAILURE in 17m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ec743eac05840c199bb9e9f4b2257de : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4bd7aaf650eb49b39075a7662a5f9aed : SUCCESS in 21m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8b901814cc24fd0bd33a91744c8cf0a : SUCCESS in 20m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ccc725759ffc47028d957adc4978938d : SUCCESS in 19m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d4cfcfc263514f0f865b86900a915e7e : SUCCESS in 19m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5a36a8b72914e9b8fff2d5d967db3f1 : SUCCESS in 23m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/120cf635fb574d29aece4db7c4acaa4d : SUCCESS in 9m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f25600be02724b7db62f7a29d188f179 : SUCCESS in 23m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3502641b7f474210a3c727603dfb1b37 : SUCCESS in 9m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b08654058014525a51de59024d41ee8 : SUCCESS in 56m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffd01499e35748a983e0a05db348f3b2 : SUCCESS in 1h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7611802d4ff44430be649f26d40489b2 : SUCCESS in 45m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc9c8424576247e5bd26ffc38e0d83ca : SUCCESS in 49m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e08772fd6f0475dac01093524252eab : FAILURE in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7192e8ffe44c189d08d685e6f4b526 : FAILURE in 8m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6560aa497db4df2a58cf253498f4d64 : SUCCESS in 29m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb68672b350143689184ff88814a0f9b : SUCCESS in 54m 13s\n\nWarning:\n  Comments left for invalid file swift/cli/info.py\n  Comments left for invalid file swift/common/ring/builder.py","accounts_in_message":[],"_revision_number":1},{"id":"ad65ff5c72bc9e2aadffc5ef8c3200a56caca499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-14 18:45:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43e6f7d7dd713f94f561ca4d744da7dbbc3be4b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 19:36:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4a6b25f03c44198efc098755f4d483\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecd0b6b5e9d04f028400731561c66501 : SUCCESS in 47m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bde7d7f373145238a171dba64eee1f7 : SUCCESS in 41m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9791cafeb8c9476fba2dcc5fc19a74f7 : SUCCESS in 48m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff96b87880a94e77a93d07e8cc282b24 : SUCCESS in 36m 48s","accounts_in_message":[],"_revision_number":2},{"id":"713412776f6056e77c06826a967054fdee64af44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:09:19.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d0fb735810741139edfea6746ce0472\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8aaabbf0d43a4753a1c7769dbc5ebe71 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc9535919784917bfacf00825e49bee : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/6a38601241a14446a7ec76b19067008b : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/630836e07150472e815a75e0e7cf967d : SUCCESS in 45m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22e889365a4641b6b6394f6d991d83b9 : SUCCESS in 1h 18m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/804abff84860437dbe94f0eab64090db : SUCCESS in 23m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edf2c7bdb4f6455a85af00b299e4af48 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7927b047ebc84475b22036bbeeeb3ad8 : SUCCESS in 17m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a75e1335dbd64eb4aced3faf1ba2ac02 : SUCCESS in 11m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24a3d034e5f94ad89c7a2ff588bc91c9 : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5165d3cc6d6d4cec8ad9e9027a1ed044 : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec6a434cde548b9a227a8376a8695d8 : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c807791da88e4e62a81d0d53af409d2e : SUCCESS in 20m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/54fb629c5e124ce7ba4c9b12e315d944 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6c2ea1f766840f589800cc6002eb05f : SUCCESS in 22m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc3898e1b8f6498f8883a2fcf6d595d4 : SUCCESS in 21m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ad0f844e85f4549b52171c31dceb216 : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0e1516d31117431e87b7f8f51a2a533e : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6eed61a1500841eead895076cc1811df : SUCCESS in 20m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62b325a7bc6e4a56bbc608d602eef5e3 : SUCCESS in 9m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f58f2f6272434438ae1db37039314233 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/121d3ca461054acba8df818e83c37e32 : SUCCESS in 13m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cc57dfdee0a34eb9b4111dd162c93c9b : SUCCESS in 59m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ccf6e3c21174dcb9a8647a2409e8180 : SUCCESS in 1h 01m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df8885e960ab4757bd7444d55305eea1 : SUCCESS in 56m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0143d67cdc2d499db2e66a827a03028b : SUCCESS in 46m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48da9a04102d4f639281762c10d982c0 : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0fb5164cd1e49fcb9d4568afa3a4b85 : FAILURE in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84c8c168455a4fe79072099c4fe427c1 : SUCCESS in 21m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38d17d2972e84577b2106d8bae1a0dd4 : SUCCESS in 58m 37s\n\nWarning:\n  Comments left for invalid file swift/cli/info.py\n  Comments left for invalid file swift/common/ring/builder.py","accounts_in_message":[],"_revision_number":2},{"id":"21d8b32b991d96e1f6cd025b56cee8ac1b0afbca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-14 20:47:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c0a13430cb98b2a85fcc8d5a89e348469cfbae6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 21:35:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f1099a6c634afc8b47e4f943ac9f44\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97a214f75f234d73b1f7c4e17564cbf5 : SUCCESS in 33m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01ed5a4e40aa4b5d9c31dd9fe4cabdcc : SUCCESS in 35m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4dc2f2ce6e548cc82ba09843a14efbc : SUCCESS in 34m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc85c8d1c6ac45fabc52ea83123a44cf : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":3},{"id":"25f09e59d6b8a0ef7d6265a8f862487181d427ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 22:15:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4563e0d6399a4b7dae25206f21657a1f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78b40750a26848b9bf574e1602cf3940 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bfdc65fe2564a769e4ad3e4f871fa26 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/8dd82115ce734b2cafa631e73116c3e6 : SUCCESS in 42m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f3f8331aaca4268adf26c3354a4b944 : SUCCESS in 59m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5335bd1879c849379200d7e3806e3e14 : SUCCESS in 1h 07m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c0b14fbc3945269c53121819e95497 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc3351d7226a4f6f86685c9305ec2ba6 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b6b1bddf1084c93ac674a082e930358 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a5b2bbb15434cc9a8b08fb00250d1f2 : SUCCESS in 11m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/230123dd0bb34b5f82e12f0d7c000835 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23ac1de0ffb74010aba34d2e48157389 : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8b3bdaed499467390ea2e676f317b3b : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1678a12f5e244be88d400d80b4ac064b : SUCCESS in 20m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d4d2482f02b4e64a929c83cecd6aeec : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/992fb252456f47168843f98d10d98a06 : SUCCESS in 19m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9718d09a5ccc4cba88a409d16e6e20cd : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d2ace06a3994d85888fa928cb9d39ea : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8aae1435182940ca88b3db39a13e2e2c : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ea3bb4baf6f1408d8b38018d24b29172 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/675ee02d6613443882b717bdae584e35 : SUCCESS in 9m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d1fb0a09d34479a986323c8347268d6 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f477316664d447f0b3221094dd1b069d : SUCCESS in 9m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d42a5ab5bb8c45e9abc800f07418028e : SUCCESS in 1h 04m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/640824b585a54ea3bf1b950328c2aa77 : SUCCESS in 1h 22m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b7a2acc641f4aebbc698aaff28850bc : SUCCESS in 45m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f55fdf48c6ed49019f2ed6247b2c8245 : SUCCESS in 40m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/309d1a12b1a74705b1bb37b4e65974c1 : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eedfebb7c17d45fcbf756e10e19cfd45 : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3c9e14f10804206a11966d258c3fc2f : SUCCESS in 22m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/129875f0025e43b691b18ddcecf33835 : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":3},{"id":"1e99fcf0bae2bd74f878cf9af9264604c2d5f421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-26 22:39:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2f399a88454843ee7ffd5d8c11686e5de063d0e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 00:00:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/171cf6da2bd4452399c19ac04256d906\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af7d95324dbc48bc98a4f35810c9187b : SUCCESS in 51m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2f4c7c156574c68befb2cc1d19c1348 : SUCCESS in 1h 19m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e940e2780c254a7f8bcfcbae45847867 : SUCCESS in 1h 04m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b24bc5f6d29416395f59e6129f2392f : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":4},{"id":"df1019eb56cb0600e00eb8912a2677d3b21cb2da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 00:05:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf782405b9c477493ba623d68a82189\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26a15a121990477886321bd3a64195ed : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ef6fb571e14c57b49e5e944f772811 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e00f7f5fb87544a1b3829fb1b2283c58 : SUCCESS in 58m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e311daea47c483391b417d252bfdbb5 : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2a4506df3db4bc3a234ecba970b2d32 : SUCCESS in 57m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1afc16ec0b50456b899e0a9b7e882f63 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/741887162f014e1c917a4edd25d21bd2 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdcf282b510e42c3aa5ab13b90f47302 : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d176ea7b2bf4458830da3ef9f4a130d : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c17bd6c6749e4c08ac98d728d925445e : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcc246a5587f4cc5a9440c0fe95cdcc0 : SUCCESS in 14m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29392dd668e747b1827ef5709fa1a03a : SUCCESS in 15m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/68956ad72ec54e79aa7e1a540b534b72 : SUCCESS in 15m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdf0853756b842e6ba5049fd1bf60b3b : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7bb4614648524753886ef82e7168aaeb : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d258f97599be435186a9738d87ec87e0 : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/64ac24035c5c4f0a8c34b7f72b99be13 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/714df5b6eb084f4a8a63eaeb2823ad1f : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f792ab8097504c44a9285682f1f8f498 : SUCCESS in 14m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b54d906f6b04ca5a14427ec8a9472bb : SUCCESS in 12m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cbea3866094d4a8fb3b0c038cf84c8f8 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2092c464db748a7937e743f77bb014f : SUCCESS in 13m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b1c638f101eb46ff902d62dadd7d077f : SUCCESS in 1h 02m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f1074429a1f474191101a50bad0a741 : SUCCESS in 1h 02m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cad02c4e9d824bc9b83352cfc4b6b78d : SUCCESS in 44m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a4bff29beac48eb88a91cf93a7f7847 : SUCCESS in 49m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ec36085ea7b45408f96f5c7452eb03c : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13710ce0557b4fd0b13e51381cdc0e33 : SUCCESS in 7m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb02ed92a7eb4839a820350de2f3bdbb : SUCCESS in 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82709b9273d34e63aa11246cf7e1fe19 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":4},{"id":"9773fc8e85d08350630e52f3339aa7fe8b753226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-21 05:58:52.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"88be99d3095c1f06d60d8e29d3c88104cd420f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 06:46:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46bdd90d7f540c7b2417d914a339416\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4604a7eca0e472ea7991f464f9b6f7c : SUCCESS in 30m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00ed93248d6d4d03bee0d2e82b36006d : SUCCESS in 29m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8d7ea581796463dbf713a396320984a : SUCCESS in 44m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/585802501b264270add3a7cdd6dab997 : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":5},{"id":"06837ca56381b22bf634a8a3639b08256b10895d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:06:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b91eccf52b9947878839c4752f781f44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f860587c8fe409e81e2360a35be5545 : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02788053f63c46a58700c8ccf9afe40e : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/427be703caaa45599a216cef8046512b : SUCCESS in 49m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1b60b9f72bf48e49d1f66cb4590e5a9 : SUCCESS in 42m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae21aace903e4a4c904683154f6e7e54 : SUCCESS in 38m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b250e19afd147f99074c99c5b1d5bda : FAILURE in 21m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cdc9afd218544fc987bc91d9523f698 : FAILURE in 16m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f298bbef4214980bd1b7d5c6d94fb42 : FAILURE in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8a862dbdb6343fe9c98244028fbd085 : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6156ca5c2ac64d71a22d9f5d8fc10d6b : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ad4e4bf3ac447cfb117436f7a8280dc : SUCCESS in 14m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bebfeae893b404d8c737a2ff51d9ca1 : SUCCESS in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/495e3e5d4c454325a4de1208137d0bf6 : SUCCESS in 19m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad3e93a8bb924f2a8df86cec6590f12e : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/512b92ff7bec4bd98abce501ffa610f7 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b315fb0c3d84905919f5178dfba29e0 : POST_FAILURE in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/50a5db9bcce94294afe3e1c926029058 : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/408ce800866d446696f0b60bf400e81a : SUCCESS in 16m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bfa366149ff844e699629baa76802a9d : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8503d574f6ed4577aa1c42222eb44f2d : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ca3ad4167217409c83b3cd87e476cb28 : SUCCESS in 12m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/27742deee2f54795a21bb4b73669e5fd : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34a3ddf34726456d9f3165be18ca9d30 : SUCCESS in 53m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/861809cdb427428c9ff6e0d6ebecf8fe : SUCCESS in 1h 00m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64b1588188204244a144eed3c3b967e3 : SUCCESS in 45m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41c85f2e6c0c4aa0ac0046117897968f : SUCCESS in 32m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e9142b9765844cb8c0a41fccaaad6b4 : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530c0f5318e444a5990f8919c5f0f156 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad618423ed0841e188885ebf2227550b : SUCCESS in 20m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2034072e6844704a21d38acd269fd35 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":5},{"id":"e5f13b02b53fea33277011cc6d6768e2765d1f10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-11 00:00:32.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"a10309edbff2d9540c328567651d5cfd28dc53fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 01:10:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9693d3e084e546d4acbfa246e27a68f8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/698e5c37666d47a1a4aee8497592c8e4 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7530892ec6f4754ba0a60aaba11be15 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/8bed121c14794568b5f8628395db2a6a : SUCCESS in 50m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21188025a4e24068a68498ee70423737 : SUCCESS in 41m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab294bef9bc347a4b137ed898a5aa554 : SUCCESS in 40m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9001c4c4a09b4a499d69fffbd2701c18 : FAILURE in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed5900730df7449f81e71adcc0181418 : FAILURE in 19m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/020cbcb981c84f569e2a02a0b7fe0cce : FAILURE in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b1330ecb9d65419b8157f0bc88187b76 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cbee4eb6b23c4556bee28a59e203b3e1 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/87da45d002e24bd1b1726600a8bd49d6 : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dceaa52b30284b77925c67913549c31e : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1fb23b7ea65549959f8a54c76b494bce : SUCCESS in 11m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/80570b5dc4e3411fb31a76bea6c472bb : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f28d85165c2844a1b532d5da2673771f : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b719be357be84258aa556157e04c59d7 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/37431ab29b454f1197a5341705b38491 : SUCCESS in 12m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b8439096930445d89de4e8ab090d43b : SUCCESS in 18m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9e02f92e49fb41019cc7c9f5be9d86a2 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62d002dd438b4f32a7b316f93bd60925 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/471344efcd4648c6a644bf19165e584d : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07d457286de7414e85d223df18da0fd0 : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3dd81be94cac4747a60e7ea9eecff04b : SUCCESS in 49m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fad7fbe4c6a842e09787cfe997938558 : SUCCESS in 1h 03m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efeac4b4d2dc4312830ec8c5862510fb : SUCCESS in 33m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45b3dc103c8c4fb19393dec899992560 : SUCCESS in 40m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/267581a734534e04af61bd9ba196a08e : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ebd7b26041437890b6f67ffacbf3d0 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85fa2dfc132e4b2fa651a1431862499b : FAILURE in 21m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a095557414e14230b27f9973fc3974bf : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":6},{"id":"a19a5165887bd3df276bc97b9fcbb8332ead0bbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 01:13:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e383545984b74de89fc4ac02fa26b1f2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9504f7a0536f4b979480b65db5af658c : SUCCESS in 46m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4725e00fbd6d4246b8a005afc60f4163 : SUCCESS in 1h 01m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7897c5f5359748e984d8c8398511cefb : SUCCESS in 47m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f4ba15878b94d8fa663f2d289fef60f : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":6},{"id":"8fcbb0ab5070822e625369c50220ac693c8a4d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-23 01:12:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"78a883579dd920e6d2321b7fcbf3ef0933c1b3ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 01:50:56.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95f482496a73419b8483ba1e51da849d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eadc49a057be4aa9ae9719c11e859e65 : FAILURE in 31m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/058914eb0928423d9925c4885443e8ef : FAILURE in 31m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9b9b12dada141e48fef4b58ce93781f : SUCCESS in 31m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ea375d6e078458082b246a800f9548b : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":7},{"id":"f56852b87bf44b81bff83b34ee66186076d372ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 02:17:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a59b7765586a4e59baa225517220165a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/afccc54d2135464b8a42307eed4d5e56 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4e4a8d32bf4436aced92ec418796a3 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/ec0eb62165da462086315c1c9beb3b39 : SUCCESS in 52m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3badd525ecc470da3e3ab0fe1c57d47 : SUCCESS in 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7972405fa2e24b8a965a553c97b660b8 : SUCCESS in 58m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1eab69d6f0e64a1aa72cae9316e41909 : FAILURE in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebf0dc6c703c4bbba7649a9d0c9f12a5 : FAILURE in 24m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29f3eae525af47b2a35b65cf82cbda04 : FAILURE in 21m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a9c37a98716411c9d36fd55473e6a63 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f868a9b111f4140b9e2b30082df68d2 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65e73bdc2c304fec827f91db00e58c1e : FAILURE in 15m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5883a6007021425a836d9e8133b54245 : FAILURE in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67d2908a419a46af91e825dd53bd183c : FAILURE in 18m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd67d1d0d3b04450a80eed0d05033c78 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3eac3f94493849a6afeed874e5201df7 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f85f72c8b7942f291f6c9dcb6142843 : SUCCESS in 17m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4f71166efae74ba08d2a588c87cbec61 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c47e7ce0ecf4164adec422f5a7a3892 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a063abb75efa437cacb8b89798bf1dde : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4091fcfd5d6f4d8da08c3e386174a89b : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6b0fc35aa86416a803dc4d87235a7f3 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a81123f82b141689202776325add144 : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8923245b2add40cd9fcc8fa4d0df7492 : SUCCESS in 53m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/40a14141aa964f5999c6677704727b5e : SUCCESS in 1h 03m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2160603da884630b660be9709e0019d : SUCCESS in 41m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eec53638bb8a4659b7e2894e7156e578 : SUCCESS in 41m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2f94172b876464d9d6839df3b76aab4 : FAILURE in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad4cc99d10b49129f18d0d8c19f7d6e : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/570b055242084fd8b4a6885b38d2b85d : FAILURE in 19m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b711fca9d9934419a726614470355741 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":7},{"id":"e76db3be78be8b9aac05bbf4a5ab8793acbf8be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-01 00:11:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6fab41b5a5ef31153288d03718150731dd49b217","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-01 00:12:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4199375737b1ba261c91c65921b5d90cf498f540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 01:12:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea79db8d07b34c10b3d20269657a4ecc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4b6ce283b1440deb34007aff6456364 : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ba8f1e19704bb2a4588601dbaf5e65 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/0f95363e5dc34e8db0ec8f2417451668 : SUCCESS in 41m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b435385c09b24fc9841d6025d20c8c64 : SUCCESS in 36m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5825051e36b3437ebbfc074b5fa2451d : SUCCESS in 42m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/358a1f1233f0431b975a1a285fa3f8cb : FAILURE in 18m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37dcc3a1fd3d4a1799e555509df4ba38 : FAILURE in 15m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c37c80c1001d4495a9e7f8c789281eff : FAILURE in 19m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/846b8baea4a84dc79b469cb6d71f057e : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/483ae6be85124b4b9f2dce3cb4a8dfac : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42caefdee6fd43d695800ca901c86694 : SUCCESS in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bac68935582a4ce39556c22e27d618cd : SUCCESS in 12m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35f3b4f2db8c4b0d88d534a7439885d2 : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e418ea952d61406cb25de821d40654f2 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc8a9061bb864843a5e2265710dfef16 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4709166614c64040a6facabbe8fadc2c : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0c5bbdb4774e4e3998a4a3fa45dacc40 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a78883e4e7234d649384b051b73697cb : SUCCESS in 15m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/275c220097a649deb3bd7149040ec810 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6a57a256f9c4f049dbafad26287f44a : SUCCESS in 8m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55dd66ac62eb4d8da7a6e573bbb9097a : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c039faafbbbf426d8043c41f8089e667 : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e0412459b83b431397ddf60fbfeafde4 : SUCCESS in 55m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/88d39a95a45a4605a8351e1603dca507 : SUCCESS in 56m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14cb4b2486e94c93847ef7691789be93 : SUCCESS in 39m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d4ae3828cf248e892fbf6b7fb2dc437 : SUCCESS in 42m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/670dd1e5ccd84295800fc828c3499d8f : SUCCESS in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c49c3954b04a57bbf454d79e045501 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0aec1083a4984603a3984536b46f7e7f : FAILURE in 27m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f49d8bb0e4e4866baf0348d5ea6c1a8 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":8},{"id":"09f68e0dc75d0988648afbbc0805bfe24e945644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 03:21:43.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/373211c7660547148d1df147d1c97539\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d1e6c0ec56a435e8eaea414e3b25056 : SUCCESS in 1h 01m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69f575e09d9a42d3855f7a6140ad7de6 : SUCCESS in 58m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/956be9d892f341ec92ae6dc667e91b37 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2713373dc9a74cb9a50274da3a092645 : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":8},{"id":"48d9f7ccb9dce4242fadb9cde6f13ef4c16b3c16","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-28 01:36:04.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"24edac3c3fa16dfcdb54070fd4eca43a61975182","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-23 21:20:18.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"28713851fe287d31d7c9c108c7f44365f8c6cf7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 22:52:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82d866caa8aa425892a3b0fdae1c03ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e19f8b0dd68145a2a161cdc5ba95ff36 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef22aa5ec9a4f4d92ade0027311c7b9 : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/df996b14ca4d4e6c8bc03a1dfc1e8cbf : SUCCESS in 1h 01m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00e8d2e2bbbf4438a99e2e6d201cb19e : SUCCESS in 1h 02m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d295c4bb0148422589ac7ed02b94e73c : SUCCESS in 33m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca3dbef730da4a2f851c4f4bb7e3212d : SUCCESS in 47m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee7e78dd1cd54987bac20a5a16faebfd : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33c83e739ed7461ab649b2203cb47123 : SUCCESS in 14m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98324b6d351e419f960e137eed79ee0d : SUCCESS in 23m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c6a6b5e66d048488387da2d47e95ad7 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fae80616daae454c9b0231c5d0b19c09 : SUCCESS in 11m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6ed09574c234e5eb41650a781e472d1 : SUCCESS in 19m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a2514820df74eccbac329bd7459ef82 : SUCCESS in 10m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cb9a09de44954a88991742b3d1887438 : SUCCESS in 19m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ab820800bde418dbbae18721e2859de : FAILURE in 6m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/84705bd262de44589be52153b7806d48 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b328bef2ef948ef8c86837f20b6b75e : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b80671396bef49aba664ce9343083e98 : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12c450951190405eb5831d49024a8a8a : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d1f67dee72f74984b8348320d869d78e : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ad5ca8112aec4bb19ea0fd6c2451d710 : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a5765f430e54d18be6996458ea279d6 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35f69cf7755e40c996a84ed364eb6ff3 : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13c7514bf9f24e4e980bafce48f8afe8 : SUCCESS in 7m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eaee7f4a04f0421a87f47ed23363320d : SUCCESS in 51m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64a8c19563474cbfb38d7f808372819e : SUCCESS in 47m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bdb7667fdc44a149cd5cbcbcc2ce9a3 : SUCCESS in 31m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6e1d0bb09f0496b9cb86b099c0558f5 : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293489b38fc445c78456d26eb7dc5949 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ed29e06940642b9aa4f113e4a5c040b : SUCCESS in 15m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/533811dd4b6f4c3bb86a4221926511f0 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":9},{"id":"35de477a658cd6e0c8a0c4eb1e7fff0266459c8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 00:22:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3838d3c4374d4e40b2a886d934e10532\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5d5a6f53c13454eafae8413d84bfaa8 : SUCCESS in 2h 35m 58s","accounts_in_message":[],"_revision_number":9},{"id":"94aac303d6e936c51134bb261c92279014e2ee67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 19:09:19.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dff167bb4dc8089eeb8e66e428ebeb85fdacab95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:23:08.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8d1672609b24681aa0e4ed47bc86234\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff76b338b76441938c5116787db0a653 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3513c79cd9c5426db0b5d0d8b72b0791 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a0932cf2fac94b6ba742751f14abf4c8 : SUCCESS in 1h 00m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98aa087de3024b30970bf5519666e5cc : SUCCESS in 55m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e940fb4ad9a4b76a1791b5b31b70713 : SUCCESS in 45m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/917499ed0835406e8ea074ebd5d25895 : SUCCESS in 1h 01m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9cef0f83e3e4e1b881cfeca70cb9f88 : SUCCESS in 11m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b25301b8176d4e4fa3a270fc752e8097 : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72e560f8aabe4e3fa0fac6391d792a25 : SUCCESS in 11m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56c7c1ebdbed40beba4e6036cde7c1de : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14c16ebdb6a443bf9f0eef2a6c964023 : SUCCESS in 11m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/444a644663b14952974f70dad22dfcff : SUCCESS in 19m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4f19404fcd7b4394b6488d4747183e34 : FAILURE in 2m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/521fd80680d9432a84db3d99d793c60c : SUCCESS in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/61297acb5c214b1bae939c2bf0340dae : SUCCESS in 20m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/11c31a0eda0149559dee44cfd64a4876 : SUCCESS in 11m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e26e44d713724cd1a660e84fc53d9e8e : SUCCESS in 20m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/579bcc04b4cd496496308d3f75b91270 : SUCCESS in 11m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25df4ffcc64d409abcbf95f1e039815c : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5eb67f2257674d5b8ff1d3adc1dfcc40 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a80a900edaa7471fb0a384e4be1b9f78 : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99a55e0f2e7b4203a9949bd3dc28fca9 : SUCCESS in 8m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6bf71dc2db04439b1ce13699c4d3a0b : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e2b4d604ea444e64b0ff29ad3e1fe67b : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff3ab742dd1c4813afee47dfabd2b559 : SUCCESS in 55m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e01ee4ce289945ceac56e2b855b42a38 : SUCCESS in 53m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e78ae161f0344a499cd364efce4c0e2 : SUCCESS in 50m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d529e0ccd44032b068b89fd6908e03 : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07609e8a8b1340b7b8da7142457c58bb : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a543693f03a4b93be1ee7f32bca69ca : SUCCESS in 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/927767a7785547289b17439f3c30eaea : SUCCESS in 53m 32s","accounts_in_message":[],"_revision_number":9},{"id":"5d9f09f5b0b4dc63414376fd64fb643acb5b492c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 21:06:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dee72af6c314d599bbe8d7a38ba8568\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe3950351744474aadd0998a0213d856 : SUCCESS in 1h 55m 29s","accounts_in_message":[],"_revision_number":9},{"id":"1a7b4f523cf9090a50f270c4c9cd5121b4de3d59","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-25 02:34:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2a211e0f2f9c057463f515afc4b441bc52ecc19c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 03:47:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b1d0c82c0e14568b527af835f5836fc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/295a6351d4f5466bb19dda9252443254 : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9af2123b234fa5982838e6814c6953 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b472a67a3dbe44b9b4ec0132cebaa0c8 : SUCCESS in 50m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b05d4322c37147949f64e4bf2ef92345 : SUCCESS in 53m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ec347c12f1046648580908befb733b0 : SUCCESS in 53m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f07a0796ca34cdfbd60da93a560fef1 : SUCCESS in 48m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/281911f10d2b4dbd85eee0ae7bae9540 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b00ee3da82c41b180d8fd73a40431ed : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e17c483a6024495a7e32f05f1fcb218 : SUCCESS in 16m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12f66b48aa27447db8bb235c9aeee7f8 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c5379afc20948d4b7a2d8d98821db27 : SUCCESS in 21m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dfad1f9d7204464abbfb6980170811c1 : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/55ef655f09d34182945122b34c5d28a5 : SUCCESS in 18m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/31d5c8ae2cca4bf5b17e1c698f929310 : SUCCESS in 15m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25fcb46874934de6a031c8c8287801bf : SUCCESS in 21m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d1402354e7e46c0a2c5603153d68b96 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1ab79d19b814b64906d6797589c26f7 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d89e289ea0984bd28d112defa072ec02 : SUCCESS in 10m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e7c3e71d2f3e4033a0ecc4849350b64c : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d14529bc53754f93abd6e8b874a8f12d : SUCCESS in 21m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e28cbc89448d4a18a99736fd12274760 : SUCCESS in 22m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6634d2f4a32942eca2f4386173698931 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/544f4dccecb34d279ff1190055e881c9 : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9380fc9edfa4da4a3e9e73254d82b09 : SUCCESS in 9m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94397af8afbd4d81b39123b15c487b72 : SUCCESS in 55m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2afb0a0bedbb46738750452fda1881b4 : SUCCESS in 53m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0607b13ee53403ca0d87aa207e6999f : SUCCESS in 56m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16c90602056e40fbadc9f82d95412825 : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd83fd04566e4ff2939694f278b81447 : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/467d3fde226340ad81d47069deb55320 : SUCCESS in 23m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/370e1f59638a4e50b123dadc13802d0d : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":9},{"id":"56b708ea0570245ec7a0ec61bb4d7b3d9884f6d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 05:39:47.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db4972bfc894a6ca63aaf0602877968\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a653dc18c4cf4a2d9930fe2d9ab6c6b4 : TIMED_OUT in 3h 00m 55s","accounts_in_message":[],"_revision_number":9},{"id":"0603713514738b6858c2020835bb8a66eb5b6664","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-20 22:24:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fe21cf04a3bd7c9e4bc69c25ffade87edab00b40","revisions":{"d6dcd189a048dec5854a5aa395cc560391e605cc":{"kind":"REWORK","_number":1,"created":"2022-09-12 20:57:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/1"}}},"commit":{"parents":[{"commit":"e33fcecab655584417da87df5c2fc732a57cce87","subject":"Ring v2/history follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e33fcecab655584417da87df5c2fc732a57cce87"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6dcd189a048dec5854a5aa395cc560391e605cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6dcd189a048dec5854a5aa395cc560391e605cc"}]},"branch":"refs/heads/master"},"8cc6632a3d7ef3f7ac0c1406a1a483bf3c83fd1c":{"kind":"REWORK","_number":2,"created":"2022-09-14 18:45:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/2"}}},"commit":{"parents":[{"commit":"c004b791a1015a22eaed07af25d16a0711d408eb","subject":"Ring v2/history follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c004b791a1015a22eaed07af25d16a0711d408eb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-14 18:44:04.000000000","tz":-420},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc6632a3d7ef3f7ac0c1406a1a483bf3c83fd1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc6632a3d7ef3f7ac0c1406a1a483bf3c83fd1c"}]},"branch":"refs/heads/master"},"8f0f5beccbfe3fc59b1a068c837520186acf85ad":{"kind":"REWORK","_number":3,"created":"2022-09-14 20:47:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/3"}}},"commit":{"parents":[{"commit":"bd58b05fa7eca41886400a910a213114b273aa5f","subject":"Ring v2/history follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd58b05fa7eca41886400a910a213114b273aa5f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-14 20:47:03.000000000","tz":-420},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f0f5beccbfe3fc59b1a068c837520186acf85ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f0f5beccbfe3fc59b1a068c837520186acf85ad"}]},"branch":"refs/heads/master"},"05ac31d5e47e8748828bbf01c95dd0edbec46443":{"kind":"REWORK","_number":4,"created":"2022-09-26 22:39:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/4"}}},"commit":{"parents":[{"commit":"3d86afa04d2cc69ef8510fe1c98880affcdedd08","subject":"Ring v2/history follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d86afa04d2cc69ef8510fe1c98880affcdedd08"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 22:38:46.000000000","tz":-420},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05ac31d5e47e8748828bbf01c95dd0edbec46443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05ac31d5e47e8748828bbf01c95dd0edbec46443"}]},"branch":"refs/heads/master"},"c79a2d613f15af5582a48c8400153a8d82906772":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-10-21 05:58:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/5"}}},"commit":{"parents":[{"commit":"cedbe893cbe815c4b471a93ab6c5bbfbf07761be","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cedbe893cbe815c4b471a93ab6c5bbfbf07761be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-21 05:55:56.000000000","tz":-420},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79a2d613f15af5582a48c8400153a8d82906772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79a2d613f15af5582a48c8400153a8d82906772"}]},"branch":"refs/heads/master"},"2536f846daeebf15cba37eeeb03f398b67b69978":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-11-11 00:00:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/6"}}},"commit":{"parents":[{"commit":"16d00bd570b708befc8886b008d29fff58d8efcf","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d00bd570b708befc8886b008d29fff58d8efcf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-10 21:39:10.000000000","tz":-480},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2536f846daeebf15cba37eeeb03f398b67b69978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2536f846daeebf15cba37eeeb03f398b67b69978"}]},"branch":"refs/heads/master"},"93b38b8a3c8940012727af6bd0bd33f40751accd":{"kind":"REWORK","_number":7,"created":"2022-11-23 01:12:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/7"}}},"commit":{"parents":[{"commit":"16d00bd570b708befc8886b008d29fff58d8efcf","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d00bd570b708befc8886b008d29fff58d8efcf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-23 01:11:35.000000000","tz":-480},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93b38b8a3c8940012727af6bd0bd33f40751accd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93b38b8a3c8940012727af6bd0bd33f40751accd"}]},"branch":"refs/heads/master"},"9427b96d81e7fa3b58ebba37ffeb084c62a6b3b0":{"kind":"REWORK","_number":8,"created":"2022-12-01 00:11:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/8"}}},"commit":{"parents":[{"commit":"16d00bd570b708befc8886b008d29fff58d8efcf","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d00bd570b708befc8886b008d29fff58d8efcf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-12 20:56:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-01 00:10:07.000000000","tz":-480},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9427b96d81e7fa3b58ebba37ffeb084c62a6b3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9427b96d81e7fa3b58ebba37ffeb084c62a6b3b0"}]},"branch":"refs/heads/master"},"fe21cf04a3bd7c9e4bc69c25ffade87edab00b40":{"kind":"REWORK","_number":9,"created":"2024-10-23 21:20:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/857234/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/857234/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/857234/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/857234/9"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 17:15:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:45:53.000000000","tz":-420},"subject":"Add swift-ring-info utility","message":"Add swift-ring-info utility\n\nChange-Id: Iefea826ed3ae766163ba0bccb9ace521ada0ae0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe21cf04a3bd7c9e4bc69c25ffade87edab00b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe21cf04a3bd7c9e4bc69c25ffade87edab00b40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
