)]}'
{"id":"openstack%2Fswift~858277","triplet_id":"openstack%2Fswift~stable%2Fyoga~I4d807f3b3649a45dbb2166bb8e911c45fb9cb701","project":"openstack/swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-10-12 22:30:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d807f3b3649a45dbb2166bb8e911c45fb9cb701","subject":"CI: Add nslookup_target to FIPS jobs","status":"MERGED","created":"2022-09-19 09:44:28.000000000","updated":"2022-10-12 22:32:17.000000000","submitted":"2022-10-12 22:30:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858277","meta_rev_id":"ca9649ef1d1938411286352527e495b409a20e9f","_number":858277,"virtual_id_number":858277,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 22:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-12 21:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-12 21:27:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 09:44:28.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-09-19 11:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 21:27:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"42f72a393a9f0a18f886d9b16e38697a578e9fd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-09-19 09:44:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dff31953c323b79e5ee481cf959f2d4808183870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 11:08:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f56d6ee4f9aa4db194b75c0853f8bd07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3f2048948a41b2923540dacfc4601e : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/c1ec51b800a548058010eafdd2d716ce : SUCCESS in 1h 08m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6862fe3838094e9c9e8ece4369d796d4 : SUCCESS in 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8145e8352f754576bfc658ddff355da7 : FAILURE in 1h 18m 54s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d08088b6f31346d2b09fb6a6894c39cf : SUCCESS in 24m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3db10c3ecffe4199ba6610186f0b6b44 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8c1a2b64a4c443799d3c1d7c9d88b0d9 : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85b3f0be452e4956a0dcb86852616532 : FAILURE in 3m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aafb33bc94c94fd893b7b4019d44fae6 : FAILURE in 4m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/443535a5b5104b6882613b7408a8e1ae : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57aab3e58aaf42f2ab02cfd4735a2b43 : SUCCESS in 17m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d18fbbf7fb77402f8b24108d4144409a : SUCCESS in 17m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa72fe5a78bc45fabca01a4902132d83 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6dfe5cb37874d1fb28eda8ccdba74c5 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c604ffe29fe4b1cb2c1114efc534f8e : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4db541980b9c4db88b1315e28d401458 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f3b867dd5970426b81a3bce89d52103d : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a2cfcae5c7db4887815f6395db2218ee : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d186b341d5e0438c9d9f4e3ae164ba5e : SUCCESS in 8m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/310f7497e0a94503a6348914627fd222 : FAILURE in 8m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1de759c33ebf4413962a592a16d574d7 : SUCCESS in 55m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3faaa9c5746a4f7aba570c699576efb7 : SUCCESS in 58m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64e2d3736f114e09830869d2015de467 : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4195047acdd14762a8b3d1e8159fe791 : SUCCESS in 41m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ff398cf3c0e4ea7a737750415a9666f : SUCCESS in 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27462b73548643548034949c1e93f6ff : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1743b9eb73d4adc9689d1b4551cab80 : FAILURE in 12m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ef42fe31aef437da130ce585dd06840 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1feaf34e964d24adbae1a3ac8fbe7285ef87dcd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 12:30:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05320e90f3c34c278b753f3a8bcb50d9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd23acdec06442fb858ed8433cf9d8ec : SUCCESS in 47m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c8cd622919f4a838687cc7124f38b34 : SUCCESS in 39m 57s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/56b69cd904204e218d3d3e6c0d9a6d3b : FAILURE in 2h 44m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76f7d81dbb2544ad80770fa1226d1b4b : SUCCESS in 33m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cb8bd7e22f34f93a209f3d36264cafc : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e738321ea0b2176a28282d4cda9dcff435b837fc","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-09-20 14:47:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"363df169ed6bd99d1c12d78a7c6e7c67cd77e712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:31:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38bbf90d4be94a4590d9132644d1bb50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d68b55f1c04ac9bf83d5566aa40be0 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4e34699379a34e4bbe65ad7eb2e8bcc7 : SUCCESS in 50m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b36593b496646be8f6483255b1af98e : SUCCESS in 47m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc68947746e0421c846f65c7d28fb317 : FAILURE in 1h 16m 34s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0421d651367b47ee95f36e2ac36f1a7b : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b8fa4f687d594e5f9cc88b49efd0b057 : SUCCESS in 24m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a3b019c16a9b476d828760345bbd822f : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b4a0c02d57e4833a69bf1507e473519 : FAILURE in 4m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bf35d0d57ae54374adf447bfac93c473 : FAILURE in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5fb9e90c23043fb87f2719c98131430 : SUCCESS in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28f78bd1c52e480092d3fa7758c8db41 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d9069d6ac1148a29ec937d0cd11e388 : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d736cae4b9743cd9319424578d9f751 : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fce4ef84fd1144eda1dce0f69ac3ef33 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6cc2df7332344fb8a78676ad03d88bd6 : SUCCESS in 20m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/15b26a4f9e534e1e97bcd353cd80afdb : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1dd739a7dccd497abe72cca34a3a92c1 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/09e7459fefad4a8f9bf5cd777532458a : SUCCESS in 18m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a9b9b333b9c5451581649c497dabb18d : SUCCESS in 8m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a0377dcd45f43be828bbf0d6aab1af2 : FAILURE in 9m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d07dc78e145f48fabe68bdee3aad906e : SUCCESS in 54m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3df22177decd413784998a9d62f3e581 : SUCCESS in 1h 03m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67bb1e6d32fa4bc3a0768311abe12bde : SUCCESS in 42m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b2307b28860472a9743f5db4ef2c4a2 : SUCCESS in 31m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ffb90d99a5f43c897055b7c65641532 : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0957cb9bf7614f688e8b52ae11b6f837 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/897a333d1c0c4fb390fa2fa45bf6f853 : FAILURE in 11m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dfdcd4bb6f848e19cf3785a53d2d806 : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1516709ee4e179d6b61f22d81b2b782fd725d5a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 17:56:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f336049c1e4ccb976d11cd5a490bd4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3711ff95b884aaa9c1ee0bae32a2eca : SUCCESS in 31m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6988bfe97c8c4a398318d3ef53d7e34c : SUCCESS in 32m 37s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/2fd274097ce9422da4f477ef1fb6597b : TIMED_OUT in 3h 00m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eafd1be1bd9c4952b813198fad86a8ad : SUCCESS in 33m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a29c570603a440680d9bdd122ee306d : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3f4a884b683ad48d3d93035f2b60db4f092f732d","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-10-03 12:03:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"902176f4c41fd8ad9367adb8300997b439f8c54b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 13:13:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a78692171fb444bdb70c042dd694ee95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a479cf9d154a449b1a9bddef19087d : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/acac172b8c63496d96535c5eba50b927 : SUCCESS in 46m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6768815e87c64bcda72c21ca80cfe26e : SUCCESS in 54m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f97072d77d149edabeb3eede1dc2ad7 : SUCCESS in 44m 05s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4b91b4f10a4a4c4baf9d805aa83fcd02 : SUCCESS in 23m 35s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/99e9dbf9e94a48d894e0b5cddab4249f : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1323533b9a954d708ed44af2094d9794 : SUCCESS in 24m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d515631891df4260b5d978747ec21790 : FAILURE in 4m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ccc1f87358b54ab8b9b2663f9ee24865 : FAILURE in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee617198d94842bcb5b8f1a9533862d8 : SUCCESS in 20m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54759326d9f042c78d9dd10f54764503 : SUCCESS in 14m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a7d0e172797460eb77176d1fb6cae54 : SUCCESS in 18m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ad7dedd79784540b22fa3c96d65013c : SUCCESS in 20m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/773433e4c1794edf85dedac707bef5d8 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79c614850f744285aab7400bddf665f2 : SUCCESS in 21m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/081b16ceeb584f2980aa5868e4bc1147 : SUCCESS in 13m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/37111bc45ff24c93b08cca606eb333e5 : SUCCESS in 22m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/681393915bd346e08d03019fac19d64c : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e40cc8172fd452c8ce2f4e58c013b6f : SUCCESS in 9m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2174b96d4bf84fadb4f8a41bbb18285f : FAILURE in 8m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e6467b9488e14dbcafe3e9a3913c7608 : SUCCESS in 51m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/df0238487e0b4d7aadb42adc81fe67c8 : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6a3e097b9444e2b80401ffc21530ec7 : SUCCESS in 34m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4eb0575dab5f43fcaf5c829ced18e36f : SUCCESS in 47m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b058d0bf023483690fd5ab71b162a6d : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9a50e0344d4951b1531741d070a016 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eed652f7b9a94de8ab27f3efddec0510 : FAILURE in 10m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/716f905d02c647d9aa6cad069f6e7bb2 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7a9db71e3d8b70e16a06239368d2c745e03a7959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 15:15:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3085fd1c15f42348345be8a5f5bcb47\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79dffaa780324fa1871eba0c9849b426 : SUCCESS in 37m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1434af3f8c746d48da2576f70c7fdec : SUCCESS in 33m 50s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6c107fe549c84851a1b3b6dc849cee5a : TIMED_OUT in 3h 01m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/38a606b8d1bc425badf21a535717ea10 : SUCCESS in 40m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b5f7e730f2f45fb8456281650abadfc : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a94921a2d541ce63c7dd3dc3b8c5a968b71b1557","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-12 21:27:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e100c00e8a21349fd279a2af9fe6814b160c2bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 21:28:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffad2bf9e6fcf8456553ab65a094495e9768be33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 22:30:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0179eaeb815a4763b8e7ecb878d4f8ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a1c3e882464ae597cc7e64668df907 : SUCCESS in 7m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/c200678483f44a7983f0183ed02f93e7 : SUCCESS in 38m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7254c90331444141be2466396d7b5b25 : SUCCESS in 46m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3fbbff73b1942058348a274ca05c691 : SUCCESS in 36m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/391fc186085c474fa635ad8dc25c0606 : POST_FAILURE in 4m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d4ebdb146d5d41c99cd68835d85e83f6 : POST_FAILURE in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/652ad616d28d4923a96a731b9bffb9ed : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30faea9bb20846bfb70267b1cc44c43a : SUCCESS in 14m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b85b764757844481be28b51c6d3c7a68 : SUCCESS in 15m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c765b77f47c4efcb4515da1c8d21b2d : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9218be97890d48a7bd232967d91c3123 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/885e9f4d4c134a2a9fd7f8e515b90ce6 : SUCCESS in 17m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5673e071286e4164a396805d40dbcf06 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6ab7743e3c9401ba335da49b0fd1852 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c66e7214862a4f9cb5865f232f360acf : SUCCESS in 14m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b76cd6ef5fd44139e726e2b4d9e986b : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7faff36aaf2426685c38ff089a92baf : SUCCESS in 55m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/689e216baa5142ada95bcb7eef4725ac : SUCCESS in 1h 00m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3685f281879247b0a389d1d0f39ccb38 : SUCCESS in 37m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/348b358087c34549aa73d8d6ee9879dd : SUCCESS in 37m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19bd1b29a039400a827af1879a8f39ee : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708d1bf6e2534f6d9d09401b133a2a6d : SUCCESS in 5m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adcee1c3e2a048eda2cd2700f5dffde0 : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1f7d76f578e097d4d5d24efc5cafbd6abb97cd6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 22:30:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca9649ef1d1938411286352527e495b409a20e9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 22:32:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddef8f5d6b434ac88803a2f82754a10e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2699c6f523948f4a0d6c47acd879853 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3268f949cfc144a1a775e36feb1f553c : FAILURE in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ec71da1868c7458291834535da8ad866 : FAILURE in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2","revisions":{"3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2":{"kind":"REWORK","_number":1,"created":"2022-09-19 09:44:28.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/77/858277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/858277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/858277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/858277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/858277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/858277/1"}}},"commit":{"parents":[{"commit":"00763d5a00896a28b6665e88f9ea4d4a9fb50a83","subject":"Stop partial()ing hashlib.new","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00763d5a00896a28b6665e88f9ea4d4a9fb50a83"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-28 21:58:09.000000000","tz":-420},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-09-19 09:46:40.000000000","tz":0},"subject":"CI: Add nslookup_target to FIPS jobs","message":"CI: Add nslookup_target to FIPS jobs\n\nAt some point, this became a required parameter.\n\nChange-Id: I4d807f3b3649a45dbb2166bb8e911c45fb9cb701\nRelated-Change: Iebf474c9351e4246d7ab2072b48a50e93dbf0b94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
