)]}'
{"id":"openstack%2Fswift~858278","triplet_id":"openstack%2Fswift~stable%2Fyoga~I8b0bd7ee1c649294ac3795cba801c9e07bc3d51b","project":"openstack/swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2024-02-20 14:13:53.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-12-21 17:03:43.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I8b0bd7ee1c649294ac3795cba801c9e07bc3d51b","subject":"Migrate CentOS Stream 8 FIPS job to CentOS Stream 9","status":"ABANDONED","created":"2022-09-19 09:44:28.000000000","updated":"2024-02-20 14:13:53.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f8df4768ea08dfb6e4609281e69091eb7e64f27","_number":858278,"virtual_id_number":858278,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-27 16:39:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-02-22 16:28:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-02-22 16:28:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 11:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-13 09:51:28.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-02-20 14:13:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"e338d304b37d112913c9e754d139541577f67226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-09-19 09:44:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e05968ce86e6a191234803313715ae5d7974845c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 11:49:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a01925d53984cb69d8b41703a9cebf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/accbb08f2cbd4ccc85616f9f3eb8b01f : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/014b9cb8725a4ac4964866667ed6ea72 : SUCCESS in 1h 05m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52f12ab170b04fe381a4a966e2cedfad : SUCCESS in 55m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81a5b3054a504b45a9ee9498e28076eb : FAILURE in 1h 26m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/040209ccabb0499caab959222270f72c : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ca66dcc2eef41a49e1537219c6f9a5e : SUCCESS in 29m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d2217fd84de4af9bdfad1bc3998edd6 : SUCCESS in 21m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7826343a1dc043ce9af6ed5c5f876548 : FAILURE in 4m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e056bea385e04f49ac1e9e0d30125d84 : FAILURE in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e648a171c5344ca48c02c17fc5f3d71d : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6dd33066c92740fba23b12ddc1644f14 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bca913dfae924e74ae731c6380f7d6b4 : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57c6298b29804e7f84dd5795c2f4aae1 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0c659fd2f454cd2bed3dbc469c134aa : SUCCESS in 19m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e6701d1a99d84d30b2f675924eb9b563 : SUCCESS in 21m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1337b6101c24b7199b01dbbfaa7dc8d : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/73e17a3bcdef46fab12674e52066a83d : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bf1b9662b79f4b909ac24da69f9835a5 : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/40881bdf41e745a3906501496c439a8c : SUCCESS in 10m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a82bc1dd438472cb45d6028b3e31498 : FAILURE in 9m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a8ee2d9aa8a84938ab5460dc5682df26 : TIMED_OUT in 2h 01m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/42486ddab2aa4ed8bc627b9709453d7f : SUCCESS in 1h 01m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d97400c6a7214291895dc2ccb50d2281 : SUCCESS in 44m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bc5d693776747f2a5ce555390db8c7c : SUCCESS in 44m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08656583daa8435b8c1e301197c77253 : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e39664c8b4e04041bc29b346d0d76af4 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/385ac439e5904f778f9073e38160643d : FAILURE in 11m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30e12b58b69443bfbc06254d2993e0b2 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"880d48c278082db9e393f14a25e710ffb5f73f89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 12:54:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfffd8d342e84a28a56874ac35e311ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/498221bbc3f14ab0950df1e49f161752 : SUCCESS in 32m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2891cb3612d74237b6b6d13a5a672763 : SUCCESS in 38m 12s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/201caa8963b54ac79159d333018f4ca2 : TIMED_OUT in 3h 00m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8781d270a3804892812e878a637a227f : SUCCESS in 43m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b258d91896cd4c9296bf75bd7b44eff0 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"30eb8dfa9432a08ed257193faa30ce08ff23ed45","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-10-13 09:51:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f60a8a61a16098702cd69ccab981fcf427935bbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 10:58:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559ca4dda15c4e538b13a12c96ae61dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82c758bdbfd4440827f3a25f42f1a9b : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/c0f948e20d284674ace1a567367a4a09 : SUCCESS in 42m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70ed727eba9f4fdfab15b03dca1ecc92 : SUCCESS in 34m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/958058b67d254bbdab161b42cb651453 : SUCCESS in 54m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34dd8f9a71b84b469f399b19147f6bd0 : FAILURE in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec2671aa05fc4a7dab98e23f9477a158 : FAILURE in 17m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9b4fd0c69574acda99de037ffe86e4e : FAILURE in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db17a1223ee740718f3628beb502cc4b : FAILURE in 4m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/72d3bdc9e0564a32b261659a48444fab : FAILURE in 3m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05d677d69a01488b8d67372604a70c0a : SUCCESS in 19m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d1fce0c85f4b719b953a6f79302fb1 : SUCCESS in 14m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46582a9e3cc84e1895e62e5571e25259 : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06c52ce27f13454d8321c2879ee5c3e8 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4fe82a026faa4865b68b4c8fe3d7b753 : SUCCESS in 20m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b1d42e94c6b48f181f2c3cd01c2c86e : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fffb5d19451744eaaf36c75e32ddf333 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/119629fdceb4403a8c40363d7350340d : SUCCESS in 14m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e1db635b98c415d9b339a827d022a2b : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e37450ba33e34a08b35e9774deb5ae4c : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8917f0938da42f3b334683fddfa74a9 : FAILURE in 7m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d71995d667254925ad91f303983e4dbe : SUCCESS in 53m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2f843edab5d43f2bc32acabb82b3805 : SUCCESS in 55m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48de99dfcf9345598110b45c750c6b87 : SUCCESS in 46m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d09765b63638498b8e42785be8e5b84e : SUCCESS in 39m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94777382e5f2491d80ce83de6be5669a : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc393638c89424ea9ae149900b5c582 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8572943013b48b5aae69853c3a84993 : RETRY_LIMIT in 8m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb459a5616404df3b0b9f96d116c89a0 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d8b5e598a87ba2874d5a62f571c12d3a00316839","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 12:55:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d36087caa53e433e8fb4f7a02c3802a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b8647e967344a13a8d84698ddc571d0 : SUCCESS in 47m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98423a682378408b9fe0100fe0047cdf : SUCCESS in 1h 21m 10s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/493ccb97ea1944fda8d6d8cc6172e454 : TIMED_OUT in 3h 01m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d2155cd5f024d10bb3f1c3fe1f852a8 : SUCCESS in 38m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1300cf9588b48a1a6cce053622359a9 : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bc5362f7d52dbde8ba4545720a3d9cb9db0ace40","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-21 17:03:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5068dd0a3dd122db58fcd52ef10c64ec3fe3500a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 18:14:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ee0adbcf784ac19f2d796385e13af3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d2b04198ca430780d14470e8af6a0b : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/5c22d1be97c14ffcb89c463084f886c1 : SUCCESS in 1h 02m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ffe12ba3df404a0fb01f43d8f52ab634 : SUCCESS in 1h 04m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3b4069e13f2402eb587d1de52afd1a8 : SUCCESS in 54m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae6300d4a11a409faf9f620d2dc63d83 : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2371ed266eae4c929fd346eaec441fa2 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1fd532169e84ac3b0b43f51aaf555c6 : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05282d26815d4895964e9dc12ce8434f : FAILURE in 3m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8ca1ff52f6ef433a88ea815f26df4917 : FAILURE in 3m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/998c4b7ee91143e3ac6b146e12627124 : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e953330b5ccb43c9b8ed70d45b7e3978 : SUCCESS in 14m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3f76e5ad7c94917b9ca79ddbfacb630 : SUCCESS in 12m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac7b2da505b1479ba693c0d4a50ade7e : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/101e500216f64f8b9049c6469162c741 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7cceee129814ce7adecf26207c80a27 : SUCCESS in 16m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/09774c7057c54600978e3f47d18410ca : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a44d6675925a4c2c9bb6f6b42d916fb9 : SUCCESS in 13m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/51b98d5bf4d040c7a1d3683fbe8ad3cc : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f1d783c84af4411a2feeec39956962d : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a7eb3362ee864e8a8f4aef1123ea567a : FAILURE in 6m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e903450870a47919748b726141e7180 : SUCCESS in 38m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b466fe362d245b4a8debda8324c2f24 : SUCCESS in 53m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efff5568b27b4434b28eb79a474d889e : SUCCESS in 39m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1aea1c2d5d94478dbd481965af7338bb : SUCCESS in 34m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84e0f1795fc743019506b2acf0028a5e : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d6eaf4c8c14408a44ea0cc4e5547d5 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/061070abf3d54d1e82051d8d1139f570 : FAILURE in 7m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f6d91f74d7943c8ada8130b09e910a5 : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"006aa30e0a3650011d43488b72490c32037baabe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 22:18:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9200168405240329f67dbfa6b753c8f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77f5f729962b4497b28947d5f47423f0 : TIMED_OUT in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1d425e3afd64c7c8d2eebcc9e554e74 : TIMED_OUT in 1h 31m 02s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/df66670c5aa74b00a96b18cccb805d09 : TIMED_OUT in 3h 01m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4288013e283480387b6fcd5cd6ffb42 : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f07961c3b254426bcae13e6ca1d5481 : TIMED_OUT in 1h 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"90d72aaf9fd515400628904dff858290cd8fb699","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2023-02-10 13:39:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"999e993940350c1efca2de3c1835cece1905a0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:48:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45e61d90fa1347c593449f383e73ab50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1446c693e142464998680f91686f6e68 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/457fa08dbb344ebb9c1e21f365408717 : SUCCESS in 1h 05m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3385b314f3494dd58d4e91da5a8823e6 : SUCCESS in 56m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c355a4876fe84984862611c2dae893db : SUCCESS in 54m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82237da8ffaf43718160c10deb828e52 : RETRY_LIMIT in 6m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/744cac9afde443b5ae5e7d3d97c117b0 : RETRY_LIMIT in 6m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5ae78886c824b09876e5d246f5be6f0 : RETRY_LIMIT in 6m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/447087ad5ab24ba29dc33892afb5bc14 : FAILURE in 3m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b15f83ce08644efc8547cbfddfedc5f0 : FAILURE in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81191377148c43f88e36f923be40f729 : SUCCESS in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c73a9c3d036e4f23a3222b884cc01906 : SUCCESS in 14m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/68ac49b086fa47a386ea82bfdd6699bf : SUCCESS in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33a00c8f3f264d49b8016c793037cfa7 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f9543fbba63416eb18da8d7194c69e9 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5aaaed9329144811a216355f81c52fd7 : SUCCESS in 17m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4dd4b2e877cf4a29a48698a4490f8c03 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b8c318fc47614cf8a7215e25a61b3c42 : SUCCESS in 15m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1c5cf63083084e6b847c4edecf9f32bf : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31251c2f0882437e80ade007f19e191e : SUCCESS in 9m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49b5d5b1c86b45eca63b34e2f9ef58f0 : FAILURE in 6m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c19ea1c3971c48f9ae54782bd59e796f : SUCCESS in 52m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fa49febde8e4b2fb29c63081949d8ff : SUCCESS in 1h 01m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/251e1b4008284e5fa64548d5e49b6df8 : SUCCESS in 53m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf329a60d41b494eab2a8ea4d8cd0835 : SUCCESS in 33m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d0fdd9fe5644f594db35ba1a477c26 : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14487724b59f4ea1baedd823cd57c47a : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51bd91bab8d14c1990575cac10e76aa3 : FAILURE in 7m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd6c27454709465c967c91ce7ef242a1 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7ad5600cd79cf9611467ecba8244892265a598c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:49:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71ec22337bd84cb29e49a378267f95bd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b438d0124594eddab7ef16341063898 : SUCCESS in 17m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/915f9647de0f496d80d13e7f19933f4c : SUCCESS in 14m 56s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/76db699f55e04152bac19b3e04eb63e2 : SUCCESS in 1h 08m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/79df5a0d1c844a09a6e6c3a32a28d51e : SUCCESS in 17m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f700ad232c54d2b9a66b0b5e42f5795 : SUCCESS in 16m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d6758fc8bbe06c194cd987a4ffe6dd4cf3422efe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-22 06:17:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f65f53c81d24a150c3f7ae4275f98c11a6c1edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 06:18:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edece0b6cf932a4f5a4ff18e43923fad050a2ce8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 07:52:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e0fe63f7b72460dbd8688d5564dc87a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48688c06b000431abf4c72dfe7f687e2 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/86a6c25eb5fe4d13bf800c99de1dec1f : SUCCESS in 44m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/959a1222b8ea41d2a5a958bc3136b2ca : SUCCESS in 47m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb0639bef456407699fff09c7f906aaf : SUCCESS in 59m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ffbf6496031b49caa07ebacebd0c0786 : POST_FAILURE in 7m 04s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c0849d77fe6243b9a6de68dff408d902 : POST_FAILURE in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d6cec2c9e2841a3926d833130f41604 : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8028d86b4e84adc99a77d921a620cdf : SUCCESS in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f2c3acba25a45d4a0f70e121404dd26 : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a42ef8394cc44cca16663c57a11e557 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8b2adc24a364a1abcb30c19aa7a6025 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd562b1d8c9b4e54b9ec2496e416f296 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/329e4b57a64242a3b631a10a9dc7bf94 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6cab688481544ac58e1edcc15a1eda23 : SUCCESS in 12m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96004ce9a5b24ed98eba5f1660ff0059 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8c91cbcde4b43bb868417082b16a4b0 : SUCCESS in 14m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f2fb10c879847d3a564d228df825eed : SUCCESS in 58m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0656671c2cbc4a4ea424430517762c0e : SUCCESS in 58m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7173ad9f5434476bd7d7fbf562491fd : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/649f84db110e46eda8106ae5fc0aac07 : SUCCESS in 45m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dc5d33a40db4539b475bf4d620d1f53 : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c586fc63f01c46089759aca0d11e3402 : SUCCESS in 4m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb3b0b4c98642b493712d6b07f15d22 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"539fefb39d54554bccd683d8b60731bcb097a40a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-22 16:28:30.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48a28fae170b40c4d32ddcdfc7d04d2286741094","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2023-02-27 15:30:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"505d0feb1ba7cde9267d9ccdbfd9e88b382df47c","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2023-02-27 15:31:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31912b6ce788fd0e2ad165ebafe717f09a9d63ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:39:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461f4b2c7fbe49ca808f1ab265288eb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376f76bfaa654b358e9cd2ab71ef38fe : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/1cd86336b94e49ecae8dd8f435538d4d : SUCCESS in 45m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9324de83e5c40428c1632e21029c1d1 : SUCCESS in 42m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3063bc7a6802466da79f48316419cb0a : SUCCESS in 50m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9719f7892e1647fabdb2ed85fc921853 : RETRY_LIMIT in 6m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ee751b9cff14420aead3ee9381445d7 : RETRY_LIMIT in 6m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f82a67cc770e4e48b5c13e5f076dbdb2 : RETRY_LIMIT in 5m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a53c3bbaea45499587064a1645aae775 : FAILURE in 3m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c1fb772a98a74108ba9b0e2a27077266 : FAILURE in 4m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2bf7722df39b4d64907cde5da80164ea : SUCCESS in 16m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9361408957c74f4aabce3f8deb396a72 : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/87a3b27179f348b0ae4078bf6551aa3a : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c3fffe2ed2c4566a82c925bcdf861b0 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9cb95a7066ff4df6a68094038ac94ef3 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4f6f678953849cca3bd56d7ddbe8dd8 : SUCCESS in 16m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c8183045c8a740f8b6f2cc05cca8098a : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbb40e745d354c4b81ba36635d820b6e : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9afea7311e68489d839dbe2ea7ccf5c2 : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d15294b602b4dc38a4e0866315f1d8a : SUCCESS in 10m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2793821a7b194c868132ceb46a6516df : FAILURE in 9m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1db4cc2742214858897cc1af90593f3e : SUCCESS in 53m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1593d7ba4ec6420e938b377d2ed3edf9 : SUCCESS in 57m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bb79b05566043aebe0d1c4e60f02f81 : SUCCESS in 42m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4005285cc2064f01a0349a9a6dc6be6b : SUCCESS in 46m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03943a1152e3412db776a794c88dacef : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/158d019e6f7d40f28846b412f8dd2fc0 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5dbc468d9b1f4df8ae678a111a202657 : POST_FAILURE in 8m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da885c8eec364e9ca72ed7693c52c798 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fd9942c2d9a0e5bf6335e5cf86683ad065f78aa4","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2023-06-08 12:16:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56610a9ff0f09cb78ede3d9fd126bfa772ebab41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 13:31:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dffe639f5ce431aaa7ffdca76115ec2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5521c23c1c944664b95ea52cb625096f : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/f2f9780533c64ac2901ab22d93072cce : SUCCESS in 38m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9d77b0905ef4dd3aa70a82fca9cc62d : SUCCESS in 54m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff3f015ea5684a6eb3b39d0037de264e : SUCCESS in 1h 01m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82e1f9b6d516421883a48257668a2ae0 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/207cdb8d308944fc8c854337193376f3 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4df10a14bf24753ab2844367c66c3cc : SUCCESS in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6551c966a5ac4ef89af59326d78c6e1d : FAILURE in 3m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21a35737df944fc6a6c89975c71f1038 : FAILURE in 3m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b710e568ad84433d954f813042dbf044 : SUCCESS in 16m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c86cc930024445079bcba0d3eaf34d52 : SUCCESS in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8f6962009fb4acb938a59d72ccd6a7e : SUCCESS in 14m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d5166e29595416f9ab3f8421b5a2d1e : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86a24b414142407194cdec41adc3a9e9 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9a70cea1908448481b367a67ece56d3 : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a9454c7684794ea49dc00d70cd59c84e : SUCCESS in 11m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6aa77ece1ed742f3a438d8429c918e6e : SUCCESS in 14m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27610e32da994cfb8748a12909a98240 : SUCCESS in 14m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64b1571f13224824aa5a6575e2ed61aa : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef1c5c6e173f48a2b2005e087d4394f5 : FAILURE in 7m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6f1a6f80180a4e2e95c12f14aa11d755 : SUCCESS in 45m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/73dcb62461c84ccebc424fa2ac8e19d0 : SUCCESS in 1h 00m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dc670efe9a643c29c0849b039cfff5f : SUCCESS in 34m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7249a4a536104f428652682d263816d6 : SUCCESS in 44m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf9f42418d274e0f8ad063743c568b00 : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a95d29097240f5b569ed8ce0d51696 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/336f6e5663e54631987259070be049ef : FAILURE in 8m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5ff6623ffaa4bb991722ffe0284ce57 : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6e951b1f19a44bc70d63b041699b8b92a2a2ea05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 13:34:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b7994263e84d09874625bce98642dd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c491db0e16664ce781817fc5f5c36904 : SUCCESS in 14m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdc788cceda24a2a86e2d2ccf62617f7 : SUCCESS in 14m 24s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/cf81e12b5c73417798203e788b926ddf : SUCCESS in 1h 14m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de14a113c66245e18416da369f47d3b6 : SUCCESS in 25m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/54a0330a813743dd8efbdc122495de55 : SUCCESS in 22m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7f8df4768ea08dfb6e4609281e69091eb7e64f27","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-20 14:13:53.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/swift is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23a74ba0b071b2027604ded9dad8ee4b75142d43","revisions":{"23a74ba0b071b2027604ded9dad8ee4b75142d43":{"kind":"REWORK","_number":1,"created":"2022-09-19 09:44:28.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/78/858278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/858278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/858278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/858278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/858278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/858278/1"}}},"commit":{"parents":[{"commit":"3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2","subject":"CI: Add nslookup_target to FIPS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-08-16 12:42:46.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-09-19 09:46:40.000000000","tz":0},"subject":"Migrate CentOS Stream 8 FIPS job to CentOS Stream 9","message":"Migrate CentOS Stream 8 FIPS job to CentOS Stream 9\n\nChange-Id: I8b0bd7ee1c649294ac3795cba801c9e07bc3d51b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23a74ba0b071b2027604ded9dad8ee4b75142d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23a74ba0b071b2027604ded9dad8ee4b75142d43"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
