)]}'
{"id":"openstack%2Fswift~858942","triplet_id":"openstack%2Fswift~master~Ib8be30d3969b4b4808664c43e94db53d10e6ef4c","project":"openstack/swift","branch":"master","topic":"jh","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-01-21 02:39:03.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-01-18 10:26:33.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-10-06 18:44:47.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-20 20:01:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ib8be30d3969b4b4808664c43e94db53d10e6ef4c","subject":"swift_proxy: add memcache skip success/error stats for shard range.","status":"MERGED","created":"2022-09-22 14:27:32.000000000","updated":"2023-01-21 02:40:34.000000000","submitted":"2023-01-21 02:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":148,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858942-jh","meta_rev_id":"76362584334e8367273a7255f1d157d9aafeba05","_number":858942,"virtual_id_number":858942,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-21 02:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-20 20:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-20 20:01:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-22 15:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-22 15:43:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 17:35:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2022-09-26 04:49:07.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-09-26 04:49:07.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-09-29 15:56:29.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"500fc32abc9e5c302014fbf288f55c030085b093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-22 14:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"146712dd8532d12ccf5c5a8295588a6fc9584809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-22 14:28:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a21fc3ab11306e42973b7c0b46c4eedaaa07bd2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:40:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4cf290e4ce34d54b04d7491dd5792f0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e00e8c0d7fe46718b57182df7f57b2f : FAILURE in 32m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f281575d7ee84107b76c9bb3db5bd5c8 : FAILURE in 1h 00m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/853d4a406df54295bed8afbae8d2fe59 : SUCCESS in 33m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d783cc6181c746cfb02873eed4770109 : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e9fa858260089dc19e0f9c5268a40acc6b7e5bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:43:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf664714dad34faf94af8d666571b68a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f40edce5bd74297bfcadd25f35da4d6 : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/0d16782178bb4261ad09eba1fe886b6f : SUCCESS in 1h 08m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/709940040b1241c2b3b0675e13fd95b4 : SUCCESS in 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5987be7b6c474ddbb37df9511a7dd494 : SUCCESS in 53m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/935095fc2c2f4a41969e985153cf04bb : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77654bfb0acc4a6cb0113a69af56be52 : SUCCESS in 22m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/996e500b185447aba3c22b30cf309ad2 : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/052c840857de453dafd81b6676d251bd : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d3c47381bf174a51ab85f195135309f9 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd3720d5fa7b4116ad76a7ed8a11def8 : FAILURE in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6c90b145a844ff6a1c874250d684660 : FAILURE in 15m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d9d8fdc7a93d4610a20e5c2dde9885ba : FAILURE in 17m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b3a6e7242d14d92a9e125bd37b9ce48 : SUCCESS in 20m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3215179e7146411d9bb5bff17905e2e1 : SUCCESS in 19m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b3edccca3d34f64931dc7fe1c3632e4 : SUCCESS in 21m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/929f14e60663495d9f10adebe295d187 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/743083949cd94e70ae98496f77cddc13 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b30678971f084edb82b5b867aaf303c7 : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/028c80f7ac2b4fc9b7eb956b07c6e500 : SUCCESS in 9m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad1c5e661724457ca3922a23ee3bb9f0 : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6a08cf637c79414c9c2327e1a227142f : SUCCESS in 10m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e1190541e4247cd8dd17749386e941c : SUCCESS in 57m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6a84c1e14fc47adb34b8d8fb64a1385 : SUCCESS in 1h 04m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f3928ad5a1446ee86dc6878d5ec8c0a : SUCCESS in 46m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8c0ee516d55440cb1318e69d5e3057f : SUCCESS in 50m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b07eeab1b5c44509a0293a51731cac93 : FAILURE in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f985486f671471fbc8ca62a16a22932 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c875956df8b4d44a45d52d98cf0c8b8 : SUCCESS in 30m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99e822f78b394725be953295054742b3 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":2},{"id":"048fa581f50bbd1706b4b30ef9479e06bb2b1956","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-22 17:35:46.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"860e240eb619e391280d22915f4caabbc6789820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-22 20:20:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2179f44bf7d6eeb9b11b030995775771196c3342","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-22 20:24:26.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eb48c539502fdcb1bb25783584545ad3a02029ef","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-22 20:26:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d240e151f1c182eb3d5e7d5b166a8a21044945ef","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-22 20:30:08.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d327d6b26015ec21751523c95e3d28d69f0adb14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 21:03:29.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e69f13181354097bbd5a0dc8e1b56f2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35d2988d568d4a57a95fd7974da1bce7 : FAILURE in 35m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a58e7779f35a486985e930d8c3572875 : FAILURE in 35m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e041c49029dd46c4a7003d972655ab79 : SUCCESS in 36m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/01a6994051744b1f862613c42586958a : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":3},{"id":"338c4476aa1ac42812a7bb1b01a503f44ffadf57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 21:38:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06a3d8541ed849bca14d4fe2402c867a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d75cb3727b5420ea28ee332ea168116 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f993ee8db16741a49c29f8bdb9ee82da : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/654b2c3b40564d0e905c73086f1510fd : SUCCESS in 51m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5e7c41b79984f62bb20a09c9c6b5aaa : SUCCESS in 1h 00m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d98aef5bc3f54c32ade382b86e3492d2 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98a819d1a7fa426b9e73b760dee2622a : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5e5fe6249df468a910b120106edf4e5 : SUCCESS in 23m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eaf0a08869a346f1831f1f7903b3222e : SUCCESS in 11m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fbd99ec4a37c472e91086729cba1ca08 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14c5029150f049bab3a9c40823b3f4c7 : FAILURE in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/632997c8619647d9b40f7dd5d5693906 : FAILURE in 15m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e92360b96e254e3b9129b7f407c461d8 : FAILURE in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/52f4b1181a9d4613ace45db013967b53 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/219a647220914c738045bb950b492a3c : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d772cd5199d24eac8f4c3f6e706abe85 : SUCCESS in 18m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1992ab0420434acb90a8385c10e01f54 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4c73a998eb6b42e2aca124e4669652e5 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/573ce1cc27fa4a689f5321c5869debd4 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e37af9c336bd434b88506499113d7376 : SUCCESS in 8m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a65b01b2c50741a498c1cf5a3f166972 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6b5fd86d2e44bbda32ccb741db15149 : SUCCESS in 6m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7965f4ecfd214eadb5bac7280a5df441 : SUCCESS in 57m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/795940c1649a451baef9e9964e986ec2 : SUCCESS in 53m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a93de285a58e48a48941e2e872ad8aff : SUCCESS in 52m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04a8efe0804942a194bb9df87119e597 : SUCCESS in 47m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2c0ff6e0eb8443793344d99b9926140 : FAILURE in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6409693081485fb04a15672c4f0f40 : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f69d873c14d4c99a2ab6684b023bcd8 : SUCCESS in 23m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f98ad09aa2a4982affb3242d45aa921 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e985c2b46ccd834c0a8b4d190ad7270f18a0e997","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-22 22:56:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"62ddfeb6a12fe328f085a515cc12a78f58bd2a9e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-09-27 15:48:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c354dbf0352f585f33477ed30de4dfbdebbd1734","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-29 05:13:45.000000000","message":"Patch Set 3:\n\n(16 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4bae192fd1c4296c0a05a3ea151c260cc9e8a3a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-29 05:14:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ea18e7003123c09256fbadd5d37daaf68855cbf9","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-29 05:18:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dfb11cefbf8093fe1867423b5660449fd888284e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 06:08:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a730548d84248068453a732bcf829f7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9deacfd93e24d2f9a95421270adf1f5 : SUCCESS in 47m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a888b8fb1efb42b4ac36878d019d57ae : SUCCESS in 48m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dedbc2a8aa44c64ac874f92849e377f : SUCCESS in 43m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba5643f3140f4355ba88632a4d622bf3 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":4},{"id":"d33c5be253c096fc84a0a6d1d26c1dc61109af80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 06:19:49.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1beaefcb23c44903aca19bd634dda04d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064ae3718b464dac938c392d8b8ad22f : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/2b2aea45a47843dcab997f53adbb2664 : SUCCESS in 1h 01m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28e3ed3de0834eb68e97376451c95fd8 : SUCCESS in 52m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2584dc108c28403e80dc5f0ef6d7bf50 : SUCCESS in 38m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e16a7b5312f423293bfa81a0b94de11 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/128c63baf4084898bdd1fb6fc292c833 : SUCCESS in 22m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/220b8c4b90104ba087053c3d4113322b : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd406259a292499fbcaf3e1fad822350 : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/75d9e79066564666bfa99ea17208e166 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e17ba5d02e4047849d6bfbd7e1f84598 : SUCCESS in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e24a46433d1472c8ac40b065db33dbc : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3dc2e3caea3242e39593796f7a6363cc : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86c7455a02fa48c1b46da171f3144986 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b16a9d57cd6463d81083e4ca3edebff : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0be272a20b9b4b85ad96204c75e27ddf : SUCCESS in 23m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d1197d8e222845c391f367d75e99281e : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5ea0002cf84d43498dbb77046ee2aa88 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/21c7aa6eaf7e4118b21b8155579353c3 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83adb3be2ff1408cb346619667b66dc4 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/251d6c8fb5b84dd9904b2c071e5bf6ba : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bdc5d3115b1f49d0ac7923540dc8d0c5 : SUCCESS in 9m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7bc00ac81144dfbafbbb5e3ea30dd73 : SUCCESS in 56m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f252f1a33585438cb08f786008b5a911 : SUCCESS in 59m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32dc7a7cfb3942f2b6956c653718a0f0 : SUCCESS in 39m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a87f687defb41d38965e57aae637220 : SUCCESS in 46m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bcc1217f1de4eee84f0af3064602a6f : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba4ab40dc6494fe7afbe5752aa476a41 : FAILURE in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aaf6854c61ba49a2b24b30c6fffa210f : SUCCESS in 30m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eeb20879b5241dd872f8a6f7ceee8e3 : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":4},{"id":"43dd51f1a2a0131c66080c98d5bb4f2358a08283","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-29 15:56:29.000000000","message":"Patch Set 4: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"06c487b14dad13b82ffec23a2a4680884968c07a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-30 05:58:27.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"23c32a99d5c0aa00ea861f01b3cbe528f1da481e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-30 05:58:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"39fdd2c0439a995af1afb84f6e9531233bff977e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 06:32:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df6d66fd271b41a1872fc8241b0784e3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79118ad4628e4f1091f0d39a3c98dd16 : SUCCESS in 28m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2806d7d0941244fabc9dfbb08eaf6b18 : SUCCESS in 28m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21febc7c3448499686f7599dd2496373 : SUCCESS in 28m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8aa5030f24440888e738af6a95a568a : SUCCESS in 27m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ec93b30dc17348754cdb3afe2092ca2fab066efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 07:00:10.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b4ad475848847f8ac90ea3a3b584b24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e044a6b998463caa735356fa914ee5 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/28bf9b7ce3d4405d8cf878e2a6afc12e : SUCCESS in 52m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a33be2a63b5e433d90a791ffc47a8898 : SUCCESS in 46m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ff6a6524287411da1428407c225e6b0 : SUCCESS in 49m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5bdb3ec7ef841549b0d295c6d25cc18 : FAILURE in 23m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3b27f6d30b14be3a38494ad37ac01be : FAILURE in 22m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/795d0c4a794a4ebab1e32dd3f53de0d8 : FAILURE in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1064a135b2394ebd8c60427290b9b7dd : SUCCESS in 9m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d2241b3c74074a9483481910cd9627ce : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03275fe50217413aa71802c5f192f3aa : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d017420ebb764e53990c0f01bd971f63 : SUCCESS in 19m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cf50e92c785c451cb1d7c2b1ac8aaf8c : SUCCESS in 19m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/69af214f5192485cbcc07f2ca73cfaa8 : SUCCESS in 20m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/230b734f80634724bdbd8644672bc884 : SUCCESS in 21m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7bb443b3482e44b482bcfd87b04f6632 : SUCCESS in 22m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4f84baeac4e24352a2b975c6862d3b7e : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/afb269c053fd4c6f905dab6f72fdd803 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c395012ae8694750899940cff24b42f5 : SUCCESS in 21m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b66755b68df49e68b69413b47f55fa7 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4f317e466ed4aa9bce634cacccd10cc : SUCCESS in 21m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b87cddca3da495abd9f185eb038ee25 : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e16bf4fd22c546538cbdb3bb52e32bc2 : SUCCESS in 50m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad832a4bfabe4c42857c07e284d8d722 : SUCCESS in 1h 00m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd428883b39f4200b8a8e0bbf7578831 : SUCCESS in 33m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/860e194d4936442ca7e861d711e5a2e4 : SUCCESS in 34m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60bb074a812b464a9d4cdf495deca76a : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d71f22e67af4a3a9589a788dd0108d6 : FAILURE in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8c9cd2c70ec4fa5930309cf4cf5eef9 : SUCCESS in 25m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba8a3507deab4a2dac8f2abcd5484d12 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":5},{"id":"b5dca4622d21940415ef46b7f197ec3e6373b428","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-09-30 15:14:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5f03f83d6a716aaef7c695b1dc1c1434ead4daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:16:38.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40ef7beaf9dc4d948967bf571a809f74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca0f39933754a8da0749b4d7c09428b : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/084574617b034abd8471c3b41b36809d : SUCCESS in 51m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de87a9b0c2c34151a69a2fbdb199b136 : SUCCESS in 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/250f34632d77454ea39074cab5f04bb6 : SUCCESS in 47m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b948b204eccb4dd899ee8225ad9d2e2e : FAILURE in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c550517088124418b7122c9d9df983f0 : FAILURE in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba71402979334157a41f14e4b20d639e : FAILURE in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4556e22ede824aee87056af6e6455d2b : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0a5ebfd0d124d93a850609034e83427 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35539fa13ef741ff94e22df08f25d3cc : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4a93dff665b43bd8449273d631184fc : SUCCESS in 14m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f44eb73d256f4ab696897f5fc04a1b7b : SUCCESS in 13m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/465ef52b8cfd4901a313510c35366735 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/847827a9181b43ddb78c38ebb4ed4efc : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a31aeed896904358accbecafdf40b6f1 : SUCCESS in 18m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ccd3a157b5ca4b08883b594049dbac65 : SUCCESS in 13m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77ad8065b502456fb865897c669b756a : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74bfa98fa91d465c8dceb62e72467c71 : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b508167d457a4f9d949902294c5792f5 : SUCCESS in 8m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e055c211eb54bc48a3a1c0a7505546e : FAILURE in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5dd4957cda2a44338768a3240073f953 : SUCCESS in 9m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35eb6fdf90fd4e51abe99b808d2c702e : SUCCESS in 49m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ae18de928a4421581bc23e861c5347a : SUCCESS in 59m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5595295ed8a4ba789eba5a22b724d0d : SUCCESS in 45m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c33ab6c7650645c583be8dc867f3f1b8 : SUCCESS in 34m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09b863e9c51c45babdfdb5a165beb62f : SUCCESS in 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dccdb27e386d478c9dec764637ef7da9 : FAILURE in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34371449ec534d528c5ccb636702a755 : SUCCESS in 29m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc9a3e776cc942d5b55151155448d523 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":5},{"id":"0e5ebb1476bbe1ab0b2bb1295eb80b19531d777d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:47:51.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab5bb7539f94b8ea9423c93956ae72d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56fa9acffa824c0f874bd9c32abc2a82 : SUCCESS in 1h 23m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2301a382ecf41ba8e1e769816f25494 : SUCCESS in 1h 18m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/224f3b62a12f487aa1c823ce187c3534 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d617ee7b9d324939959d1745ee7f6d0a : TIMED_OUT in 1h 31m 01s","accounts_in_message":[],"_revision_number":5},{"id":"f8d8d4fad7191e56f907baef32ab350bded7d6de","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-30 17:03:44.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6df050de5e82fc2d174daf5902cb9ac032a24c2e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-03 04:40:23.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3e46869ef33607c4bcd3699f39cd51d9e3512059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-03 04:40:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b1a0588cc13f8f143153daad3c5ad6f2a00f8fe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-03 04:42:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"98b98a6ac6568932d2d696d01f9e4a35efa7cbe1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 05:46:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc0b84c9c7ac45c8881d81f7caf946ea\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8ac257f82724b5fb8329d2a42600faa : SUCCESS in 29m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7b4d5a6e77e450ba9ecee58adfe60ec : SUCCESS in 53m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/710281c1197a4b028bd2f125897188ff : SUCCESS in 30m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/29c0eb30e7fe4b76a85c7d16b06f35be : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":7},{"id":"86f2f4d967c8e5d0d1cfdc7e008d221ffcdb537e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 05:47:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2cfe757a9754005b27350e0898d5a60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1300a10fee4cd8b7bbc4baef19ba97 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/16bd6eeb8b2e45e9851d1fffe8a4443c : SUCCESS in 39m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/454dfb3a76ca414a9466ef4bf4cea5ae : SUCCESS in 32m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a61b5a46e8934809a07b55cc393315ab : SUCCESS in 36m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19a402c8498a442cb9cd851027d6b502 : FAILURE in 23m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7510e806c7c84e708094e56cde25b737 : FAILURE in 15m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3338c841c0aa40eaa44903040dd479a1 : FAILURE in 24m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7be353071f8d407aadce485bfdd15852 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/262ef13d20474769b466ed4577c2192f : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea35724fea374b8091e4fcce5b0b6059 : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd1b212599245dda908becdf2866a42 : SUCCESS in 16m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c946c0d432c465998b844c45447b3b7 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e5f34da4e4b4dc88741296fd551fe8f : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f119db402fed4ce69884a519d34f6e4b : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40abe8354f2f46b19034517236d86636 : SUCCESS in 14m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/57c12bd8622945bc994a909a6d271b24 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9c9885fa73c487eb1ab9fad9fbdf298 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48d6a88b0fff4fbdad5b150a43976bc9 : SUCCESS in 13m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0a4e04cb9795436694ebb344fd86452d : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edc9d775fc974dc78be4038ceb3ffd66 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d42a718023864505a20e60b2799e813e : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b805655e8fc49aca882ce5a7c7e3e38 : SUCCESS in 1h 03m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e580d0081b34309989e7c8e556190b3 : SUCCESS in 52m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd53ea0a677b429e9f0da6c49314c15a : SUCCESS in 30m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7235815ea27b43f0b0aef18407d37265 : SUCCESS in 47m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4edc460daf6848adade8be4117cc724b : SUCCESS in 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b9b7bbdaa7463cb50098571e028fd9 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d610baa270be44f6bdedb7fada54d498 : SUCCESS in 21m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/803400b6c0f841379b473da558655c3f : SUCCESS in 42m 46s","accounts_in_message":[],"_revision_number":7},{"id":"dd3e3f7ee6eec580728455136b17a5904a1011e8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-04 15:28:56.000000000","message":"Patch Set 7:\n\n(10 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5aa438cadc642356dd96a3d76f0da5598783d120","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-05 23:40:10.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8dace965a4536466e65133440a09ff503c75804d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-05 23:42:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cfb7b4801d34a10b1f7fc8de15f7d1e9eec599bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 00:30:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35c185ca2f44599ad54e871fa2452d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4aaf46526744fd992cc8abfbb64b2cf : SUCCESS in 44m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/475927f6d5e34470aa2f63e4ab62d889 : SUCCESS in 39m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe4573bbdf8f422299a6453316398829 : SUCCESS in 42m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3104787bae24cb8a922cccaf1775f7e : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":8},{"id":"1e5ce552971c62225c2919759c542fc300959e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 00:56:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dca1e0e65af24fe6ae22b68df7498a9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673f1fa1588c4372add4a6e541f84603 : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/22cb6072ce9e40349db2634752b0ebf4 : SUCCESS in 53m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9dcc1ffdde9f43328b54a2f74714a118 : SUCCESS in 41m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fde30af3f4f4ed1825b56f3de23f6b0 : SUCCESS in 51m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fae9f7c543084efe966bfff638a4d7d4 : FAILURE in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e85b05d53d240519b65c8b52fe632a2 : FAILURE in 22m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a297c9aeccf4c12aa0bb6250993dd30 : FAILURE in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad0acfd72070489bbd3520c2c2422d55 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f4679f59a3fe41cb8cc71948eedb0d8b : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cca195e20f414fb1b5773740d4329d75 : SUCCESS in 15m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea4df61fb8314aaf8fbfdf32fd553828 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3051f14cf68c47d190e4603349ef41b3 : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ef297b3e86364b0e8d99b60783455ce7 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e097bee44b0b42a4a22718a0c95e185b : SUCCESS in 20m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ea61f2190294b7992924067a36430ad : SUCCESS in 21m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9dd5311590434f359ec965314f5c51b2 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/874bdc1b93a048a3a24c93bbbe0f2033 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2a41b0029a0741c193f1e884077d0cae : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c847919403db4703bd28bc5c708e8f4e : SUCCESS in 9m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/194b62c9c8844740b64c228d5ab024cf : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1509fa03ca414ff8b53717902ca1f53b : SUCCESS in 9m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fe2e6961a6fa43b8b509838e6d22c560 : SUCCESS in 51m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b948419f3ea4db09930d3079939cbab : SUCCESS in 1h 05m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9bfeef545b41417aa927022754390e28 : SUCCESS in 38m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/132fa520cff9453590dcc07bac95cdb4 : SUCCESS in 36m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e636aa235efe457286712d8654f4be37 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c142b835f393407e9e51da6bc370cdd1 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d89082f95be24f92ad6e1350dc7859ba : SUCCESS in 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c4bd1efb5694ff69b808b813d77e774 : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":8},{"id":"39eb77bc1cb19ab43f9e0ae30c07941218093432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-06 17:36:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"500f9a0063d05cf7fd88c9d378daefeb20ddd4c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 18:24:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5041cb46fd3649fdad4cafca4e6f6615\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae3e1d9704184fd9a551bf4134f0387c : SUCCESS in 37m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60329167b32c43a0bf706b84181ba323 : SUCCESS in 37m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/65184f187d514d0b825b292e5ddcc8ec : SUCCESS in 38m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/20604c8f4e8e41ed9d0eb2e9f155c7b8 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":9},{"id":"4364e8f5f1f93192035d40ec560dedcd6072a54b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 18:42:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b85e4d73fbd541d9bd65debc17a44356\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be04f904a69d486d8309ffb3e5425804 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/aafec3e545874b26874bddbc0707adc0 : SUCCESS in 59m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d26a37b4a814f81805c2b99ebeeb95a : FAILURE in 29m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7399140a967240c8ba4734b61e0acee6 : SUCCESS in 56m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d674104acbe476aa83da6e5f40418bf : FAILURE in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc66ecf538cb4c35a7b932ea27eabaaa : FAILURE in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a35e8ca686994077a5a9007bcd9197df : FAILURE in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/414b4584a2134961850babd0f240e392 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f72d585fc1324426ab6cfcb78f28479d : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2adcea032b264ba1ba4705b27840f1b5 : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4f44cbe51a140fa87f96bb6fec9dde3 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/190dea43daef480c9bca452aef9c0386 : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e9e495c6e18d40d88c526b000cdbc418 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2d05e10205d44989a9c0878e7f249093 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e98a5e40d13b44e78f94362f52eda718 : SUCCESS in 15m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ff80e5075b7742b290549d6a4d6baf82 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bfd103ae08ee484985cbd703a6ef2396 : SUCCESS in 13m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/36bf89a66a0d45dabf399287c956c490 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9ceae87ca034189948212ae209fd67a : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/80c78f4588754edea54d273a38d5642c : SUCCESS in 14m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1df0a8cd478441a495ca15b35fef9186 : SUCCESS in 6m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2389c6f27f4f404fac5dc676d4b0490a : SUCCESS in 49m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4fd48051ca36404b8eea6664aceffa11 : SUCCESS in 52m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/589c8054cac8459682be524ad1d2f098 : SUCCESS in 42m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7e4fd7b7daa4ef1841bb9aa9b501778 : SUCCESS in 42m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6de2a0ef9054f1d9535b332a8f40264 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926c456e2efc45d2b4d242300fd27257 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36afaf24c4c247578c8c8704c1b9d338 : RETRY_LIMIT in 6m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3f8f7df4961431585b8c37abc1b34d2 : FAILURE in 16m 54s","accounts_in_message":[],"_revision_number":9},{"id":"939aec118f59a1a83152b24fbc3d43fa26d17d9b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-06 18:44:47.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"dd05a73cd01f7f316948d7fc87a55ad00fd5b97e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-02 10:22:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8beb924f3d44588a489e08e84ff87da1b7965df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-02 18:07:15.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"5b19b38aae0962469fc81cf18211b3a7bf172341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 18:41:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c00361cb0bf547938a77e88eb07dae87\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f7a14624f3b40dbadd6ca3dcde207a4 : SUCCESS in 27m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/546cfea3f95c49a1be7e83e114e5d711 : SUCCESS in 27m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6fdf7fd780f4e1ab9e4068b34d58d8d : SUCCESS in 27m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e23ffa9f49804097ba83e939fcc0dea0 : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":10},{"id":"952e7dfbd25f7c1e7218303de760674a03b93905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 19:10:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb259686ea74338b0b1e548db90159f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4308f89caab340e1bdef4c4055a5255a : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/4a3d1840d94c4fab998dbd1d016c3bf8 : SUCCESS in 48m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59da237c70c14799aaf56ef7d3816213 : SUCCESS in 39m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/530e4ea8df7a4b0d97f00d247c50bd49 : SUCCESS in 45m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0ab0e2d4a6d4e30b40d0483f46b4361 : FAILURE in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63d65a6621974d7e8317c10c975919f9 : FAILURE in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59483d4f77794eed9a9a0d7c79543399 : FAILURE in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2bbdef63df4476187c61775f489f433 : SUCCESS in 7m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/efe7f9a5332b4c6eaced82e984fcd209 : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f2432034f5c64a9798bed571a94e2320 : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b25dd838bda41498967f36184c5b328 : SUCCESS in 15m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9570fb814a5d4762b09687084a319e62 : SUCCESS in 10m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6b5ae5c1ace46b9ac9d591e8cac8f7b : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/428697fe7bbf45f68c0b6318c5547f54 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b186ed86102d473fa3d729eeec8d557d : SUCCESS in 13m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/733873fbbc464ef3be9e5626e2bd52b2 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b311f434daa549ba9932aded312a1999 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7c1431b547374f529a2df77ad9119c53 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c059a30813243848cdde48b6ffa1a58 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b452af6ac3c4c5c8d01c69dfc9a9ae6 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07ef9d2d2f754270afc8166994d9c8c6 : SUCCESS in 7m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a0bbc0b0ca18405ca912a916fc29824f : SUCCESS in 57m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc21c129ae174a72a04233413798808b : SUCCESS in 57m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cce1745ef6184d23aab0d81f33d902ba : SUCCESS in 52m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97409371e8e94ead80fd82fcf89662b8 : SUCCESS in 33m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd2bde7b765046d88cbfb6d3d12a51ec : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c2d038f15a4b408645ac05590e7959 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab1e1457e84a4a8f8563c468907c3017 : SUCCESS in 19m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ac6172eabc44e759294eab1ffdab38e : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":10},{"id":"4ad2d56f162df556c915c8c78bc8b23a107057e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-30 10:49:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a8aef5c0c0589eb82e5569e33fd02937008d896b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-30 10:52:38.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3104fe02f3bc914da200c1b0b8553c9a096bedcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 12:18:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1d776f78bbb4c19bafbd876f658a78f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af8fdaff3764fe0b53e935e265f185d : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/67f0ee900af5404d8e34554abd99aa18 : SUCCESS in 43m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6445d45b4bf4422aaa89e42a3f98557d : SUCCESS in 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09ca38058fcb4ddab3f047f72637a3cb : SUCCESS in 47m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8c6e3b02e3e4dc2928d35422c1bdffc : FAILURE in 16m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c2e87782d404b5ca5d8bbf8df33d316 : FAILURE in 17m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6397078a2ff44defb2e68ca60314cda2 : FAILURE in 20m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cef7a50489cf49f2a91128eecc03eaf0 : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/76790d8ada2645f1aa10d5a9316b6fdb : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/375f9e0a57ac4a66afd28c8f77ea4c9d : SUCCESS in 14m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2af2962771cf4bd2994ef5ba16ec2ab1 : SUCCESS in 18m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/409145253fa74bbcb7f905a39af98178 : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae60d5a10438485d971eeb731251ad6b : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d50bdaa9b3ef467dba35fbfef6f1a21a : SUCCESS in 15m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c102be9873b44c39c2fe0d9042a7547 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ad668a28c8241aea915f66493ec55a6 : SUCCESS in 12m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/099b70ff78454341937462d1f0b42d67 : SUCCESS in 13m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/528ff56abb224d518e3c3c791db62f66 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca8d0094ed2948fb942e665a010dc388 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91ae868a4b6c47debd84172b144679ff : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5124371a11f542a78a5f3c151a2bf199 : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c033091b983f4defbbf4dc553550d924 : FAILURE in 53m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/71cadec6d08347cb92215822bb68c1d7 : SUCCESS in 56m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf17806b1fd74a229e1f58c5dacd8a41 : SUCCESS in 41m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41161bc3526c42d88c79a8e86363121f : SUCCESS in 35m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa722ef9bdf54f2bbd2e0d6068d9fb67 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26eec4636b0c47f887ea3850d1f2d25d : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f6a61fa708c4a47956d302850acbdc0 : FAILURE in 23m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f6288d6e536440d9ec901ed50c8cd05 : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":11},{"id":"4e1f5b1fb5b57456cb4add7fedeff0a178325f76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 12:21:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8099abd56a1d411988c7636a48c4a6e1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bc184b0aeab46d9a7bc5cd787382a2e : SUCCESS in 1h 03m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33750462ab6c466086283b747f47016b : SUCCESS in 1h 28m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d97d70b088724affbc79c5a352cd84f4 : SUCCESS in 59m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9a3993c5e9b4b45a5162e522f748d70 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":11},{"id":"1188bf205c88288c01153ae700ace9a9b47d2caf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-02 22:27:26.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"f16d21250241cd6a7572b90860cd6d9a0839bdc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 23:19:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92288c41efe420192f85feb7f89d93a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fde97127854d7f8d86340c97105020 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/cd022a9ee61a452190c101834e0d44d0 : SUCCESS in 43m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ea31a7d36e94566a2d47ba0960c7c1b : SUCCESS in 32m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d337befb6f04d0c90c5ab93f8c3bec3 : SUCCESS in 37m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/531d01a21a1c45e28012a8c8dd436965 : FAILURE in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/280cf2bbc0b74a8489cd0f18ab814379 : FAILURE in 17m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b46e4e4d0f4c4d2099b1dba4f6efbd36 : FAILURE in 18m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a0b82a63d11447b9952012461863e3f : SUCCESS in 7m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9d12852660d74489b102eb4bead77308 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ef021d08c48434fa688716a3074a7b0 : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5222aba15458470f98805da0fe4ccfc2 : SUCCESS in 15m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9718df67d7a24966aba922255ac203f3 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35723602c3b84af0afa30bf25b35daed : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dcbc6e23d7f045be9006d7eafddfda33 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02c177a8f116463b89405c7d93b987cc : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/093d63bda1804155bf0cbb2213b00bfc : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c9b2ea778ce04eb79821b7af32cad02c : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48d171dac7fc455197b6eb4a52d65a33 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdd10b18dc6b4cf5bf13bd947d3ebe6e : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb2c8d4fe0e0448f93d0276dfdca17f5 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9c2963f96484d61a1256a6c2ec8a951 : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a500191dc4d4563bf2a298902056c4e : SUCCESS in 44m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/46ed594d56244d858fd9672bb28cacf3 : SUCCESS in 50m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef485e670bac44f7a5bfeafced0da4b2 : SUCCESS in 40m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8dfcd56f83ab487b9cbd392751f58b1d : SUCCESS in 32m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73f2e6ddce0f4a5eb3c373ac11cc43da : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045ccd158c104994ba9b4e760837d17b : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e3005914d3f4e29ad0fe25e36b1cc60 : FAILURE in 22m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/372734bd14cf4582b2a3b99987e1ed3f : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":12},{"id":"6386c87bb23e7cbbe740a7d68c3008ebdcab02e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 23:24:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b92942b2c014ede8e4a6ab4ffccc292\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f43ab4128c644c508169b8533eeab56b : SUCCESS in 51m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f366a2edd91949a48bc4196af734cad3 : SUCCESS in 54m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e3a7ffd079f4efb878bdca79b8c859a : SUCCESS in 46m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c0a92faaeea4eb58f250cf850a0c572 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":12},{"id":"4186bdafb080bb57b5d8340facc50c009d536783","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-13 12:21:43.000000000","message":"Patch Set 12: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":12},{"id":"cf1719307e3ea6f79003b80d6c112f767e9247ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-13 19:04:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"908f0d850c9ea77e9ad4cedc0bd9b2fe0a9753e0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-13 19:05:06.000000000","message":"Patch Set 12:\n\n(13 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ea3fb36421af9b9adbf6dba8b0cebc1151f4b919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 20:32:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a4f00bf0d804f3b9350b34eefa50855\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ae0c81768940c591da293a342c5f58 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/42c711b5ca5d4c8195e3acac131faf0a : SUCCESS in 47m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63078eb7568e40eebf500ef1b11d4fed : SUCCESS in 55m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccd3f42ebbd94429a9799ec56a17bcc1 : SUCCESS in 56m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/151ab2a6063e408c88067169b967ae3b : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d451dd66d0b9442e93cdf979b218ac76 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccf51f9ed4d04ad68f2c44cebed5a21d : SUCCESS in 17m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/605106b271604986ba4a5ebcf69f2f1f : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9a5aae26f6074e269edf12bef499eeb7 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/480ec92645ff416a944ff34b9fac2835 : SUCCESS in 14m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c67582da362413589c8959988c44b7f : SUCCESS in 15m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/609b7171fd38463b8fb44a8878b5d762 : SUCCESS in 16m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/451cbfff2e794360820b79450a94a68f : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e4d69dfab5aa4e03b706c8d9d27210db : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/93ff661fd79b421fac134c90b222bf27 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/322184a5a0434d8383a4558161821695 : SUCCESS in 18m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a1eafd48eea04fdea49fafefcc7fc86b : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/30f422369d974e3ab0629918b36cdac0 : SUCCESS in 21m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0dbf7ad91122467aa44c04971d2a46bb : SUCCESS in 16m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/968b48caf3c04eeca7d45e948c283146 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bfd2bb714c3441ca0dcfffc060a59c4 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/26827b16407041018e89ff9894051e77 : SUCCESS in 6m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b0d91a7f8e86445e959db0dfd73c2f6b : SUCCESS in 1h 24m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/53e2398da49841048fd9c87c9fb8ae35 : SUCCESS in 1h 20m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb0c9fb21fc14221a4e11005a13d9df1 : SUCCESS in 53m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6202bc3c32e94b97a5c3ec550e7601ec : SUCCESS in 38m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c64deada17ac40f4ad278b9efccee0f0 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54fa413e47c941ae8263f67629f08f75 : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/828870fc908c404699eed745c90dd56e : FAILURE in 18m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fdb9acf590f42c0b691bc252a65f8e8 : SUCCESS in 45m 06s","accounts_in_message":[],"_revision_number":13},{"id":"e32662affe62df0efb9253fc150a08889a3c68e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 22:40:45.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd613341e258434aa4c5eac17f1543f5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63084500e4f041e0a9a6d852bc8d3351 : TIMED_OUT in 1h 31m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba2ed84736874919a2403d09a7ea8bae : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/056f0f71b31b448abbf2760abc0fe4b1 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ec258dceecf4c9587197b3c803596b2 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":13},{"id":"76a84456f89c1404739a0550608f99e50227fe38","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-14 01:25:23.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8b5f2820fb1c0d89a864f9831229ea912bf073c2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-16 11:45:38.000000000","message":"Patch Set 13: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":13},{"id":"2ffd89a920b7fd89be8e1e5375aecd5ab2b71ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-17 03:45:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"27b460cc6be0b6ff84a1f08f2bdd944489afbba2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-17 03:47:44.000000000","message":"Patch Set 14:\n\n(6 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e4b3bf8d749e7a54b5573cd24ac0bc67c9a7aa72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 04:51:17.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922949975c1a4abf8a27a008119d9b7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a92adcf1c12947f0bafbd8130ad279f2 : SUCCESS in 1h 03m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63d2b58e830e40e1aef38bf4e0f3516a : SUCCESS in 1h 00m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/06b3246157484e5dac88edbbe7809959 : SUCCESS in 57m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/57b3583c071c48758d706d4f28754f89 : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":14},{"id":"ad26d6559a610227b08924ce351afb217f16ca64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 05:17:56.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/400fa8b68ed24e37a391d9a3d24005a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d52bf06c72a49cd92db65914d964df0 : FAILURE in 5m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/13b1873da83a4002b45a8c6b09ad91f5 : SUCCESS in 54m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f8e531b2f3149d58255bb748af578c9 : SUCCESS in 40m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a980cc4296c4bb086b5efd1fa928c94 : SUCCESS in 43m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4de69bc8d2ae4f3083f23e6bd1c7881c : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ca99767da364843a5bd1d0761e2e3ba : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0406094bb0a74af9a53e446a985e1895 : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be334f8e322149e7bc26ee4899c090dc : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8bc8e7ea15f48ed9f9122e39ecfde6a : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/705f752d7edb46c49e2b1fc2f57de296 : SUCCESS in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4465c422c43141a88a2aef62b1eec11a : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14aa1f0f2c9a4d7d8aadc1e2f5780d52 : SUCCESS in 13m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4ca26fca845640ee94dca726ccd3d8b8 : SUCCESS in 14m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6955002670ed4a08bb62d8bf723428d8 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8bb8c6f7e4ec43e3a047c67868793ed0 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8b20f7bcf0684cffaa2ea39a6ada41dc : SUCCESS in 15m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e35d67ef22f437f81c5d023baff3523 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ca3ea883a32e40a6912e39a797eaa8e3 : SUCCESS in 11m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/42d51a8d43f145e98c86ce89f05791d0 : SUCCESS in 13m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e5051e15ea34fcfad05ee19a600624f : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/992660345c4f4fe2a4d0c25708f9b3a8 : SUCCESS in 14m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/77e2ad42f9ce40f9b9d2e5fc0f9dd1e1 : SUCCESS in 9m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e0c541889c1545d3bcd7949fb5cc10aa : SUCCESS in 1h 12m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3de722aa65fe4d50a259bdf36c15755d : SUCCESS in 1h 16m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb994c92fbe547419590848c00dccc30 : SUCCESS in 48m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60792ef21b8146d882470b78bca30194 : SUCCESS in 54m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55e8a560c73a46fe9a07d255ad67eddf : SUCCESS in 20m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/633e84b2195a4272a1ff0a7dd520731a : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/658ab36377ae4994866dda33e7fa9e2a : FAILURE in 20m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37947534e1864d689db5a36071482727 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":14},{"id":"5a9d15094e2943fb8bc7e061b953c2c06a67d5c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-17 18:44:25.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"01d9a10d7269e0ab2659d345b62f08ed1db4ba0e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-17 18:46:03.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b68bf33520a6f1c0d3cf84f610ec21c18412f752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 20:15:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30964313a8734ccba5c5641c1d5f80d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191353923c7c4e5aaa4a34a7a9e08e8f : FAILURE in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/684c2be25f074e43bdd78cde0f544ca8 : SUCCESS in 57m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/253351ea79cf4090ae8c90c2dde3a931 : SUCCESS in 57m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19aadcfe348745b28afe10ba5c89987a : SUCCESS in 1h 01m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ff1490b0e854466bcd8fb4122e44796 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b987f84c613458082e7bbd3344471a1 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed586a347e614d00b4b3c8924e72fd10 : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f42c75eae6bf4a46bbd99d7b5b91ab08 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e715bb8d3c914720bb0248c76290e85b : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7fa99cbf942d49f589acb3e7ccb30446 : SUCCESS in 14m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a603834d8b1d46e9b4b839f06bf3961d : SUCCESS in 18m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2fcc0928d7b445099f2ced5b7b9fca9f : SUCCESS in 12m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97518ae651fb4aab96c5594164a173e1 : SUCCESS in 17m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d472d54c49849f78648c9dc8b45dfbe : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9271b4cecf984871b94cece4ea3da67a : SUCCESS in 20m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d9ffc74e70e43beaa7a8aee5f51432b : SUCCESS in 16m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb955528443a44588402408b02a5d239 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5bb0f96fe33e484f9e73df4f22bd8545 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/66f66ec8e167440a8a7b0fbf771f0f71 : SUCCESS in 31m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa1a2f1c9ba24417935335136f020b7e : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/30b937bc0d37451d9ffd17a1b22b02c9 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13206d803e6a4b0c8ef5db6bef113d90 : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/274ca7265f0548b6960af0b63ad5c337 : FAILURE in 1h 16m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5f53d66fffe476aa1d7ec11d1e4eec3 : SUCCESS in 1h 26m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d64474e77aa4c849d35d4ab010b757a : SUCCESS in 47m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61fba74f0a2d4d1f8d620e2f43b09d3e : SUCCESS in 46m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88230617a68d4d0f87e1ccf2717aca87 : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b712ef35094641b320143bcb12b30a : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6454a24c9dc49939b3ad1817f75f3fb : FAILURE in 25m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17ab0cdc2fb148d0a0638127130ae5dc : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":15},{"id":"3cb11ee87776b2430c6ffa182a1f5ddc85417948","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-17 22:09:23.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"98f886f48cbc79de131ad25f1d2ef65c92b005ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 22:26:45.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/872c44a3b23f4c9a8ef209c12407bb99\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ed78bd9e95f460cb57052ed669229c2 : SUCCESS in 59m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c76cd02825941d0b8ff1a3718651a6d : SUCCESS in 1h 18m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eddc59c5b9648a2bb46027b10405d06 : SUCCESS in 1h 26m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/56585a978c1f4e1280139869c65267e0 : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":15},{"id":"6c99cd4ae6c7c86a80fca4b330a4ce76cdf7ac96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-17 22:28:42.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"056720f6a1e687528d3fd58bf96ecd715b9fa42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:13:45.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fda2eb34d064e7d85b4fc5e315545ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce144771960a496eb6590621aaff9b44 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/d2afb32b6d184af897cf10015dafd065 : SUCCESS in 56m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f71869b235640b8a2136441dbe5ef5a : SUCCESS in 53m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13cadb197a2646b1827a9ebd5072d7f0 : SUCCESS in 57m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c940f489a22f490a93ce2079c314f3b1 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/049ebccc82264e8aa4587ea9cf380af9 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bab33ef95358480ab4b9d5aee0fe9890 : SUCCESS in 17m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a92990340d8a4d5d87b0129a39bcf5a0 : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/841a535032e34ad29074e5fed9fdc07a : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28c8e9556eeb4e9488667483343dd374 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ae0ecf882de4283833c99968fd93c4e : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7192068c840d4c32abc5ddd8cb2b3133 : SUCCESS in 13m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d557f9b8dd29497ca19023c76395a7ae : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a3a9c4d669b43ed8534ebece5fe7473 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/102fb8c322464f29b6b221fc4ee1ccbc : SUCCESS in 15m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28480994a15d483fa7849e07c3767ea0 : SUCCESS in 15m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dc341277aece42099fc0074fb34bb92e : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/490e711aa29e4ed1bdd8f4e08f84652d : SUCCESS in 15m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3926205dbcd442dd813e7fe16fdacc40 : SUCCESS in 13m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f39b73ec83a647219d3a7611c402f3c3 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b559ac6142624f61a78d520a48fa55c6 : SUCCESS in 14m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/09c131ed422b494e8ffdfe3667e54f46 : SUCCESS in 9m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6de83fc1fa674ea18453a08e6e0f0857 : SUCCESS in 1h 07m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae218d534e574d09b551b600d451e337 : SUCCESS in 1h 34m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae917b81e73a485abd7f6563d9e2753b : SUCCESS in 39m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04f35fef85154cdb9b6e3f6a08047cc4 : SUCCESS in 1h 01m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/666ddccf9cee4fe99e4285f55acfab92 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b51f2d5b0f24e6bbeb925a8f9f9ccde : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f073270d430846429b32c142b11937cd : FAILURE in 23m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d208c49d42ee4d5f90a28f723c6ad84d : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":16},{"id":"dbefc9ac0a1553fbf0214bd708a436b71b8a267e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:37:51.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b9992d4dea43b1a2ca6c02146e0d5a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37bfbf12870a42d38c1c52c6451f5623 : SUCCESS in 1h 08m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a0246056b5348b38258bdf9f3f22c05 : SUCCESS in 1h 10m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbf65e5c597548f9aa1c94c34eb8acb1 : SUCCESS in 1h 06m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f543733ea25438bb663dea5761daf56 : TIMED_OUT in 1h 31m 12s","accounts_in_message":[],"_revision_number":16},{"id":"a0a44d1fa27c442940fc7f278432b8477496b033","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-18 10:26:33.000000000","message":"Patch Set 16: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"14c1a57985d8130a19b970c672a1aac8b8863042","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 01:18:24.000000000","message":"Patch Set 16: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"d74831a3a7f6bbad390e0dc0e32268f55e8a13cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-20 16:37:16.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"925b17dde599eebec354f14c57d340ecd45e1771","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-20 16:38:04.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"521e9e2622bd56ab675ebb2668a006b282979683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 18:00:58.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4324e7258b04840978aa12f548cc632\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b94891b35984b2ab702bc6e3b618d1f : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/67ce28692b6642adb6a59497dc0028a8 : SUCCESS in 1h 05m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2600090fd2cf40a9b22a498a7784d56f : SUCCESS in 47m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cc0d90bafb84957b5612c719731fe31 : SUCCESS in 57m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2a9d87a0b944cdc82ba3114ab242ddf : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7aeb096980d3496fa6098822092f81f9 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d996cbc81974e1ebb22961886328f13 : SUCCESS in 17m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59979e8d570c4d54a391e4dee8ff9b94 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b35b9b89b5b462699ece9f9fea6cefe : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5159df70b60e4b3b9fb072da4678d8f3 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c91d7d046b2641c991b51143493864ac : SUCCESS in 14m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b51c0847d870401981f28d2ad5b3a94f : SUCCESS in 16m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b942792274fd400da1d83f5507af4e37 : SUCCESS in 20m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b78b7a2a88324829bd31d87fe261f90d : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/53f517a446d84196be833d95b743f0b1 : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfe6a264c7904257b375623e32915e2b : SUCCESS in 36m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc68001ece3d4d53b42882fcb0a9d8ac : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/931a0aa79f2f40aabf8659b4fd949990 : SUCCESS in 19m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/79d74b96903a4e93865b6469a47367bc : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/004975a22f8a4074aef296f5cb51143a : SUCCESS in 8m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e56b961b7384905b7a5053507127d4e : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ac778d13187428d8b9d6f09b17c6891 : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2426a99a92b14b59aebbd31d4f646b5c : SUCCESS in 57m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/09c74575bc3b41559bc75048011d9a98 : SUCCESS in 53m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3343f32c32a3406ca79ecddabcba7f87 : SUCCESS in 39m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13c0fb1cdeb34f589b064bc64df04a3d : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f9f20c26f6345e9af3895831b3f0556 : SUCCESS in 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b86b9096b34bd79a1601c08b42b668 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b53ca3b5f3d489e816085a008f7538b : FAILURE in 39m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88f1fdc5f404408282905ca7739a7c8f : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":17},{"id":"24d0b244d3e710c7cf1184a85f7db371d87fcf16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 19:38:06.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0780bf42a8e049299ecaad52bbb618f1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb9eede365374be996b4808afd5ec1df : SUCCESS in 1h 12m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17f30ec30aee4bd2a57707c23167b5d6 : SUCCESS in 1h 08m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/35d3fd07adab411d84a715b1d85c7ff4 : TIMED_OUT in 1h 31m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1db3dcafc9947c983b48597a1b4b358 : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":17},{"id":"a0a11416a6740fb66cc3bf0ec7711035e4828188","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 20:01:47.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6a0fd4790bf593201cdafe84042d4b1a775ddbb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 20:02:45.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"04330da1653741568386bdc5185a81e647bc545b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 02:39:03.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b37c031943546e09e08b2f8ca670e1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e0468fe325406ba2aa871e2dd77bb7 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/5d4e4cb3baa54d6eb0d1951f653e49af : SUCCESS in 57m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/78ee2ed9e8e3465d8e8ba99d28ef9740 : SUCCESS in 54m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d55884b4ec044c27b31b7baf11b4704d : SUCCESS in 55m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dcdcff6510e24a30ae357b89ae5e4d0a : SUCCESS in 9m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ee769e5b1f614b468f58899b728c9b53 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7bc2dea89a1497187df268550dcea2c : SUCCESS in 15m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76935f8ed3bd4ea4a1da689cae86b778 : SUCCESS in 16m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6bfde6c81da540ffa860572e4f3e19d6 : SUCCESS in 15m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/27bbde4429674302aaa966f37908e573 : SUCCESS in 14m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/872f4deef75a474d96b876a6ebb4c57c : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99e3699a37f74451930fd1b78b1c169b : SUCCESS in 15m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40ec49f58c784089a8078a0c1acb32ac : SUCCESS in 16m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/affcb43fcaf640e48896deabf0339b0d : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7bb30bac26b74cf2809d4a20b87a57ff : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d6c5eac164aa4e6a8602c002f9d4a29a : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f35817448f9480d87f6d77de9020c88 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d845604b545d41b2b632c05cd659b5b1 : SUCCESS in 6m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9d45205d12984d2999513ded65c9adb1 : SUCCESS in 1h 01m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef0018ec1fa1497b957f16c9fe2eea57 : SUCCESS in 54m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13c0e2a0f6304224879acd7b5dd5123b : SUCCESS in 52m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21cf8466100546ee944c89f872414219 : SUCCESS in 46m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b537e6958d01442c9127c9cfce725954 : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956a604fdeb1423fbae7765f1e9b95c0 : SUCCESS in 4m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bae4df60833c4916a0743c123cb3d152 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":17},{"id":"51520598c86916ae2fe6a13bd66fe5acd85e2a3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 02:39:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"76362584334e8367273a7255f1d157d9aafeba05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 02:40:34.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df87e78b3cd74ed9a9cbeaa0e8298264\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a167c513c5a741c5a6210591a50901f9 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2e2c7092c92c429f8df2674957fee276 : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d3890bba25d9465180b9fe04d23b5bc5 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"2f32d07707a8603c12a13b9d1a0bff6622f398cd","revisions":{"73aee405636c1d961923080ea2223183693bdd4e":{"kind":"REWORK","_number":1,"created":"2022-09-22 14:27:32.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/1"}}},"commit":{"parents":[{"commit":"c96d2bdd529f3d203dc19db720a825e2c6888331","subject":"Merge \"sharding: Skip shards that can\u0027t include any new subdir entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c96d2bdd529f3d203dc19db720a825e2c6888331"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 14:27:04.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73aee405636c1d961923080ea2223183693bdd4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73aee405636c1d961923080ea2223183693bdd4e"}]},"branch":"refs/heads/master"},"d3b9c44ee6a9b575ded77f148ced94f4e7962eb8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-22 14:28:09.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/2"}}},"commit":{"parents":[{"commit":"c96d2bdd529f3d203dc19db720a825e2c6888331","subject":"Merge \"sharding: Skip shards that can\u0027t include any new subdir entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c96d2bdd529f3d203dc19db720a825e2c6888331"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 14:28:09.000000000","tz":0},"subject":"[wip] swift_proxy: add memcache skip success/error stats for shard range.","message":"[wip] swift_proxy: add memcache skip success/error stats for shard range.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3b9c44ee6a9b575ded77f148ced94f4e7962eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3b9c44ee6a9b575ded77f148ced94f4e7962eb8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"de9973d5babbc2eadec72f76eb70f49a7c58dde5":{"kind":"REWORK","_number":3,"created":"2022-09-22 20:20:32.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/3"}}},"commit":{"parents":[{"commit":"c96d2bdd529f3d203dc19db720a825e2c6888331","subject":"Merge \"sharding: Skip shards that can\u0027t include any new subdir entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c96d2bdd529f3d203dc19db720a825e2c6888331"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 20:20:08.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de9973d5babbc2eadec72f76eb70f49a7c58dde5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de9973d5babbc2eadec72f76eb70f49a7c58dde5"}]},"branch":"refs/heads/master"},"599b40855d3b688be26d6c189b5610c78cbe9ef1":{"kind":"REWORK","_number":4,"created":"2022-09-29 05:14:05.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/4"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-29 05:13:12.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to be:\n  1.shard_\u003cop\u003e.cache.hit\n    This is still same as previous.\n  2.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/599b40855d3b688be26d6c189b5610c78cbe9ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/599b40855d3b688be26d6c189b5610c78cbe9ef1"}]},"branch":"refs/heads/master"},"10c3f29fc58c8a35977e6410c721166e73cff03f":{"kind":"REWORK","_number":5,"created":"2022-09-30 05:58:42.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/5"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-30 05:54:14.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nGrafana dashboard will need updates to display those metrics correctly,\nalso infocache hits are newly added, please see above message for all\nchanges needed.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10c3f29fc58c8a35977e6410c721166e73cff03f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10c3f29fc58c8a35977e6410c721166e73cff03f"}]},"branch":"refs/heads/master"},"036fde83fd5e4bbcbb88e6a4c8fc0a5be7788821":{"kind":"REWORK","_number":6,"created":"2022-10-03 04:40:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/6"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-10-03 04:39:58.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nGrafana dashboard will need updates to display those metrics correctly,\nalso infocache hits are newly added, please see above message for all\nchanges needed.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/036fde83fd5e4bbcbb88e6a4c8fc0a5be7788821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/036fde83fd5e4bbcbb88e6a4c8fc0a5be7788821"}]},"branch":"refs/heads/master"},"0088f1d9a9fe7cc520765f1c9e2829cf660e7fd1":{"kind":"REWORK","_number":7,"created":"2022-10-03 04:42:43.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/7"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-10-03 04:42:35.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nGrafana dashboard will need updates to display those metrics correctly,\nalso infocache hits are newly added, please see above message for all\nchanges needed.\n\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0088f1d9a9fe7cc520765f1c9e2829cf660e7fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0088f1d9a9fe7cc520765f1c9e2829cf660e7fd1"}]},"branch":"refs/heads/master"},"59fbcb9da30d5be081637578d42be600255f1e27":{"kind":"REWORK","_number":8,"created":"2022-10-05 23:42:32.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/8"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-10-05 23:42:20.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nGrafana dashboard will need updates to display those metrics correctly,\nalso infocache hits are newly added, please see above message for all\nchanges needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59fbcb9da30d5be081637578d42be600255f1e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59fbcb9da30d5be081637578d42be600255f1e27"}]},"branch":"refs/heads/master"},"163a671f10c512f986234cec55db8a3ab72d7d08":{"kind":"REWORK","_number":9,"created":"2022-10-06 17:36:53.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/9"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-10-06 17:36:29.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nGrafana dashboard will need updates to display those metrics correctly,\nalso infocache hits are newly added, please see above message for all\nchanges needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/163a671f10c512f986234cec55db8a3ab72d7d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/163a671f10c512f986234cec55db8a3ab72d7d08"}]},"branch":"refs/heads/master"},"f3ab054b41c5cbb67c0579eb07e6fe722364a2e6":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-11-02 18:07:15.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/10"}}},"commit":{"parents":[{"commit":"429702e30de95ca3f0188b89d51364f6b47d5121","subject":"sharder: refactor _audit_shard_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429702e30de95ca3f0188b89d51364f6b47d5121"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-02 18:05:03.000000000","tz":-420},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ab054b41c5cbb67c0579eb07e6fe722364a2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ab054b41c5cbb67c0579eb07e6fe722364a2e6"}]},"branch":"refs/heads/master"},"e841815543d142549f1cb2405de716e29a02c858":{"kind":"REWORK","_number":11,"created":"2022-11-30 10:49:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/858942/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/11"}}},"commit":{"parents":[{"commit":"a1939cba037926400cd777059ee511481916cb08","subject":"proxy: add test for ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-30 10:31:54.000000000","tz":0},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e841815543d142549f1cb2405de716e29a02c858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e841815543d142549f1cb2405de716e29a02c858"}]},"branch":"refs/heads/master"},"83348f1bed916545ca14e8a65ea46866bfb1caba":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-12-02 22:27:26.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/12"}}},"commit":{"parents":[{"commit":"d503ddaf14b8ecf569a3b1bd01f3098c91140dbb","subject":"Merge \"DB locks shouldn\u0027t squelch errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d503ddaf14b8ecf569a3b1bd01f3098c91140dbb"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-02 22:23:57.000000000","tz":-480},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83348f1bed916545ca14e8a65ea46866bfb1caba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83348f1bed916545ca14e8a65ea46866bfb1caba"}]},"branch":"refs/heads/master"},"b1aa0bc894b24ccad16a24359251b0587b2228cc":{"kind":"REWORK","_number":13,"created":"2023-01-13 19:04:48.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/13"}}},"commit":{"parents":[{"commit":"d22a75405297a148d7dc1e2c8d703678df065f63","subject":"Merge \"tests: Remove references to soft_lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d22a75405297a148d7dc1e2c8d703678df065f63"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-13 19:04:30.000000000","tz":-480},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1aa0bc894b24ccad16a24359251b0587b2228cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1aa0bc894b24ccad16a24359251b0587b2228cc"}]},"branch":"refs/heads/master"},"382e7141c0e77435bb3707036433ffd309298f3b":{"kind":"REWORK","_number":14,"created":"2023-01-17 03:45:32.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/14"}}},"commit":{"parents":[{"commit":"d22a75405297a148d7dc1e2c8d703678df065f63","subject":"Merge \"tests: Remove references to soft_lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d22a75405297a148d7dc1e2c8d703678df065f63"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-17 03:45:22.000000000","tz":-480},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|skip|force_skip|disabled|error].\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/382e7141c0e77435bb3707036433ffd309298f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/382e7141c0e77435bb3707036433ffd309298f3b"}]},"branch":"refs/heads/master"},"71d4ca709515a86ced8a0285fd50dea9b1e87ea1":{"kind":"REWORK","_number":15,"created":"2023-01-17 18:44:25.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/15"}}},"commit":{"parents":[{"commit":"d22a75405297a148d7dc1e2c8d703678df065f63","subject":"Merge \"tests: Remove references to soft_lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d22a75405297a148d7dc1e2c8d703678df065f63"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-17 18:44:17.000000000","tz":-480},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|bypass|skip|force_skip|disabled|error]\n                    .\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      bypass: metadata didn\u0027t support a cache lookup\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71d4ca709515a86ced8a0285fd50dea9b1e87ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71d4ca709515a86ced8a0285fd50dea9b1e87ea1"}]},"branch":"refs/heads/master"},"f1679b59616e018145caf0b0c9cbdb3af0feaa66":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-01-17 22:28:42.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/16"}}},"commit":{"parents":[{"commit":"dd9b687f715f0ef6d87b420a0bad3fe8f54f0954","subject":"Fix docs build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9b687f715f0ef6d87b420a0bad3fe8f54f0954"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-17 22:28:35.000000000","tz":-480},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|bypass|skip|force_skip|disabled|error]\n                    .\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      bypass: metadata didn\u0027t support a cache lookup\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1679b59616e018145caf0b0c9cbdb3af0feaa66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1679b59616e018145caf0b0c9cbdb3af0feaa66"}]},"branch":"refs/heads/master"},"2f32d07707a8603c12a13b9d1a0bff6622f398cd":{"kind":"REWORK","_number":17,"created":"2023-01-20 16:37:16.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/42/858942/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/858942/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/858942/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/858942/17"}}},"commit":{"parents":[{"commit":"dd9b687f715f0ef6d87b420a0bad3fe8f54f0954","subject":"Fix docs build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9b687f715f0ef6d87b420a0bad3fe8f54f0954"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-09-22 06:07:22.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-20 16:37:06.000000000","tz":-480},"subject":"swift_proxy: add memcache skip success/error stats for shard range.","message":"swift_proxy: add memcache skip success/error stats for shard range.\n\nThis patch will add more granularity to shard operation cache or\nbackend metrics, and then remove some of existing and duplicated\nmetrics.\n\nBefore this patch, related metrics are:\n  1.shard_\u003cop\u003e.cache.[hit|miss|skip]\n  2.shard_\u003cop\u003e.backend.\u003cstatus_int\u003e\nwhere op is \u0027listing\u0027 or \u0027updating\u0027.\n\nWith this patch, they are going to become:\n  1.shard_\u003cop\u003e.infocache.hit\n    cache hits with infocache.\n  2.shard_\u003cop\u003e.cache.hit\n    cache hits with memcache.\n  3.shard_\u003cop\u003e.cache.[miss|bypass|skip|force_skip|disabled|error]\n                    .\u003cstatus_int\u003e\n    Those are operations made to backend due to below reasons.\n      miss: cache misses.\n      bypass: metadata didn\u0027t support a cache lookup\n      skip: the selective skips per skip percentage config.\n      force_skip: the request with \u0027x-newest\u0027 header.\n      disabled: memcache is disabled.\n      error: memcache connection error.\n    For each kind of operation metrics, suffix \u003cstatus_int\u003e will\n    count operations with different status. Then a sum of all\n    status sub-metrics will the total metrics of that operation.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nMetrics dashboard will need updates to display those changed metrics\ncorrectly, also infocache metrics are newly added, please see above\nmessage for all changes needed.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib8be30d3969b4b4808664c43e94db53d10e6ef4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f32d07707a8603c12a13b9d1a0bff6622f398cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f32d07707a8603c12a13b9d1a0bff6622f398cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
