)]}'
{"id":"openstack%2Fswift~859097","triplet_id":"openstack%2Fswift~stable%2Fzed~Ic4f48eb0eb3b542a0b0ee0bf9c5f1090efab0860","project":"openstack/swift","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"Ic4f48eb0eb3b542a0b0ee0bf9c5f1090efab0860","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","status":"MERGED","created":"2022-09-23 14:14:35.000000000","updated":"2022-09-23 18:52:23.000000000","submitted":"2022-09-23 18:49:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859097-create-zed","meta_rev_id":"7d0429a9327a23c18c65e71b587a96c6ae643c7a","_number":859097,"virtual_id_number":859097,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-23 18:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-23 16:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-23 16:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-23 15:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-23 16:40:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"8fe80651616c087eecc1e646e516dfb2ef796250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-23 14:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cadcd101c8c5d7d88f428f2c554f25d921771b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:27:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18b4f8bd2f3e4c97b1ff760d5a7e498e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d433c7c60b4b33af7de6b18d7c3346 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f2f5b3795584344bf58d3687a750e24 : SUCCESS in 6m 15s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d40cc003849548a5a9a9983e45f18fba : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8a2645065d1f43d4b0558eb75ed4d8a6 : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/370d622defc246eb82124a05c2e0aa53 : SUCCESS in 22m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/702c28bbc45d4fd4a6955db2fcda16b5 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ae65e094f44d4183bc109bf9ec45c062 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f4c9c76e6665439b88a7882031fe420b : SUCCESS in 21m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f91be54da174b80b55e3e5324a55c33 : SUCCESS in 16m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe3fe4003ead410180a05e7541fb68fe : SUCCESS in 14m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e61f84ecdc674fe79f4385cf9b8e670b : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71cb4bef35e54ae9be0243601c1ef661 : SUCCESS in 22m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/621a4d51df0046159cfdb0b9f97a5939 : SUCCESS in 21m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/df04a5144d1b4646a422c9a5e9113d76 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2a95917a00a46b7bd720ea8373ca673 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7add8486426c45d18c81383af59af1af : SUCCESS in 21m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcf7f74821ba4e2cb9af1c43272063f2 : SUCCESS in 10m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b98fd478695b497b854e7c951fbd3910 : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/06f9bc49aafd4f11a9bdd73692f3444f : SUCCESS in 8m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7a656b85dec40b6b479ece9d3a2a25f : SUCCESS in 1h 06m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2707ff8bd42493e840aeaf84feb212a : SUCCESS in 59m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e711e1dd410843ed945b69a29f609de4 : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c466f9b9d34ee7b1f398b48921b0fb : SUCCESS in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1ab1442c8ec413fa04a68c8e9d2992f : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6db1e205a8566421bea3a94e8d410409da40109f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 16:18:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aea16aee08445ddab13dee86a04e4e8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca420c44f14844a1bff466f16c3e2cef : TIMED_OUT in 1h 31m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2798266e4d1c42618c9dae03147828ed : SUCCESS in 40m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/705360fded774368a97924fa09393e28 : SUCCESS in 1h 04m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/249553c90bd44705b8e5d395d7bd1494 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4189d9d9085541a9a6e885a71a85523dcfb0bb95","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-23 16:40:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3d8f6b9afb72bad87de6678693bd46ad1d2831e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 16:41:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2af2fb3ef4b6c1bc8a67486946d03de8c9a3e9eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 18:49:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a4476d39434fa986f1fac7e9b265ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb43027a161d4105a2f8bc01e4658d9d : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a07f94b4feee4a58970519908bdfeb7d : SUCCESS in 7m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cb1c12528df34fd69aa3c9745eaa9ff3 : SUCCESS in 10m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/013c1b0f3c2841bcb99c2f3c5635aced : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b98274cee03a4c339b004645fe70d208 : SUCCESS in 13m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/204a67f972ac44c7b16686e0835511a5 : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f9c05c340494adc964998dd9feb570f : SUCCESS in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bd4d8776d6f4465be13a945cc3fe716 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b5503e2c6794a528c407c50b8bddd32 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/334cf355218d4674b90e5d9d20e2aa67 : SUCCESS in 21m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f0f33ac58f5b46fd8828db6d88c418f0 : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/33050bd561004d21bc31c00b90cc7177 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7c0a63d9c8294c318d3560cd2f8fa68d : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/477438c0987d4ed190d02ef275433f95 : SUCCESS in 9m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c3aa53007eba49f991ffbbccbdcc8fa3 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e9c61579eec4b2695eb9cafb0be0378 : SUCCESS in 58m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a07a355a5c564fdc819ea24671078d19 : SUCCESS in 47m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca981bb6142f47a4a432f1d30689ea9d : SUCCESS in 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb34e0d3b754f989253e105e4ce0a74 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a53123959ce404aae7d7f7b20d522b7 : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1e1d8bbf214e9a012821ecd76197fff5ad0b35b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 18:49:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d0429a9327a23c18c65e71b587a96c6ae643c7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 18:52:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d31d8e1f45104d6c93676b9633a6ef73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5faea5bd65040f6af9f47b66ec9a8f8 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af18bed82d634950aac1fd362245960b : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bf4a5bce47f34d87a2a31fa3d78e531f : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c27b921f2cbf45e09856731e6ff889d1 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49d35e0c32f65724d1cf3af3e3cf0a3586f2f8f1","revisions":{"49d35e0c32f65724d1cf3af3e3cf0a3586f2f8f1":{"kind":"REWORK","_number":1,"created":"2022-09-23 14:14:35.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/97/859097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/859097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/859097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/859097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/859097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/859097/1"}}},"commit":{"parents":[{"commit":"86c46d6145b68bbcc57f5e1fe88280b743f4f930","subject":"Update .gitreview for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86c46d6145b68bbcc57f5e1fe88280b743f4f930"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-23 14:14:33.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-23 14:14:33.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","message":"Update TOX_CONSTRAINTS_FILE for stable/zed\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/zed branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic4f48eb0eb3b542a0b0ee0bf9c5f1090efab0860\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49d35e0c32f65724d1cf3af3e3cf0a3586f2f8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49d35e0c32f65724d1cf3af3e3cf0a3586f2f8f1"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[]}
