)]}'
{"id":"openstack%2Fswift~859348","triplet_id":"openstack%2Fswift~master~If01b36949fac0894881191629ef3ac9aa06188a0","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-11-14 17:47:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-11-14 17:47:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If01b36949fac0894881191629ef3ac9aa06188a0","subject":"object-server: Add API to just consolidate hashes","status":"NEW","created":"2022-09-26 22:39:04.000000000","updated":"2022-11-14 17:47:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"124caf86eb55950a7401dba9d308496bb48a0408","_number":859348,"virtual_id_number":859348,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-11 01:35:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-26 23:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 23:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-27 16:04:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"e8ad2733c7976efa1c3b2d6c721f1917d9ca7175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-26 22:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a35ea474979deb49660ae213cca89401d19eb5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 23:23:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdcae0aabcfe4845b0edf57747d6f847\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9654fa5537749cc80189ce1e086c72c : SUCCESS in 31m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca8e263af641406eb3c386d1feb982aa : SUCCESS in 32m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a34b80ec1e754b27a3798a273debc083 : SUCCESS in 31m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/322388b72bca4750844e2101ac1e82c3 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"66e9ee6d3d071d2de7e3eba855bfd9aa4278f9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 23:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61d674262a3d40a781ada106d8516a73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe96fd043a30470eb102054ff079c243 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/4fc3e319d38f4834b8449840e03e233d : SUCCESS in 55m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15b9736b74a44210aa6d0b89b5f4c8aa : SUCCESS in 42m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba3b250bf71348b2a0fc0a1bb89ccb7b : SUCCESS in 42m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83b8f07da90e480f9009fc6c520d8cdf : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f6cddb5e6ef4fbfa9da82e4705f1966 : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8dfd3837c91f4fd48bfbb12f098abedb : SUCCESS in 18m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd68303f9965449eaf37bd35def6073e : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/118557dd22a5420981a2adda6571cdc3 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f5ad45d8e114c8987cf7a298af450b1 : SUCCESS in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/874d1b24b24e47cca5b18313f8694247 : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f7a5f39af72475fbef3a874141468db : SUCCESS in 12m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f7b0c4deb6245cbba3a4a394e2194e1 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9dabb63d6e7445eca36de415fdc85343 : SUCCESS in 21m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5cd3d679842e4c88958bc50b9cb1c52e : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44b261b7fdd84d43a9e9428b38a125b3 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77bf5efb490443d6b05130763ee70d0d : SUCCESS in 13m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fba9f540b891422fb0bc62fddaf3d9a0 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/321469a33a6b4c4284a54ff1e0184a8d : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f139576fe9a42e49a76933c3b852718 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/74abb7fcf04b4899875361a1a0312157 : SUCCESS in 6m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/77db9a9430014fff8a55a6899a1c1496 : SUCCESS in 1h 00m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d6c9cf00d8c48e49a7c029c1e63b0b3 : SUCCESS in 55m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15b266261c2344af8a897d160825b8c4 : SUCCESS in 34m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f22614a54c94b98a527ed3787465f48 : SUCCESS in 42m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780be56b2cf14d94bcfa26bee1538448 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be4eff6da1c409ab7592c509068bf70 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7824da6e42734d219f9e5163378939fe : SUCCESS in 24m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76d29e81996243688610de1dae2d29ca : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b1bfa0e40fb241f7256ed2a4d5a157e8128a61d0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-27 16:04:07.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9620693cd86589ceed52d69b5b66f8987ede8a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-11 00:00:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"06979b859d77d7ddeccd2829597da134f2c06bb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 01:18:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6114ad3aff2343049ff39f248c96090b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c0ab8f3a34b4ae9880610165f5f3b7d : SUCCESS in 46m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc334b9bdcda4f918b83ea02c597e1d9 : SUCCESS in 39m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/16412b55c87645dabb249979637c26af : SUCCESS in 1h 04m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/38934b56c9d14a7da226948d92e26554 : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b4c27fe889d29b065e3046e1f54a98556c20e5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 01:35:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4286757a344470a091701ec8d75d40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cfd582adcec4e06aa98fb93a05636eb : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/5facea80b7864ee8a6d1d7253ac83f4a : SUCCESS in 48m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3378ecfcca3446eaab63a16062e110d9 : SUCCESS in 42m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed6e8843ff88462ab92b9e43fb49411c : SUCCESS in 44m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe7816fa5c6a4c0cbd4b7f03812ba51c : FAILURE in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e03fb4e45674ea580d012cd6ef28e1c : FAILURE in 22m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ffd279359144a8ba0092db0dbbdc7eb : FAILURE in 16m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/446e7de7270648d6a615c2acd338f161 : SUCCESS in 9m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f009210f3fa4c91b96f93dd84b15448 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93f1feaaeddc4a6684bdd14bb7fe6efa : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ad1315668f48c3ba82f3f3c6b6f67c : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ac3c0fcdcb546f8b63568e1a6443ab1 : SUCCESS in 15m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c0fd299217049ea87928702fa48b450 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af5e7e15e7af4a6388212705896068aa : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e67c42acff154a688346675ee84dee81 : SUCCESS in 17m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d54725fa80ee46c6957b085c1f65afa3 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2101405c159c47389f5ef93042a1c412 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e20775fce1b244db9b81484f254631b9 : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/172d92511b294771b94f718829e625e6 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ff7f536da6a4437846396a9a4218397 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e52fe14990e42e5bbeb69e08906e837 : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/25632af3e98c44ce9a14e87785959cd8 : SUCCESS in 50m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/20ea4df36ad64abca4e0ab2257c92a29 : SUCCESS in 59m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a6460e287fb49bfb623a813fd906a68 : SUCCESS in 43m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f37524713daa4eb5b6ad8a29aec5808b : SUCCESS in 45m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22462c3844bd4efabaa61d3ffd3d322e : SUCCESS in 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67cf4516957847f1985dbc19f1e6588f : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b000644ae5534f02b137089a88b6db3e : FAILURE in 19m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01dd4f2ac1754b90a9da87787f03bec0 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":2},{"id":"4529cfd6927eff1190be361c0f6f66847162f97f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-14 15:57:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"124caf86eb55950a7401dba9d308496bb48a0408","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-14 17:47:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88eab597cd5670088dd906d1e14ee29b83cae7eb","revisions":{"1fcabb76d68aca775043b18b2c4292e6f98a4a95":{"kind":"REWORK","_number":1,"created":"2022-09-26 22:39:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/859348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/859348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/859348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/859348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/859348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/859348/1"}}},"commit":{"parents":[{"commit":"2f9a6b639de8f381e853e55a5141f5bb29521883","subject":"proxy: Use last-primaries table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f9a6b639de8f381e853e55a5141f5bb29521883"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-16 16:56:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 22:38:46.000000000","tz":-420},"subject":"object-server: Add API to just consolidate hashes","message":"object-server: Add API to just consolidate hashes\n\nChange-Id: If01b36949fac0894881191629ef3ac9aa06188a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fcabb76d68aca775043b18b2c4292e6f98a4a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fcabb76d68aca775043b18b2c4292e6f98a4a95"}]},"branch":"refs/heads/master"},"88eab597cd5670088dd906d1e14ee29b83cae7eb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-11 00:00:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/859348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/859348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/859348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/859348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/859348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/859348/2"}}},"commit":{"parents":[{"commit":"4166a618ef601a8c84f5899ba5422437daf35cc6","subject":"proxy: Use last-primaries table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4166a618ef601a8c84f5899ba5422437daf35cc6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-16 16:56:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-10 21:39:10.000000000","tz":-480},"subject":"object-server: Add API to just consolidate hashes","message":"object-server: Add API to just consolidate hashes\n\nChange-Id: If01b36949fac0894881191629ef3ac9aa06188a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88eab597cd5670088dd906d1e14ee29b83cae7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88eab597cd5670088dd906d1e14ee29b83cae7eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
