)]}'
{"id":"openstack%2Fswift~859349","triplet_id":"openstack%2Fswift~master~I1ded891087907ecf95c183abadbbc974ebef4a54","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-11-11 06:40:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I1ded891087907ecf95c183abadbbc974ebef4a54","subject":"replicator: Use last-primary table","status":"NEW","created":"2022-09-26 22:39:04.000000000","updated":"2022-11-11 06:40:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"5130dba1d732631e9da63ca8fe9bc122c87c9150","_number":859349,"virtual_id_number":859349,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-11 01:07:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-11-11 06:40:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-26 23:23:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-27 00:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b71bfd69a652570ca1fc8f4c9af88f1c2b87e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-26 22:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e861a5b9cf7426116fab8722a0f4027d22af6752","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 23:23:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa55ed8952284753914e3da77dd9639f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d4630ff73a94471a04ade3fdf092c6c : FAILURE in 31m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a9919c304654dd0a25450ee5c2ef584 : FAILURE in 31m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/44edf92f5ae74af1bb9c72f438cf454d : SUCCESS in 31m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f74216c6fab1489b824b36d6a7993a85 : SUCCESS in 31m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e7376e95a669080d855e5de2a4318e73fe94d505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 00:37:50.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f08faf73a26a4c29b281e0540e605b6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14b36605af446b7acdb6267e722a3f7 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/69a9e4f062584f4db987442711ffc56f : SUCCESS in 43m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0026b205da224739ae5d291ec80edee1 : SUCCESS in 49m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a91d0e42e1fb45fabfd1a442ca15f5ca : SUCCESS in 43m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aea3fa5eca8c4cb3bd3d54d0af249256 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7dd26deff14f4b42b9046e17979166c3 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75b622ec9a8e473f8e83093615529f12 : SUCCESS in 16m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6649d54bce5d4238817e6f3ffa562165 : SUCCESS in 10m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/70d478724ca24c28b20f1d9e3ab588d7 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1246c89882dd4504af7b304afff11fca : FAILURE in 14m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41e12f6e5823438b8f30469842871ed5 : FAILURE in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd3755939e2947549aaf776821e7611c : FAILURE in 12m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbe37f10c0954b5282d8a6cfe1f4a76e : SUCCESS in 20m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96554a8cbbf444f19a356f63cde6b164 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/32ec92eb397b443bb9572a84217e230e : SUCCESS in 15m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/430abf78f5c3484e84f31e8410ce0117 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7089c2743fbd42e38af2351bb98daaee : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1c8174f0221a487d8dd9cc70c273b696 : SUCCESS in 13m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/79249a88a0f345beb0fe5d01078efb17 : SUCCESS in 9m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/740e637f75124cc49cecad4a69fbb0cd : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/899f19c6dbaf434ab68928aa6e60cd3b : SUCCESS in 8m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/469ec676ffb34a088703eecb6cc1ca7c : FAILURE in 1h 23m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb675d60d48a4984b5af8295447ce6ea : FAILURE in 1h 21m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b09a605b35df43ad86e55d71a979f54f : SUCCESS in 49m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ee7bf7d6ccc41b2b76ca7ba7584ec53 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db3dd59a15374c468373ab519cb88c07 : FAILURE in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76cb8e2264f41c4be8db243ede22fc3 : FAILURE in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8aa184028344ef6893d6108c238c8b0 : SUCCESS in 28m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8360c5cc74742c6bfa6a6c50a65a938 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"30338a535b0c27d60c8e874cb9a4e78ca6813d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-27 00:47:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47b4ce4b1bfabf7bb1cff63ea05779fcfe3fdfc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 01:22:47.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/320883b2367c4fea97da7c20526dd405\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba7b165d7f434430b124e31fa181d89f : FAILURE in 28m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc209ac3e0a2446bb8f21ebdf9679793 : FAILURE in 29m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0dd0db2d77d74fa89dac667cdcc87f7a : SUCCESS in 28m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3df36fe098948249e6faa2ac244cfe6 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":2},{"id":"786074555e96ee3afa2c8da8fb078452967f252b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 02:03:41.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b313b066e224f189fb24dc3b8867c5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723101d3195b47a7931b1778264baf13 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/31225916695e4852bd536f6beae36e57 : SUCCESS in 1h 08m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9702019ee684317a7753a4e6f66f704 : SUCCESS in 1h 00m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56cc4187183849c19ccd2f800d96b312 : SUCCESS in 57m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb143f6bbde14edabeb1bb8a44910cd0 : SUCCESS in 24m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/044e6062164343a692d93be5d6da045d : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2543a3265584682b3045bf76614bba6 : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49c13437fc8f4fbf8ed4567dcde92518 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dcf891759c994b8897c185cd7ba06277 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2bb9d9cb6a034bfcba0fdaf44ee4d85b : FAILURE in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2efbe0fdf44de7875b5b636d1dd8c9 : FAILURE in 21m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57521040019d4fd3be6d04514931746e : FAILURE in 18m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ccff4650c88438d9173cfa77cabcdf6 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/056b72e063d44ebc8ef00bf67bd720ff : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/375dd5a17e3a4ba89dbad0122cfb0425 : SUCCESS in 19m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4eb38967acd44e8491f3ebdcc2743594 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4251f5d6eced4d27871199069b46c8fe : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bbbedea1f4514999a85798298f589dbf : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7f87c6f94c742da825476d33906c37d : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/afe80b7983134c4ea37e1eb456929ad7 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/268beedc8f564922b281bcff444f5196 : SUCCESS in 8m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/896ce47099d84d27a025c2cac9c43203 : SUCCESS in 59m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/51660026644f42c3b49d6325aac8c350 : SUCCESS in 1h 00m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/558872408f1947de920ac1042ee45cd3 : SUCCESS in 40m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/487a3b1dc16949a4b82b317db30c32b9 : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47503202974f4f6db09ecc51c29098a9 : FAILURE in 22m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a369cda5981b4d238c3fcd9efca8d04c : FAILURE in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3dd915c3d4014d7c9f5259d758461158 : SUCCESS in 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dbe7a0a958a4386aebca24a8c5daa8b : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d2f85c1f9f302971ca3474ec1d1973a5e1f827f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-11 00:00:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bb23a437ec1eccaa2e58dca8237227b5812f9db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 01:07:43.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb96075e63084f4db5dda6a0f5b09b88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fdc91ccc2a74c75ac33b6feafc9fa9c : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/1027c73d429d48cd83165e5b59b66c3c : SUCCESS in 49m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2766bba1a8a14838aa1f7acdacd9c8b9 : SUCCESS in 48m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5244971859c44201bc7f69372d8d6f84 : SUCCESS in 46m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9436f6b7376941c8aeb939bd83665a37 : FAILURE in 17m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/426dccb64390461793b89411468614a5 : FAILURE in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e6c219703e94b7581d15cfd7683d6fb : FAILURE in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b08bce01d25540b1ad71645141fcb800 : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/52840512a5844652bf7a3886cd3a55e2 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5ebe5db2826470aab9b24327e7bc93a : FAILURE in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e32ae0ecb364376b2407e1c2611fd01 : FAILURE in 18m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0e6d3f564934897bebe62236296db08 : FAILURE in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14bd2e7efa0f4209aca6979ef0b25a87 : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff807cf0c4914375a252827c31dfdfff : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a41e8f822f9447c5b27690bc90dbe61d : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5cc004daead94665a7b4fabbab7d2a4c : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/921268e52036420a9d6f6579c402dcfe : SUCCESS in 18m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0dc0cb3b8b1a4272ad1162f5d60c289d : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/737ee8d3c16e4b4f990a1c0414175872 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f31b2c411cf45a3a19c541ef296b416 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a66c25a3c5fd42efb144d4740602f406 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9eb4418bc00641cbac1857020c989205 : SUCCESS in 1h 00m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/63c9c2f6d15f4d1aab645885c21e59fc : SUCCESS in 1h 03m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32820ba777ea44cda714805b0e3dfc6e : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/940fc93091a946b6847a1606b688b29b : SUCCESS in 41m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2ccdba2827542e8ad33bc065bed2a69 : FAILURE in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd5580eebe574d75937892dbe06d584b : FAILURE in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2363a96e44a143c7ae76b4875f704796 : FAILURE in 19m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f09e91df11d043c79206470b07afaf0a : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":3},{"id":"1dac5079abee160de7f42c6e238a5fbc70de5294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 01:12:08.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff5f434c0a7748a1b09f788bb1038028\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edd83ebc237b43fbab8bedca37382f36 : FAILURE in 58m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c87bcd3f0e9455e80ec35765a6e6a7b : FAILURE in 56m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e06413f7b1946388a7c3e18f1788a37 : SUCCESS in 58m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4f5a9f3dfab427a854429c4c4e9353c : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":3},{"id":"5130dba1d732631e9da63ca8fe9bc122c87c9150","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-11 06:40:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4be194c5ae5bd63c73e0437e8e436951e491835f","revisions":{"fc9591771d1b65dcb30755b2bd930670c2e168a9":{"kind":"REWORK","_number":1,"created":"2022-09-26 22:39:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/859349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/859349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/859349/1"}}},"commit":{"parents":[{"commit":"1fcabb76d68aca775043b18b2c4292e6f98a4a95","subject":"object-server: Add API to just consolidate hashes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fcabb76d68aca775043b18b2c4292e6f98a4a95"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 20:25:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 22:38:46.000000000","tz":-420},"subject":"replicator: Use last-primary table","message":"replicator: Use last-primary table\n\nChange-Id: I1ded891087907ecf95c183abadbbc974ebef4a54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc9591771d1b65dcb30755b2bd930670c2e168a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc9591771d1b65dcb30755b2bd930670c2e168a9"}]},"branch":"refs/heads/master"},"6e2ec26589342fd112cd7b533cbdca4d15ea527e":{"kind":"REWORK","_number":2,"created":"2022-09-27 00:47:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/859349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/859349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/859349/2"}}},"commit":{"parents":[{"commit":"1fcabb76d68aca775043b18b2c4292e6f98a4a95","subject":"object-server: Add API to just consolidate hashes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fcabb76d68aca775043b18b2c4292e6f98a4a95"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 20:25:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-27 00:47:16.000000000","tz":-420},"subject":"replicator: Use last-primary table","message":"replicator: Use last-primary table\n\nChange-Id: I1ded891087907ecf95c183abadbbc974ebef4a54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e2ec26589342fd112cd7b533cbdca4d15ea527e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e2ec26589342fd112cd7b533cbdca4d15ea527e"}]},"branch":"refs/heads/master"},"4be194c5ae5bd63c73e0437e8e436951e491835f":{"kind":"REWORK","_number":3,"created":"2022-11-11 00:00:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/859349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/859349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/859349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/859349/3"}}},"commit":{"parents":[{"commit":"88eab597cd5670088dd906d1e14ee29b83cae7eb","subject":"object-server: Add API to just consolidate hashes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88eab597cd5670088dd906d1e14ee29b83cae7eb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-26 20:25:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-10 23:58:54.000000000","tz":-480},"subject":"replicator: Use last-primary table","message":"replicator: Use last-primary table\n\nChange-Id: I1ded891087907ecf95c183abadbbc974ebef4a54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be194c5ae5bd63c73e0437e8e436951e491835f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be194c5ae5bd63c73e0437e8e436951e491835f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
