)]}'
{"id":"openstack%2Fswift~860000","triplet_id":"openstack%2Fswift~master~Ib28196191275022fcb74d2365910240cc7c61c3a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-02-09 05:19:28.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ib28196191275022fcb74d2365910240cc7c61c3a","subject":"object-replicator: Count sync failures in update_deleted","status":"MERGED","created":"2022-09-30 15:43:19.000000000","updated":"2023-02-09 09:16:04.000000000","submitted":"2023-02-09 09:15:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860000","meta_rev_id":"353479b8c666be9c8897f05913277b58f4ab5bd2","_number":860000,"virtual_id_number":860000,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 09:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":1,"date":"2023-02-08 17:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-08 22:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-08 22:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 16:27:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-30 16:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 17:31:27.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"4703771813303c4ef800df8fe7230da02fd77e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-30 15:43:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a813990b8086f4a4e08938c961f3474af80970cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:27:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37a7c40f27c24531983ab113e73a3cf9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70190419c7394561a218f16462dd8285 : SUCCESS in 37m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5431b1a4adfc4ed5aa4ec65c462621fd : SUCCESS in 37m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/49131270b8ea4f21803704391fb9b2d1 : SUCCESS in 36m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d3ea90cab1e46d6af85fb734f0fbcb9 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0cd9ae4bd6c3967e2c4a1988b7ec3cb669f1dfe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:51:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b413b8b7c734700af13df82922b4fec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c29d50f6464f80b6f10b7821e34029 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/ddbca284d2af4c91b00a6489a84e7f15 : SUCCESS in 39m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/373b41f80547455db409e724dc5028cb : SUCCESS in 53m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2641115288546a8932b75345bdeae29 : SUCCESS in 58m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e635c4f54cc4418681353b15b506b0c9 : FAILURE in 15m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62fc54738a174717a897f9b5952d576d : FAILURE in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae4760fc23f44fa4afb40d853effc5a7 : FAILURE in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a3be560da974ba7898cc3cce0230ac5 : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9215cd02e50c4fb0aa3460a382c22c97 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9062e99fb1d4b8aa9c5c570b6830e19 : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcca670392774f9fa3735c73da5c085d : SUCCESS in 18m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b74357282f114939beafb082c0aa9979 : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4834d182159141f59c44d29fc1551beb : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d33357d3eb84a738a3d11bdc9e75f33 : SUCCESS in 21m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a460407aebc74f6ab1818329d9982ed2 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7baee0c83c9044dba3420ed2d069113c : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/32001af28b4b4ca795a9a50cfa1b0e7e : SUCCESS in 18m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a3b16800f204023a0cc90561d20e632 : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a71920806a3a483299347abd99d78811 : SUCCESS in 8m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1086c407ff0744858edab29de80b20dc : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/971388a068e544838a95b5083c387fb8 : SUCCESS in 13m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/111dfa1f11924af39cfa4008070e8160 : SUCCESS in 1h 01m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3c5a64a0fab4152b45e6245015f3dc2 : SUCCESS in 1h 00m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4954c438e1a24cd29ca4b583bb621543 : SUCCESS in 50m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8908fabd97be44c0a8da87fc7a392538 : SUCCESS in 48m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9173039504441f098ae77d54289ece3 : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c913f8dc1e4c422ca477bc00bac6a981 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1620f25e5db40e6b2296842f4302b64 : SUCCESS in 21m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da716d1911d148978c49bfc80538d1dc : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5ad99c0143804eb761fd707a1f838cfe4386fd59","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-08 17:31:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51c4bc1994b395c8a939b90c9dbefae88e796ccf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-08 22:26:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b4404934092b62e08a3ed38296490d91ca11eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 22:27:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8969437e468449560471c2d1d78ba602809757e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 00:37:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccfc769fff204af3adc681da99568e78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36e030a6884438c9f7c992e130f52b9 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/16d853d32f314ab4852014ad7ab4cd7b : SUCCESS in 1h 05m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8225939c085343ab877e8b101b98c105 : SUCCESS in 48m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/106a58e553824a3ea9a98b06d9ae1353 : SUCCESS in 55m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/65e9f6c53bac4b6c9f689f67a4837b78 : SUCCESS in 8m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6194ee1fcdb947deb6f1b3a6026d8e0b : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97f71d1f61314ccf8537f739a7ca8a2d : SUCCESS in 15m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76014362204f4aaa995b9d4fded7ea29 : SUCCESS in 14m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d07afa9c6573421ebb72a34d9b3d4306 : SUCCESS in 15m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/56023946315e428ea53870f896458a6a : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c52b8602ee74c6ca88ce56d641bb5b7 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ddde51c7b7ac4e238d952e9fe1b5c265 : SUCCESS in 18m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/837c8e49f1a843e1b11b2d0bf721c7d2 : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c29cdb39f69742b88fffb78664ecb04e : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e16669fc60c43599d65035a4283d4c0 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8bd1e0e694cd43b8afbab51b89d22a9c : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e48a7982d3f84fe79e118e7d824d308b : SUCCESS in 6m 47s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab69c55cf5b44fa380c463ed5b0f90d4 : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b8680156dd9497a86e4b720ff0d08eb : SUCCESS in 56m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ebb3638df1240eea78481d7d8832fc0 : SUCCESS in 1h 12m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c066eac8008445195bf4ad143a24e31 : POST_FAILURE in 55m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/526790e7a3194b4686efc39b26732315 : SUCCESS in 53m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7de054bcb7fe413da1e1d140152af9fc : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0efb44994f5946f29776b3ac3e09e978 : SUCCESS in 5m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa3f13bb10b14005a07365fdf2258270 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fbbe10947479925a49460a1e2d4ad814c3e9c261","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-09 01:39:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44853390ccabadbe66e311b73940ae61e7e4f0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 02:59:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd802d25b58f433189b165f04e3be323\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ac914415f14a81851f684d2a802069 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/13fa245313ab462684ca0f01c1f92b3f : SUCCESS in 1h 03m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f44dfa017cd74e6698cb3eacdec3b56e : FAILURE in 37m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f6a99f6d55a453cbee52f80c9d17511 : SUCCESS in 1h 05m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4def75952973460686d83da7d610c7cc : SUCCESS in 28m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e93fdd929c549af8b1881073c6b8950 : SUCCESS in 23m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d813624fb61b43ffa479978d8cf4adbe : SUCCESS in 28m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91756c7f608e45a9973b27b80bbc92ae : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a9537739f83487fb661032180844fe3 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/323a35354a7146138366e4d194dd01c5 : SUCCESS in 20m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b606fe5a6c64f109cabc1aac0478420 : SUCCESS in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ab6dae00b694eb7a0260b4d54856a46 : SUCCESS in 13m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a3eb7e2722824c90b7a0a95e0c73fddf : SUCCESS in 14m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20de9814357e448d932ea0ee1622506d : SUCCESS in 22m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/726fbe87b8384688af0d345f6437027a : SUCCESS in 15m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73bb947ad6364dd495814a06e35dd21c : SUCCESS in 15m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/26223986b49c489983cf2c2bb3214191 : SUCCESS in 22m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/695dadd63db44a6aa95929cdaf6d9146 : SUCCESS in 11m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aafb5b93fee146b9be7f6300532c64c2 : SUCCESS in 21m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50069cc4e5c049fbab3e3b2d6a04bece : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8d4c39c22354214b585df4b5c5ca115 : SUCCESS in 22m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9aaaa859b3d44d708f18b882b9eff900 : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b9d49766922b43578e78438a7f2ef6c6 : SUCCESS in 50m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce4d42312c874d4c92a5aefdfa2bd5e2 : SUCCESS in 1h 00m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b17871a282b4880940c89c7c6551c3d : SUCCESS in 53m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39c3578cd8a04aa28133c57be65d3417 : SUCCESS in 44m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f8e49f28f5f48bf89c49ef641ebf491 : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50afcf1fa374e9b80004662d0960469 : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2005bac263b4ea78374f68dd71f34b2 : SUCCESS in 26m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d10b74573c474fa6b08b896821c8c7fc : SUCCESS in 42m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3a9a61af8f64563ec7b89174ed78356e1a5b96c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 03:11:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91105de10c81493dabe20c393247017b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b5342062ba446bb8b57a8e437a941c7 : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba2462e8f64b41a1b80e9636ed6f4fd9 : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/51208fea338f4bc985ab24cbbc405827 : SUCCESS in 23m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2f11e1f822d44bbb530df5c9f7dc48b : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"91cd4a63b9a66f92add1cacab29d05a4bf30e599","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-09 05:19:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78064c2fbe69798f4df8694ccae9ee6fc3c61fcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 05:44:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb987d1e04704f4db6e11edc3068e3bb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3edd3eee2c38427f84a3c322adde16f5 : SUCCESS in 16m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8247eaeb3374c9b8b78726b54d162d5 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/46afc742f14f47c9b74455bd4e96ed4f : SUCCESS in 24m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa3c2ddad4204d8aa34d5e7ea05873b4 : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e5d099a1d6acd651c8062ceafa90ae9e64f034ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 06:43:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e17920249644c182b444a509ebd4c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705f1f4bfe914b619fd813dbfaf6221c : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/ef2bcb823a4e44fbb89c24ccbb16c49b : SUCCESS in 44m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e333cca767f404a8dad551265ab7cac : SUCCESS in 54m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6329e412d6e4aadaaa7a621a44094b3 : SUCCESS in 47m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad6fc21dc1fa41be95552c6647ca4f21 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ac89c4ec7c14a60b48653109df4e462 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4f88b0c9f524d5aa0b178b0365f816c : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a92ae9c5629c4746ae8747e7b7f8d7a8 : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7c9a0f4c889f467da686a5b575a22618 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88674473dc6e4a19bd6baaa7b47d759f : SUCCESS in 20m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb56c5aff95d4608bf7e29c7eb78d44a : SUCCESS in 20m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efc366e5f43d4c6b82bcfa095fa2a25a : SUCCESS in 18m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/48059de9f4014778bd96067f0acdee7d : SUCCESS in 12m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6af9658425af47e9a41d8f6f38d9a382 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31b96c08e3b04c44aa7f568c434faeff : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1f56892ff0a412cada001648afae555 : SUCCESS in 15m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5528071dddd040e89600a2c511a39c38 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db72e76e835e43e984e5986746393d22 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff2c130782154d49b074e847cfaa99e5 : SUCCESS in 24m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38e70f1e0b6143bf8f5eda6d94a24abb : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3330ff0930349f1a4d4488e86fdc3aa : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/001785b0a4424aa4b2bdba091a16b46d : SUCCESS in 7m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8f5c9c8109b94612b1bea6e12e0723df : SUCCESS in 52m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f1c51dbe2f64af780d3d2fa1dfc2346 : SUCCESS in 1h 12m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ba5a20cf9344cbfa06834c594c4a804 : SUCCESS in 58m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f00761c5cfdd4ea190c0b65b264d9e5d : SUCCESS in 1h 04m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ff6eb780a7b42498f1d37cdf683ecbd : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57776ea452f44db4888db91ee67e4e2d : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5071582eb5144c4394bfc68149b1e4e3 : SUCCESS in 19m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a4d776bb79c40d285aed951578b6a5d : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":1},{"id":"48a4cdd32daf25b14be270343dd154a80d32d840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 06:44:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfa8198038814ad873c377e48110dcb3aa14d28a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:15:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe5b1d016644441bad3d13d852e887c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1504d50f0d1449d4968ffbe1510e545e : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2e93d6cacefe43c895c6bf330e6d6f48 : SUCCESS in 51m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ff5a43be65d54769bd226f93244dc703 : SUCCESS in 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54a29ab33e854c138b18be807d885af9 : SUCCESS in 47m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7db4e913a86f457e85711324d1d424cb : SUCCESS in 8m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/da1b95428f4442daafa6b0f0c4e40ed4 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/484377b33caa4c06bb6441132131951b : SUCCESS in 24m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/173dd0652d6040ddb1a4cb1cb624b0d1 : SUCCESS in 15m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c708dc8ddaeb422783844b9d7038c9ba : SUCCESS in 14m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/934592735ac5495e982b818c19b82403 : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9747a2646cbd41a98c93bd098d7b1c6e : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b210a14663bd4d56a6bc11a694d768a4 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25e5e80e32dc48ecba97730a40bb6349 : SUCCESS in 15m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cde9d0b0e21b46a991bba4ec51b0c6fa : SUCCESS in 13m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f3d1f3ecde784a188d9fce2e5ae4870b : SUCCESS in 14m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e43e6175d3f64fc5a69c7c22591cef42 : SUCCESS in 13m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a65e4fff7884e0d9ea4ce920cc0e3a0 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2dda732bcf2c4c9687df59b54ff33505 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9589f4641ebf4829a3f46fe47c092f29 : SUCCESS in 1h 18m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/67949f41033840a599359ca428e2d294 : SUCCESS in 59m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f2701f4588b46b19a439e62e27c76ad : SUCCESS in 1h 00m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f37ea612542a4f839d4128a89d45ea35 : SUCCESS in 49m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc26cee4264145859f09f937f3a2b248 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b33eae147e94f009244291facf6befa : SUCCESS in 4m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf7020341e914e40ac8d50c30f615f83 : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"43e640986f20c905c51608de07e012ca9b51b05d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:15:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"353479b8c666be9c8897f05913277b58f4ab5bd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:16:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d551cc4406d2455ca953b89bf95b699b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25befd26d144672aaed437d60c7d360 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c77a2659554c441582846db37aaf9d0e : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/07ee55e504cc447d9f4dceeb1abacf97 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c87278aba8985447a4a1746e66b4f785a602f8ab","revisions":{"c87278aba8985447a4a1746e66b4f785a602f8ab":{"kind":"REWORK","_number":1,"created":"2022-09-30 15:43:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/860000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/860000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/860000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/860000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/860000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/860000/1"}}},"commit":{"parents":[{"commit":"99b0b3d1ec15e93a98dd7ddd1c24e1da9650f64b","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99b0b3d1ec15e93a98dd7ddd1c24e1da9650f64b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-30 15:42:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-30 15:42:43.000000000","tz":-420},"subject":"object-replicator: Count sync failures in update_deleted","message":"object-replicator: Count sync failures in update_deleted\n\nPreviously, we would track the failure in the failure_devs_info set, but\nnot actually use it to update failure stats unless some other exception\noccurred.\n\nChange-Id: Ib28196191275022fcb74d2365910240cc7c61c3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c87278aba8985447a4a1746e66b4f785a602f8ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c87278aba8985447a4a1746e66b4f785a602f8ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
