)]}'
{"id":"openstack%2Fswift~860793","triplet_id":"openstack%2Fswift~master~I9f802538afcd8f5dc65a12e88db2ef560609eb2f","project":"openstack/swift","branch":"master","topic":"test-py310","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-09 08:45:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9f802538afcd8f5dc65a12e88db2ef560609eb2f","subject":"DNM: Test py310","status":"ABANDONED","created":"2022-10-09 08:16:59.000000000","updated":"2022-10-09 08:45:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7299d238c648567601ea37f5d3eb5679476197f3","_number":860793,"virtual_id_number":860793,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-09 08:33:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-09 08:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27f4823b3a18b7b15b522caddcfbaecbf459d68e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 08:16:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2a7d78621a73b1e7149fb8bc3164991de4c0180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 08:17:17.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 647, column 31","accounts_in_message":[],"_revision_number":1},{"id":"7db42794764f71deeaf21b645faabde07af48196","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 08:17:23.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 647, column 31","accounts_in_message":[],"_revision_number":1},{"id":"d01efed77cc64535749caea3dac74a59fd0c1cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 08:29:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dcc886d1745c15016866841ad442ad9f41d952e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 08:33:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7d8d06b493545d2a405411088a6669a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e83eb5299764eb9b1410aa3a62f10fb : SUCCESS in 3m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5bc9ebdeada04ac7b9dc371c83400095 : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"7299d238c648567601ea37f5d3eb5679476197f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 08:45:49.000000000","message":"Abandoned\n\nduplicate of https://review.opendev.org/c/openstack/swift/+/850947","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b816354ec2c8444dca7639cbebe8e7a59154ee1e","revisions":{"90fe243ea80ec61b12e537b91b95008fb3cff9f0":{"kind":"REWORK","_number":1,"created":"2022-10-09 08:16:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/93/860793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/860793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/860793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/860793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/860793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/860793/1"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 08:16:35.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 08:16:35.000000000","tz":120},"subject":"DNM: Test py310","message":"DNM: Test py310\n\nChange-Id: I9f802538afcd8f5dc65a12e88db2ef560609eb2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90fe243ea80ec61b12e537b91b95008fb3cff9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90fe243ea80ec61b12e537b91b95008fb3cff9f0"}]},"branch":"refs/heads/master"},"b816354ec2c8444dca7639cbebe8e7a59154ee1e":{"kind":"REWORK","_number":2,"created":"2022-10-09 08:29:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/93/860793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/860793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/860793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/860793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/860793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/860793/2"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 08:16:35.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 08:28:55.000000000","tz":120},"subject":"DNM: Test py310","message":"DNM: Test py310\n\nChange-Id: I9f802538afcd8f5dc65a12e88db2ef560609eb2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b816354ec2c8444dca7639cbebe8e7a59154ee1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b816354ec2c8444dca7639cbebe8e7a59154ee1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
