)]}'
{"id":"openstack%2Fswift~860866","triplet_id":"openstack%2Fswift~master~Ie05af464765dc10cf585be851f462033fc6bdec7","project":"openstack/swift","branch":"master","topic":"proxy-log","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-02-13 21:34:49.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-02-13 21:34:49.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-02-13 21:34:49.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-02-13 21:34:49.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2023-02-13 18:57:59.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-02-13 21:34:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie05af464765dc10cf585be851f462033fc6bdec7","subject":"proxy-server exception logging shows replication_ip/port","status":"MERGED","created":"2022-10-10 18:12:41.000000000","updated":"2023-02-13 21:37:58.000000000","submitted":"2023-02-13 21:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":147,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"860866-proxy-log","meta_rev_id":"80d9db549102e959209e75d14fadb0842259fb4e","_number":860866,"virtual_id_number":860866,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 21:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2023-02-13 18:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-13 18:57:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2023-02-13 18:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-10 19:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-10 20:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-10 20:31:26.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-10-10 20:31:26.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2022-10-10 20:31:26.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-10-10 20:49:15.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-10-14 21:54:06.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-02-03 21:53:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-02-09 20:53:37.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"67d7914c28c1dc80918c9e3b724415af398233ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-10 18:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0e49933660896faad7acc9cbe3e91fb71fb38a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-10 18:25:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91d4b70fc847cfff39f016eead8ec98178654b8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-10 19:07:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9fe98718dab10a480e0d06236765badb015940d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 19:58:44.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f9d70db3c9e4d7eb8ba5050898fcb1e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f50aab91b7684262964df6f28d052495 : FAILURE in 50m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e69f76060d1c4da8ada51130b3342fc1 : FAILURE in 28m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fea553533c584cc6bee48d2d6d904180 : SUCCESS in 45m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a36edf37b554b0db8a677116c0482c7 : SUCCESS in 27m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ef3bdb9e0a6cc5ac005e8508d483d6b094847036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 20:08:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cb3338b811844e3856f230eba9d7cee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64769e8bef14f99ae34f34e6bd47217 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/896d1dad0c4d4c77bcf780eebaaa300a : SUCCESS in 48m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0feb2c9cfcfb4e2ba9aa118c1d48a814 : SUCCESS in 39m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95abfee6ca974d5fa6f491e90c9516ee : SUCCESS in 44m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd4f7f95f57b44ed91125bed530e97e5 : FAILURE in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca8e2391234e4b5db734511d10518760 : FAILURE in 14m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11cb5f6878624520a5f5b54da014f016 : FAILURE in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/397149322f8a43d7ab4cdb52f9e4eff6 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69e303897eb2460b9d05763977b968d9 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3cb062e80f943759c418e5bb105e913 : FAILURE in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0968e4563b954bb29c778c17515e90f2 : FAILURE in 20m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c31a3017abe460e90892fd20e5f0fa7 : FAILURE in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3356e7aa9594f168edea089c2ff3bf8 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/464ced1e556f46148d0c3cba73df770b : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7046cbb56bac4296b6b4800e76d80814 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf46ec41eaf4496ea8e9f71dd2b2d236 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/000743bc18104ddf84d4d9266db06acf : SUCCESS in 11m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/12d269870ff247eaaaf8d14a7e753643 : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de9db7bb104f40a3b7475de9a1ec5999 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a64e81918bd466ea3f4854585197358 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95a94d4cb0fe425c9a93bcb035ff84be : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d59fbd07d474ca79cf344c363baa27c : FAILURE in 54m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fbb94f647154aaa9e745684ca4953c5 : FAILURE in 52m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b976d371bdb4d0f84a61cd75a83b3af : SUCCESS in 50m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/315e80a5268546849bd8ad2f2f200c6c : SUCCESS in 28m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c12dc41b0d634bf986c2d461cbf7c1e8 : FAILURE in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e864ebab44b445896ce33ded24b51af : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14704f95074b48f889607c2a91b69feb : RETRY_LIMIT in 7m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc9be78ea9274c16a14479416fca840d : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":3},{"id":"48008c5ea6a04c64d5e3cae676e9748b31a91c92","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-10 20:31:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f522de05ae282704963681398b014d89d06239c9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-10 20:32:24.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"6a592a1a6761d5434d9dc28cc2ff27faa3e24a59","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-11 11:45:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32211a9cc52033597219817d4615634265f98d94","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-11 13:31:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6b16fc523bba53867c6228d05502d1285a3cbc69","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-14 18:17:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ff247b9e710bf79342cf683764aa5b6d28fd68be","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-14 19:56:22.000000000","message":"Patch Set 4: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6aa93b2b599302efaf8ce27716e49a007f6ce0c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 20:05:55.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73511c284a74c3ea65fbfb305c56f7e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/148b07feed8842769a6e409ab1b65533 : TIMED_OUT in 1h 31m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55bcbfbca6654116a9800df14e3d2bf0 : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b64366a6cb746c1b5bad8b20aff20ef : TIMED_OUT in 1h 30m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b79a92431eb47f0854fb4953e45e348 : TIMED_OUT in 1h 30m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7808f31da6bd15db61b52197b661aa4261eb9773","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-14 20:25:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12dd2c88aedcf2d2035b2d55b3c862dcd67d96fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-14 20:27:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8f74713d24070a4cbce3106043f3d795f95f1914","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-14 21:54:06.000000000","message":"Patch Set 6:\n\n(9 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"b47940f589745713e18813beef8f6670bdfc21ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 22:04:18.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d195ebc048f43efb7aca25a6b1d1308\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ced449ebbd0a4f0c9fc1abd3123427e5 : TIMED_OUT in 1h 30m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/763b9f22375e4948a5e4048a7cafc568 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b848de2aefe74b1ebd0581bb667cda36 : TIMED_OUT in 1h 30m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8d890f5dd67404ba015e52a38ef7095 : TIMED_OUT in 1h 31m 11s","accounts_in_message":[],"_revision_number":6},{"id":"fe40bdefcf2606b4bbd814238791d9c4b263bb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 22:57:34.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/432e6d4c6cd44904a7be2eced2f23ac6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f5851bb8a8483dbf0f0ad2b444a94c : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/57893b6232404105a62f276b295743e3 : FAILURE in 34m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aebdbdb8a5b54b4da9f084a91e53684a : FAILURE in 21m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/096463af4a664254867b698c9466c8db : POST_FAILURE in 18m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ec60bfd8e6a4b278fa10491d2379739 : TIMED_OUT in 1h 31m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d4d688d2b074b10ac9d4cd2937823ef : TIMED_OUT in 1h 31m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b85b460f91b4ad7927b97035f09aa34 : TIMED_OUT in 1h 31m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/006f7ee057834c59a32bf8af427a2a8b : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9cffae790f094939950af5ef50ae65ab : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/656a90c378804edca17f304a6c913d7e : TIMED_OUT in 1h 30m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44b49610ce1d4fdabe971269b58662f3 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/993a0f75b0c94ade953b4af1c9b7a4c1 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06082813c27b4d7aac145b74986d8afe : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82129c5273a4499aa6ed0125f5e9180c : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be3bd174b07f48088ea01f93ed8467b3 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/246a144658954429ab6f8c7b5d7b8485 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2a3016690b264d9b8721851a3607c801 : TIMED_OUT in 1h 30m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b6a733045594a0d91fd82c9d888f624 : TIMED_OUT in 1h 30m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd30bba8fcfd4923adfd55c4e29e835e : FAILURE in 8m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d124d048aa3242fc8fd5241159da678d : FAILURE in 6m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1346404f8b824d09ae4e67015704c122 : FAILURE in 10m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/accd358a637b44c18811e48f95060277 : TIMED_OUT in 2h 01m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/07ce9adf8dec4983b72f6b3d22eea321 : TIMED_OUT in 2h 00m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/253141d09716436896a9024827d3eb1f : TIMED_OUT in 1h 31m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b34d24f2b24d4274bd09bd2ded95fbae : TIMED_OUT in 1h 32m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a8f40ec0a7947bfa3b2943a350c2166 : TIMED_OUT in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cd1de890394c8cb6ea6a46e516daa9 : FAILURE in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8116b34fde2943208b174fa14e4bcfdb : SUCCESS in 15m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e3d5daf70804e7998d2d384f841e9b3 : FAILURE in 23m 06s","accounts_in_message":[],"_revision_number":6},{"id":"2b53a87ff52734d83d55011839e8cf51ddf9a034","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-14 23:13:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"23c9044f132ac6926cb147191f6fb8f0af352f4b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-17 09:05:50.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ee5b272eca8e87920da619764537d8e798936aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-17 15:31:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1e0b745dbe66b4f0d93bdb9fd24462ea2c093343","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-17 15:32:47.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a6816f03da163669e6deb0c6fe2a3a6f0def9ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 16:39:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b0ab367f7504e06a16f10d33e82d0c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fccfc3f5c974b4992eb32e068ebd4ec : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/d464ecc9f8ba4577b5c1cc8116ebfcb7 : SUCCESS in 37m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2bb24cfdf15b46c6bb6694d62d2acbf4 : SUCCESS in 32m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/137a873441b74006a059cafd08ca6ee5 : SUCCESS in 43m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a4dfb3b09d34f12bbb5ea86b782a541 : FAILURE in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e8a5082dafa4459ba46a982971dc8e7 : FAILURE in 22m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d1fda666c3b42e68b3426dc3fd3590c : FAILURE in 17m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fe7635510ed476189856fe7234b4b53 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ccd069c4fe894d56b3e77aafdef399db : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ddff436d9fd41f59b925a3b3b137436 : FAILURE in 4m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/288094b4da8444ef85a18ca4a52b45f2 : FAILURE in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/192320f21bc846d9b28c8fd0ec5244fa : FAILURE in 19m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e4c57f5c53b4b0dafd6c68426b02937 : FAILURE in 4m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d45ac04ed2994944a81d8f09ef846459 : FAILURE in 4m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a1003438fe6f47fa9ad9bf4ca0214b63 : FAILURE in 4m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65861ee2d9684d4a8a657a7a6702224b : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12bfdf39595d4e2e9266430c96ad6c38 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/056069d0ff604a01a9fdf84159d7ce6e : SUCCESS in 18m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/443fe8c113dc49d48c772db6eea9d2b9 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1c367891e5a43819ce3ed4ea88e3003 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b4119cb3a6dd43698a6b8a2ed851b6ca : SUCCESS in 6m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4fc0c9c994c14dac8285d098c9cc8a80 : FAILURE in 4m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1970f34618e42a8825d603c72ee6df5 : FAILURE in 1h 06m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fbf3fdadf2804c73be1a3ec03619b12f : SUCCESS in 33m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47733f53d28b40128c30fd1fee5b31df : SUCCESS in 43m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d5b4eaa47f14ab8940ee749a2749609 : FAILURE in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f92e99872c674caf8a323fb44b1c9019 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f26e13ed7aa64cc183da15fa77f0005f : SUCCESS in 24m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63cbc68694694e04bcfbe0d00487a69c : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":7},{"id":"cd5c20645d9f35ca9aeb5dca2fddd47d4074b746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 16:58:17.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35f822b47ccb4741b4255506ac17563b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee5853a5fc3048b7b7acbb9e1307772b : FAILURE in 1h 24m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4319e5c2dc724461837c981761ee8c77 : FAILURE in 51m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f3dc62e09f946d58fe6275abc07c803 : SUCCESS in 1h 12m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/79f68783541249c8a2cce3d730e8b5b6 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":7},{"id":"e4a173ec60df30d22af71b0d5afd2475bb6a1d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-17 17:19:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0f165379bdba5089fd3df99321c1a98cac5e57f0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-17 17:47:46.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"daee8df2cf147736ce1bc491cc73259516e9058f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:21:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2190ed2b2c8548ff887de6be89653cef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075849d58de34d58b9a12889a4713d75 : SUCCESS in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/55e92680d47e40f8aa9807cb21979261 : SUCCESS in 41m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/592d1f6894ec409b991f013250e8d523 : SUCCESS in 39m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/099a85ba8fa54e8db33de1bc554ebc80 : SUCCESS in 40m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3f40897d19f48c9ab843898b5b84c2d : FAILURE in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87ae8f03c27e4c86854b68743a85e852 : FAILURE in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d81fcc0184944b0d98af41c84d6f15a3 : FAILURE in 18m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0df6d5726484dd987033be0c5cdaaee : SUCCESS in 7m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89023b4861364a53bc66788f9700d781 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93ccb6f8942d470e8022b0207658b12b : FAILURE in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a4d54a82912475e897f8af2380ee285 : FAILURE in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2bd1c9ae74944de38dd773fc99ac10df : FAILURE in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f587fccf1ff54aa49596c726a8fe7fc0 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d07daf2edb945f595994916b3e2e4c1 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/23550658b5cb405ba984bbed306fc1cc : SUCCESS in 21m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ea93759037a49c9b57274f3a36dc1db : SUCCESS in 20m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b00a087c51074d68920dd1aa61eaa575 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa2ed511f264405bb73b48787591479f : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13861e3d406a478dbe4c61df55936748 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66901a8bf4c24a51966da84036f441fc : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/74f331c8aedb47348f0534d35dc5af35 : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b99b06dd76904606a54b716ad0033b5b : FAILURE in 52m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7369eafb27e24a1f9687161e1e475490 : FAILURE in 55m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3b4f549bcfc4a3c80255f21e152e6ff : SUCCESS in 42m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77f09f9d3bd64650bf0364e65212715f : SUCCESS in 41m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5545077b79404a4388ddcd53354695fe : FAILURE in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b571152dca947b786ccfcb1422efb59 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83d80789c38448f29b83fc5b8d45d7dd : SUCCESS in 26m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9029bef5d6284e74b3d5c84fca7cc2b7 : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":8},{"id":"91e3b9f28c776feb6f454a74e3cb4a2c58466243","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:32:12.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c79e4786ed3414e9230ca9306126282\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c83666294aa4d63ab84c3eefa75e028 : FAILURE in 33m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97b8676746fc40cb9c3347495c280734 : FAILURE in 52m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c75c803033c0427d9546f9d9c7b13a72 : SUCCESS in 1h 10m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/54a7c5857a6847168272c8bfdc9eb54b : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":8},{"id":"d2991f3bb5960b424596c9ad9f151ea463d488b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-17 18:58:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"840108f52929549b93cf4201f9f0459cd739c2ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-17 19:03:10.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"e1002f37af64c0e90843f5573cc0cd056362a779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-17 19:05:48.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7531dc3de156b52749f221aaadb4ffbe1fb456f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-17 19:06:43.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"7ffd9e9efe130ca5d03cfdacbdddf70e1f2a1551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-17 19:18:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9b1682e65f85a454d3d5686cf696b5a3e41db8d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 19:52:36.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87ed5ac8f9d84ec7a117fe745a5e43f4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca33da26c04f41e2b9f014146a459539 : SUCCESS in 32m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7817d5ed625b4e068f8284b430fa5ddb : SUCCESS in 33m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4424ad667f6c4446815fcec3479782fc : SUCCESS in 33m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f04b14c9f74a4de78af90dc9badd72c7 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":13},{"id":"461a119178ad38226dae571ab09c946c285a3e0c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-10-17 19:57:25.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"aa47ca838b5f5322d9eb5b6cb5dcad84ba491039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 20:23:48.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25690c2f94c24cab9200ce4cc86b617f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecca0a214a1441f3b8c1ba76f5d66fc0 : SUCCESS in 7m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/a5ac74e48fcd4770be58b0dba4c49494 : SUCCESS in 40m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c916f5dada694992812f4a3d336608d0 : SUCCESS in 34m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df3be10a2a914344b07489bd44c19e08 : SUCCESS in 37m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52c4bd08450f47f7857db95a2c480d7e : FAILURE in 15m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73a1fd15698840c484d8731f6c7c7974 : FAILURE in 16m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f8c56b1fbbe493a9f5e657f18273891 : FAILURE in 16m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a3ad3b6cf5c4fe48f9479cf10bf6445 : SUCCESS in 8m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5720ce1f39ce4469b1c29f60cc3c7cc8 : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/22017800dde2472e9b1b28b3d5636737 : SUCCESS in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bce3ed36513e44328ae8b82961f23947 : SUCCESS in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb41923117674c859c4e0ecd0a553309 : SUCCESS in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ab4408e56c44a2aa59d7687cae18439 : SUCCESS in 12m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7aa79896a3b5455cbd22f0e57ebc0f20 : SUCCESS in 13m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7be9367f8b874e4e9d367bb46b918d1d : SUCCESS in 14m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/62b8fed961d34be6bfd6fb534b2b0bce : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/08c5ff8f0f3f48f0b73fbf42b664f296 : SUCCESS in 12m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f0d6289f5ee94566a54bba6868aa7def : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27350b9aef844b129c6464404e596b1c : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e68518d0464f4d34b6a67329d3f41f13 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f780a85f6d64349b7ce150ece2dd58c : SUCCESS in 6m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fedc68a2cf724af283cfd70b8df6c80f : FAILURE in 40m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/35a53762bcb14ea4a514cf8e7ca5451b : FAILURE in 53m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26d13916d9664aea8bd4929b5eba9865 : SUCCESS in 45m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c079068693a438ea48acfa3c97f83ad : SUCCESS in 41m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d1eada8f0945fca0d466796030799f : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47227f2f622a4bb582ec24607b16e118 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b715a5e2da6f4f818c3cd09a9f9f119d : SUCCESS in 20m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d97fed6bdc846cd9b3d16fd7e85528b : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":13},{"id":"566a2aa121fc3295b1abf8c76dfb31e822d438e0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-18 14:14:37.000000000","message":"Patch Set 13: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0921829c5230bb41f59f57391037ef6099a1b326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-19 18:45:31.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"64b0ee4bc440e6217023d4824993c5ae09d48f13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-19 18:47:26.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"3ad2b3eb778780aa0356db323ebeb6b529f3543f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-19 19:19:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"655f5eb8dd25b5ef3dc96f4887c1877864c6d0c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-19 19:24:49.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ab9d46bd9b1c27975106d2a083fb79eb0b637602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 20:37:24.000000000","message":"Patch Set 17: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dff42a150d9649c984ff7b6e1b129118\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1342030838d41dd8f369f5f10428df1 : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/99807e6d2c79432fa114cb4e983824cb : SUCCESS in 38m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9078a24743a4c3eb126a05bacb3258a : SUCCESS in 38m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d88fb913eef34203b6e47a708818c36a : SUCCESS in 41m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbe7dc07a2db48869f7f3ac0b8195979 : FAILURE in 14m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0acabf1daf14f0e882433918d6a3825 : FAILURE in 14m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7342235aa943457a9986fe930b6cd7af : FAILURE in 15m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79c1410c483f44ae93335a0fa60acee0 : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1cc8ccae76224e6c847cf8d5b665f281 : SUCCESS in 4m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ed70bb70ffc4c94a8837dad271e5c78 : FAILURE in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a435e7eaeee44d3b382f118fb131357 : FAILURE in 12m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99250ba2db2c492abef8b0d50f7d2b00 : FAILURE in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f1ff69665b504fd8bc32723cddb94bbb : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea0ab763bd774303931e2a1943583dea : SUCCESS in 15m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb7d9dbc0b414ca3987c67caf3ba52b2 : SUCCESS in 17m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9722fcd4dbb34d998ad90ad24aad075e : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cbcdd0adf1e6416aa54db9dd03360644 : SUCCESS in 12m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d78f4bf3f6745dea51afa504d8b37e3 : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bacf0b25c5624456a4cf2c7121a09931 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e17c1ba6119460a99b5b0ae0d94570d : SUCCESS in 13m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b54944ee8424b479b8fb5ac2519a54a : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/38d1eb83730747a9bfcffb654eb872fe : SUCCESS in 58m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5365d68785794e368649cd8fff6aeeff : SUCCESS in 45m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fa074a916de47a1bc69a57c86317940 : SUCCESS in 51m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f52f658dd30249df91782e4f49affcfc : SUCCESS in 43m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af88058f8001402bb18d1b2d633b7ba1 : FAILURE in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975b0e0253054b00a1619591d9e35fcd : FAILURE in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49bd8567ba9a4537b08a70e75b9c1b86 : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ef14354589d4c25bb9226af78f40cb3 : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":17},{"id":"a4c5507cc8c03a1b151fe41b2c9740fd8265e70b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 20:58:10.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1dd3a0254f548a1909fd65a66e2288c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4201d3e462c4267a795616657fe30a4 : FAILURE in 46m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbe82ef3544949769b3bfe3481524008 : FAILURE in 45m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e91556f555974c9e8cfbc97c53e81efa : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/520d0d7d5bb845508937a48e60e2cc65 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":17},{"id":"b15546ddb464575abf1bb51087969f6a665332d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-27 01:59:33.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"c6122ebce6648c437d74fa01a5eaa7778cf2f6cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-27 02:53:42.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"7b829e7df23a580f160d58a3cac419ab20a679f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-27 03:25:40.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ad145ace9f11ea2595d5ee7e0745867082b12484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 04:23:59.000000000","message":"Patch Set 20: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/462796f2f35a4417a3cb9e4180261d98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf000c5ea8b6466192b9273aa4fc52ab : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/d030e9dbe02140639f60adf77527f581 : SUCCESS in 41m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94a434e982bd4b3d8fdaf39e6645469f : SUCCESS in 36m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1fc92484cfb4f95a182e23d7580c92d : SUCCESS in 38m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f280f0443c804a09a94b31fa56107b19 : FAILURE in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31608f9452de4eb48156632e455c4ed6 : FAILURE in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b41c5663366948339350a1f4572f795c : FAILURE in 22m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f674a6941d9747a3a2ac9d1a5b979a61 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1d5d7ebf4044b41b717e8fbf2cafe1c : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94ded0a00bba492d9912ab9f666d2d4a : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/028aaa583516473fa82c45772ba07fb2 : SUCCESS in 13m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/405e515cc1934d09b456bea3fd7738d1 : SUCCESS in 12m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/915574bec20748e48dc1e541bfc91596 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e7189db146024d7aa29bfe5aee026ee0 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0725db12de94065a364a154cbbf0ad3 : SUCCESS in 23m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c05a5cc75024c5ca39c12fa64b1fa8a : SUCCESS in 24m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d3d3bf08655e46a9a0d7a2bf3cd276f3 : SUCCESS in 14m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ad194918d2754de2aceb14ad63b28026 : SUCCESS in 13m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e9a48f3bccb449249f3ff90b996e8be3 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/337b7eaa37d341aba72cc6bec3d6ff87 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2c7e65f2eadf4f48b2c080f3e17a2b9a : SUCCESS in 13m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/788bf55af82a4799885d92c6746fc289 : RETRY_LIMIT in 2m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c8e793e8d2b498ab9e0e9ec0061f541 : SUCCESS in 54m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76a425149c874390a746bf611d9e03a6 : SUCCESS in 51m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/237b4451b6e047bd91a29e6f2dc8f04c : SUCCESS in 31m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39cead0ad1c645608a2e23dd7a679a86 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d370fd20df3e4e7c95a39b8897110e4c : FAILURE in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb0bab8986d84c61a3784006628c1a3f : SUCCESS in 23m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a52232d15834b8b82a0ca4e687e5bba : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":20},{"id":"7f7596b59d30bbf69e605ab2337d7618a61a5554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 04:52:48.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59e783d7376647678c374fb135741cec\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15c90b0bfd034544b73c9758b04057f3 : SUCCESS in 1h 15m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73f356ad98c54c618ca0fad81368acf0 : SUCCESS in 1h 12m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a5add6f34d14a6095a302f1dbc9df62 : SUCCESS in 1h 15m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a14c76effaaa44f78f531b418b784f51 : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":20},{"id":"49c38f44e870bb4469a5874bf744a681ee476a98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-27 16:33:10.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"55ed7af54efd4fa6ab40819ff8f6f4e2100aeda4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:32:34.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634cee6a513040c0b3ebaf8f8b599654\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da7ef04a970d400085819c9cb6b44ed8 : SUCCESS in 52m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4918ea0be244efc870002b6161d6787 : SUCCESS in 51m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/10c56570b935480d8bc4102d3800e356 : SUCCESS in 52m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d12e2c7f9fd240b1802e340cbcaddd99 : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":21},{"id":"c228a790e8b90b8ae2ba2ee0296c429c975e0aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:35:40.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/206ffde5613c4149b6f92434be9cb19e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e731c8171714398be3f2438996e1e03 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/83cd867c047d44479b74d911bf3ff410 : SUCCESS in 44m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/840e4095282440ac80cdd6ed016ec2f9 : SUCCESS in 40m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/747c52ebb8fd4dbabe4d7b7a1fae4461 : SUCCESS in 40m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/684cf7d4832b46659126bec2e3fce683 : FAILURE in 16m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/139abedd63d24a5ba8b48523cf2293a7 : FAILURE in 21m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67238e805467475d873c7cfbfca4b0a8 : FAILURE in 16m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79ae9caa7d1f4f7cbb56e739e329b9a2 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a48cd9cf87a94f0da92c650a011fca65 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e40282fdada4362be38c98f9e0d72e2 : SUCCESS in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9cf46bff2e8e482fa4da91f1facef680 : SUCCESS in 13m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/756a8769d59e457b81e21ca06b4b35d9 : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a730945b5cb54a989af8c55740492fcb : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7c872780b07461dac65df049607fef6 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/002a3f83d3c44364afce3d4e58601bae : SUCCESS in 20m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5403e65af27744a0b960570b52897b99 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/08f0b7a89a8449feaebc557813e5181a : SUCCESS in 16m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a503173a6538489eaee4b67c715f1622 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/962ea26e52f14b76b0805e1d55310d3e : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d78c0a7f51f4a8da3e6048ea3def782 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0501b4d79354424b7f11b0f7c7de4fb : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94f63c70c01d4fb7b79f07108e8b71ae : SUCCESS in 43m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b185f79a530f415ea418c1b87531c9a3 : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6440219c4712424d89fd841c4f675f99 : SUCCESS in 41m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85d1907f8a8f446d8cf72021b11f5384 : SUCCESS in 41m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e2d5301cd8845d38deefa7ac019f2e4 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6585b05465e4e5e8d7f60b1cd17cff8 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72ecdb8178634d088fe58185b9b12cae : SUCCESS in 18m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c8b66926be3495abd8a6487404c70fc : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":21},{"id":"d996f7e2b71bd89109849025f4d5553078bca4ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-10-28 20:10:19.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"1c1f8010eece4eb4059d87535c1361b31736e644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 21:08:27.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c15274daf34dabac151284ea80f45e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a02d3da248a4ddfa332690ed410b8ea : SUCCESS in 55m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2a6cbcc27a246b2bd471529ea6e1b16 : SUCCESS in 54m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d44bbe9fe80944bf9ce9dcce8e8655db : SUCCESS in 28m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/88d94376c0e741008db8965d632e2584 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":22},{"id":"f10db16902430d0f0b4cbaeb29bd5d7dc7f017ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 21:17:45.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80c21807b6294b5db05425f7304e2bb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed9d1ab252d4a1392826034e1cc7efd : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/3ed0fc71ac2c4ac8aad6f301ec7980e2 : SUCCESS in 40m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce1591ad6ee44feeb5f457e20354f58a : SUCCESS in 53m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4a305c88faa4989aa2bd8feb731d5f7 : SUCCESS in 54m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/049a17205e8c49a6bf50628ef4b0c2a3 : FAILURE in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b2a8f2a98764c008d0a3f992cfaf0d1 : FAILURE in 19m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f21fbe963e84698a60f522c0541ec5e : FAILURE in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6904892ec12242f6b82ed4a86c752159 : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7360344aa2ab453a9d37831592b83c3c : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/abf9345d07a948e8b0b9955146244d3e : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0af2163b5b9f4fa6985e5383cfbaacc1 : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/90776cb2e26248418a7e957a13fb5cd3 : SUCCESS in 15m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad369bbd1bac44c6adf75a79a434f443 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/842fc6e8198f4da2abc0d376ae717907 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79994199533e41af9c334c436f6637a0 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99af7210faa2460eaf9c18909cf5b6f7 : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4aa158ffe0e24638bb99e503db7a6131 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/733c972d42eb436c9ca790ee94b1bc6c : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a544b142b674e0ba02ad2436eafff33 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f72c0f48d114cc68a9e640b5313232f : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a4532e9be25a48a8ad94edf4bae3cabc : SUCCESS in 9m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa120338ebc049219156d93c24cb170f : SUCCESS in 54m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/062002a533784a2ca610202a2d52c47a : SUCCESS in 1h 06m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec331a63c3dc46d8944f84d76c9cf8c7 : SUCCESS in 35m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab0b9a583c9142bd832112b1b46315a1 : SUCCESS in 41m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27b2f63ec63a4202b1b4a3247af22c20 : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a9c0fc25d748efadb028f8d8056c7e : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6aafa97aa7cd4d1a929b08cc1f3e1d6b : SUCCESS in 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76cf3a5b2c104f86acf5a80f86f5ba5a : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":22},{"id":"4f4f7da7fdaab6df3cba727caf1270b66fd01437","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-02 20:59:16.000000000","message":"Patch Set 22: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":22},{"id":"719bbb9661d09a3f1dce902d3d5620a0bfef1cfc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-02 21:00:44.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"a7fa786b23e8430c51a8ed2d5a1bb582ee5344ca","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-02 21:01:59.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"7d7b308ccc384f6cafcb23e74d10d3a113689afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-15 18:52:04.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"d1955e3966a677cae63af95c5541d5a2e2147f36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 19:43:24.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b51863bb004c5c958008ca47d7c962\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac6fdf39077e4f12bccfccd8f2c36107 : SUCCESS in 40m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0332d34cc1c4a5a82eacdc98e61c0e7 : SUCCESS in 47m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/463013389f9942f6b0e9fea8c1083f27 : SUCCESS in 40m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a4ff5a5588f4aac97835fcbc29d7f8c : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":23},{"id":"28d28d9c689ab258f2388a8ec44e4ff2c6b2392d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 20:03:53.000000000","message":"Patch Set 23: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c1da556c5b94e24a2501dac899ed42f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e848f8cc8740b3b0eaf7f4be307e65 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/75910912d37d420cb886220182cffc13 : SUCCESS in 42m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/883062587f3044f698dcb98787779301 : SUCCESS in 36m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e68aae96c1a4fbbabdeb968a4b46128 : SUCCESS in 41m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/817df44729a04104bcf7983e39c58b04 : FAILURE in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0f15253b49543c48d50f930eefce11c : FAILURE in 15m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1883d0173e4a4572ba71c57f6d2d00b0 : FAILURE in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6773054a9bef4c0e996d99172c5df35e : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8898d747750a4dc4a1613585d55c59b3 : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/715a73e7c55249a087c9d4fcf680c401 : SUCCESS in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f533632ec0e4b959d479700a904a44c : SUCCESS in 16m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c073893752134d66bcaa73b43c739014 : SUCCESS in 15m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1c04a500133476e9590d1d261ee4912 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/826915f2f189438b8c31d8ce34a37528 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f49f9183f7a4779a5cb9b8f072a73a2 : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/214d8e3fb448432cb007477249b8297d : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4f70dc13db0a424aac0d29e17bab39fb : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/da9ab0e5f64f4a73ae8293831c963024 : SUCCESS in 18m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b09cac1c0484a05bd7ab85db7e0aa50 : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3a3c38415b94aa5b51bb41613769223 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb21570a902e4d429666f428bec9f2e5 : SUCCESS in 8m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a8dcf2dd68b949d0a9079c7b52103d2c : SUCCESS in 55m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f65c408248f4a5889a8b781fa7f61a4 : SUCCESS in 1h 06m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c764453045c34b0fba06e149cb618d81 : SUCCESS in 51m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba2d3e4f51cf48efa44b9c3f5be57708 : SUCCESS in 46m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/343a9da07a8e43f1ad13ec9c61a2409a : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0057fe8ec0aa4b2789cd0e6c5f2cf7cd : FAILURE in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b67a539a72846f488d8796b662c6c23 : FAILURE in 19m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a5b9f1374f94b55af5477cbd8b05e9b : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":23},{"id":"50e8838a0fdd4cc23a0b34b9e09f842fa2ceeb69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-16 19:33:09.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"189fe17f4a97fd54961153f13b7c0a48d8e8f1fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-16 19:46:51.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"1983657c5e961ba350522eb47b214e31d72b869f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 20:36:53.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e57092bde1e47d4b8ba3eaea684be41\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5d9930f61d44dca95375e87884740e2 : SUCCESS in 29m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5345f16cc0a475d983a07522a9ad3ca : SUCCESS in 29m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/53609ae5f6ed431b97aaaee819caf82b : SUCCESS in 29m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e80e73de9e504284b1296c7a7832ec7d : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":25},{"id":"13b9726c6b9cd35f1dea6b6f57ff88acff21c9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 20:59:39.000000000","message":"Patch Set 25: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a3eb2361fb6478f893487f488421c06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff8370424284a4db85bf73a8b4ddb81 : SUCCESS in 6m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0bd7ba7efa1547c8bc8ce8da325795be : SUCCESS in 49m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d8518c887c14f8995eb0f78a44dd0eb : SUCCESS in 40m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6b7ae1ac0eb42bd913efc9556f41b84 : SUCCESS in 39m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edfcefa00064438489f8ec469e2b1a41 : FAILURE in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2649029898204dc59c8829288729dfb0 : FAILURE in 23m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfdd09500d7e494e847647a907bc3d5a : FAILURE in 23m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55841896b0704d8ca9cf08e93829ec15 : SUCCESS in 7m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fde2d0bc82aa422797bbbdf530278308 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9adab3e848041cda6657fb9ce09cf7b : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0aa3a40766784ec0a439564967efbab3 : SUCCESS in 12m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d0462c33b5ba4f5aa17b891e72d85cd8 : SUCCESS in 16m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/045fbbed9d84436987aae8a04335d82c : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbaced14ea4542008241c3e3a46114d0 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5dadce0dcbbf40199f6087f8bbc85685 : SUCCESS in 20m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/299f290eaafe4a6da81f103166e44495 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ecc7ab93807f4c2093dff870688322e4 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b5778ce73e84d899fc3bfb6a7ea9c55 : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c4c16a72abf4a6281400c58973ae8df : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/155cb96136c145a091266d8c8cfc4627 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96cd0cdc8a4b4fe4858404f335d169ea : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cb751f1499e34586b6ba408fd1b6b1ac : SUCCESS in 57m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e026283c7da4361a2134e53b1a9336b : SUCCESS in 58m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1cfda29a4b84ff983d0e03471f24c5a : SUCCESS in 30m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68001bd76f1f412b98175a1b5cd3bc7c : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b8d1b3dd7f84775ada0cb69a0be0e74 : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa81b254fe9f4efb83e103fdad0e1779 : FAILURE in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b62f1f17ed8b47848391a2a173db6c89 : FAILURE in 26m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e58934155e5a430dafe66c2b343130ab : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":25},{"id":"c2e590dab9f4e9dcfb72efe1c1ac7fe2f9e6933f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-17 23:39:45.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"d73881c523d6ba453ba52991566f8dee96ea2e6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 00:19:59.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71620e0114fb48cda31e2a4954cedf69\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7a2e3e157fe4e23af4fb31528ee5c0c : SUCCESS in 32m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40fa06544e2f4251af06e8a7cf5f1d30 : SUCCESS in 33m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5980b4b3d7a54b0bacfe8a97886d8c65 : SUCCESS in 31m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e80e86b0ec6f464eacfec1b0deb7d423 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":26},{"id":"36eaa1103dfff6ebe48352b18a9be40a1fb21087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 00:44:18.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4259e71712e54d5690f823cc4be3c72e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae1c51fc435b47679c1f88b82b732793 : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/371ca47c464b4898ba5318b25cd61895 : SUCCESS in 47m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f1b1abbb734404a8a18fc4ce9497f48 : SUCCESS in 42m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66e59ec442584bfa9d73e34ce1052aa7 : SUCCESS in 40m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d25728cbbf14d108322c52a8fbf0da6 : FAILURE in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04443e70c00148de9b7935c460ca381a : FAILURE in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e400f9839c6b410cae87f6bf88b991cb : FAILURE in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f626100b76284566b96f75d025ffcf4c : SUCCESS in 7m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a5ac789d5e944a09c16f8e5e9929a06 : SUCCESS in 4m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ccdd31b049cd4c35a00fe81ed9cbe085 : SUCCESS in 14m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e859c06b006467eaf9f213ac3dc6549 : SUCCESS in 15m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b00d71351d6f412abfb043d385ea70df : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/657a30539a684193be09018586954a75 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43a788ef9bc04b4fb6eac277bc03d6b3 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ba5c4fe01084c7c98e47375ff41f7b8 : SUCCESS in 20m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e7e7cb7b2125482fae8c29e83aa1acd3 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/000468c22a9c4251b376eb4fa9c6a553 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d823188682544498fc43fc5efcdb296 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19060ed7db6e4d389c8f3de375d6525b : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4954458bb0464a5ab8e992979740bc1c : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0247367f74274ec5a79867f531eb5418 : SUCCESS in 6m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9029253351cc4573a0e3c17af11d93d6 : SUCCESS in 1h 01m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/21b19b9b62e74a7486818060ff4fa854 : SUCCESS in 54m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/688535014a5e47dda0e4cffe8777f639 : SUCCESS in 43m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/326a6afd0db04e8b8aa5acfa12ca3a88 : SUCCESS in 40m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16a721f435b4416a889353607e3821c5 : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7c3a33754440a9a2ee04afd0083676 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5777c0c4538e4a8e95df9474d1ac4afa : FAILURE in 25m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9df335a3f6b6468cba7dfcb819df8fda : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":26},{"id":"0eaf6f101f83ffc5c273761a48715b64515df861","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-18 16:58:31.000000000","message":"Patch Set 26: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":26},{"id":"7ba207bc40552086f450d66323664bb35f0eec33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-21 17:37:48.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"0548933dd7e6a3c8d51057f911f0b35bd72cce45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 18:46:06.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba8e92143b8446aa7805acdfed50813\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1caed21cd5ca4570898e185e5f6130ce : SUCCESS in 29m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ba8164cb6774c68b363756c6b379068 : SUCCESS in 58m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/031320468f28450a92c2216982384216 : SUCCESS in 1h 03m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/019dfd55cbd54d549ad880ae25d6da2f : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":27},{"id":"1097dc4cff059d6c23e6f5a5a84f22f0a396f3f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 18:53:25.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a76df0cb319848f9a36a1cc94ce565a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723f9ac6a131404dae034fc66d8a6727 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a145bb21bfc14d7298c3e297d92df02a : SUCCESS in 55m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec8d5e3f44e940a1a20b41a1e4aa3692 : SUCCESS in 54m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cf8ab811af54cd4b38579dcc002a677 : SUCCESS in 1h 09m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc6b990aca5d434abd08df53d4368dc0 : FAILURE in 17m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4cf5ddee09e4d5ba1bc29f67650a8a1 : FAILURE in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ecc91bbc3c44927915e23656b1f4eda : FAILURE in 17m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce2bc7104265422bbf5e44b64a61ec8e : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5d65ae6c90547c49bb5632eaa1b8b1b : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d34143e648247b7950da9fb9ad78fd6 : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dea461355e8e4cb796e3c198d4a2558e : SUCCESS in 14m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4fe711ac27c0407889ddff0b5085adae : SUCCESS in 13m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75dd0952784e4c73a1192155423d2b8d : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13bddac7a80d4e0787df698610271422 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d362c38648146e5869963a0e4d01c8c : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7782fad2c76a4af3867b0241972a0753 : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b6aab2da3704d758767a31119db97cb : SUCCESS in 14m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f5a2edd5a7749a188a54823216d557f : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd30a926ea444cc695d907dabad667f5 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a535edb21d141d69e7672b67d0b9a89 : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6658cb8ad28e4c48aba8c91991d15559 : SUCCESS in 6m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96d3a9de09ec417cb761c4f1da539a52 : SUCCESS in 49m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/765a83575f454b5a8809ca75f9eecaf2 : SUCCESS in 1h 08m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf3d870ab3f641f29ed0c5cae2cff4e2 : SUCCESS in 43m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a58a8363ec144c198ab4adc196969979 : SUCCESS in 42m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/703fce8b4167425da5d4a333c729485b : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bd9175717a457aba2dc405d24f2f29 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f51ed759e99c45c9b4c26b28b0503cfd : FAILURE in 23m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94d878e1031a493386a868bcbc7c199f : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":27},{"id":"8d5a74b6ae43621ddb6822e043993c823b90726a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-28 10:52:26.000000000","message":"Patch Set 27: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":27},{"id":"69f1e89937b67db22ff8a221de61e8caa66e8f72","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-28 23:50:43.000000000","message":"Patch Set 27: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":27},{"id":"9894f9e5a3e3ec4371d01f324699f695edaa653b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-30 02:33:45.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"f92da1055a7bc2c4a8ccb818c0fb60d087314dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 03:53:31.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26183fafe0c4416981b553fd43028d35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f7939e6fb248d594d57808cebdffa4 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/e136ebac15aa4654a764cb088498a440 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1fc5234dfed4f9687ebbb1c5ae91cb0 : SUCCESS in 49m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80ac6ca11b9f4f2dbdac63a81915087b : SUCCESS in 1h 01m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2eee6eeb4634474a88f6af8c5894fce6 : FAILURE in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65289ea532234591883fbda32477ffc3 : FAILURE in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9797c5078a5b44f990ebb9084bf7d4ad : FAILURE in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c07816af30cc40e39fcf7b72114ecd38 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/efd259586fba4f1d96b1bc39d04e0fbb : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6d9e8632bc842f89ce12baa8be3f27a : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21ef86dbb008484083f76abfa1f760d5 : SUCCESS in 16m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab7a425976654bd2ae15b3af1634f324 : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6157ca28c97944c7b1d68d263d3eff35 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1442585cac3d4f6480b2a6716c49ceb5 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4aea4722209d4512a2180c760c998cbd : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/37b963157a964f7cab2c8bf1d449aba6 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/72b5c06e05e34361b32d5e998e1a71a2 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e601db30ae674755bfba8d7ce1d18de3 : SUCCESS in 16m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b50f8f63a634878974d98f659a6ef83 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49153cb7ddda436a810bc7d8fdfcf551 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/78c6df5df776496f98e5896b018c08b4 : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d785c5af431401c93be6e5a15248cac : SUCCESS in 58m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/12158403b0844c5a90b695de2eea895c : SUCCESS in 1h 07m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb61759c202649a4bcf3aa6a22a2741b : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1a7009634c44fbba6307b24d37d67d4 : SUCCESS in 43m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a524ad394452479e8f2a0fc26a924b78 : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7b585b02d2438aafffd7963d59f7f1 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b77caebe90a94dff9ccd6716aa558f9f : FAILURE in 19m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ed259d711554b5bb280ee6013faa1e5 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":28},{"id":"5dbe388db5f7c223fe746f6c8382f4e45204ae63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 04:19:49.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83040e6e83ac47b6b073fbe6caf82399\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a0f54e617d54f76ae8a2f491b413696 : TIMED_OUT in 1h 31m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb9f84bece3c4662a52092535f6ece1c : TIMED_OUT in 1h 31m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76db519e57524ad095dad2029d6a4319 : SUCCESS in 1h 30m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0acf24cca8b04be4ab56d40082255f1c : TIMED_OUT in 1h 30m 54s","accounts_in_message":[],"_revision_number":28},{"id":"dce40bbc4e5f7ee7ea88474b319abfc30e89fd47","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 17:53:55.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"add7c332ffb427d8d94b8c95de88ea589c5cde70","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-01-16 01:26:50.000000000","message":"Patch Set 28: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"89b66df614c51edad43a8b080ab5f56c1ae19955","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-16 10:19:54.000000000","message":"Patch Set 28:\n\n(5 comments)","accounts_in_message":[],"_revision_number":28},{"id":"4dd7e18b5c25d5bae38f6c27cfced1282cfc4bb4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-20 15:27:03.000000000","message":"Patch Set 28: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":28},{"id":"ba1f4f9d0061b08a084897b66218defc09a31db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-03 20:01:21.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"c288efe27997840c2d16b2f17fe2e1022e5c564d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-03 20:12:51.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"66d139e67eca6099c3b4faa54e8187c8f2230ad6","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-03 20:51:17.000000000","message":"Patch Set 30: Code-Review+1\n\n(23 comments)","accounts_in_message":[],"_revision_number":30},{"id":"004c7919f8e0ff380cc20b4bfae972d6d7e9a5b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 21:10:00.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3815f0ff05874090ba05edbda6f2b3cb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/995f2031cf6844449004c3c44e099811 : SUCCESS in 13m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e57ad4c6ecd4225a059e07ff86277bd : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9074c6e849544018ce99a1071af6487 : SUCCESS in 55m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/68aae5dad07b4f5cbcfe979671527426 : SUCCESS in 21m 21s","accounts_in_message":[],"_revision_number":30},{"id":"d9d8fe8f9bd0c96211cda741f817f25753b9043c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 21:36:14.000000000","message":"Patch Set 30: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4019ae9b1a564205990949df37bc6936\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dca28b2c61343d3b57cc14fcee600b3 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/c0935e8ed58243afa2a2a03e8285b5a1 : SUCCESS in 53m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08337721bf704da09c5252f6a9ae9f2d : SUCCESS in 44m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff6033bc8cf245648814573ea028697a : SUCCESS in 48m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44e3a5a1b8dc45cba675e0e5de294253 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74a19fe1788e4da388cd81ebf4efeae4 : SUCCESS in 29m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad704d527bf34012a8d2a2458e9befe4 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00146012b50d44cc85da32cdd27cee93 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/04d25a3ab2b54011b8e217f18d2dfb18 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05a3328845ac4704b396dfc474f22e53 : SUCCESS in 20m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64143244a35346f2b75d3bac27d6057c : SUCCESS in 15m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/255f3c8f162d4c2db883b59656d33434 : SUCCESS in 20m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1369d90239da47a8a3c3cf0a330ec82d : SUCCESS in 33m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/87d08480cd234b12b3a8ef6a2069d24f : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6204c59628442afa4c0a4078e7f52a6 : SUCCESS in 20m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff0c1103dd6742b4bac1dd9f0cbe3ef6 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e960ffc546a24120b3f55f27984e5f90 : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/553082b3e01f417186d1f3049df54afe : SUCCESS in 24m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/59f91dd73cec4137b6d34fdd7eb66e48 : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a5efdcfb39e94fbc81ca5c968743755f : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f238e250a1f45e1a42b5b5ff22ccd26 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/265eef3eec564f25a749ba2f52185090 : SUCCESS in 10m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad383f3317c548a3980f0fc0f9d4feaf : SUCCESS in 1h 06m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1088a41462d949a58eca1c918edba46a : SUCCESS in 1h 03m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d838b94550544d14af37ce65b20ea458 : SUCCESS in 1h 08m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf8a7de684fa40a4a3a4760d09e4f4d3 : SUCCESS in 50m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f93cc4aa6b5640419275c94f4c7ddd26 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91533db050864e8a9c67a960e650047e : FAILURE in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb93d71d5ada42bc99a1d06a35b94259 : SUCCESS in 22m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d23c20877ed7441490cac088bce7555c : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":30},{"id":"064314b255a34aae96df5d5429194a25f5ea7975","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-03 21:53:13.000000000","message":"Patch Set 29: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":29},{"id":"f5c6989983388f0b9e963a69b1d5d976b0a827a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-03 22:28:04.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"77ed2f3458b7fe3aaab8cce3682ef17f0f0b9e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-03 23:33:15.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"2f8a5caf63ff31266b8a0aaac8a36908fd70c13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 00:41:57.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2dee7cf42a44d6c8415efa242f5047a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895d842863dd4428b3fa5b53c81d6f68 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/c5f94b80b9e34636819892a2f7a0fadf : SUCCESS in 52m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49f88172032b4066ac0769b4e97e47a9 : SUCCESS in 45m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2310cf53c86b43d48231d5305b6770a3 : SUCCESS in 54m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17dc7c0508fc40459dac73fa9922b5f0 : SUCCESS in 25m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51564e01e5f441e39638525fb3067473 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ada260da55cd41a5ad42bc729454a576 : SUCCESS in 26m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e8356fb9c444e1ca1fc2c6d7d6489a3 : SUCCESS in 7m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/820f2795b83e4787b8d47800f0594134 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7c8d366d03540d5b9e310bab36045d9 : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac418ab2ecd439691cc027f0faab300 : SUCCESS in 22m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eab63193dd154a8c8fd3a87c89d0a173 : SUCCESS in 15m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bd78a0394bb34fcb958b5146f2227258 : SUCCESS in 16m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29ad13d86371474886c0b269981d1b7f : SUCCESS in 23m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3039447a152446ea4d7941b27f4f3c3 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f233058489f45b0afbed4ce4fb246bd : SUCCESS in 16m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43476d46345946098b05b263422bb2da : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3ba18c0036bd4eed9e1472b2a5aede86 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6f92989656474a2e88142b21fb213971 : SUCCESS in 25m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d3fff3162e3443e91e764fd31941817 : SUCCESS in 5m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/548d19510a15461c958328fcd648cdec : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b8aed8b99854f7cb31963707b4d6388 : SUCCESS in 6m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/08ff64163d484ce3b1dd7cdc2bc040d1 : SUCCESS in 1h 06m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc6d3cb2fd614caeab103daf31d25d95 : SUCCESS in 46m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd83b1865d414582b77699214bb0f604 : SUCCESS in 52m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c4092bf1e4945828525a50932944a39 : SUCCESS in 35m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d32a5b9bcf84889a1ad6cf08e46bed9 : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab82527114f74b158faaa415cefa1656 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/afbc854eaf1047eda5375c4f7974f5c4 : SUCCESS in 27m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f5b1512dbf34a7c9c85f518f8eb43ee : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":32},{"id":"28ecccc166075ba49809c7b782e830a2964ac91f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 01:00:09.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a042d4a107439db7c2fbfeab978e93\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/962d35455f8e4c8686b5585246de0c93 : SUCCESS in 14m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e18936c68acc4beeba6a1c17a0aa6c30 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c40c92d150f49c3ad03657fd166bf27 : SUCCESS in 24m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7a3bab1ff8f421f9055a5d81b4b5ac6 : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":32},{"id":"0064a07dfecd7d39eaa9a019f810d27f1803dcce","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-07 10:33:06.000000000","message":"Patch Set 32: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":32},{"id":"5d50bb545f908c2eac86c47dfa8d6c472c662580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-07 18:28:20.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"6e250b6eba9e6a89aa990148b50d3f6d7a981a84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 18:52:12.000000000","message":"Patch Set 33:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bbec232792846b4a4bbab46426366b8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14b87079d7024ad089981e7a0ce9374f : FAILURE in 16m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a79bd49739b42248f5fddb5714b1f4b : FAILURE in 18m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c019a04f6fe941e185b9c82bc8e1f1c9 : SUCCESS in 22m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/176aeece429a4b068c434a43f48d5ed6 : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":33},{"id":"ff308de8b70cb07be0f6c8c63c95b12e1a129fa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-07 19:17:56.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"d92fbb5687ca6979e74a5c1009aa4e6f3a446a62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-07 19:18:51.000000000","message":"Patch Set 35: Commit message was updated.","accounts_in_message":[],"_revision_number":35},{"id":"9f40e0ac469c5264e51a43ba12e447a9938f56e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 20:25:44.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0c3470f3614718839ec99dc6273032\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21f8451ee18a4953bc780101d00a4ca9 : SUCCESS in 14m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3619059d51254a0f8240f8b35e103953 : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c285b1d2759f4505a34c6750cdd96d6e : SUCCESS in 58m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae47f392b3bb415fa7731f58efb8169b : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":35},{"id":"364123679870bab1e30f053a10d1840087701909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 20:35:49.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7198f4f888ee4a64abb084c8e3f2e985\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a423cadb021e4601b238cf707586f081 : FAILURE in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b46cebd383fe434ca68e8e338edc934f : SUCCESS in 58m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e321e66091174d7eaf1890e961476500 : SUCCESS in 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4388ee59fc5e48c297ec296a9e22ca81 : FAILURE in 56m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bdadd22d84d410faf7b5c18d17fd21f : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b4fe33be6f441cf9fe409da8210bd12 : SUCCESS in 28m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec075f340e9349f482858c63e48cf8dd : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f12767544d34260875a8ab7bd9e2009 : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f158f50ba5374e39b6527ffab647fe9b : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f42407af05554190b1659c3b61042528 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcbd4a8c51334f6292a87676593ef7b3 : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bb6f679a6b44b1791869dc028bfd82e : SUCCESS in 23m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/87cd7a7316c64c2ea97ac8a617ad35aa : SUCCESS in 18m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89b34bfea9c34a54b85204a27b8ee9ac : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f9d3eb143624d28aa1813e7e1ea5507 : SUCCESS in 26m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/69c8b4a49d4c49b5a9ea1ed9ff3849bf : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ffbe238e54d4018893d56695dfdfc92 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/84b301b0f1a1470b9262b917014eb649 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99a07b4effbc4e289bcc21631080fe7e : SUCCESS in 25m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f4f3a92c279451ab5aeed1e64fa1ab7 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c78ae6c4c17f4e19b21285910a498c93 : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d53e87ae67f43f6953e1a4e119b0d72 : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1388944bafd41228280264fb7526207 : SUCCESS in 58m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ab1d1438ddd4acb90ecb50e144b6b07 : SUCCESS in 1h 15m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1f392cd0b1e4f7f9d821073cf7945a8 : SUCCESS in 48m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f09fdad4eb04a5c80b07260f4325a12 : SUCCESS in 1h 01m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/501ffa8bb3d1467ca91e81e4ddfa093d : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764357e9f6084145bd93714277cb2e56 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ebfc7ff98b941ab8a2b5f43c14e96e1 : SUCCESS in 20m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/475a90ea75694fa7866b87aa1c5bf47c : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":35},{"id":"2357595cc336f4152a3e980eeb71597bea45281e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-08 11:14:16.000000000","message":"Patch Set 35: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":35},{"id":"6e7df18bcd1a8bac3092f504434024aad9b1bb8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-08 21:11:04.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"6947aff8c88ddded69b3e9162f7e31ffa74a24f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 22:44:38.000000000","message":"Patch Set 36:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86a0214ac1024d13bd998c6cb65d315e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bde6bf19635047a2ae13ce02b888f47b : FAILURE in 19m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ab4d297dfe7473ca5e9194b23ad23a0 : FAILURE in 22m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/977e2fafeb404ace866c5b1902eb9852 : TIMED_OUT in 1h 30m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b7fca760f2a49178634dc9dee758d81 : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":36},{"id":"120e544d2cbae4b78da4191e9d1fab5f1cfc987d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 22:50:59.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d1b90d1becc44a7bdd8026911269dcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59fb9f168c904b0ab707094c470add55 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/fca585a9e5384041a4a60c241754c756 : FAILURE in 1h 07m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/711f19dc922446d28d4ef3eb5ff67f47 : FAILURE in 23m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e45ff068e6a94ecbbd20cdca307367e2 : POST_FAILURE in 21m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffec8630764145eab043f3e1d0f74eaf : TIMED_OUT in 1h 31m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d256def7f21d48bdb72c1e80f134cedb : TIMED_OUT in 1h 31m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f496aa97302476cb524718443efbae7 : TIMED_OUT in 1h 31m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef90bdcabf164dedae8b85cbd6abb2c9 : SUCCESS in 9m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4dd436003492480ea116442c435ca27b : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c47492f55557437bacdb593e6e4deb6e : FAILURE in 23m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c68a131c1e2476bb0cd533cda646097 : FAILURE in 23m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c18bebdca8e4f838cfaf77c787a0517 : FAILURE in 19m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/68ff3f7c3f24434e868040652683399a : FAILURE in 24m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea0ddd5ef76f4414b52474867826979e : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/104f0c66ed3842bb996f272fa8c0a1c2 : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f560a2129cd4f50b8c1fae0b6b51df1 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ebf60417b7a24d60941d421873e72359 : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e770435e89124f3683cd19e36257f530 : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/03479a6ac69b4b9e82c0ee585ced2d28 : TIMED_OUT in 1h 30m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c565ec2aa97f414b8805e6370b6681de : FAILURE in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f81f1016dba249adaea5814018b48fed : FAILURE in 1h 03m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95e13dbc29c14f19900d7b3831e43864 : FAILURE in 9m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a0f53a37a5ad42eeaf43699facfc0eee : FAILURE in 1h 29m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/33e4edd024d84dcda75894d95e1caf3a : FAILURE in 1h 37m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b53f38fdd7164752ae5e8fe7fcb9b4fd : TIMED_OUT in 1h 32m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2814aa30a5404c00aca2ce7a16d49735 : TIMED_OUT in 1h 33m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87d7346e2d634ded8a9a76ded73ef333 : FAILURE in 32m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c5fbd5777848a19c229888e4f26331 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29102510c1764123863c71233c087c3c : FAILURE in 38m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdaca6d41c4a441fa95c3c0702b6db2c : FAILURE in 20m 03s","accounts_in_message":[],"_revision_number":36},{"id":"a6c9428cd1aeda3a975ab14232850f685c5700b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-09 13:28:29.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"839656a9de0e69ff2c2df5b6e2baae8610e45a70","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-09 13:28:36.000000000","message":"Patch Set 36: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":36},{"id":"95a6dda5b4533a7f67d406bffebdf3536f6eaa53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:42:08.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34fd932f6d434eb5808d0f15c9f23560\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fc2432a6eea4c0db0677400bfca272e : SUCCESS in 17m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d1869ab1f994342908a5586bcac6823 : SUCCESS in 58m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4e4675780e74c6aaf79c63ed6217ee7 : SUCCESS in 20m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c58479c35e3244e28b7fcc1e13912906 : SUCCESS in 1h 11m 20s","accounts_in_message":[],"_revision_number":37},{"id":"a58d6d546748d1abaa2b532c6a7d570ce29ebaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:53:41.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8eb7abac4fa3445f824be7fa3b83e634\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3968431554f4482a767f0e20a3fa7ce : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/ccdb4e0c74934e9a9875223ef589a9e8 : SUCCESS in 56m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f046542105e74df9bd4d2aeea652410d : SUCCESS in 45m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80f18f4443974f4c8e56bbe2e16a6ad9 : SUCCESS in 1h 00m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/233271fb6ff348b9b97f54de052cb5ef : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d61b641c54da4738a0b4758a068b8d6e : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64d756597ec5487286c22b6ac0482088 : SUCCESS in 17m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b73e02741be42a289208840564edb35 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c8330a90f24f42e78c857042e40ea538 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/120714829c5b4ab784411a716221a246 : SUCCESS in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6152baa3b8c468e924cdccf372b0c1c : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9b523659448243638a333b3eb7e59e1e : SUCCESS in 16m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3bb19c774f1445d792b0797b1c942ae7 : SUCCESS in 15m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4341cffbbb6149f5b885e59213b73650 : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/591def4bceb34d3385c8cd642a01707d : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c9044c9aef04f4381ef8369b8fc1327 : SUCCESS in 15m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/79dc8618e46e4d74ba8453cebba73a14 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd91e6e431e2418bb15f4e4b8bc67f17 : SUCCESS in 17m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f1d672241b64246802c99f7eda03b71 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/faed6ac40aad4c069171543cda1d883b : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62e8d906374f476eacf9adcc3908905f : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2badb0fa553487884c4146222d97e90 : SUCCESS in 7m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6de742b2c5f4be69a5cabe52ab0f543 : FAILURE in 56m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6227677d65044ef3bdabbf40a04cd65c : SUCCESS in 1h 10m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fe25aa3103f493aaff0404fddf21078 : SUCCESS in 42m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb22078d279d4c8e8918717562cc1b85 : SUCCESS in 58m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08c94621de464459b35c8d6f01812ea6 : SUCCESS in 24m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d72f75ab8365425c95523e8bd88a196b : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1886a93143f4d1c861171178df7bfbb : SUCCESS in 27m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7196b6bb574746b5963cee3a1fb09146 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":37},{"id":"fc9628ff5bf617f5401de3acfa4f72d80677d70a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-09 20:53:18.000000000","message":"Patch Set 37: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"bcaceca0cf84eb75a8e8ac1b49b0fe418cac3da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 20:53:37.000000000","message":"Patch Set 37:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 860866,37","accounts_in_message":[],"_revision_number":37},{"id":"2a4505d2173c44bcd1187d6c16d9db384e74fd22","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-09 20:53:37.000000000","message":"Patch Set 37:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/swift for 860866,37","accounts_in_message":[],"_revision_number":37},{"id":"e1248695e27cd049847e9a110d6dce98cb0d97b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 20:53:37.000000000","message":"Patch Set 37:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 860866,37","accounts_in_message":[],"_revision_number":37},{"id":"a863a8a7c2f90ecd4da66c0896948d173ec29879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-10 09:35:17.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"820dcd981f93183a59641a2c6050c0bde7a17fb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 09:56:25.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc454af94e6c46d79b066f7750ef0958\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/446dbd51d56b429398f12b022627ae17 : SUCCESS in 14m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5215d3e8c5143f9a2aa3acbf904c67f : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/34df238548fb4b83b78bc11e2b7252db : SUCCESS in 19m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac0216820e6f4d24854360a1657065ba : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":38},{"id":"1f665db3c9163be31bff2717ef12fb096c684481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 11:18:38.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3527810472a14c53a43549c7590ccb48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a64ad73ee5142888fdd724863c4eb4c : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/aee99c516be342fd945eb244f5374f8d : SUCCESS in 49m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8bbf401b5bf145deb5c580e593cab744 : SUCCESS in 36m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0998f5094c164da3b239043c54cdddfd : SUCCESS in 50m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2814aedccb794a08afa55ad5abb242f2 : SUCCESS in 25m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15b4817ede0c4964aadad90a914630d9 : SUCCESS in 24m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f04cb59514a14bf88e8cce29fed4c31a : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03f942e5aebc4b068c8b0833eb0ce141 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ef64e07b2a2d4a9f99d8ac291870bd25 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5da81e3eae64d11abe1c60aac687303 : SUCCESS in 23m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ab50708c61e4a759b1baeea08a0259e : SUCCESS in 17m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2396057e0364909a923af78e1267941 : SUCCESS in 28m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/34162b1f633e45ef81f6814b23dc73ed : SUCCESS in 22m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fc31f808606d4b0c9e3864c360df9dfe : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5862e3da66149368b2e0da535f51535 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a4844bb3a4149d2a17e86d4cc66e623 : SUCCESS in 24m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6df774510e654625a83a81feabb46f53 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9c59878e4b194b42840e819c6301f5fb : SUCCESS in 22m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48ea7ca9e8284821ad3a1a93dec3a7ac : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3321274849d54e7196b54c764a689a8f : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ca31a5eda6044c739d2af617e0ea5498 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5a5ff807858645a7a741b503c8d7e0d7 : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35066f6309f94734a5e3b72b4497f83f : SUCCESS in 1h 29m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e5fe1a574fd4023bd165c6a108a975d : SUCCESS in 59m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0483dc54b344f979eebc2bb284c7a5e : SUCCESS in 1h 35m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9880266ffd45484eb6942bf1f6e15a75 : SUCCESS in 48m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a731164c444ee5bb3437f378347b58 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa399e0300c04ba8890dc59bbca52aed : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a5046244f3346cc98671f144e636c49 : SUCCESS in 23m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d323986e85cd48afadf4e9efe8282f9f : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":38},{"id":"7e027bba111fa770cd0aa7837fa16844dd540fbe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-13 18:18:41.000000000","message":"Patch Set 38: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":38},{"id":"df273bd94d54663cdcdfcbe130e1a4b089cb0288","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:19:43.000000000","message":"Patch Set 38: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":38},{"id":"962af8a4af3820813369d012c6bee6b071f94c6c","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-02-13 18:57:59.000000000","message":"Patch Set 38: Code-Review+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":38},{"id":"e79b61c2cc932791e5a1ccbde6defcdfe9499f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:34:49.000000000","message":"Patch Set 38: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9129e7389b9c4679bb0f642250615309\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71bbe3fa98f640a7af8e350f6fd21edb : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/942ada0cf46b4c00bb37eb99d7cfe035 : SUCCESS in 40m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/37e24849dd314022963dd61dbf3f1a81 : SUCCESS in 37m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd8ac6ee87bd406c89d62ad965d69c96 : SUCCESS in 1h 00m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0b78594d78984a17a9c0be1e9c2a1fe6 : SUCCESS in 9m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ad13ef2baf5d4d26b0d1b14c55aaf16f : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27faa33293fb4561bdf2e95efcbb5609 : SUCCESS in 19m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f94294cab5034d2a965f8f20c8ca2387 : SUCCESS in 14m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/827b9b26c8a544068794d5a44ad2b610 : SUCCESS in 16m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/543d5759228a4d079f16e6e49eae43cb : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0832017223494073b0dcea8c976eea31 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbe4e51875c74aec8559c119992fd0f9 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08c7279dad6c44f68dddc7737d81b65f : SUCCESS in 18m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/053da2aa241e4687a02fecdba5ca163e : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d527be81ca8492cba3ee51dbaba662d : SUCCESS in 15m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48f1cf5cb35c4d3d970fcc988c4b8b74 : SUCCESS in 14m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3874deb00d754dcc867adca2a2c27c59 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04ab841120ce49aaa30548063bcadbef : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c64a2070e3a418b9a5c4698edc8feee : SUCCESS in 57m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd98bbc096884a50b922f8f5ed3ed8a2 : SUCCESS in 1h 05m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3594b60050a540f3b4d1e4b39f2e3c58 : SUCCESS in 40m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/948da6e6156e4f05865bf8e2a31abf85 : SUCCESS in 59m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76d1e43bd10b4c7db29fd27f36b25342 : SUCCESS in 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed764a7f77fa47b08b17516aac231680 : SUCCESS in 4m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eec378c89e5b49ef825ce918a917562b : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":38},{"id":"908c0e3f58812d1e8d71e8566c3b54f13d3592e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:34:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":38},{"id":"80d9db549102e959209e75d14fadb0842259fb4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:37:58.000000000","message":"Patch Set 38:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da630604514b47fdacc83320ee77c179\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b902088f154e16a053a7b94273cbe3 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/241e408554094364a93ed6c5f88a3d84 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e70cb60ac5f04fe489c0b38578fc21b8 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":38}],"current_revision_number":38,"current_revision":"9ec90d4d56c598ab5567128fd59dca8c54fe8dc4","revisions":{"d66eb8d16e39d038e1e3f1ecde242c580b1a6ad0":{"kind":"REWORK","_number":1,"created":"2022-10-10 18:12:41.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/1"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d66eb8d16e39d038e1e3f1ecde242c580b1a6ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d66eb8d16e39d038e1e3f1ecde242c580b1a6ad0"}]},"branch":"refs/heads/master"},"09e3592d69028d0542c7ae87708cfcef4746f578":{"kind":"REWORK","_number":2,"created":"2022-10-10 18:25:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/2"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 18:25:40.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09e3592d69028d0542c7ae87708cfcef4746f578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09e3592d69028d0542c7ae87708cfcef4746f578"}]},"branch":"refs/heads/master"},"cc764b4faea4bba5827cf130e441492e66cc6a7f":{"kind":"REWORK","_number":3,"created":"2022-10-10 19:07:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/3"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 19:06:57.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc764b4faea4bba5827cf130e441492e66cc6a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc764b4faea4bba5827cf130e441492e66cc6a7f"}]},"branch":"refs/heads/master"},"50d46e3d91ac34f66044488fd9568b4a2b47d55a":{"kind":"REWORK","_number":4,"created":"2022-10-14 18:17:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/4"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-14 18:17:00.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50d46e3d91ac34f66044488fd9568b4a2b47d55a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50d46e3d91ac34f66044488fd9568b4a2b47d55a"}]},"branch":"refs/heads/master"},"7f3622e1a79daf4130e9f49600e9f786537d3556":{"kind":"REWORK","_number":5,"created":"2022-10-14 20:25:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/5"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-14 20:25:11.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f3622e1a79daf4130e9f49600e9f786537d3556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f3622e1a79daf4130e9f49600e9f786537d3556"}]},"branch":"refs/heads/master"},"a015001ec7207dbb5ce24d2e783e1cacd6e7b8bb":{"kind":"REWORK","_number":6,"created":"2022-10-14 20:27:27.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/6"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-14 20:27:21.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a015001ec7207dbb5ce24d2e783e1cacd6e7b8bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a015001ec7207dbb5ce24d2e783e1cacd6e7b8bb"}]},"branch":"refs/heads/master"},"fe4e2d9011aa8d05fdb0ca1da1dad1645594bc98":{"kind":"REWORK","_number":7,"created":"2022-10-17 15:31:48.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/66/860866/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/7"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-10-17 15:31:42.000000000","tz":-300},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe4e2d9011aa8d05fdb0ca1da1dad1645594bc98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe4e2d9011aa8d05fdb0ca1da1dad1645594bc98"}]},"branch":"refs/heads/master"},"57c2f9481d293262ce94b8686c9927f6cc2421ed":{"kind":"REWORK","_number":8,"created":"2022-10-17 17:19:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/8"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-17 17:19:30.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57c2f9481d293262ce94b8686c9927f6cc2421ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57c2f9481d293262ce94b8686c9927f6cc2421ed"}]},"branch":"refs/heads/master"},"dbaf0e75c279a2901c1ee280532e81a6b23726e4":{"kind":"REWORK","_number":9,"created":"2022-10-17 18:58:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/9"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-17 18:58:45.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbaf0e75c279a2901c1ee280532e81a6b23726e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbaf0e75c279a2901c1ee280532e81a6b23726e4"}]},"branch":"refs/heads/master"},"81799bbbc4126213372db0d5ea998af0391fa6b0":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-10-17 19:03:10.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/10"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2022-10-17 19:03:10.000000000","tz":0},"subject":"proxy-server logging contains replication_ip/port","message":"proxy-server logging contains replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node by looking up the header value\nfor x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81799bbbc4126213372db0d5ea998af0391fa6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81799bbbc4126213372db0d5ea998af0391fa6b0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"941a78ea109c8dc42349e97b4867115ad1bf9f04":{"kind":"REWORK","_number":11,"created":"2022-10-17 19:05:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/11"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-17 19:05:44.000000000","tz":-420},"subject":"test proxy-server logging contains replication_ip/port","message":"test proxy-server logging contains replication_ip/port\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941a78ea109c8dc42349e97b4867115ad1bf9f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941a78ea109c8dc42349e97b4867115ad1bf9f04"}]},"branch":"refs/heads/master"},"e89e5135c5bd8b3fa3f31a1f2e89e34ff70ddb75":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2022-10-17 19:06:43.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/12"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2022-10-17 19:06:43.000000000","tz":0},"subject":"proxy-server logging contains replication_ip/port","message":"proxy-server logging contains replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node by looking up the header value\nfor x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e89e5135c5bd8b3fa3f31a1f2e89e34ff70ddb75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e89e5135c5bd8b3fa3f31a1f2e89e34ff70ddb75"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e0848435462e5d4cb4f31fa11af1deb888b45731":{"kind":"REWORK","_number":13,"created":"2022-10-17 19:18:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/13"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-17 19:18:23.000000000","tz":-420},"subject":"proxy-server logging contains replication_ip/port","message":"proxy-server logging contains replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node by looking up the header value\nfor x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0848435462e5d4cb4f31fa11af1deb888b45731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0848435462e5d4cb4f31fa11af1deb888b45731"}]},"branch":"refs/heads/master"},"1e026a180cf2f592e2381b8360b80ad595ebb24f":{"kind":"REWORK","_number":14,"created":"2022-10-19 18:45:31.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/14"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-19 18:45:23.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node by looking up the header value\nfor x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e026a180cf2f592e2381b8360b80ad595ebb24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e026a180cf2f592e2381b8360b80ad595ebb24f"}]},"branch":"refs/heads/master"},"2887ff33bb000628ce4a399c852ebdb7139d049f":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2022-10-19 18:47:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/15"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-19 18:46:14.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2887ff33bb000628ce4a399c852ebdb7139d049f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2887ff33bb000628ce4a399c852ebdb7139d049f"}]},"branch":"refs/heads/master"},"083541ded61e7831bfb6ecd6c027466ed854792c":{"kind":"REWORK","_number":16,"created":"2022-10-19 19:19:29.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/16"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-19 19:19:23.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/083541ded61e7831bfb6ecd6c027466ed854792c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/083541ded61e7831bfb6ecd6c027466ed854792c"}]},"branch":"refs/heads/master"},"5e20f78793fc852379e4fb34380d15dceee74174":{"kind":"REWORK","_number":17,"created":"2022-10-19 19:24:49.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/17"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-19 19:24:43.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e20f78793fc852379e4fb34380d15dceee74174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e20f78793fc852379e4fb34380d15dceee74174"}]},"branch":"refs/heads/master"},"e1ea8e6f0f087519bdccd29b2ddfc56e7dbfd88c":{"kind":"REWORK","_number":18,"created":"2022-10-27 01:59:33.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/18"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-27 01:59:12.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ea8e6f0f087519bdccd29b2ddfc56e7dbfd88c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ea8e6f0f087519bdccd29b2ddfc56e7dbfd88c"}]},"branch":"refs/heads/master"},"cc8eb1e8cd0c9a5c9fa53c65297b28eb7c447953":{"kind":"REWORK","_number":19,"created":"2022-10-27 02:53:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/19"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-27 02:53:36.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc8eb1e8cd0c9a5c9fa53c65297b28eb7c447953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc8eb1e8cd0c9a5c9fa53c65297b28eb7c447953"}]},"branch":"refs/heads/master"},"b1aa06e49ec66bf55cecf174d047f8fa21134715":{"kind":"REWORK","_number":20,"created":"2022-10-27 03:25:40.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/20"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-27 03:25:32.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1aa06e49ec66bf55cecf174d047f8fa21134715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1aa06e49ec66bf55cecf174d047f8fa21134715"}]},"branch":"refs/heads/master"},"4396137451c770a80d937ee9669196628ffdf6cf":{"kind":"REWORK","_number":21,"created":"2022-10-27 16:33:10.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/21"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-27 16:33:04.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4396137451c770a80d937ee9669196628ffdf6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4396137451c770a80d937ee9669196628ffdf6cf"}]},"branch":"refs/heads/master"},"639a909be3997a957e6d210b5004331e955ee313":{"kind":"REWORK","_number":22,"created":"2022-10-28 20:10:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/22"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-28 20:10:11.000000000","tz":-420},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639a909be3997a957e6d210b5004331e955ee313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639a909be3997a957e6d210b5004331e955ee313"}]},"branch":"refs/heads/master"},"f1b53d6a3ee802cc4e313594c4d72bd6dfd1ac92":{"kind":"REWORK","_number":23,"created":"2022-11-15 18:52:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/23"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-15 18:51:43.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1b53d6a3ee802cc4e313594c4d72bd6dfd1ac92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1b53d6a3ee802cc4e313594c4d72bd6dfd1ac92"}]},"branch":"refs/heads/master"},"15169a4cb75bf7e54fa92212cfee699cfd618063":{"kind":"REWORK","_number":24,"created":"2022-11-16 19:33:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/24"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 19:33:02.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15169a4cb75bf7e54fa92212cfee699cfd618063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15169a4cb75bf7e54fa92212cfee699cfd618063"}]},"branch":"refs/heads/master"},"d2101a9de6ac8b14d340aeb02aba3f7b9c82b5df":{"kind":"REWORK","_number":25,"created":"2022-11-16 19:46:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/25"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 19:46:45.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2101a9de6ac8b14d340aeb02aba3f7b9c82b5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2101a9de6ac8b14d340aeb02aba3f7b9c82b5df"}]},"branch":"refs/heads/master"},"fc7275d940b7149f92bae9efa63c63be15a4524b":{"kind":"REWORK","_number":26,"created":"2022-11-17 23:39:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/26"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-17 23:39:36.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7275d940b7149f92bae9efa63c63be15a4524b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7275d940b7149f92bae9efa63c63be15a4524b"}]},"branch":"refs/heads/master"},"14b1f87bb7da1905a4a91723e8f3bbc907710336":{"kind":"REWORK","_number":27,"created":"2022-11-21 17:37:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/27"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-21 17:37:40.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14b1f87bb7da1905a4a91723e8f3bbc907710336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14b1f87bb7da1905a4a91723e8f3bbc907710336"}]},"branch":"refs/heads/master"},"f8e17d7859ba82804c588dcfc7b5a1dffdd3bd65":{"kind":"REWORK","_number":28,"created":"2022-11-30 02:33:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/28"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-30 02:33:37.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8e17d7859ba82804c588dcfc7b5a1dffdd3bd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8e17d7859ba82804c588dcfc7b5a1dffdd3bd65"}]},"branch":"refs/heads/master"},"08c0eb8454aaa9b76fe62031c8ac446d3ef8f6bd":{"kind":"REWORK","_number":29,"created":"2023-02-03 20:01:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/29"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-03 19:59:39.000000000","tz":0},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08c0eb8454aaa9b76fe62031c8ac446d3ef8f6bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08c0eb8454aaa9b76fe62031c8ac446d3ef8f6bd"}]},"branch":"refs/heads/master"},"5974c5b6d726015ab5b149f46f6f3b23b0499d59":{"kind":"REWORK","_number":30,"created":"2023-02-03 20:12:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/30"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-03 20:12:23.000000000","tz":0},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5974c5b6d726015ab5b149f46f6f3b23b0499d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5974c5b6d726015ab5b149f46f6f3b23b0499d59"}]},"branch":"refs/heads/master"},"3c61aded7b12e9d9ac792e7d908dd35c2ad3eeee":{"kind":"REWORK","_number":31,"created":"2023-02-03 22:28:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/31"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-03 22:27:55.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c61aded7b12e9d9ac792e7d908dd35c2ad3eeee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c61aded7b12e9d9ac792e7d908dd35c2ad3eeee"}]},"branch":"refs/heads/master"},"c5b77a0bef1bdce703eb0f36582ca34edf380548":{"kind":"REWORK","_number":32,"created":"2023-02-03 23:33:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/32"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-03 23:33:07.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5b77a0bef1bdce703eb0f36582ca34edf380548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5b77a0bef1bdce703eb0f36582ca34edf380548"}]},"branch":"refs/heads/master"},"7d7502c236feb2ff072e0834b2cb2662e5077776":{"kind":"REWORK","_number":33,"created":"2023-02-07 18:28:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/33"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-07 18:28:12.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d7502c236feb2ff072e0834b2cb2662e5077776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d7502c236feb2ff072e0834b2cb2662e5077776"}]},"branch":"refs/heads/master"},"d1433a8f620b4ba67512e57fe6cf1e7f65d9f0ca":{"kind":"REWORK","_number":34,"created":"2023-02-07 19:17:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/34"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-07 19:17:50.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1433a8f620b4ba67512e57fe6cf1e7f65d9f0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1433a8f620b4ba67512e57fe6cf1e7f65d9f0ca"}]},"branch":"refs/heads/master"},"1da86d5d79e00c486128cbc4fe911cb4f562f7ca":{"kind":"NO_CODE_CHANGE","_number":35,"created":"2023-02-07 19:18:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/35"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-02-07 19:18:51.000000000","tz":0},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network, Request is a required param for node_iter.\n\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1da86d5d79e00c486128cbc4fe911cb4f562f7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1da86d5d79e00c486128cbc4fe911cb4f562f7ca"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c167d3ce003da99f77b7c7988834f3ed0f6a734e":{"kind":"REWORK","_number":36,"created":"2023-02-08 21:11:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/860866/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/36"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-02-08 21:10:49.000000000","tz":-480},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c167d3ce003da99f77b7c7988834f3ed0f6a734e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c167d3ce003da99f77b7c7988834f3ed0f6a734e"}]},"branch":"refs/heads/master"},"27763441486a472491e2421463a7ba8ceb02be84":{"kind":"REWORK","_number":37,"created":"2023-02-09 13:28:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/860866/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/37"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-09 13:28:12.000000000","tz":0},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27763441486a472491e2421463a7ba8ceb02be84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27763441486a472491e2421463a7ba8ceb02be84"}]},"branch":"refs/heads/master"},"9ec90d4d56c598ab5567128fd59dca8c54fe8dc4":{"kind":"REWORK","_number":38,"created":"2023-02-10 09:35:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/860866/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/860866/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/860866/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/860866/38"}}},"commit":{"parents":[{"commit":"d9bf70ae2b28222f62470bba5f6877bbbe057532","subject":"Merge \"s3token: Only replace access_key_id in account\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9bf70ae2b28222f62470bba5f6877bbbe057532"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-10-10 17:27:00.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-10 09:34:59.000000000","tz":0},"subject":"proxy-server exception logging shows replication_ip/port","message":"proxy-server exception logging shows replication_ip/port\n\nAdding a \"use_replication\" field to the node dict, a helper function to\nset use_replication dict value for a node copy by looking up the header\nvalue for x-backend-use-replication-network\n\nChange-Id: Ie05af464765dc10cf585be851f462033fc6bdec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ec90d4d56c598ab5567128fd59dca8c54fe8dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ec90d4d56c598ab5567128fd59dca8c54fe8dc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
