)]}'
{"id":"openstack%2Fswift~860937","triplet_id":"openstack%2Fswift~master~If1ef202b41af124637091a3527320b80734948fe","project":"openstack/swift","branch":"master","topic":"p-zuul-ansible-user","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-10-11 17:19:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1ef202b41af124637091a3527320b80734948fe","subject":"playbooks: replace ansible_ssh_user with ansible_user","status":"MERGED","created":"2022-10-11 10:50:10.000000000","updated":"2022-10-11 17:20:59.000000000","submitted":"2022-10-11 17:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860937-p-zuul-ansible-user","meta_rev_id":"73ccf42c6c3bc4e56f10ed811eddafb9d8790de3","_number":860937,"virtual_id_number":860937,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 17:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2022-10-11 13:18:19.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-11 16:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-11 16:14:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 11:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-11 11:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-11 16:14:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9a4effdb2e4023148931e88b2c47abca7b8b9fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-11 10:50:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9277150d0c8109f227b8d6e9c696c34f40e636f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-11 10:52:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3727b0a44565e1448c55c4c48d85b707202b2729","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-11 10:52:31.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"7ffc635edcab644d966e2dcd48224d2c78ed8d6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:27:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f31361baa58543b7a02ee7a73a8db32d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3acaec8b9bcb444cb32e9b37333f99a9 : SUCCESS in 29m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca688fbf93254d67afc3f879e02535e1 : SUCCESS in 29m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8f3a30dc3b54b338ddcc055bdf8dddf : SUCCESS in 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea0549276f654fbe9ce143464be6ce70 : SUCCESS in 27m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6e4d80a9dbb7c11b03d2b6beb64d628436f31778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:52:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d825acbe254859ba777b7821315aa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ba827d105c4d70bda90f4144097684 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/5ddcfcb242774556a20761037f809134 : SUCCESS in 48m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14bd11adeac448d9bed96922ea37467d : SUCCESS in 45m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0b30091656a4923af36bdfd851a67c6 : SUCCESS in 47m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21b12a47e56943d988ffbcc4a6293e8a : FAILURE in 20m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9bd461f69d54e6e977e01500ae25ecb : FAILURE in 23m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80c6be8e9dea4927adc4cf33f6da3b2b : FAILURE in 23m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42be2cba4a0a457bbb444761574e2f4e : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4c2af1b7db214dc596208bc323459992 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/613945e98f8e4b51bcd0ee877fa641b8 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/699830d3e724470b91081c0331588bb2 : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36c75ff91a6c4a80892f70ae947f7dc8 : SUCCESS in 18m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b5c3691d3e942cd98bf604bfaf5a626 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4961043741d415ca5fe88acffdef001 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02e8fa11db8a40a19aa4060b38f51114 : SUCCESS in 20m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c9170c4efbb04e559903676620086017 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7795084c3df2411990fa17b27cbf5b80 : SUCCESS in 14m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bca36c616439458db57144d5589ba047 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa81b5c9ce384eac91cc602999910034 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/19a00ae4b78a4d4b946e2347774c5090 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f5a8460ec6849d5a933856756759a13 : SUCCESS in 6m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0fa1f750284b4a2a8aebf10192ef4dca : SUCCESS in 51m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f8dab54b0d74df8b124e88daf4fdac6 : SUCCESS in 56m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8500ba0f90e040b687096c1fa4a20fae : SUCCESS in 36m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/06da1854e5894fc9acd2044b7044aab9 : SUCCESS in 32m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/169a3e9cdf344abfac1805ca1194ee53 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da8e5a884fd4153bfb319e6dc6aa972 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31aff84077c5434c9f3b766cafb78a46 : SUCCESS in 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ef862f619ac4b9bb3a57694c9f8e175 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ef8cf3b614d65e0ea759ddd434851bb95a53d2d4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-11 13:18:19.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d4e7977416bcda9ecfd3f3701d339eb3e940637e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-11 16:14:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a505156b36152c68687e629e329364ca4f82108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 16:15:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1039107ead478c8be5ae71f8065100422dc0b0ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:19:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93bed22491db44a78825fb1e34e36199\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5d25bf1ad5488d9641c2cf2c5cb7bf : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/c810ada31af847d7a3191959c517e9b9 : SUCCESS in 37m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23f97fc7f1a04cb0822a193f2240ca25 : SUCCESS in 46m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a91bf67005bf4644b1d7483cd31dbc71 : SUCCESS in 36m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2e05854e1bf84fa48ea01e4fb61a9d14 : SUCCESS in 8m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3a1a6ed7177b494091678fdae5f06334 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc275c1a9b3e4abb98614bfac0113ffa : SUCCESS in 13m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d60efe26d8ab453ba58db389f3b23cfa : SUCCESS in 12m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df713c2bdd424e15833cba3141c172c0 : SUCCESS in 14m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cef415f6d6ad40199899c907e27d8297 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98ec4d89b13e48d2aee9d7ebce877104 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d12c32247cd840efa468095004425521 : SUCCESS in 15m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a6a52186a444e2e999f27a8183a8198 : SUCCESS in 12m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ca064c2489d34fe5b6c08a1fcf176c69 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/948dbbcbe4d142f5a668ffe574b05d0f : SUCCESS in 13m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69fc5986283b4dadb8fabbf9030db92f : SUCCESS in 6m 01s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ac78ed8d69a49df8f2fc27c3decfe77 : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/82c9e6783dab4cd8aeabacdad4b75a91 : SUCCESS in 55m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd4be547045246b0954040a835d5eedc : SUCCESS in 59m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c80c5a261b5452c8f3a938fbc900b0b : SUCCESS in 35m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47f17a3a2d5f477c90b326397e3c8496 : SUCCESS in 46m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41828c94fa984a2b9c798337b6873aac : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a8dfc0c8e94e1387221916f7f40d73 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6778f2a78d04623ad4acab0142fdbe5 : SUCCESS in 19m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64ff87280d2c4a888e3241648de6b0cd : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":1},{"id":"91af3e22c60557aa2715a45ebee5993bde9732f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:19:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73ccf42c6c3bc4e56f10ed811eddafb9d8790de3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:20:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0ee66204b34e508cf9c67b4b27e8aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f66e5dbd044622976210314cbe6d6c : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b345d5ab90044f9da0c2856400bdf40b : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c1fb565720ff4c8f9699d3e66b8a8ac7 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07b5f139a5f6353ec40a52047dd8456a820d3518","revisions":{"07b5f139a5f6353ec40a52047dd8456a820d3518":{"kind":"REWORK","_number":1,"created":"2022-10-11 10:50:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/860937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/860937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/860937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/860937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/860937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/860937/1"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-10-11 10:27:03.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-10-11 10:27:03.000000000","tz":60},"subject":"playbooks: replace ansible_ssh_user with ansible_user","message":"playbooks: replace ansible_ssh_user with ansible_user\n\nansible_ssh_user has been deprecated since Ansible 2.0 [1]\n\n[1] https://github.com/ansible/ansible/blob/stable-2.0/CHANGELOG.md\n\nChange-Id: If1ef202b41af124637091a3527320b80734948fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07b5f139a5f6353ec40a52047dd8456a820d3518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07b5f139a5f6353ec40a52047dd8456a820d3518"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
