)]}'
{"id":"openstack%2Fswift~861022","triplet_id":"openstack%2Fswift~master~I7785a73820031aeb4469f4d54e590a7db77532db","project":"openstack/swift","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I7785a73820031aeb4469f4d54e590a7db77532db","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-10-12 03:15:33.000000000","updated":"2022-10-18 16:38:34.000000000","submitted":"2022-10-18 16:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861022-zanata/translations","meta_rev_id":"4ca31ae1d75bd94e12d2e5778ec4b85715c5457a","_number":861022,"virtual_id_number":861022,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 16:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-18 16:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-18 16:28:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 03:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 16:28:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"6af756f03a7e1b57444a72896442feccfbfa5c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-12 03:15:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60f9bc440b03134d6e635bec5a6cda4d4312540e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af591f75c91c4303b2510f4755b1c130\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb619baaade49ba9d6187e6a561b9c7 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d26058315c34f85a8142d01d21cd3d8 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e32decaa3b1540c078c0d17d0f740d384bc0919a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-15 03:36:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b9de7998d0c95f3d4e501a88acc02b176996a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 03:47:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c066c6a615944cfadb806d57a1b4444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031b2babcfa246889053bf1beab12e15 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c5f92eac04545a5bb66084ec946db09 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7e0b792924c213dbc422c17e39e0203779fdc984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-18 03:45:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a18b3232f544e954ea6295b39245ce49d9a3279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:52:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d717519922a406e808c5454d7cccf60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca02838c9eb4e9cbed342e429e3a4e1 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/110437f8b7ef4880911c78b36eecd04a : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b99f2cb8d1560e80d6b6cbe0837423403ce5570c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-18 16:28:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a851002e4872be7e2adb28fc2c1242b696ce65f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:29:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7aeb22a67c101628e6256d83d152640a4ec5992d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:37:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c65277995b6437caee70d1dfe5c9280\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc00dae965824dc587376acb84469dbd : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e848700009241d2af25c1474cd14ba5 : SUCCESS in 5m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb4ad247a36e44f392721525f8f01f0c : SUCCESS in 8m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/53ba5ecae8b64073a1483311a7a4f352 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ab0c97a5802253ecdafee64170be4f5b9df0ffac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:37:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4ca31ae1d75bd94e12d2e5778ec4b85715c5457a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:38:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb85884823e4f18a94f8d6e28bc93c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e762deeaab64047b1826bdbbc0ef796 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdb6b5968d3f4b96abc71ee771b648dc : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/31b1528464864e1789bc41a09229d389 : FAILURE in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4cfc8adcd4b04dccaaf7c40512fa22fa : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79e8f56ac41093e337dba3b60c3a21b2c26b59c8","revisions":{"e5ecfd098f786346af6173cced51d2b7a451e60b":{"kind":"REWORK","_number":1,"created":"2022-10-12 03:15:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/861022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/861022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/861022/1"}}},"commit":{"parents":[{"commit":"65a1f4a2ff6badd4015c278d1cb1468b51113466","subject":"Merge \"sharder: always merge child shard ranges fetched from root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65a1f4a2ff6badd4015c278d1cb1468b51113466"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-12 03:15:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-12 03:15:29.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7785a73820031aeb4469f4d54e590a7db77532db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5ecfd098f786346af6173cced51d2b7a451e60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5ecfd098f786346af6173cced51d2b7a451e60b"}]},"branch":"refs/heads/master"},"e78e5c13c0a7258cccc04ee82c57bfba3a8a745b":{"kind":"REWORK","_number":2,"created":"2022-10-15 03:36:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/861022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/861022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/861022/2"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-15 03:36:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-15 03:36:38.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7785a73820031aeb4469f4d54e590a7db77532db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e78e5c13c0a7258cccc04ee82c57bfba3a8a745b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e78e5c13c0a7258cccc04ee82c57bfba3a8a745b"}]},"branch":"refs/heads/master"},"79e8f56ac41093e337dba3b60c3a21b2c26b59c8":{"kind":"REWORK","_number":3,"created":"2022-10-18 03:45:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/861022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/861022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/861022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/861022/3"}}},"commit":{"parents":[{"commit":"675769bc5e3eb9a263651f67cb68a483f391be7a","subject":"Merge \"swift-drive-audit: reload systemd after editing fstab\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/675769bc5e3eb9a263651f67cb68a483f391be7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-18 03:45:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-18 03:45:21.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7785a73820031aeb4469f4d54e590a7db77532db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79e8f56ac41093e337dba3b60c3a21b2c26b59c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79e8f56ac41093e337dba3b60c3a21b2c26b59c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
