)]}'
{"id":"openstack%2Fswift~861271","triplet_id":"openstack%2Fswift~master~I0bdc98765d05e91f872ef39d4722f91711a5641f","project":"openstack/swift","branch":"master","topic":"tempauth-fernet-tokens","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-04-25 23:33:56.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-30 22:37:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0bdc98765d05e91f872ef39d4722f91711a5641f","subject":"tempauth: Support fernet tokens","status":"MERGED","created":"2022-10-13 23:58:25.000000000","updated":"2025-04-30 23:46:36.000000000","submitted":"2025-04-30 23:45:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"861271-tempauth-fernet-tokens","meta_rev_id":"2f73a6d0a1132943fed314087084bd98bc6c86ca","_number":861271,"virtual_id_number":861271,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-30 23:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-04-25 23:33:56.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-30 22:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-30 22:37:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 01:01:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 22:48:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-06-17 23:43:34.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-04-30 21:29:58.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"8c1f4bbde5a86f04211b4970281866e68ccd0a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-13 23:58:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b04123b6994579e811681f4ce587ad34fa3255c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 01:01:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4cd4953d68841b195858a94f0915f20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47920b7dffa44b0b4986f6adfb74995 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/e4cc8245a0aa41328b9576fe451502e6 : SUCCESS in 59m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3af56e2513f64f64b7a2bc5c15c0227f : SUCCESS in 47m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9429803b2ab415d991a3387b83fcae6 : SUCCESS in 47m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccb6fe2b8c714d5482f21751a10dfb3a : FAILURE in 23m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71617df404b44b16a5758d69f21d27d5 : FAILURE in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e10b7c98704c43f9a115b8cd2bd19b79 : FAILURE in 24m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c83fb9cbce543b7a2c01db946173fd7 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f7ba346edee044919315d362b42614b2 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef29acbaa3be4b079884c5186c22e05a : FAILURE in 20m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d65a6a8afd694e04b092a296111bab07 : FAILURE in 19m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/038aa49758dd43719e5669eb0c443713 : FAILURE in 22m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41168ef8b77741c191511d550138526f : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc75731167ce4dd7ad222dd44e4eed1d : SUCCESS in 21m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8c04173b8fab43f9af9698bd96fff834 : SUCCESS in 21m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dad59af1d1d740b498fdf90c3c72b3ad : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b98846487bb34c8abfa20ecaec5d70b9 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80a1f0a23cb0485e914429031893fd8a : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eec22b2de3af46d68e0d76f825b6821e : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66449704a0de4f26ba6d5e052610d85b : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/859572af38d74534ad8e5055d19d74e1 : SUCCESS in 6m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/afe855cce28d421cb206f81761ae6e11 : SUCCESS in 52m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae28682aa22446fb90947bdd7caf8133 : SUCCESS in 46m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2c8f31006244d8d8194715424efb923 : SUCCESS in 32m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25de1a8b7c834e56986b3984d582f137 : SUCCESS in 29m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b901f2ea8b4c4144b87e2b1a9950afe1 : FAILURE in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfaaadcaa96d4c7e9a37e3d65a6dcba7 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/089ff5cf32ef47e2b34f98872bce18ea : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/719dea43c6a24214a842c0f3045120bb : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a92c639faf7888c1d67b8cc79ad47cf326de0da8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 01:05:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96f68522259b48dab72bb4266ffa8c6a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de8bc2b378f841a59d1ca5a4d0d1c395 : FAILURE in 31m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1da38a092a264edc921d096d271c1a7a : FAILURE in 31m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7895a811f43a45a6b6ecd76b83159584 : SUCCESS in 31m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/18c17ee082f244938f22d06633ce5af4 : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d2372497da262f3b00a55a28c5105a460ba63069","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-10-18 22:48:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f58258a7c83bd809563946f04a0c436b41759400","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-14 20:11:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eab929a22fde76af1cd8c7d8c686d558a3a54e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-14 20:11:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b116790437cc8c923e70cdc8b8a3f4595fd1ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:31:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ef009871644d4f9033f98a34ff9368\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/757b5d1315694939bf570a70792bc454 : SUCCESS in 16m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/194538333fc44a4aa9dad4ea9c0849d1 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7f76e0a73ef4940bcea70243861cb62 : SUCCESS in 18m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c6b1777cf1f4f08af92d9148961cac4 : SUCCESS in 18m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a0fc394de5007c9f59fd0fb8ba3434129496f84d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 21:24:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cd4be8df0ea4dd897f08b1c8ec5e6f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2e8908391949b28c95768f058314dd : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/376f5e24d0ed4fcdab87acf37aefd278 : SUCCESS in 52m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e6169c1120b4577a9c19d1a274c43a7 : SUCCESS in 39m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2a69a92d62a43589d07a03dce875618 : SUCCESS in 52m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b744a9f1b4fc4f6f9da4bcd61b13cf3a : SUCCESS in 33m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c755485cd4434427bc46b7b04a955721 : SUCCESS in 31m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f95436e83952491eac2909b21bb73749 : SUCCESS in 34m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e06abc0386ba447a8216ee9d895cdfb4 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4988e9d346c049f290ab1b82544ed39c : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/425e7026f199418caa3332fb90cfd996 : SUCCESS in 34m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa16c524a08d4bd59d84940a54201f82 : SUCCESS in 30m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/194a72bcf51149efb58021762eda1e54 : SUCCESS in 31m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ec1ae572b6aa4373b327813250901849 : SUCCESS in 30m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bc4ab3b31b640d396c48eb02a4dbc2c : SUCCESS in 31m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a78b8f78e64a46b5b349e2545d5ea791 : SUCCESS in 35m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e380672ac304b7e857be4a897c0d6f0 : SUCCESS in 32m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c9124458c98b4c8aaa096d56cd9b9fea : SUCCESS in 29m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7baee3225da64a44b01d4aea0e329a4c : SUCCESS in 30m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e36466543a643aa984d53cf7099b98e : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a87ddf00d95a400a8481ca0697b34ce9 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f7d553e8b324146890b9717705a8cdf : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/50169eaf9d3f4a93bdc6cbde0f4b96bb : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b34d3806cde145c1b1de35d6e3480603 : SUCCESS in 56m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d741e1cd6817401a83cc4e994131eb41 : FAILURE in 1h 05m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08f928271eb74f76919917c0cfcb0c08 : SUCCESS in 37m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/263082a1022f42db95bf23cd67210bc5 : SUCCESS in 37m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4680cfb883b948d38b512096a5b4b14d : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6ca2b231a14d8fbfd757d42105fa6e : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1079b36d863d4a48b27d7a3514b63f51 : SUCCESS in 20m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eacae3f03a32464ea744b132174b6fe7 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e2d53a2f984825563a9540f1eefc02402741152b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-02-15 00:53:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5758a809f601d63785579f3804a1c250d35c0d2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-15 17:13:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b35ec778f4bfa61ea53fcc95897ee01c9eaa264d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 19:18:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"01a662eb45bdf30490ec680299e91d4295d50531","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 19:38:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/825d731056104e558de0eae80b950de9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ada8eef09dd445318860b81f3ef24504 : SUCCESS in 15m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0608217415f471abe72b04adaf09d4c : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e4feb03b65e4fd5821f8624e27c9742 : SUCCESS in 17m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2eb286116a4f410ab57145f3f817eca9 : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":3},{"id":"5d552f67f16ff88eb11709caf084e15c8215aa79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:43:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6828ca7788224f6ba57ef3daf4e806ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6344630544b7493db45ce3a786a66741 : SUCCESS in 19m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/13dfc615ce834175a3f3a3eec2476e00 : SUCCESS in 1h 13m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c99e5754e761497b811a74597d6577a0 : SUCCESS in 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fd4455ef044725a442a8bd81acc39d : SUCCESS in 53m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23b484ff06b9419382ba478b1693330f : SUCCESS in 23m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4177dc7eb67e4d78a43ff8bb1eee0c11 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0340f4cba3494ed1b0ade3349b3c900d : SUCCESS in 27m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a513b69c738430696a8ef6ce1ae3f26 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f76d335e50c64ff595dac888a3779f80 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ce6c27c718549d0baafdf74e19adaa3 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cddb3900312407a9ff7348f28a4132e : SUCCESS in 14m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/225a935a70e84c99ac6fad1190816438 : SUCCESS in 22m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a3a267615e944e66b6f8ab4cd73a0327 : SUCCESS in 16m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d44f7fbe63ed41f99552d019613f669a : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5918d6934204bed8cc6fffdcdf3d91e : SUCCESS in 24m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9ba92dacb844258bb14a2757d5a1259 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3e70cf905d59423bb777411adf413a39 : SUCCESS in 21m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9bdc395c08fb46899868991e689d2f10 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f0da399004e4109a5ed40b1a10cd2a2 : SUCCESS in 24m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e2906f068114ba6aeb5a1b4feada5cd : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1334286cc7ca449283e110f639118bad : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eeb5834813284cce864179ccbbd957cf : SUCCESS in 7m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/368c99dd886e444e8df955e243b223d5 : SUCCESS in 1h 11m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3890d2ba753c4351b3e72e3d45b9a593 : SUCCESS in 1h 08m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8c5cb72eaec4e2ea84ebd22050c2d12 : SUCCESS in 58m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/891a90844b284ae2a0fb53b0327a9b78 : SUCCESS in 59m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1d9b172a36b455882d1b5f84b8b5ece : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc850aaed686415d911767c1e0e05604 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/179e090ee6ea4d0c9597f02b3689f294 : SUCCESS in 25m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd8dd83ca89d4e2bb6c47510d3ed3519 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":3},{"id":"baadbeacd1fb9c76c8a2f77f9042c5e4279f2fd6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-02 05:48:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd7554b674159412913a44fd9e17e37353b218c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-15 18:36:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"306552f0c1fd80f2dd36baeec1bb1f4eb4723746","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-15 22:52:52.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8262cecff4332e3a52b74dcf82882cbef87beb9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-06 03:58:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2711250e3e42eea5f23cd0e244161e9278c14f13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 04:56:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94fb653a25f346f38be72e28177d8769\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e0723a1b1984197a6f9bc2c7596b378 : NODE_FAILURE Node request 300-0020917287 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5793fced0b34772a3be44d017862188 : SUCCESS in 52m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9e4be7c77704fdfb3f22aa0b12a807f : NODE_FAILURE Node request 300-0020917289 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a21a4bfa2b3e4818b29454a237b38e38 : NODE_FAILURE Node request 300-0020917290 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"81f58d5155247f008d48458966d3dd572c083f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 05:09:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b532216eeb3749c4ae85f79395d31174\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9fa10b99884c318e4969d6c6ce975b : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/acf99c3d67d2407fac92e6d447d39247 : SUCCESS in 59m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/186263f2505b4cc793c73e3af77e713e : SUCCESS in 44m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df86c79c4671408bb73023d3bfa5bbe9 : SUCCESS in 1h 00m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0a758e062d1460abdbc2fb139bf22e8 : RETRY_LIMIT in 4m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2a1a564fe5f4176a9d73c6c0d0a409c : RETRY_LIMIT in 4m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4046b50fa374a48834618c92671c3da : RETRY_LIMIT in 4m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba1ab6ccccfb471d83460d030cf4337a : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12cef49274264b19a28facf4eddea442 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae568eaa200644ed8331125a0e1a5564 : SUCCESS in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27a7193552644f699fa8c540629865fb : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c7d2129c76f4e6c9e4a55eab547b5fc : SUCCESS in 14m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d80ed07db9954fa0bf1a9dd8fc1364ab : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8855d1aa4d744b498736dee04b908f0e : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e49ffbeb90d46acbdfc8a4d4f09f843 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6aa209818aeb4466bb25e0fad928ceb7 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/34d6953d61e54c5da98b5cb599f52e7e : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4f7c03b9f4184c10b7a0295fdc9de11f : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b319fe40a2df47e7982f0be4d735a148 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1806889330754b4fb7147ec7c849c826 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a8bb13804284c8085f068895ac84be8 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/97817119b9ca4dac831812869cb071ef : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce451f85aa64498dbe32d2b1590c413b : SUCCESS in 1h 05m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ecf986db8c84044ace1419056f60f89 : SUCCESS in 1h 02m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09a6895139a64e269b75398d47832116 : SUCCESS in 55m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/beec48faab114c1f9913477e64d26664 : SUCCESS in 54m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ca4b4d215814d7d8985f940accd914f : SUCCESS in 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81140c6d92845e78bc631d597c8577e : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd26e62c279f4faaa71813f350c04a0c : SUCCESS in 18m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6172207b80754da0b03411b9abce281c : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":4},{"id":"2a761e40c11ae60a0df0142fbb997961426aa5f7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-13 22:29:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"86869276415bf3dc54ac18df96719cb7ace0261d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-13 22:29:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8a7d5b6a57b9112ad72a6413f24998804c14dd92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 22:56:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e566f7163624c0594516b0bc0cdc7bf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f49667cb41842e7a7bcf59e1030cb13 : SUCCESS in 14m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21c3f8e978af4ed0b58bf29f0b05d021 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0479e86bd4d0486b8e16a3eac5df8b4a : SUCCESS in 22m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bf8d1f870fc4b9aab6a64def9b56b2f : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":5},{"id":"1499aebc1647f28eaba69f1d53e3fd825c951fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 23:48:56.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6561165ed933469aac65080b5d856899\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152c26a8ff2547c68bae61653ad9dcab : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/33ddd77cf77b4169ab02339c70dcabcc : SUCCESS in 1h 06m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ada83c765aed40ab851684b397f07c46 : SUCCESS in 1h 00m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c428193c96ef4353b86ebc61db346288 : SUCCESS in 1h 04m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f191b821aa404bf9a416333cf88e2dd2 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0aac7efded9042c29362351f5f1bfd90 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e309176e1b0d4dc397dcaa5f4cea6d83 : SUCCESS in 19m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c25278f65044c6cadb6cf3cb315ba56 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/35fa80131dfe422d9a40bf8ed25e7bf1 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5315aa1c09c34b7f8f72808eebc480d9 : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/843080b562ea4080a7462098115b30c3 : SUCCESS in 18m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d5fbe5e5800048f088e15a178ec49ddc : SUCCESS in 17m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5683bbe4f45644f4b24db1a6f54a706e : SUCCESS in 18m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bfc611b623884dd4aa8cfbf2bf9e3c17 : SUCCESS in 22m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/94185297c14e46ef8c9c2095cc97ed28 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/972843a258e64f879de46b986d7834db : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c339df50593460ca4a7211d7aa174f8 : SUCCESS in 18m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cef034c477374bd9a69467557238c1cf : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/459103a3bcf44459afb90e0c50050f2b : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/906eaaed9445446fa29cf8dc25588462 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/087f124453b8423db4ac4dbc58e38e7f : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f84e9376a9ce4fe19ce30ada968440d7 : FAILURE in 14m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7aff048e180545b0bb5870c37e61bfed : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af6641570f3e44ae9c57b604aad7867a : SUCCESS in 58m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d06f1e3870b849fd868eeb7bbf514726 : SUCCESS in 35m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff4c5b24160c4a7dba51c49c4bf3b4ce : SUCCESS in 54m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/694c316d221e48a3865a658114654c5a : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8567d6a29d704068a89f690e8acbd94c : FAILURE in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9e098b55adb45f1afe6b8f10632ba63 : SUCCESS in 17m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b26ba69418b342fa9cfd80162a4ead6b : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":5},{"id":"dd77c7bad37eb7876b9113b73f87d8e58ac47584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-17 18:14:29.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4698abf9fc4aa31eb71025e4b804114356ef512a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 18:36:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406c874d72aa4fa493a3b318c9832554\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1ff686e9b584971940fb7b8fb50a9cb : SUCCESS in 15m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5c8bd85ff0f459c87f4e50193b2aaa9 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/14fe98e98671431a80ae19266a4fc031 : SUCCESS in 20m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ae780fb89d9477b8a03fb3e4642310c : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":6},{"id":"d7484b4b0366dedb881915f6ca1997984ef5a39c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 19:28:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c32e4006134b269ba5706991d2218c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41138d8bf11f4e25bd5920617cd1a0d7 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/99e692a88b0f4b299b8b30515ad4ea0f : SUCCESS in 52m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3779516cb1c049d8b14799bfcfe8a72d : SUCCESS in 54m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cd2c935efc84f37b629ba35bd6aa131 : SUCCESS in 47m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/290299e47df74d35a93af93d5bc0bec9 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b1aa729489142e3a296510d12a0ecba : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/505f232279d54f50957346f75d3979a1 : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58c3100330484be69c3a31361121a044 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0935497b249840e0bfb995078521a32c : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0d41147fa2a42d582fa32ca2335208c : SUCCESS in 21m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03fd6d35ece44b6f9d7ae35439ca43ef : SUCCESS in 18m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e1c2cf6db6845dfacf0a1a645206fd1 : SUCCESS in 14m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3da8fd4ccf1e426b873f00515d024c06 : SUCCESS in 16m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/abb179427f7b4ba3b5167891bdb063a7 : SUCCESS in 23m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a433404c0e0847dd9166e49d2dd7205a : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0ab1d31115c4aa993a8bff05adc7864 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/329604382732468182ffeafbc4206e60 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0bd26a32ccf34ee2861a6a896ef3eea8 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/89002c91967b40faac575563d05f89be : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/88115b47cadd462bbe1202559044ee1f : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eaaf8ca0a46745929f9cf0383db88242 : SUCCESS in 5m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fdb5981b24c6486a9776279fddb7f232 : FAILURE in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5682feb60c634383a869fb8dbad72916 : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1094ac9cb97f4dd097d91e98d11b2231 : SUCCESS in 54m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/673a6ac70ba749d1bd04d6bfbfd2c86a : SUCCESS in 52m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5ee18a2649645049e88d6cbc19a0bf5 : SUCCESS in 49m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b86bfd14994f0e85bc36d4d58e1106 : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2254db2d68ab4218abb8c23ad429c01c : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3d5a32b6bbf4af5a0814b58f09b5242 : SUCCESS in 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c7786b1993a41938db3fbbf28f6f0ef : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":6},{"id":"73487fa192e492484925975ebc671e84232b8f5d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-06-17 23:43:34.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"888f724ef393885569aac1bb457de6c75e2187b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-24 22:14:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"01207981b109a0036a8d766cd6fd51350e2ec560","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 23:06:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ca4d1fc10e43b3a71ab74bab9531f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/35a46f9bd40e497ab57a2d6585fd6719 : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":7},{"id":"b51f4170b48b34d456ac6d4b733985ec34bce31b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 23:13:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf761ec086384b6b973dbd7ed7315336\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36fe0b2418b44e4b3839fc3daac1c83 : SUCCESS in 4m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/941a44b4ad074b0ea70375a18933c733 : SUCCESS in 39m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9217e48e2d2a454287d3bf3874c2aaa7 : SUCCESS in 41m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b568bba5f51d40d4b4cf5b285c0c5cb4 : SUCCESS in 37m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3cb199bd48c450d8f5ef0fa117bc10e : SUCCESS in 45m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cbc5018511f44de85b613b33de658e1 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6ed77adacce4ca28d896e88d2c7053c : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/312597dc888e4e72b203888ec3e15a26 : SUCCESS in 15m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e9b54aca3084b12a2324a8bf4b2e758 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/949be743d82c4558830968a6694ba8c7 : SUCCESS in 13m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0dd7e099fe4adf841f7f8879a90d48 : SUCCESS in 13m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fabd331d67ab4fe5bfa835ff58e3a246 : SUCCESS in 16m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/48af27b5e0994466b74e83958fc83909 : SUCCESS in 17m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/77186d7c777048e5bc763b5f1c1245d0 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d453d74f11b4e84b3465c23bbdcffd0 : SUCCESS in 7m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7949c9f3a314466aa8e15094a6757895 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a3c3336d4d94e5e9a4196f7fa5fdaaa : SUCCESS in 3m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2051c45ee3ca4b63acf542939c4404d6 : SUCCESS in 14m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/47956fcc8f82420485e0a5cb9fd9c791 : SUCCESS in 3m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b790271117ed460bb08628ce9f2cdaa4 : SUCCESS in 53m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dde538910e2d4b0f940b03de35183ee7 : SUCCESS in 17m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a2f6635cde84fd6b422c22c4881d2e3 : SUCCESS in 47m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6124d631bde48f2b947a836377edf3d : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1108f36729fc4bd487597e076521c399 : FAILURE in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5646bcfee3334ec7a8b1de052a3f14b7 : SUCCESS in 20m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3a931fc710f4bab946106f8e69357b8 : SUCCESS in 32m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d1f0f3e87f448018cdef4d6bf0cc1d3 : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":7},{"id":"95f5abe3ac288fb371249037da942f0e9175de7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-25 21:49:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bf7128a53420795b68648e01e54b2e2666fa12b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 22:45:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/563aa34d4bf74a65bee4830b1d958dc7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1d24638068146459ede04342fed781d : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":8},{"id":"fcc65552e5e3a78fe45b065b59961ed1288e3f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 22:51:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183713917d2c400aa7ee8ee40eff7a6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1543d7d8149466a89cb098dcefa628c : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/bd101e04cff44b94a23647ea1c9092e7 : SUCCESS in 50m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62816b7074874a67b7fcd7e0b9ca5636 : SUCCESS in 26m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07a8eb796c364e07a9c1b8f7cf511175 : SUCCESS in 50m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/485f41c9fe8245aaabcdd8254b4fcea5 : SUCCESS in 1h 01m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b88198d175e46bfb0225a0d54692bd0 : SUCCESS in 10m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6d920ad8df245d9a41448c218b3bd41 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2efc659f9c6e482987b7e588f727d608 : SUCCESS in 16m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d2a2284eba64005917b6b1b2d681cf5 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b66b84b897a41fab33316120c9ce049 : SUCCESS in 13m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/688a1e2927fa40f5b67a50ed3c7e6d2f : SUCCESS in 13m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/93f8affa1fb84158b18cc668b4683aca : SUCCESS in 18m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b4a664a1c62e4ce1840bfe237cda9a2e : SUCCESS in 17m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/02d6044385074ece84a3b716d7138a22 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e034e98d68ea4e74855afca15ae81bc4 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b9fa22545c4f4d0bafbd900b833d4b52 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a96d1465dd8d45a5ad42d2ea7817ba46 : SUCCESS in 5m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88fb12521f35429c8f57e89c7655256a : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a7cb3fb61044e26a485082622fe2786 : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dddccf4b3afa4d6ab1a1b11f956e9f12 : SUCCESS in 51m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90ddf2c1b725489b950bd86b68827375 : SUCCESS in 32m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7277b19fded0457b80e987c4ff82aa00 : SUCCESS in 42m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/089e88b9da5545b482b162a70cd00949 : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9015ad155e434ca8993f8904f6387272 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05ff10a3df2d4453a695a6ddd42cc98e : SUCCESS in 20m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6ad7682fe2d47f7bbe39fb39b0b6745 : SUCCESS in 37m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8de8f99dbcd4bd3ba3efa87a73d0ff3 : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":8},{"id":"9f4e321228165dbdc9f33fcd48658c267d3828a4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-04-25 23:33:56.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b99c25ebf9932688463f30d251b43d97e0358596","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-04-30 21:29:58.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3bffbae0c5837a386d7ad570ec6bc116e8f93229","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-30 22:37:53.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0484bd2fbbde1bafeec26e7bddc4fc60eda2151d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 22:38:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5fa5b13a45ce70fac53513c8a55b38b17296893c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 23:45:48.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15fd0615da2c4f2bba5ae4d121b2eebc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55590fd66efd4dcd8db35695e7f2ffb0 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/62a4c21bfe79485fa87a9f5dfbeec182 : SUCCESS in 46m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54f7958369e44d3d8b2e1be2f47b6cb6 : SUCCESS in 43m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3aa9045251524b16beb61183688c325f : SUCCESS in 38m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99b1ed6ec40a4e32bfa7ee40034e2cd2 : SUCCESS in 48m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2262956cb19b45ae89d99bb4c0afb0ae : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ffbf950e7114a539c1e089758134db8 : SUCCESS in 18m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32740aeda4954d90856f3f5d4eca912c : SUCCESS in 19m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/955ddf1fd035495db1b1b3477e6096dc : SUCCESS in 16m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/24f21803da464be6a7701677247f4b7e : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b78e8a7f0164226a3052f2f1d3fad44 : SUCCESS in 7m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/84907761f65d4aeb88c689212379d5cb : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66efab4827eb4194be4108cfd1f802c7 : SUCCESS in 5m 19s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f77fca8562bb45da92a3acbc6d0da34e : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e7a0519d25c478591759a18f62e27a1 : SUCCESS in 46m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8864e9a99ace4640935c54b0f03bea15 : SUCCESS in 47m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d86cf3aa9ce4c1f87a69df9b0f6dd0f : SUCCESS in 45m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05bd1a7f52344acbb5b3d3e61019fe8e : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b41f3497f814473926f8758eb0f99ad : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/175f2dfa85584508b048394dacf9bb48 : SUCCESS in 21m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72e179580a7c41129b8741eab6b91f1a : SUCCESS in 50m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9ca5a1f64d8470e829b7b5a8764d944 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":8},{"id":"3863c90935cc0c7dfc9f121712c56c2f4823c78d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 23:45:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"2f73a6d0a1132943fed314087084bd98bc6c86ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 23:46:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98836db134a4f4483055d4103545186\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ba61245d024750ab2080d48f24b115 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/62257bea90dd488ea85012859fc12c25 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"74030236ad1ba955f3904917dbc405027e5d6905","revisions":{"66402337f69ef8fddd9c388094d7dfa9e6a37fbf":{"kind":"REWORK","_number":1,"created":"2022-10-13 23:58:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/1"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 23:54:10.000000000","tz":-420},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nAdd a new config option, fernet_key. If set, use fernet tokens for\nprimary storage rather than memcached. This allows generated tokens\nto be validated even if memcached is restarted.\n\nAdditionally, opportunistically compress groups before minting tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66402337f69ef8fddd9c388094d7dfa9e6a37fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66402337f69ef8fddd9c388094d7dfa9e6a37fbf"}]},"branch":"refs/heads/master"},"7d2cc7092e8a74c6be7654035ae0fc49f32bd293":{"kind":"REWORK","_number":2,"created":"2023-02-14 20:11:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/2"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-15 04:10:09.000000000","tz":-480},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nAdd a new config option, fernet_key. If set, use fernet tokens for\nprimary storage rather than memcached. This allows generated tokens\nto be validated even if memcached is restarted. Note that the key\nmust match across all proxies in the cluster.\n\nAdditionally, opportunistically compress groups before minting tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d2cc7092e8a74c6be7654035ae0fc49f32bd293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d2cc7092e8a74c6be7654035ae0fc49f32bd293"}]},"branch":"refs/heads/master"},"4ed7cede2534bb0b5608207a99941ae7044c1f43":{"kind":"REWORK","_number":3,"created":"2023-02-17 19:18:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/3"}}},"commit":{"parents":[{"commit":"63e618b232c5f1a50e16baa0e685f4e25716f737","subject":"Test x-backend-replication header with POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e618b232c5f1a50e16baa0e685f4e25716f737"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 19:16:52.000000000","tz":-480},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nAdd a new config option, fernet_key. If set, use fernet tokens for\nprimary storage rather than memcached. This allows generated tokens\nto be validated even if memcached is restarted. Note that the key\nmust match across all proxies in the cluster.\n\nAdditionally, opportunistically compress groups before minting tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed7cede2534bb0b5608207a99941ae7044c1f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed7cede2534bb0b5608207a99941ae7044c1f43"}]},"branch":"refs/heads/master"},"bc0866c8c5fe9cc0f614f676b9c1a7fe03e5f332":{"kind":"REWORK","_number":4,"created":"2023-04-06 03:58:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/4"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-06 03:58:03.000000000","tz":-420},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nAdd a new config option, fernet_key. If set, use fernet tokens for\nprimary storage rather than memcached. This allows generated tokens\nto be validated even if memcached is restarted. Note that the key\nmust match across all proxies in the cluster.\n\nAdditionally, opportunistically compress groups before minting tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc0866c8c5fe9cc0f614f676b9c1a7fe03e5f332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc0866c8c5fe9cc0f614f676b9c1a7fe03e5f332"}]},"branch":"refs/heads/master"},"310ac792075a75574463a98fd6a72dbd203c6e64":{"kind":"REWORK","_number":5,"created":"2024-06-13 22:29:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/5"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-13 22:28:51.000000000","tz":-420},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nTempauth fernet tokens use a secret shared among all proxies to encrypt\nuser group information. Because they are encrypted, clients can neither\nview nor edit this information; it is an opaque bearer token similar to\nthe existing memcached-backed tokens (just much longer). Note that\ntokens still expire after the configured token_life.\n\nAdd a new set of config options of the form\n\n   fernet_key_\u003ckeyid\u003e \u003d \u003c32 url-safe base64-encoded bytes\u003e\n\nAny of the configured keys will be used to attempt to decrypt tokens\nstarting with \"ftk\" and extract group information.\n\nAnother new config option\n\n   active_fernet_key_id \u003d \u003ckeyid\u003e\n\ndictates which key should be used when minting tokens. Such tokens\nwill start with \"ftk\" to distinguish them from memcached-backed tokens\n(which continue to start with \"tk\"). If active_fernet_key_id is not\nconfigured, memcached-backed tokens continue to be used.\n\nTogether, these allow seamless transitions from memcached-backed tokens\nto fernet tokens, as well as transitions from one fernet key to another:\n\n   1. Add a new fernet_key_\u003ckeyid\u003e entry.\n   2. Ensure all proxies have the new config with fernet_key_\u003ckeyid\u003e.\n   3. Set active_fernet_key_id \u003d \u003ckeyid\u003e.\n   4. Ensure all proxies have the new config with the new\n      active_fernet_key_id.\n\nThis is similar to the key-rotation process for the encryption feature,\nexcept that old keys may be pruned following a token_life period.\n\nAdditionally, opportunistically compress groups before minting tokens.\nCompressed tokens will begin with \"zftk\" but otherwise behave just like\n\"ftk\" tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/310ac792075a75574463a98fd6a72dbd203c6e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/310ac792075a75574463a98fd6a72dbd203c6e64"}]},"branch":"refs/heads/master"},"f2360ecc999483aad240173800788c45b5840790":{"kind":"REWORK","_number":6,"created":"2024-06-17 18:14:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/6"}}},"commit":{"parents":[{"commit":"46e183df15d0a98d2595e583d2a50c98aba153e2","subject":"Merge \"add bytes of expiring objects to queue entry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46e183df15d0a98d2595e583d2a50c98aba153e2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-17 18:11:28.000000000","tz":-420},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nTempauth fernet tokens use a secret shared among all proxies to encrypt\nuser group information. Because they are encrypted, clients can neither\nview nor edit this information; it is an opaque bearer token similar to\nthe existing memcached-backed tokens (just much longer). Note that\ntokens still expire after the configured token_life.\n\nAdd a new set of config options of the form\n\n   fernet_key_\u003ckeyid\u003e \u003d \u003c32 url-safe base64-encoded bytes\u003e\n\nAny of the configured keys will be used to attempt to decrypt tokens\nstarting with \"ftk\" and extract group information.\n\nAnother new config option\n\n   active_fernet_key_id \u003d \u003ckeyid\u003e\n\ndictates which key should be used when minting tokens. Such tokens\nwill start with \"ftk\" to distinguish them from memcached-backed tokens\n(which continue to start with \"tk\"). If active_fernet_key_id is not\nconfigured, memcached-backed tokens continue to be used.\n\nTogether, these allow seamless transitions from memcached-backed tokens\nto fernet tokens, as well as transitions from one fernet key to another:\n\n   1. Add a new fernet_key_\u003ckeyid\u003e entry.\n   2. Ensure all proxies have the new config with fernet_key_\u003ckeyid\u003e.\n   3. Set active_fernet_key_id \u003d \u003ckeyid\u003e.\n   4. Ensure all proxies have the new config with the new\n      active_fernet_key_id.\n\nThis is similar to the key-rotation process for the encryption feature,\nexcept that old keys may be pruned following a token_life period.\n\nAdditionally, opportunistically compress groups before minting tokens.\nCompressed tokens will begin with \"zftk\" but otherwise behave just like\n\"ftk\" tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2360ecc999483aad240173800788c45b5840790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2360ecc999483aad240173800788c45b5840790"}]},"branch":"refs/heads/master"},"8c03888a8c86ba1321569767e25f9722d2584b5e":{"kind":"REWORK","_number":7,"created":"2025-04-24 22:14:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/7"}}},"commit":{"parents":[{"commit":"64bd1acb9ec0b6ecc9a2f1c56c053c0de6f5efe6","subject":"Merge \"Clarify that get_ppid is Linux-only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bd1acb9ec0b6ecc9a2f1c56c053c0de6f5efe6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-24 22:13:00.000000000","tz":-420},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nTempauth fernet tokens use a secret shared among all proxies to encrypt\nuser group information. Because they are encrypted, clients can neither\nview nor edit this information; it is an opaque bearer token similar to\nthe existing memcached-backed tokens (just much longer). Note that\ntokens still expire after the configured token_life.\n\nAdd a new set of config options of the form\n\n   fernet_key_\u003ckeyid\u003e \u003d \u003c32 url-safe base64-encoded bytes\u003e\n\nAny of the configured keys will be used to attempt to decrypt tokens\nstarting with \"ftk\" and extract group information.\n\nAnother new config option\n\n   active_fernet_key_id \u003d \u003ckeyid\u003e\n\ndictates which key should be used when minting tokens. Such tokens\nwill start with \"ftk\" to distinguish them from memcached-backed tokens\n(which continue to start with \"tk\"). If active_fernet_key_id is not\nconfigured, memcached-backed tokens continue to be used.\n\nTogether, these allow seamless transitions from memcached-backed tokens\nto fernet tokens, as well as transitions from one fernet key to another:\n\n   1. Add a new fernet_key_\u003ckeyid\u003e entry.\n   2. Ensure all proxies have the new config with fernet_key_\u003ckeyid\u003e.\n   3. Set active_fernet_key_id \u003d \u003ckeyid\u003e.\n   4. Ensure all proxies have the new config with the new\n      active_fernet_key_id.\n\nThis is similar to the key-rotation process for the encryption feature,\nexcept that old keys may be pruned following a token_life period.\n\nAdditionally, opportunistically compress groups before minting tokens.\nCompressed tokens will begin with \"zftk\" but otherwise behave just like\n\"ftk\" tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c03888a8c86ba1321569767e25f9722d2584b5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c03888a8c86ba1321569767e25f9722d2584b5e"}]},"branch":"refs/heads/master"},"74030236ad1ba955f3904917dbc405027e5d6905":{"kind":"REWORK","_number":8,"created":"2025-04-25 21:49:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/861271/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/861271/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/861271/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/861271/8"}}},"commit":{"parents":[{"commit":"64bd1acb9ec0b6ecc9a2f1c56c053c0de6f5efe6","subject":"Merge \"Clarify that get_ppid is Linux-only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bd1acb9ec0b6ecc9a2f1c56c053c0de6f5efe6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-13 22:29:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-25 21:49:12.000000000","tz":-420},"subject":"tempauth: Support fernet tokens","message":"tempauth: Support fernet tokens\n\nTempauth fernet tokens use a secret shared among all proxies to encrypt\nuser group information. Because they are encrypted, clients can neither\nview nor edit this information; it is an opaque bearer token similar to\nthe existing memcached-backed tokens (just much longer). Note that\ntokens still expire after the configured token_life.\n\nAdd a new set of config options of the form\n\n   fernet_key_\u003ckeyid\u003e \u003d \u003c32 url-safe base64-encoded bytes\u003e\n\nAny of the configured keys will be used to attempt to decrypt tokens\nstarting with \"ftk\" and extract group information.\n\nAnother new config option\n\n   active_fernet_key_id \u003d \u003ckeyid\u003e\n\ndictates which key should be used when minting tokens. Such tokens\nwill start with \"ftk\" to distinguish them from memcached-backed tokens\n(which continue to start with \"tk\"). If active_fernet_key_id is not\nconfigured, memcached-backed tokens continue to be used.\n\nTogether, these allow seamless transitions from memcached-backed tokens\nto fernet tokens, as well as transitions from one fernet key to another:\n\n   1. Add a new fernet_key_\u003ckeyid\u003e entry.\n   2. Ensure all proxies have the new config with fernet_key_\u003ckeyid\u003e.\n   3. Set active_fernet_key_id \u003d \u003ckeyid\u003e.\n   4. Ensure all proxies have the new config with the new\n      active_fernet_key_id.\n\nThis is similar to the key-rotation process for the encryption feature,\nexcept that old keys may be pruned following a token_life period.\n\nAdditionally, opportunistically compress groups before minting tokens.\nCompressed tokens will begin with \"zftk\" but otherwise behave just like\n\"ftk\" tokens.\n\nChange-Id: I0bdc98765d05e91f872ef39d4722f91711a5641f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74030236ad1ba955f3904917dbc405027e5d6905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74030236ad1ba955f3904917dbc405027e5d6905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
