)]}'
{"id":"openstack%2Fswift~861486","triplet_id":"openstack%2Fswift~master~I9823d00e78479e3335f7a25d0d2c99b63c1f9c12","project":"openstack/swift","branch":"master","topic":"per-policy-quotas","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-02-27 01:07:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-02-28 22:44:03.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-02-27 17:05:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I9823d00e78479e3335f7a25d0d2c99b63c1f9c12","subject":"Use FakeSwift in account quota tests","status":"MERGED","created":"2022-10-15 00:05:55.000000000","updated":"2023-02-28 22:58:53.000000000","submitted":"2023-02-28 22:44:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"861486-per-policy-quotas","meta_rev_id":"118354d5b06e49d8a245692e350a0ddf44f95c7b","_number":861486,"virtual_id_number":861486,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 22:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-02-27 01:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-02-27 01:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-15 00:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-09 23:51:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-02-10 06:20:00.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ba1831f9145647bab03adc43231ff4513346f05d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-15 00:05:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b227b0bbaa0d3a56f0880fd3173d58098c65b13","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-15 00:11:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c302103582f621bd4c61f69cf9ff3c2b021e49b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 00:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/720911802ff94150953738ecd102fd03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec9d8964f534fd78484f934c94faf06 : SUCCESS in 7m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d37c49e704440758d5001aed6fd7c1c : FAILURE in 17m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e3fd1d97e2441308ab2404b993ce9d5 : FAILURE in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d4348939bdc4d629498527aa0742a91 : FAILURE in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0efc61a311c4d7d82eddef84a168601 : SUCCESS in 7m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1666251c33a483892f86bc9b3d8e01b : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42d682e1506045fb87c53791ccf20005 : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/63178332c4a144859726b725bf06bef8 : SUCCESS in 14m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3e8181b2d3a04fe2a025ac1e60f09cfd : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5a3e8ed678b44d1582d6f1392a3b8a7e : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/483ce8121b1d4d549db7cc828e9a0ad2 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0411d2cf43e4af5a2abee05bcb3bcfa : SUCCESS in 18m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4962e9157802417d805e624c09628513 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c047ea42acd4f0ca2f686c023e69cb1 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f9213532f30847c4a296f6cab431cfd2 : SUCCESS in 15m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68f21a8a9e43424d84bf2367f97b183f : SUCCESS in 48m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/153d2773df624aa9be67558f0fb840b0 : SUCCESS in 43m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/123b978353b543d7b788fb05ac161cf3 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ceccbac14f40138548f8f138c8900f : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2dc2e34c8cd4412f9bd000968403811b : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9588b798ba985a7e7cd05a4e9ce87d149032ca51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 01:32:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25e52d2275464e7298d9231ffab218ab\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/374867cc68fb48638791efe696bae3a1 : SUCCESS in 1h 01m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/885df1cfc41541d087476de6502b5a9a : SUCCESS in 1h 17m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ee3331cf99841ca90071497f14bb469 : SUCCESS in 47m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/741a1f9b6e1449249d5102eba425ee33 : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5c0478f00f90b26d9382050bbdfcf291f1b7e8df","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-09 23:51:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5d4e37cda10625c102dd8b1b04ed9314d4ef174f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-10 05:36:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0fbad1cb734ae0997259de677255c88c6f3aa84","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-02-10 06:20:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"85be716bf121d4c18712908216c8bb8e0652c652","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-14 20:14:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc43d4f5d856627549f63628967e2e3a12815ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-14 20:16:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4cc8802b002fabd9c0d4d78b6de9c80d69ec520a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:38:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82d1211bb4c24dd994f10b44dee583c6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bce8613f73794b66b06a3a876f4b3817 : SUCCESS in 16m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c67b9c6943014d408e6c5a38e26b1c3e : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2ad049f026a4c578db3a72fcd26d0c3 : SUCCESS in 19m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/136598860eae47fcbe8e5c3457e100d9 : SUCCESS in 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"964758d244248e8c24a52e3953b62a646e87e55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 21:19:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f86c1907564b26af6c0ab2ca0f1e0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a943c033a7420f81d121d1f2b7167e : SUCCESS in 7m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e53dd45a6124c70ab1e1a3ee375659b : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0336124524b4c9793890b6fedee11e4 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdf6ba69423446348f2741ffefc7dcbd : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c1f7ec51512484cb7b62084923a202c : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a2276e3c912a4b64bb9eb59a868e3901 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ed1f02c3cd946048360a59c76fc0c64 : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5aaa0f8acae4e2a8295e5e36b50657e : SUCCESS in 28m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/270c13e14762494ab564b3360323fe44 : SUCCESS in 15m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ab2e983de9f24dcaa87c22a69af95a4f : SUCCESS in 14m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9304e18c05624ce18809dc4f3d491289 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ccb148f244a7428695e4d9e97e18bc88 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a375cfe10864f7db3fe9cb1f18927db : SUCCESS in 22m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ee249006c86403c9045dfeeddfeb40b : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c65918721c14a84b666e62732a988fa : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5333b8f05dfe4d26a172d48395e76667 : SUCCESS in 18m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5233e4687d26412997eb4bdb43075780 : SUCCESS in 57m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/859a1e3aaf9d40839a99712b5ef40715 : SUCCESS in 38m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef205f32609842e1afb356a25dac4517 : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db70a9e00044f33b3b5738c8800fede : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58ed1d30617f4d0181445564311a1021 : SUCCESS in 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56690416d4f4089ecea281162a5d5d2d4212c293","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-15 15:41:02.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a79e2be6885edf20a6ab1fa04ae97c71fd75842","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 18:09:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"676d781b34f7b318dec01759eb2f3de12734e76e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 18:10:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bd90abdac5f988415a9397765cee433efd459f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 19:05:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6596104153d47f18ce79a720e92cc84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f5a6f32f7543c691eb7e0f7fdb2cec : SUCCESS in 9m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/769c004064214dcb9ab2571a6688bce4 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba692e5f9a6945f6bd7ba6736a94f85e : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/243d9a01525d42c2b2e5ccef3a126aa9 : SUCCESS in 17m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/10310250c96949da92dd9852001dc2bd : SUCCESS in 10m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/758f4597d02043fabbe54b8e0a1937b6 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fcaa98f120364ea9a023410d825b2209 : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/55872061b8a34905aa0bb1608a74e032 : SUCCESS in 17m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/843ee0514abe4e47956d8456170f0c05 : SUCCESS in 16m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ddc5f805b7ee4a26b974fe2006e7a36a : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea24f00b05714e498ab7e1ffceffae58 : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f73fbec26844190a761869b88c4a443 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1966badb1b544d62ade90a1f748e0269 : SUCCESS in 16m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e65d09279724142b11643cf80b7d33e : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0eee64e7fe1342269445abfdb2f2d4ad : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c6b13188c144694a452e4682f6236c7 : SUCCESS in 14m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/636cce9b9c2c48f2b2648ab47681675f : SUCCESS in 49m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d85baa922afa467f8a3a6591a3333a2c : SUCCESS in 49m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7161f07c7888410ebb4c443e5b0edb71 : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bfaa1921efe4f018af6be4becc8bc99 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33ef462d6f114d7988643dd6d8e75075 : SUCCESS in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"62a7d098592bbdb141e912d3f5e706bad4813da8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 19:45:40.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1a73df3a744a379be24c08efc5c354\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e273cf7660fa4cfb908cd3fa1d68efb0 : SUCCESS in 14m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24705fddac5d49bd9a1d86b48093c91d : TIMED_OUT in 1h 31m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2695cdb441c6436ea3071ad6457eccfd : TIMED_OUT in 1h 31m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7f8ef2bb96d499f9cc6b731632496d5 : TIMED_OUT in 1h 31m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9759be32efcf0fc03ff8752bde2165a713b4ea8b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-02-27 01:07:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ca14c25ed165c6df484d21b9c1776503e578cc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 01:07:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c0404a0547787de0ad140c11a2a7c4780266f3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 01:51:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97f0b8d054be4edfb9dd2cc046f82ebd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0fc28b3417407e87c44ea4ea01a4d6 : SUCCESS in 7m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/85b9062a2c3044138a9fba36ab3c78e9 : RETRY_LIMIT in 2m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1f2c6665880644cb8b89bee0cac8fd7d : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4076a143be69453c8d85a87e17bbb20a : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/152aae2c6d6441d8b0cbae76439194d3 : RETRY_LIMIT in 1m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29ea6033bc064f7ca49f820034c3356a : SUCCESS in 14m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a2ccb67d75ae4c958767ad30320b247f : SUCCESS in 15m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bd0ddb2c4b04f18b957ba89e85d73db : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/591e0ae72a6b40b4a3d23bb6a46d0a7f : RETRY_LIMIT in 1m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c62caaf7a9244dea2c97a2336ac08cd : SUCCESS in 18m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d2784de40ee4805a4cfaf1b2725a061 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e73da400b3f64a2d9e427c357bec6ac3 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a58cd092ec3e4bfc8efb991515a327e5 : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c01759edb08477c950f9e0e8de33dd1 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2267f10e413f4a90ba21371aeeac5b5f : SUCCESS in 7m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94b780d4509f4de189fb815edcc138ca : SUCCESS in 33m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17adeeb3cc0348a0b1434124fb492c2a : SUCCESS in 33m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb85ee63808d4df79a8632314ba04a59 : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78fc1e76bae4be4810fb0cc7c57328f : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"6778a27f24b14e59a45f438592b3ddeee89ffd38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-27 17:05:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23f592c568ad1083b35650e2e01beaf8e06ae3fc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-28 17:50:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a10cdecaddf305a32f7298206c492764b265a69c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 18:19:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58c83b44132347c786f9826fdd11bc48\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9c202906fb54d69bdb2cd1334493c44 : SUCCESS in 18m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cba9366fd244453b6e4d985376ea00a : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/36f25f5848b34b1099341c084d2566e7 : SUCCESS in 27m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/906b6394fe064780ac4a47b711c45ea9 : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":3},{"id":"00f134c4ef45d9f1141feeaa7e242228a0a69636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:58:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7400337bfed4fe4895462f4e90aa192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7cc32d8a024b4da79401750195d435 : SUCCESS in 8m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/112bf15818dc49bd8cf9ddc6a15b3e87 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e779c754f6a4644956b0b59945060b6 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7d9a157a82d4593bcf6f7dfc58fe96a : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f341dd2233ed41be955b51a365fff5f9 : SUCCESS in 10m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c8dcf4766b9405aa50cbeb638769ffe : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6955f572cc6b49fe8f9c3d90caf66d9c : SUCCESS in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69536cb5218347799d5ef091d0e9338a : SUCCESS in 15m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/078c969e5f9843538af06bc76c025d1d : SUCCESS in 16m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4ea63372cae74e0bb64115638d473aab : SUCCESS in 16m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b1aee6485c741de8cdc7776c07c0e8d : SUCCESS in 39m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa5c575d55c44f96be87c357b44b4f9d : SUCCESS in 18m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b6ff404f9d24c8992719bcbe9da660d : SUCCESS in 15m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d0e84074f59648bea37bc926cb7ca769 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1262762bcabf4183809b272eb44abe30 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/995d28ea6be64c6b804c4c5d0a2f4013 : SUCCESS in 24m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8238eec2ad9f4dbab5656dee5a8fd51b : SUCCESS in 54m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a80cd4853dfa423a809814fb47d64040 : SUCCESS in 57m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87cfca6daf1e4d6eb8e67a6b8c31d2d4 : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09940082cd8844f28dcfe1b2ad022994 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec5235df7d144b00bcf336769a03a43c : SUCCESS in 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c974f848aca19ad567cf58c52dec6b7eb6e3f721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:59:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3a381638bfefd195ae8290cf44bc06d784873a96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:44:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e866a658754f5bb37947784fa09ec1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba528f3b660431ea38d06fee5c2d601 : SUCCESS in 8m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9ed0733f2844425399ea13d67028b35a : SUCCESS in 9m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1c471afd48044a19b01681be2a392c19 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d7ad1f535f74b42ae4faf5d1394c346 : SUCCESS in 14m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07bfd2b6db9541e98e96c7c674bd8ff6 : SUCCESS in 14m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6957dc0d5c0f4273a4a96ee489025122 : SUCCESS in 24m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b8096fc8bd924bbb97c9e2beee748ceb : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d26fefb73fb0435e8ef46e990f222fb1 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/898866d741c24382b05a28a6c5af9634 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4553a355bc714222956258533cfde664 : SUCCESS in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d3706eb60edc482bac4fa607a88cee3d : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eae92f62cf9e4eb6b5fc8a6088ae779a : SUCCESS in 13m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2d5a771eab964b7a9134fbd9874779e0 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/452a9284b95645efa81301300c185a38 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c25cad7373bb4060b4e3e1bd89281808 : SUCCESS in 7m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d3e9093754f4873a78fd36dd5648be6 : SUCCESS in 51m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e28ecbd39e264bd8b4644f4b37065f6c : SUCCESS in 35m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e51e4765b264b34a6716649aabb1416 : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c57758073eb941efb6d4ee42c9e4e4a7 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"75da91a0131382287b5458e285da3e237888428a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:44:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"118354d5b06e49d8a245692e350a0ddf44f95c7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:58:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff8cb372cd14c3c945487ab39369267\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb15f63f9b9048039c01c8bd1e21d236 : SUCCESS in 13m 34s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2fd20b3abb854f1cb040f30dcfaf2c20 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d10326a96df54bc19ccef4668249dd84 : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e76d443e00ae5f504bac4f109317fc489b5f1bc5","revisions":{"f046e910d6ab95941a25ab6b115adeeeb1cc0591":{"kind":"REWORK","_number":1,"created":"2022-10-15 00:05:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/86/861486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/861486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/861486/1"}}},"commit":{"parents":[{"commit":"3ad39cd0b83a7f70d6c559c7b0e68a2e625be179","subject":"Merge \"sharder: always get ranges from root while shrinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad39cd0b83a7f70d6c559c7b0e68a2e625be179"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-14 20:32:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-15 00:00:36.000000000","tz":-420},"subject":"Use FakeSwift in account quota tests","message":"Use FakeSwift in account quota tests\n\nThis lets us be much more controlled about which headers come back in\nwhich responses.\n\nChange-Id: I9823d00e78479e3335f7a25d0d2c99b63c1f9c12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f046e910d6ab95941a25ab6b115adeeeb1cc0591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f046e910d6ab95941a25ab6b115adeeeb1cc0591"}]},"branch":"refs/heads/master"},"5f4a96cf505310cef1b93ea2e79ba7baec4eed9f":{"kind":"REWORK","_number":2,"created":"2023-02-14 20:16:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/86/861486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/861486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/861486/2"}}},"commit":{"parents":[{"commit":"9ec90d4d56c598ab5567128fd59dca8c54fe8dc4","subject":"proxy-server exception logging shows replication_ip/port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ec90d4d56c598ab5567128fd59dca8c54fe8dc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-14 20:32:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-15 04:16:24.000000000","tz":-480},"subject":"Use FakeSwift in account quota tests","message":"Use FakeSwift in account quota tests\n\nThis lets us be much more controlled about which headers come back in\nwhich responses.\n\nChange-Id: I9823d00e78479e3335f7a25d0d2c99b63c1f9c12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f4a96cf505310cef1b93ea2e79ba7baec4eed9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f4a96cf505310cef1b93ea2e79ba7baec4eed9f"}]},"branch":"refs/heads/master"},"e76d443e00ae5f504bac4f109317fc489b5f1bc5":{"kind":"REWORK","_number":3,"created":"2023-02-17 18:10:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/86/861486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/861486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/861486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/861486/3"}}},"commit":{"parents":[{"commit":"63e618b232c5f1a50e16baa0e685f4e25716f737","subject":"Test x-backend-replication header with POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e618b232c5f1a50e16baa0e685f4e25716f737"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-14 20:32:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 18:09:31.000000000","tz":-480},"subject":"Use FakeSwift in account quota tests","message":"Use FakeSwift in account quota tests\n\nThis lets us be much more controlled about which headers come back in\nwhich responses.\n\nChange-Id: I9823d00e78479e3335f7a25d0d2c99b63c1f9c12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e76d443e00ae5f504bac4f109317fc489b5f1bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e76d443e00ae5f504bac4f109317fc489b5f1bc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
