)]}'
{"id":"openstack%2Fswift~861583","triplet_id":"openstack%2Fswift~master~If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","project":"openstack/swift","branch":"master","topic":"PasteDeployVersion","attention_set":{},"removed_from_attention_set":{"34175":{"account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"last_update":"2022-10-17 17:14:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","subject":"py2constraints: pin PasteDeploy version","status":"MERGED","created":"2022-10-17 08:48:30.000000000","updated":"2022-10-17 17:15:33.000000000","submitted":"2022-10-17 17:14:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861583-PasteDeployVersion","meta_rev_id":"46bfcf7981dc1f3f21acfc93715a7fb19cb4617f","_number":861583,"virtual_id_number":861583,"owner":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-17 17:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-17 16:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-17 16:02:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-17 09:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-17 10:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-17 16:02:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"79b8db737257151d38b03c9c0a6cfffcf9a1f574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-10-17 08:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9124608534c6d041eb3d4ff6ba2e22e050b37ed8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 09:22:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0059758cbfd24a35880b89f4200c12ae\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/047df52646c74b34b7a48ceed8966467 : SUCCESS in 27m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd4d4e5fd4c749a197ea280024cec4ce : SUCCESS in 28m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/af22d0e15bb94438ac6fe3f419334a12 : SUCCESS in 27m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5c4068026614ddd885d3bde2e79f3b8 : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a15e3c391c94a422292043791166eda9deb4948c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 10:12:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3ad9ac61f1043479e8151dbe1a5ebff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4dfb5f627a04e5f9c545906ad4039f4 : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a2ed0305aff439fa5585ac387221f89 : FAILURE in 2m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b4735df9f56e42788ee693ac8221a52f : SUCCESS in 46m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f26a865bf65c4587950081ebade937b9 : SUCCESS in 38m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6abde29f8224477a377921a10879b1a : SUCCESS in 36m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79f9123533774adba452d70a1c7fec2c : FAILURE in 15m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11abb0b2eeec4f949e5208f5199b00e3 : FAILURE in 17m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d5c21356b564d1fa1d34a6a7fcd18d4 : FAILURE in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a760719c39e4111a74a2c52bdb658a4 : SUCCESS in 7m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f00cf9918b044d09814a59dac2831c37 : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8061c86283244ffa5003b7fa02f4a6d : SUCCESS in 12m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bef9354f11f4e158eed56688856d8de : SUCCESS in 11m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3b7cf42981b4f1dacc1aca5f32e700c : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdc99edb922644dc94d8d5a213707252 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/56cdfe6e56594037a7a2849e69350d9d : SUCCESS in 14m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c5a626b47d24cb186fec450b9e40ddd : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a840001a1de24b95a3fc5a884da91bc9 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dcd2bde2c77e4d9d80784e8d08c50ee8 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a073af550a2f4ac9baa6c23d79572da4 : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/482488e17580436ba4a97f127fa38e55 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/01c35d5d6f0e47cb98d8b3e7139ca11a : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1d98408b18e241c0b33bc91d5fa03f3a : SUCCESS in 6m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b8c284e2c8046f19d307901c53e5f7e : SUCCESS in 58m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/296f4093195a4397b931ffc0b206ef2a : SUCCESS in 1h 04m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13a5750da760495a947605c0763d7c3c : SUCCESS in 32m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78a8754b2280498289dde7122924a15a : SUCCESS in 35m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcbaf7ecdeb14606b24443f72f913822 : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a3862679d445ec84788a82b10ac2ec : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/793d0ecfde85462fb12356310ba44193 : SUCCESS in 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e67ecc8d611f435fb13917f170bb55f1 : FAILURE in 44m 14s","accounts_in_message":[],"_revision_number":1},{"id":"79287ac811911fdd565b010f6ae8f66870d775ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-10-17 11:07:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42f65961b8fb59660fd7cb0a065e2eb10d1c8398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 12:06:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8719414ed89b4003817fe75ab28c89c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24757be1ada54303b8b9bf94c0672308 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/128d8798c49249eea32555c77aa6b85b : SUCCESS in 52m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be4151df00ee40e79a184c5625e63fc5 : SUCCESS in 50m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9d38f537ad742d5bf5db1e4b71951cc : SUCCESS in 48m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da90b2d75aed43b29bb75af1c14b65d2 : FAILURE in 22m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58b733f227f54fdeb3622b4d419255fe : FAILURE in 21m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80afec6095804eacb0203b54eb419b0c : FAILURE in 23m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a30386f4dcf7462b984a111106277e7a : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/919666b87a9a4910bc76d641795aa116 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/43438563e43b45ccb789e956356287c1 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a6357fd9dc44c5ea14648fc15c24f56 : SUCCESS in 15m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad77f0e732d845faae656217ee231630 : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/743f592a5fd4424fad126868b68ad5be : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/65b52640ea0d451382a12253da16526f : SUCCESS in 21m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/21843414f63d4339be57b41516058742 : SUCCESS in 20m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b170174124f84a268cf49dd13d6d3ea0 : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9acf62bf6bb46ddaed07de4828c9891 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6f1571e7c15461892aaca528def77a6 : SUCCESS in 21m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e585991b2574369bf1502fe38908866 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b33b918c9fba48b7abbd9bc83b905ae7 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8328d1a2d0b841ffb3d9735c323dfe76 : SUCCESS in 6m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ef2cf70ba0084475baddebd1b3d24891 : FAILURE in 4m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d27ac2b05b874bcbbd41c1891339c0d3 : SUCCESS in 57m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/adbe1e0e5b08442f826ae995ce326ea2 : SUCCESS in 39m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b4f2887aefc46c198c00bf0267a5247 : SUCCESS in 39m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e096ef0821a74dea8057a85e0d69a7d2 : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd3c1f009f24a219cb890f5e47f40e5 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41eccdf7f2784b17bc14685195e3eb40 : SUCCESS in 20m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/339d2d757a2e49a780d92b9f5c8517c9 : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":2},{"id":"75402098815def26980ee27276676bcf3f71657c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 12:18:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9974fa5221dd470386164c9c516682ec\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29054a873ca9418c90c18bcafb39c599 : SUCCESS in 50m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/029ca12789ee4b038f10500b4d0347ba : SUCCESS in 38m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/acd33f751b4341d6838364b5e9da5171 : SUCCESS in 45m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/78f1a1854d8b4678acc982a5098553cc : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":2},{"id":"99e08f3a0b7486320483845ad843d4bcab5d7efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-10-17 12:45:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9db788b3e0b0e705c6486d2d1b6e67bc35e52ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 13:41:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1761a80ac9af497aacd0320904948a73\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0acfc6a53cf47af92fae43d71a06e02 : SUCCESS in 34m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3ad2cdff0064c2ab5a7f3c9a45426de : SUCCESS in 40m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7d507515d66440da1499431d1e6d9c1 : SUCCESS in 35m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e77e8916ac8d4123aa170531391e2545 : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":3},{"id":"c6f16f56860175c48a64d4d500b14e53e3088e27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 13:43:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716e28f6050140ebb6f18188eaaf31ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6d8f3849654431b0481e083c621e7a : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/87d1a100f2384f7ca7e245d52c0f25dc : SUCCESS in 46m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1cba9ef014649feae6258fc08f6d1d8 : SUCCESS in 41m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53e460a1fab84ab49723db4d42fd4879 : SUCCESS in 43m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ef27e9fd37947c7b0185df6c973346a : FAILURE in 21m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0056fcd09844f26b76519b2bd5fea5a : FAILURE in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3a197f0199b4eaba66992d47b729719 : FAILURE in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4d595bb105b435f97f8552d22c04271 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1c44798cb9f94fe7a96faa3681cec3f3 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42d0eddb15da47849918fbcba4963936 : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b07f996b1e44782969a844b09d9070f : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d5e631dd2c64ea6b0bf0884246b7cdf : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4231b3a9c1844bcab132962d2c337a4b : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b1adb6022b2406b92059533a5012548 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/031d59420a4f49959eb87488c831a3b3 : SUCCESS in 17m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe6377f39d0b42eb92a3b2b65938150a : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a62017cbff54c8d89acd291498ee548 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/540fbf4d507045de8841c12428f0c936 : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1db26744a3aa467cb3578f86b47fd781 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65fba07d086b4f32af7ea1400180bbd5 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b080f36f0c74d93bca2956252a1d19e : SUCCESS in 5m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b645d493d12140d5a7af148a2fcaac1c : SUCCESS in 40m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1c3d7169e744fb8952e61c6bfc21cab : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b097c1be83fd4d009c797ad279be3950 : SUCCESS in 31m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6da2628824254e0dba8cd8cd9b862624 : SUCCESS in 34m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4871fc79ba694ad68576aa80ed7e78bd : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d80248d10c27418a9e2144653718b06c : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dcbfbeb98e842d683b3c758a6b10f83 : SUCCESS in 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bebf4af2fa534e199377190c28fc8972 : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":3},{"id":"347cfd59fdd426f6cdb9d0531eacbe85b271c654","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-17 16:02:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa96c958f4e089ac87151c6de40df2fcb18081b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 16:02:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70c201efaf4841e5563cfade623c0a259a1e0f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 17:14:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab15e595bf7411fb449c23c2fa3bd6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1465cfaa556f4f0d873a6b5565426022 : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/41d9a7e80d944aee82dfd56ef02c5d43 : SUCCESS in 40m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcaa6094f4f545d09e5c12da3d87ccbe : SUCCESS in 37m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2be20c9941ad4faf837a66b02de1da01 : SUCCESS in 44m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d6993daa87e840b3997c6ff593870373 : SUCCESS in 8m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/90c2083dd6e54a0d93bdddd652ffaa36 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c147a0a47776454ca30c929e552f2d34 : SUCCESS in 12m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f18cc762fc04fd1b11e0544ab40ca44 : SUCCESS in 12m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6b50844848634ff9a9e26d5801191b94 : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb578248d87842f9bf52089c2ca29702 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d60437c9ed9846908cff2380776cb13a : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eeacb113241a4109b88593a4ba94e60a : SUCCESS in 15m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ee3b1e96954418dbcb669fa77cc9c4c : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/671d642c014542a5b2423ee123049f1a : SUCCESS in 11m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f4d7a8938064698aa9d36da9e34c96e : SUCCESS in 12m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a9844b0841d9443bb912540d42907054 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1767bf48d595424dbb960213f322bb80 : SUCCESS in 7m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/36cc595a9913447ca416e66c67638b13 : SUCCESS in 1h 00m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/42a06639bc9e41b38ea62a1f91d783b4 : SUCCESS in 56m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac444082791a46c99ae17d736f9b0512 : SUCCESS in 38m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0b5b6b5ce7c43f7bdeaecdd5dce5eb6 : SUCCESS in 28m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bef6a1ccee1d4763945c9ccef6ebfeba : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ddb4e1002a483384d7d34a106bd083 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e198c1561e5a41b19fb022ef4eb34c24 : SUCCESS in 16m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dbf4ac8dd524aa191d704e71bd8778c : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8166a32fb9371eb3f16b6afe28e73f1770331c2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 17:14:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"46bfcf7981dc1f3f21acfc93715a7fb19cb4617f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 17:15:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f87c06ac80e4077822ac708343a7f00\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d6803e24e648e0bd93fbf33c0467d9 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ee5dab6f94f1441d884edec2d4ffeed1 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/fd21f9654b8b47668f5e00cd5915116a : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc033154ad4a4f345258457f3ceed9143fb3d46d","revisions":{"5b371d05c9ea2da845003a6dccbd3adc373e0b49":{"kind":"REWORK","_number":1,"created":"2022-10-17 08:48:30.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/83/861583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/861583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/861583/1"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:14.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:22.000000000","tz":60},"subject":"requirements.txt: cap maximum PasteDeploy version","message":"requirements.txt: cap maximum PasteDeploy version\n\nPasteDeploy version 3.0 (2022-10-16) dropped support for python2 (as\nwell as \u003c3.7), which causes our py2 tests to fail. So cap the version\nhere at 2.1.1, the last which is compatible with our tests.\n\nChange-Id: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b371d05c9ea2da845003a6dccbd3adc373e0b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b371d05c9ea2da845003a6dccbd3adc373e0b49"}]},"branch":"refs/heads/master"},"8d5e68511765330ecca1ce085678770c5c996814":{"kind":"REWORK","_number":2,"created":"2022-10-17 11:07:05.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/83/861583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/861583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/861583/2"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:14.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 11:05:53.000000000","tz":60},"subject":"py2constraints: pin PasteDeploy version","message":"py2constraints: pin PasteDeploy version\n\nPasteDeploy version 3.0 (2022-10-16) dropped support for python2 (as\nwell as \u003c3.7), which causes our py2 tests to fail. So cap the version\nhere at 2.1.1, the last which is compatible with our tests.\n\nChange-Id: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d5e68511765330ecca1ce085678770c5c996814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d5e68511765330ecca1ce085678770c5c996814"}]},"branch":"refs/heads/master"},"cc033154ad4a4f345258457f3ceed9143fb3d46d":{"kind":"REWORK","_number":3,"created":"2022-10-17 12:45:24.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/83/861583/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/861583/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/861583/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/861583/3"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:14.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 12:43:24.000000000","tz":60},"subject":"py2constraints: pin PasteDeploy version","message":"py2constraints: pin PasteDeploy version\n\nPasteDeploy version 3.0 (2022-10-16) dropped support for python2 (as\nwell as \u003c3.7), which causes our py2 tests to fail. So cap the version\nhere at 2.1.1, the last which is compatible with our tests.\n\nEven doing this doesn\u0027t stop pip install swift pulling in a newer\nPasteDeploy in the\ntools/playbooks/saio_single_node_setup/make_rings.yaml playbook\n(causing the probes test on CentOS-7 to fail); so handle CentOS 7\nexplicitly.\n\nChange-Id: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc033154ad4a4f345258457f3ceed9143fb3d46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc033154ad4a4f345258457f3ceed9143fb3d46d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
