)]}'
{"id":"openstack%2Fswift~861777","triplet_id":"openstack%2Fswift~stable%2Fyoga~If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","project":"openstack/swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-10-28 00:01:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34175":{"account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"last_update":"2022-10-19 07:29:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34175\u003e replied on the change","reason_account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}}},"hashtags":[],"change_id":"If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","subject":"py2constraints: pin PasteDeploy version","status":"MERGED","created":"2022-10-18 16:32:29.000000000","updated":"2022-10-28 01:14:53.000000000","submitted":"2022-10-28 01:13:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861777","meta_rev_id":"1efabb49a1c501f6bb4cb145a90f302372c583d3","_number":861777,"virtual_id_number":861777,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 01:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-19 07:29:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":2,"date":"2022-10-28 00:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":1,"date":"2022-10-28 00:01:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 16:32:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"state":"REVIEWER"},{"updated":"2022-10-18 17:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f488047ff7bda740522cb5757fd0ddb5ccb247cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-18 16:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d2f48939cbd6b9eda9218b8dca1e58379488769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 17:35:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3802f981e6148eaa36d1415ca6ebfac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce5b36a99884ecbabfb4b12ac3ba0d6 : SUCCESS in 6m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/75b87cfe5d4747da8f2c90b603f48c17 : SUCCESS in 48m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/809da23848174dc2b7b37d4c8693c60d : SUCCESS in 33m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf940bb81ac045488478f3ca5099e82c : SUCCESS in 48m 37s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6266a4b9c076447a986a878b60d8f0ee : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eee62188a85a4e66bb27493160041e43 : SUCCESS in 21m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/85076d05310d4d8fa6cb16b5aeb78a00 : SUCCESS in 18m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9dce7c2140b4263a0351239424ea013 : FAILURE in 3m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ed9beaa7ede4841b1ab1162fb9fe505 : FAILURE in 3m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf42520a30374bec91d4b3a02ce5d381 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c03a5269e5bd4b00a8c055d44988ac90 : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/43c3833094d9469ba44fc93fdc37319a : SUCCESS in 15m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ff5cab6c299c407ea783bb158d47d30c : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8446659deb74bdca9fccab445abc9d2 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57062be0a701471989bdff0b8218d235 : SUCCESS in 21m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fa834d60a22a485fa47b04519db6f5ba : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f720a44d8c1144f78c9d6e906baf31af : SUCCESS in 18m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/26f62da9f71049f3b14fdfce5fc00997 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5cd46ab02aa4256bc94bf0594d171a1 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4129298faa264cf289b709b8cbff1aea : FAILURE in 8m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15f46279f855454ead3627f067b9c097 : SUCCESS in 49m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3612a1b121bf4093919d8ac990bf2a04 : SUCCESS in 1h 01m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a7a4e36c8dfe41c1971b761edc5351cf : SUCCESS in 35m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/660f43be26ed4aaaa1bc37fe7749a4fe : SUCCESS in 39m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d51e9c2d7cfe4383909a59d1f8244312 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc25cfd1e7e44d697a56bfc2d417eaf : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/529495cddc7a49c089b639b970f2c414 : RETRY_LIMIT in 6m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2747a369675248439b4d193f86e8a3db : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b264679054e6a9814c310b414fe2c48a782b00c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 19:47:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139a5da9f7c8450c95ad52a815082c99\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e20b9cf55de4acbbddd9f9a6120541b : SUCCESS in 32m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae2618ff983a40ee8f06c285f0c328cb : SUCCESS in 32m 15s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/515b4658015e42deb71b20aae7bcdef3 : TIMED_OUT in 3h 01m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b69915a1de54442f93c999aff77cb90f : SUCCESS in 34m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/da7908c111634e4c9d42be3f55590c1c : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":1},{"id":"69095d9943e97456f26f811ae5de16dc0e051c30","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-10-19 07:29:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4790de3cc4170aeee2b59ea805c55cba64949ad3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-28 00:01:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3035f7357ee3c9e23181fb0952384ba1112d2dd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 00:02:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d069b0bf4e8c06d320044164713f65f2df0ee82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 01:13:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992ae2888287474a97e3d33593f0e0f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9864c1ffe24546249293c0290311ee4b : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/34ee99e2507645edbed0112a36ab2698 : SUCCESS in 46m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c96344a20ed7413484f8be644e1bfc98 : SUCCESS in 44m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/835b2ab6c823403b86af1cb1366fdb6c : SUCCESS in 47m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/afd98426e6934ba7b7c4d6e608c19794 : POST_FAILURE in 5m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5982f29c7cdb46cca232660c196327e1 : POST_FAILURE in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8fd1845eea6544b49485bf21caa76d13 : SUCCESS in 21m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/403487f6f35a4c61990f1fd67625b212 : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28186c5b63b64764ad11c7ae79abb7a5 : SUCCESS in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8dad6911864749e294f6bd89b810b8b4 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f703a505898a4eedb5a78d8832040001 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/42242cd755e248c8923fb7583143daf0 : SUCCESS in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aa4e8d45f0c94123938abc6f57fec1c1 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b0888896d9f4299a3dde5ef26336685 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b994b90e44cd4737af1dcf4bfff319d3 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c27483452a14a36aea2ef586ed44506 : SUCCESS in 5m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a801f7320164209b9540d6dc00cbfe8 : SUCCESS in 50m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4315ec62726480e821fbd752ceae13f : SUCCESS in 55m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/73a9e2ad5c6a469f93eb53779832526a : SUCCESS in 38m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0d40cabb32e40d18983bd0086537697 : SUCCESS in 39m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cc96c9aeacc49ce889bf8f5e76a2182 : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66d1de1b4084306acd52b45fbb2a73b : SUCCESS in 6m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf99c8b6834a4e1c8a4646ae4c3c84f9 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"dd4e3d5445d2ab6f59c88153745f751c12c538cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 01:13:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1efabb49a1c501f6bb4cb145a90f302372c583d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 01:14:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6f5bb352724d8692049a2fa42a44b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018a70f8491c4dafbab20178b179cb76 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/348d309e38844789b16cbae8ab655344 : FAILURE in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/807b4c08f4ba406a9a1a67f077ebf796 : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10849f049fe969bdc8fc7306c2666230d11537e5","revisions":{"10849f049fe969bdc8fc7306c2666230d11537e5":{"kind":"REWORK","_number":1,"created":"2022-10-18 16:32:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/861777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/861777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/861777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/861777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/861777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/861777/1"}}},"commit":{"parents":[{"commit":"3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2","subject":"CI: Add nslookup_target to FIPS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e6ce930b6ce00a29ff206c4d323e2f3ea51d2b2"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:14.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-18 16:32:20.000000000","tz":-420},"subject":"py2constraints: pin PasteDeploy version","message":"py2constraints: pin PasteDeploy version\n\nPasteDeploy version 3.0 (2022-10-16) dropped support for python2 (as\nwell as \u003c3.7), which causes our py2 tests to fail. So cap the version\nhere at 2.1.1, the last which is compatible with our tests.\n\nEven doing this doesn\u0027t stop pip install swift pulling in a newer\nPasteDeploy in the\ntools/playbooks/saio_single_node_setup/make_rings.yaml playbook\n(causing the probes test on CentOS-7 to fail); so handle CentOS 7\nexplicitly.\n\nChange-Id: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n(cherry picked from commit cc033154ad4a4f345258457f3ceed9143fb3d46d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10849f049fe969bdc8fc7306c2666230d11537e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10849f049fe969bdc8fc7306c2666230d11537e5"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
