)]}'
{"id":"openstack%2Fswift~861778","triplet_id":"openstack%2Fswift~stable%2Fxena~If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","project":"openstack/swift","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-10-28 03:58:46.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34175":{"account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"last_update":"2022-10-19 07:30:01.000000000","reason":"\u003cGERRIT_ACCOUNT_34175\u003e replied on the change","reason_account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}}},"hashtags":[],"change_id":"If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","subject":"py2constraints: pin PasteDeploy version","status":"MERGED","created":"2022-10-18 16:33:20.000000000","updated":"2022-10-28 04:56:17.000000000","submitted":"2022-10-28 04:55:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861778","meta_rev_id":"777603cf512592b235785a340b821992838824bd","_number":861778,"virtual_id_number":861778,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 04:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-19 07:30:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":2,"date":"2022-10-28 03:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":1,"date":"2022-10-28 03:58:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 16:33:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"state":"REVIEWER"},{"updated":"2022-10-18 17:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93f3716bb8332253f45aef8958943fecd142f272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-18 16:33:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a13361d2707ecc60c4141865f011ca448024c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 17:34:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a18ab24a1b142fa94b9b196d07e26b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4c62a07be548df8bf7fba35cff83d9 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0e42c10b96974d2a866bc8400a9a9082 : SUCCESS in 42m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26fb7c7b309d440bb5de4bcd547653ad : SUCCESS in 39m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/028bfa0d0d5d49c9b4af3157a0783a61 : SUCCESS in 44m 29s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8df8ef3271a94efd881924c7e3d16ad4 : RETRY_LIMIT in 4m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/74cd3448e2ab495c91980f125fd472d6 : RETRY_LIMIT in 3m 49s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/77bb134aadea4deebaeb22e6af96a245 : RETRY_LIMIT in 4m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4eea2eadd144a8bb888897369860128 : FAILURE in 3m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/579911818e9e4f9bb597ac44e2e6c05a : FAILURE in 3m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0685a8d1cddb49a886c3e062513ae9d6 : SUCCESS in 12m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abcf952c3407449fb8911bb289ff6c45 : SUCCESS in 14m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/41ee6b331e9945678e28f06c15692121 : SUCCESS in 17m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/566ba5cb8ac04e9aaa96610de7a94653 : SUCCESS in 12m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c1aa00e60bf04dd499774841e45e916c : SUCCESS in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3fd260eae0a4495b400b629ffb9dbb2 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5cccf9f3ccc04a5f9ced085a6077d762 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a1609b293ce647ac9e25e1fa083fe9dd : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b80cb14ef1024b4fbe9ef892f748591c : SUCCESS in 20m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82bb58574e0145ad9b24cda85c5350d6 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0e12ce00c96d4fb3874157f179921f40 : SUCCESS in 13m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/910d9877a1ef4800b1f111b4cde6adc3 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1fa0d74570db4905a295ae70639938de : FAILURE in 7m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/332065c4841141f3b78ab2af9aa609b0 : SUCCESS in 47m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/73e8783401504fb2b4cd4f6c739439fb : SUCCESS in 55m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ccb6ec7782c544a78374837406dcb181 : SUCCESS in 42m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25d0b05c5ba24a9a9f843bb5a5a56f7b : SUCCESS in 25m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc3692de4d71408b814c07e6daa56322 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee15198d0e8a4f0c860fc2a1e6360e7c : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e0ed660a07894d9c9e0d55de48982126 : RETRY_LIMIT in 7m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cc2aecc08944dd1ab13d4b344d4d1db : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0a9c6f4392379ab37cb1a7e7f0f2a61dfc861f83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 18:46:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1876d89476a4fac8a6e686162fe2209\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00a19077791b4e03945111a6af2f86b6 : SUCCESS in 35m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d69574edc8a4735bd50c0940004a654 : SUCCESS in 35m 05s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/4510cf1875294302a96fa71413346b5f : TIMED_OUT in 2h 00m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2dd67254cfbc4be1ab1ce1f8f846760a : SUCCESS in 35m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1801bb5e910442b87b3bfa4fad6550a : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2dfbe0a3fb5a0ce52fe0761944d2e7ad3c0866d6","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-10-19 07:30:01.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77686042868365c1f1e64febf482776bf4c8d8d7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-28 03:58:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da58b7c5cf9ffc6d799939f2ae0a1b36746af76e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 03:59:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1e3cf5a4cc5dcd3f5bff53f7d5ad823c42e6b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 04:55:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4791c5eed7b74175ae48e4cd88f10087\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ecd962d4fa43969e7bc958c4111066 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7a152ee8485f46f79fa53e2e256fe381 : SUCCESS in 52m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b1a60d90b8543d2a4c5c30550f650c9 : SUCCESS in 37m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8a6672de9284e5ca4ee2a2543056add : SUCCESS in 42m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ca421fd5b00349c4b5fb0d43fc07d260 : POST_FAILURE in 6m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b4428d52ec6342268ecbfcc7856eb2d7 : POST_FAILURE in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/011d5fa01e4b41748880f65efddf2206 : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/914bdf3b41cd4ca0a6970b58efeeba18 : SUCCESS in 16m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aec9853e7fce4483bfc98b7d6430fa04 : SUCCESS in 16m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/917e21255a6e4ca6a72ddb106e4aa496 : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ccb20ea91b242168b8c86c98eb0cac0 : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/963b9fa9429d4426bbd811c660e2a6b6 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9327106b7cb44802ad517de9498e6500 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fed750ecdb6f47fe82927561ea79d83d : SUCCESS in 16m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4c5baa5f6121465daeb80a15e0df6950 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d09515b775d64b18ad21571984293724 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/04b15608e27145ea931e2a9886ca379e : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0fe74b446f4849a6bbca7f124fbb7f6a : SUCCESS in 6m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae0b32f182394288a4f2b4a3d4d4a9f5 : SUCCESS in 46m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7df1a249098f46f689c33ecfbffebe45 : SUCCESS in 51m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5084086c2ead48cea2524fb4dfd7d33c : SUCCESS in 33m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8eee560fbc5c4a509b42f3f82fb84d84 : SUCCESS in 34m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a81a10e4f4254aac90b30406a0a3c9a5 : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd69f336ea1341008e6920095a2701fa : SUCCESS in 5m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de00b613c0e749a8982c648c6e47c1f8 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f469df973997c639e3de83f53c3772b40407795d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 04:55:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"777603cf512592b235785a340b821992838824bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 04:56:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5031bde7cae4530ae5abcba71a96e6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e32d3a60054aa6a2aff3f15f94d5ae : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/df3b1413eda24871abb19f84e316eba7 : FAILURE in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9ab19714271341a0856128753c6bcd0b : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d05ddb03648d70e4b8612243bb8cd2ddf8ae6baa","revisions":{"d05ddb03648d70e4b8612243bb8cd2ddf8ae6baa":{"kind":"REWORK","_number":1,"created":"2022-10-18 16:33:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/861778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/861778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/861778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/861778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/861778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/861778/1"}}},"commit":{"parents":[{"commit":"c6c51be17a6a7e407e95388fab9583111b07b3bf","subject":"Stop partial()ing hashlib.new","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6c51be17a6a7e407e95388fab9583111b07b3bf"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:14.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-18 16:33:08.000000000","tz":-420},"subject":"py2constraints: pin PasteDeploy version","message":"py2constraints: pin PasteDeploy version\n\nPasteDeploy version 3.0 (2022-10-16) dropped support for python2 (as\nwell as \u003c3.7), which causes our py2 tests to fail. So cap the version\nhere at 2.1.1, the last which is compatible with our tests.\n\nEven doing this doesn\u0027t stop pip install swift pulling in a newer\nPasteDeploy in the\ntools/playbooks/saio_single_node_setup/make_rings.yaml playbook\n(causing the probes test on CentOS-7 to fail); so handle CentOS 7\nexplicitly.\n\nChange-Id: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n(cherry picked from commit cc033154ad4a4f345258457f3ceed9143fb3d46d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d05ddb03648d70e4b8612243bb8cd2ddf8ae6baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d05ddb03648d70e4b8612243bb8cd2ddf8ae6baa"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
