)]}'
{"id":"openstack%2Fswift~862758","triplet_id":"openstack%2Fswift~master~If8efd9a0fd18d17444a58119c525ff54cfd55dd9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-10-27 16:00:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If8efd9a0fd18d17444a58119c525ff54cfd55dd9","subject":"CI: Install pip from pinned upstream for py2 probe tests","status":"MERGED","created":"2022-10-26 22:15:32.000000000","updated":"2022-10-27 17:09:50.000000000","submitted":"2022-10-27 17:07:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"862758","meta_rev_id":"494889842fe185170b459d6fe636a6ba5abbfb1c","_number":862758,"virtual_id_number":862758,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-27 17:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2022-10-27 08:52:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-27 16:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-27 16:00:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 00:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-27 02:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-27 08:52:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"429f044d7e097fa2760bae8109962befa3198edb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-26 22:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"641b8d651a77d3e795088b8300227e03f35e98cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 00:48:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7edaccab0824b18bedb288c813f39cc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9066f38fe17545458df81d8d8e2d13c0 : NODE_FAILURE in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2774f292fef4774933e6387aef707b9 : NODE_FAILURE in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c21c83e195fa434898579ecd788fc81b : NODE_FAILURE in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd60b0c1f846418a8a73f626da2650d5 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a051661e6560c1715d6c6ee049dddc62cdee7f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 02:11:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a2aa5659eba47a9b5fdfd8a60709e69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629fe2796c1648b7a67a39aec7498ef1 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/b05cdf224f7244f690b14775b244bac8 : SUCCESS in 59m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/186fa26412ce444e926161ff02699f7a : SUCCESS in 52m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bde7adb92544849a10a978dcfbd8ad4 : SUCCESS in 50m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d3c49f76b1143b0b1ce46f8357b134f : FAILURE in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4d626197eac44bfa6cae18daea498ba : FAILURE in 16m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2b71ba7cd144589ad5c18f1f24172e8 : FAILURE in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5742b992fee240ed8e8d144df554e10e : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b8f310fd44d5451ca36e2b3fb96302cb : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51626dc836cb4f3e986d3e0b66aa94d9 : SUCCESS in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11e7c8f3e95e4833817880c6c6cc2e75 : SUCCESS in 14m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28f720cd07184faba4a58f44d22b2bf4 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9805d15708424187a6c84990176f5e5e : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8231980c7dc4e3dbbf70c66b8142407 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/673ca95b8dd7459f8b9e9aab05005969 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/678eb8f96c7e4024a659980e1c278352 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f5d5d84c615c41d39e2cdfbc860b78b0 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bdd7a01a6425495daff591d54f14ad23 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf0a7d416c7f4dbc86ef09c5fe2c413a : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d1cbae788444722b4eb12463fff769a : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37fe880d6fec4624bfc8a1e0adcb8d6e : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b1e79926c48f40e18d491304a4b6bfa7 : RETRY_LIMIT in 2m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f39282356be44f3ba2fd2627100c7257 : SUCCESS in 58m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49eeedccb37e4ce79ec3724fac89865b : SUCCESS in 44m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28b982fbf8a64629a43a735fd447fc0c : SUCCESS in 43m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89cdad11f92c46318b8ddd647e670d49 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e6cd41febe4353b83c8c7e8ccc946f : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39c47a64b210438082b17e8e8fc58dcb : SUCCESS in 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2c5a2686c314febaa7f9c9284a114dc : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"36bbaf801973614e3b728780d628164625d9fd57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-27 03:45:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"01d56b7e1d8ed57fc23edbf1f52d7db95d0a7b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-27 04:22:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"029867561a412f3fae29f0c840c762d9edd5a304","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 04:22:26.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  Job swift-probetests-centos-8 not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: swift-probetests-centos-7\n      parent: swift-probetests-centos-8\n      nodeset: centos-7\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n        under Python 2.\n      vars:\n        bindep_profile: test py27\n        ensure_pip_from_packages: False\n        ensure_pip_from_packages_with_python2: False\n        ensure_pip_from_upstream: True\n        ensure_pip_from_upstream_interpreters: python2\n        ensure_pip_from_upstream_url: \"https://bootstrap.pypa.io/pip/2.7/get-pip.py\"\n\n  in \"openstack/swift/.zuul.yaml@master\", line 346, column 3","accounts_in_message":[],"_revision_number":3},{"id":"2d2345b345109210bd9a6ab57b228feed11415ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 04:22:32.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  Job swift-probetests-centos-8 not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: swift-probetests-centos-7\n      parent: swift-probetests-centos-8\n      nodeset: centos-7\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n        under Python 2.\n      vars:\n        bindep_profile: test py27\n        ensure_pip_from_packages: False\n        ensure_pip_from_packages_with_python2: False\n        ensure_pip_from_upstream: True\n        ensure_pip_from_upstream_interpreters: python2\n        ensure_pip_from_upstream_url: \"https://bootstrap.pypa.io/pip/2.7/get-pip.py\"\n\n  in \"openstack/swift/.zuul.yaml@master\", line 346, column 3","accounts_in_message":[],"_revision_number":3},{"id":"650674a7eb94b08650d504f81cbf276553786623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-27 04:23:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dcc23b22e54fd218c9d5ce0352a163fc7d589835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-27 04:25:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"23af8b4999fac278cb79d768fb93351297100613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 05:32:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c9f84dfd34f4da9a69f56c4688e0d50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249ab8444fb844eea7bb2b64747c7c1b : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/61b3c6001e66496788ac588592b2fe6d : SUCCESS in 43m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e9912b5477249d78b3753df9d8d39d8 : SUCCESS in 35m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a63153acbfe4b49844ebca0add997a7 : SUCCESS in 36m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/021ab80da94647ee9c0dc1559e701216 : FAILURE in 14m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3ef69bf6c3245d289e9b3813c4edb13 : FAILURE in 15m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/436d7e61cd7d4900b342544a85fc276e : FAILURE in 15m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0ceb7466be2420ab758b12d32fb9543 : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ced3229d18e44b5e9e36015cee039a7c : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e35838a1b1cb4d8b8c947c42cf10b24c : SUCCESS in 13m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d6339ee5cee44468fac8b4813e85130 : SUCCESS in 11m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/899494c4dd7b4871a34381383022a155 : SUCCESS in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b95d7ef9f00b4190a7c389dddf83a736 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd5044e35da1482c9861ced65554fa27 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/991f78b0ef524fc08dd984eed022149e : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/50651293dcbc412c98cce9c47cc58c23 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/644d2e85d77f4a1197542e01da4f2119 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4e75c3d968444f759cd195f0c22aa7f1 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1aa36ecf4d844ea940e8af0d8cee04e : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb39bf76d16148dba49f9d77560650c2 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/919cf004af8341afa9f5aa556a539a8f : SUCCESS in 6m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/58914576e2f04573b6e4fa0f1f2eb070 : SUCCESS in 46m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebf76181d66a412281c2b1046bc8f416 : SUCCESS in 1h 01m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f9478ef781a4e44a5b5402ce2dd00d1 : SUCCESS in 36m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd422e57153c4987b65c851dfeaeccd5 : SUCCESS in 32m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa084ce8b6e944e1959c3d695b2dcbba : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7a9c9fc71e4e7f80d8043700090981 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac707ba1c3e145bf9c2222ad59b79ca6 : SUCCESS in 20m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9526c5bef3947bf9da15493369f6132 : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":5},{"id":"d0bfa564763e13d88bffbc4c738e00b9dc7b781e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 05:59:10.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31c0c9ac62484fff8d24ee7df818e762\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6c5588188a74f2cacf19a822ced0669 : SUCCESS in 28m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60c3fe222d4449d3b5073bdba5dae7d3 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/29a6e5e7d2a64336bcf8b2f0ac3e568c : SUCCESS in 29m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/686f8dfea06f463f9d4fbf0c659c7b3d : SUCCESS in 28m 30s","accounts_in_message":[],"_revision_number":5},{"id":"fc56516a6eb6e0c5595dab1dced973532a610b8a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-27 08:52:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0a4b7a47237430fe019dd9053ed44201c2c48d24","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-27 16:00:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5f004e80304716bd16a247a7593733dbd07160be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 16:01:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a41142549ef6343e74e23635cbdd69cf05df8f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:07:54.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0158efc33588478b95e9e40e3eda78e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d09cca0b9e04c81a6f71c1f864f98ad : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/160f0c220e8047a9ab46294dc3645957 : SUCCESS in 53m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/518270e8e5ff4f6fb267c87b4b76d297 : SUCCESS in 40m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/160bb750a2774fe8ab8f7fa4fd309203 : SUCCESS in 45m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9f0b1fee36ba43ac951732586b2e225e : SUCCESS in 8m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ab0cf8b46cf04a22a7f1a1f44f6a5927 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ba251eb1a34476395c8354f1bb4ca88 : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7599351d305b43eabc1237a4ad89307a : SUCCESS in 19m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/671ebe3452c34d70adda3444e163d619 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f1d5a421495b46c38131d4db5aa75e26 : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40ace594f1c344958b9a4fe217e77ab7 : SUCCESS in 19m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f6c6f87b8f8469192c34f9bc2de3e65 : SUCCESS in 18m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/72597327dd094dd0a6e4be8e4f87ab71 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c7e1b88fc2d4f669c45a0bc49ff1603 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1908ebbc99ae4ea39667597c2d2dade8 : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc6bc9fc731842359d398ad033961cab : SUCCESS in 6m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10f8c8f1127d466d87762f6bf27f34b6 : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4da051f9a9aa46f3919dac8ec141d71d : SUCCESS in 52m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0c7868592a34311b25f4084ce403662 : SUCCESS in 1h 05m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ad05b8038424f29adfdb6d5a0f29dd2 : SUCCESS in 41m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/071035485b26480f94cefc72de9ec3bc : SUCCESS in 41m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57c8bf29d8fb4263b60dc687a8b46bf7 : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ba0de543aa4a01b8af7b22a68fe1a4 : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/482b4b4540b3455f9721ec586f1e7fcd : SUCCESS in 21m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83cf41b697ff4154960990fbc520a1fb : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":5},{"id":"3c3f8375ad3d415ec2c4800f169acb4fda26f53e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:07:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"494889842fe185170b459d6fe636a6ba5abbfb1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:09:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc48e72451a4849b5fe369295cd9105\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07e5a17801f4306baad96e0b01fae5f : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d29a5530cb2440958811ba6da916145a : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/dd9c5da28b844ed0a537bc3d8c985b0f : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3571cdf2f77978f9a8c835a66048278ce8086473","revisions":{"05d1270c4732bef475550cf04668104bccdcc395":{"kind":"REWORK","_number":1,"created":"2022-10-26 22:15:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/862758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/862758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/862758/1"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-26 22:15:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-26 22:15:21.000000000","tz":-420},"subject":"CI: Install pip from pinned upstream for py2 probe tests","message":"CI: Install pip from pinned upstream for py2 probe tests\n\nChange-Id: If8efd9a0fd18d17444a58119c525ff54cfd55dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05d1270c4732bef475550cf04668104bccdcc395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05d1270c4732bef475550cf04668104bccdcc395"}]},"branch":"refs/heads/master"},"639569615edcc665e769e8463826a6583c2e0c78":{"kind":"REWORK","_number":2,"created":"2022-10-27 03:45:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/862758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/862758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/862758/2"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-26 22:15:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2022-10-27 03:45:57.000000000","tz":0},"subject":"CI: Install pip from pinned upstream for py2 probe tests","message":"CI: Install pip from pinned upstream for py2 probe tests\n\nChange-Id: If8efd9a0fd18d17444a58119c525ff54cfd55dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639569615edcc665e769e8463826a6583c2e0c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639569615edcc665e769e8463826a6583c2e0c78"}]},"branch":"refs/heads/master"},"3786f2f4269bfef7a9b81353228e3ca598d785e9":{"kind":"REWORK","_number":3,"created":"2022-10-27 04:22:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/862758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/862758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/862758/3"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-26 22:15:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-27 04:20:50.000000000","tz":-420},"subject":"CI: Install pip from pinned upstream for py2 probe tests","message":"CI: Install pip from pinned upstream for py2 probe tests\n\nAs part of that, invert the inheritance so the py2 job derives from the\npy3 job.\n\nChange-Id: If8efd9a0fd18d17444a58119c525ff54cfd55dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3786f2f4269bfef7a9b81353228e3ca598d785e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3786f2f4269bfef7a9b81353228e3ca598d785e9"}]},"branch":"refs/heads/master"},"eeeffe91df2dcd5af5a4526969e4af70bc23139a":{"kind":"REWORK","_number":4,"created":"2022-10-27 04:23:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/862758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/862758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/862758/4"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-26 22:15:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-27 04:23:46.000000000","tz":-420},"subject":"CI: Install pip from pinned upstream for py2 probe tests","message":"CI: Install pip from pinned upstream for py2 probe tests\n\nAs part of that, invert the inheritance so the py2 job derives from the\npy3 job.\n\nChange-Id: If8efd9a0fd18d17444a58119c525ff54cfd55dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eeeffe91df2dcd5af5a4526969e4af70bc23139a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eeeffe91df2dcd5af5a4526969e4af70bc23139a"}]},"branch":"refs/heads/master"},"3571cdf2f77978f9a8c835a66048278ce8086473":{"kind":"REWORK","_number":5,"created":"2022-10-27 04:25:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/862758/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/862758/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/862758/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/862758/5"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-26 22:15:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-27 04:25:02.000000000","tz":-420},"subject":"CI: Install pip from pinned upstream for py2 probe tests","message":"CI: Install pip from pinned upstream for py2 probe tests\n\nAs part of that, invert the inheritance so the py2 job derives from the\npy3 job.\n\nChange-Id: If8efd9a0fd18d17444a58119c525ff54cfd55dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3571cdf2f77978f9a8c835a66048278ce8086473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3571cdf2f77978f9a8c835a66048278ce8086473"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
