)]}'
{"id":"openstack%2Fswift~863562","triplet_id":"openstack%2Fswift~master~I71adba876503cd41f3e451ec3ddd8ba87ff49ab7","project":"openstack/swift","branch":"master","topic":"p-namespace-map-2","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-03-16 10:29:20.000000000","reason":"Change was abandoned"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-03-16 10:29:20.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-01-19 11:24:42.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-03-16 10:29:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I71adba876503cd41f3e451ec3ddd8ba87ff49ab7","subject":"WIP: restructure cached updating shard ranges","status":"ABANDONED","created":"2022-11-03 21:30:32.000000000","updated":"2023-03-16 10:29:20.000000000","total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"865023ae63eef76f845ce2b3be7fe900e2ac624e","_number":863562,"virtual_id_number":863562,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-30 23:44:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2023-01-19 11:24:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 22:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 03:38:53.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-01-19 03:28:05.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"8ecd3db09739bbcdfabd9b0af46d54a30412c9ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-03 21:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f3998ca03ff04bfbbe91e5efbc9bcef3aa21738","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-03 21:31:22.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fd45b9dcec976becf36b8ba9ba64e2f73c8a2ab5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-03 21:35:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd9967cf6716cb4fb4ba9a8bd6ee0cb5f4a5ba5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 22:24:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40edb0544e7a46e78bf45805776359d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3f206f01634d708257eec759786f1e : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/007607f14c1d47939ad9904532b56ccc : SUCCESS in 39m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac7b7ba024eb4cc2ac58857634ad0c49 : SUCCESS in 31m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cabb8e7078c4c9583c2dfe0417c0889 : SUCCESS in 38m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6702f0a627d45cf9314037638d27e36 : FAILURE in 21m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a05988ca438c44479b476a08b44a6957 : FAILURE in 21m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9623696546f94c68a5d08bb950ee3d42 : FAILURE in 25m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae7f0d497cfc4980b296bb0989017a69 : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c547553ad18240b283b428d4eed010a1 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab640989a93c4461a40f5133552940a3 : SUCCESS in 18m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3bbf4ca6b72046ce8f4792d2065d0fda : SUCCESS in 15m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b97118f56d81486dbae89bf0cdae7692 : SUCCESS in 16m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5abad64ed782494f94f1539d8a135708 : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11471c90dbe34a61b37a738907864f3a : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e6d20536b2949ab8612124d71a26c65 : SUCCESS in 19m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c059dbfa7ad349078b039c2027536a51 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eaa6b270ce57487e94cc72433155ee86 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2eb129b9bfb4490688d5d133936d7f91 : SUCCESS in 19m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f91fab769f4a486294ecbe2a6fad0767 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e1a4de3100843fc8bde8ef5e9c5f771 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/230eccda9b48429db358b5563fac92fd : SUCCESS in 7m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5a9893133ab2475389bc9373fc50db9e : SUCCESS in 52m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/964365cc4c014082b32cb9fd33fdb49d : SUCCESS in 51m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b29e83ec9f341a691ca13adb082f1a8 : SUCCESS in 39m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c87831dc58347b08cbcceaf90089fed : SUCCESS in 42m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c20f81cae924a1bb924256548269f2a : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90dbfed4002f439a84c0f9c324fe8684 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cca8c012f8742629f6ba1c7ae9d34cb : SUCCESS in 24m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a026bdd0ed642f6a1d0ef65e4ca11fa : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b0dc9cc5a9facccae856714a135e9ef8180130a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 22:33:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3acd748a89ec4f55906dd812a2067e51\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd707ae7bb0045d1881f527f1b31d26a : SUCCESS in 28m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ead784b909241658d35b7516e8f8c09 : SUCCESS in 1h 00m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2f3eb14c20e47748f7832cee60816f3 : SUCCESS in 51m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac7a8da0841e43a9b34e788858428c02 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"195fa0fcc902edc8a36da45422df62c36e74f7a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-04 18:35:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81c4dde99593ba5bf58aaa1d36cfcc2242496aad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:43:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97ad1fa96014304bc96270b0c4b68a2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/298f5c0770d84bcc8926fa605a6d08b4 : SUCCESS in 1h 05m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b581c882db4485dbb1c00d2c347305f : SUCCESS in 1h 00m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6720ff9d55de4403bbe5b4198e30fc2a : SUCCESS in 1h 04m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d58a0e5d462b4f18bfea27d20c5546c3 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9e7f1dc81640f28e636cb71a59d877a129074163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:43:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc0299a0801a489aa691e75b0f4b3329\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b863ba0f0cb470eaf60b9726bfc3c0d : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/6cbcdaa4994b4da2932ed5ed9dda0bcd : SUCCESS in 53m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b652394fd14b4f7192c7f0c53e3d867f : SUCCESS in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef5ddcb04f8041f484e4693bde08c85d : SUCCESS in 36m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82ba1553ee824cd38b96af0bb4b16232 : FAILURE in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04267d516d9b40fb9837992a554cdc61 : FAILURE in 21m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4b14a0ca64a4509aa4fcef12c8bf44e : FAILURE in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c173217cd3ea4de083c9a0f6bed7667c : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cfcd2c5af45d4e1bbb264c8ad0470486 : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/499ed883560e45e6aa4c24da9693eff0 : FAILURE in 19m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f89069f48d6a4f3b9d105b1c42122a3d : SUCCESS in 13m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f98ce90934334881a348b0f5c6f611aa : SUCCESS in 18m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d5225b82b234c98b7f58bd4c50bfba9 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07a86fd93262469fb2e6bc1c0036ac5d : SUCCESS in 30m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37b1a4334e88491990218f108714cb7b : SUCCESS in 20m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d2c5c992ae614191bd8d87740b19a33a : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3cc54f0c58de4b618d34cbd08f9f2903 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67448e708ef44d12939b456d11e12735 : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1746e9c5b8a04023af8841a181e49502 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/108a233878074d0d9243565ec4357da4 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/806bbc4c3ca940afa13f9350cfac64a4 : SUCCESS in 6m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/17b94b19b83d48a390e9c3e2cc88deaa : FAILURE in 1h 02m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc8725370e1d4214b4eca91067557311 : FAILURE in 59m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/210e5c951d3d405cb402a21d209b7a61 : SUCCESS in 32m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d824cccff2c487ab55b1a6dda9ce5d6 : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b33f3fb31264964add5981d761b4e38 : FAILURE in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506b92f0096949ef804f64baceec6363 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/afa7edc2618544169d57bbedb15f9a45 : SUCCESS in 18m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da58d6ef7ba7471e9ea7c0d0a21a282a : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":2},{"id":"21269d63dd5904feff910d08fc4d025996417e56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-07 17:06:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf83a4e31183b6d0af121c0f8f41b7bdd6314025","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:14:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47203d3ad0447a9bcaf85b857a9f3f2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3da2f75d554441592254d70c9930c58 : SUCCESS in 31m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f26571e341a74415a60f9363cb60e349 : SUCCESS in 31m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/682f96cdbbf44f95be1e8fe152977ecd : SUCCESS in 31m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cd3216c5cbe494a8185e8222563ceea : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e9e16253d09d46a5ba4846fbacf88beebcd34a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:39:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b21f8c826d374144b9dd381211e9840b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ced9aaffb2c4afea7e4b8b07f0241ab : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0ba8951b375c4e7cbd3ddf166354276a : SUCCESS in 57m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ad9455d87214ce09112de277de709c5 : SUCCESS in 38m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d308d452fecb4723affb15da339eaf38 : SUCCESS in 55m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97e3cf248459474e82fd052510c828bb : FAILURE in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c28c7239ce424798b8847fe99a69108a : FAILURE in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9702ca74c674b86979fb401dcf4a2ed : FAILURE in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87fa44f1def04d1db8e632df81bee881 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82249277d82a4654acae25dc4264ed13 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1e85f9c89c8404b8bcbe66f0cf2cc83 : FAILURE in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0083075929f409cb1c3e42f791bb4b2 : SUCCESS in 15m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f2fa8ae96ad4d259f6c6965d7612a4e : SUCCESS in 14m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71437f357eb441af8a414aed6a3a6f8f : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac9f729455554c36956923f08dcda4aa : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57ab6c0ab1f647608a62a4289734e3d0 : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b986171cd794a9ba7bdc1643d67c2f6 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f00ec2f3c2be41ef9e7d1139e4546511 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4cac060f1849405da28d9c5a31345298 : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a332c8a572f484fa3fd601daebbe693 : SUCCESS in 8m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4857b825ae0a4b3cb7a1439dc0b9763a : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2851d1e04f254e5a968a1f5ba0154262 : SUCCESS in 6m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ffaef618eb5a4f78a27122cbf2a0a988 : FAILURE in 1h 14m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fde0b15e17a24f458a6abb17dfa8ed6c : FAILURE in 59m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d995e1a6001f4516b4319eca8b1d9355 : SUCCESS in 41m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/499a5fc249484d6eb3d8401d6a29577d : SUCCESS in 41m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/240394df56d7427596f450bd4fe8357d : FAILURE in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519eba3d70014dca91cc7b7872873429 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c5e6fc2297542e4825840d38e340647 : FAILURE in 26m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bef1470fc84413ca08c409eda0d1c98 : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":3},{"id":"7531971953e88f1a29cd67cb9dad2bd44d1c02c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-24 18:18:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2da2097e024ae3348605bf91df6ee8d93e6e0e9b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-24 18:19:14.000000000","message":"Patch Set 4: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9d0687c7b580712528b7efd3659d90dc70c69f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:19:46.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb410842abd426b8b8225b68169becb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca9c3ff07ab9407b9d60fe301779eb3c : FAILURE in 47m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0286d2a175ef4e75995d0ceadbc7c320 : FAILURE in 51m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f01e548300a4a0d9abba1780231cb77 : SUCCESS in 56m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/52666a70b6384dd79f894025807da420 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":4},{"id":"01173aca26e64638c9dcac13e7b45490226e3064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:25:17.000000000","message":"Patch Set 4: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06a53f7074ef40ad9c4b50b00d35d45a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150f89e5927941acbbc7a5ec018cc3d2 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/bc4312b706fe4f57903c71b12143fb1a : SUCCESS in 55m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b32ee85fcbec4c2ca0dbf9bc20e738f8 : SUCCESS in 48m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d05ff9fed1b647179d1379dd4ba76bf9 : SUCCESS in 58m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f3d40f532e14942bd72dfce98e69e72 : FAILURE in 17m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81da03ea34df45e48cd44467b3473df3 : FAILURE in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08df779c80994fa2971688552a379cee : FAILURE in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/129cced4e40e46e9aa95237e28173016 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5e27bec307547cc8b9c21c10694e775 : SUCCESS in 4m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0b9c0585c9f4b1282ff54e1dba6f069 : FAILURE in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/502a1917059a4f608d690a279d3c978f : FAILURE in 14m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61ac950916fc431ab074287401ce1c25 : FAILURE in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41e2a88015cf4fdeb736453ec9452feb : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6200b7f2f92e40179427d615afd3a9f3 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a4bbd1668304882bebc2c03161464e9 : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb61a843d8a14dee8df871755bae90ed : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e0f7929dc55d4c9d8f555e7c81ac16a5 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7fb2a765c40e40f2a542c794e85119aa : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e26cb539acab46bb967cecfdb8846886 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0399421cb4594e92bf0a67564c46bfb3 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d321bd595544f7ab5358629b1454255 : SUCCESS in 6m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/acb44e125ca24dc6a1f3868fe1f8f3b8 : FAILURE in 1h 03m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c49fb712ddd648538903a03ae57ab6a6 : FAILURE in 1h 01m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/767b8304e93249269dda7aae0a62dc6f : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fdb2213760ba44a89977f8f35820f1f5 : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ae5595810494a33a6b936590e0aa127 : FAILURE in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d505df56f94407a0347a580721f845 : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/89d9af6856434c10a5a966f50bcc615a : FAILURE in 17m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8db0bb04354429883494859b1e48573 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":4},{"id":"64ea436dc14e06e3a7db24e7fb9e4168e9908ca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-25 18:25:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"543fe6aa57ba8c468d8cd9f2d610849a1fd92193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 19:34:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a19a1a6745e441ab4b56cc082dbe7ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8bb83aabdc4daea1e7a1cd864aaf47 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/8d01d7c8ce3d433a93407bca70b9c47c : SUCCESS in 1h 02m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f90212b18c646449f14f5d1eb5402a9 : SUCCESS in 51m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cf136830fe548fc888f030ba1341d7c : SUCCESS in 57m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76ca18ac03364b93bdbd0fc64290f1a0 : FAILURE in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45888dca1cc54350aafb81d938e8bd25 : FAILURE in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/792b8dbe1a7d43eab1acdf552134c8b6 : FAILURE in 17m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d537b61e780474ebe4428c668f6a5ce : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/345d6c1e9fa2476c89079a5a8de9df3a : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d294efc992e048e88da0b239d608886b : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1dbb504a3c034811b6e1fcde4df00fc9 : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b38cb4e13bb944c1af55325e3ff8ec55 : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d589a9be4f54acbb2a3792b9d817abb : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1129a66d14b438494a32811aa725fc6 : SUCCESS in 21m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f3ea5aa99ed41cb8f5648c13cbf4231 : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/74f836a3092649d792519b9706194f51 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/61d84d41e1b443e78d51d9efe488cc2b : SUCCESS in 13m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e5128fd880394fb1b704b4c1cc262b22 : SUCCESS in 20m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/071611e864b040708a87ca77949172a6 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ec7db9017d747ce93509a47ef09fac3 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/01903226415a4e83b022854411ac3e85 : SUCCESS in 6m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/547afc5f07bd4be58ca6d0d117baef7d : FAILURE in 1h 04m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b50545f2f90943eda6fb3ee54825fa81 : SUCCESS in 1h 04m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ad35057d9754d40a78ee27272dd4114 : SUCCESS in 46m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/245b6b910fa7457092e4682614095fdf : SUCCESS in 40m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c37759475cf14496abe3a845d0430a24 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c84d7bdb96f4e408b92f8cc11f5b320 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2a2075c93b146b6a739b9c247e7423d : FAILURE in 19m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65cf8c36bf61408b983ec41987e6f35b : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":5},{"id":"9f2924fad101e43f37844597e17f69288e822ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 19:46:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b8742dc5424db1bd5446bc344eac5a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/858876687b4b4ca0930e87bf6901126d : SUCCESS in 31m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d58a3d48153489e949dc93aec4863c2 : SUCCESS in 1h 05m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/77ab117890bf4d659d28e5dbb89b033b : SUCCESS in 30m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc931a0d063843d5851eb599bc81d3ed : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":5},{"id":"055265e9522a9ee7179fdc597311c3585807688b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-28 03:38:53.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3511cc698871a59fd8dfd37f478f04c136f82d56","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-28 11:08:34.000000000","message":"Patch Set 5: Code-Review-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a27b96ca1e4a2689560c160e7467d8548c1604de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-28 18:48:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5f71e4f855e71f5ce8b4b801657e521596ee25f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 20:06:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb20bd9bf35b48a4b052d1d7d5fa636c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3380c0afd146909814783400d6d2c8 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f137df558f634b6ebd54402583d2c3cc : SUCCESS in 50m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cae285a897e84a69b8092471e89a8f10 : SUCCESS in 57m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2e80e49d01f468c9fcac9d68b607351 : SUCCESS in 49m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a95d39b9b038462d8647969164f60835 : FAILURE in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6daa29c3a14a459fac7b0b73873c6905 : FAILURE in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4eb9fa2ab97f4918af6a3a1ab546ffb3 : FAILURE in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04c786410e4c4b55b3ae41f858f3fe2b : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5bbc8ed0b3ec459991fbf22d11fe9a1c : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d587943925a7471f935bb0ca5aa3e24f : FAILURE in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f88a35a04710414e84de93ff16f210a0 : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84eaaae56c214686ba3087fe67649531 : SUCCESS in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c85d07851dd4466893e60a031021f60 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d58b3dc4c98f4665b1f65dea99a6b045 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67e38f0afcd244df8724547a355ff041 : SUCCESS in 15m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b937555691304cda8fe39279338e236e : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1c3a7e1113c34845a1c3714919afa78d : SUCCESS in 13m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d217b110a50b48bd9fee86ab3abb1103 : SUCCESS in 20m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8449c7124cde40deb2a2908db67cb711 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a7495abb6ac4a8eb4625f318f3c6600 : SUCCESS in 14m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a4ed2fc6b21149aca286bde988338287 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/393f021d34ed4c0c80929144ec4c8b1f : FAILURE in 1h 12m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/073e4655f42243279f64d6a09a3285f1 : SUCCESS in 1h 04m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/827db56310f24e9c867c5c67b3d50b56 : SUCCESS in 40m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/640b563d7aea419e9d8b1183be0337e0 : SUCCESS in 42m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a213967c2bc48f6ac2bf03ced26410c : FAILURE in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2380a155bf12439e82968ccb2424b618 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/495ad72150664f699a67db29eeaeefea : FAILURE in 24m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8edc0b26b44749c7bc648b48b6138e0d : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":6},{"id":"c35067c602e9f8c99e7fe18466ea206b462c2647","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 20:11:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b823b3c152c400b83d397fa7c2b0073\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1a58c7de917433ea62677f5bc3bded8 : SUCCESS in 55m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25dda59d4b014379ac7cf64f93e46165 : SUCCESS in 1h 01m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8de156767f754b79b69fd1dc1da13a06 : SUCCESS in 1h 20m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bf7c5e8ddd74d4e86fe78e0a0574f1d : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":6},{"id":"0e618070c3aade2f6b762f88cf22b7e9aa76189e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 22:34:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"48c18cde62b58bffa0e84222b0af553ac23afa03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 23:44:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1d525f73f94c3c84d64f80520f04b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7acae9c8e6468dad28f81902140023 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/8f05504d9ddd4ee6bc37e5a873df395b : SUCCESS in 49m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1fd2827d7ed4fdfa975149e2142c50b : SUCCESS in 41m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/951692089d7246828f03dae9b42c750d : SUCCESS in 51m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61529b1d7a89469fb56a86d0983b4fe8 : FAILURE in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38b2e5bc15ea4c7fa7a8b23c978a97ac : FAILURE in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0be3a658a6594d20b231b36eefec6270 : FAILURE in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3fd6ca18b9f0471da3c067022dcd3147 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5574a65adef54b2486a93a58db74eb21 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d49aa17b318b4239b105a9f02195ec95 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca9d3e19b2724db7a81d17026ed0669d : SUCCESS in 13m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa771af592794579931dd0f8b9ed3f72 : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b94fc12e26fd47f5bb3d6ccda648270e : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f321faedc1e4b7e94d21f43d51aedb1 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2beec99fef84795bd27039808b56664 : SUCCESS in 19m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44a020fa73804874840259c1ead22f95 : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4041a011e9ec40daba6ef5c89f1cda6a : SUCCESS in 15m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3178e4b616b4d309d843bedc07b32a1 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/79e068342cef4268ae760878937af112 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb8c1fe2746a4da98daf2a47e96ccf58 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8748dbe50c7243c98cf19954c34712fd : SUCCESS in 6m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf1b30360de64214a3a2381b5201ffa9 : SUCCESS in 1h 02m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6932a789fbe4dae85862e063099243b : SUCCESS in 1h 03m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce2db44ee40945f4a8aadaa7a0dc5830 : SUCCESS in 40m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b36343079e844b93851aa32450a2deb5 : SUCCESS in 41m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43504b392f4a4eff8384f39e51510217 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1f16e180d24c33b1271dd8639f22dd : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18c42cd52bf5407fb571ccc040b7d2a9 : FAILURE in 25m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f227e5b2c61848cfaa056d9f3c16c9c3 : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":7},{"id":"733379f02e98f0296065655a701e4cdd5a7f4370","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 23:51:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eac1ed0b3dab4fab8dfdba0f06429b63\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/939c8bcd2d1242d8b8161b8a73087713 : SUCCESS in 1h 11m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f9e3a5cc3384c86bd23087f47849170 : SUCCESS in 59m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/99ae0672318e4fe8ad77933dea2f344a : SUCCESS in 1h 04m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6267950c2ab49a1a2611ca634c51932 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":7},{"id":"aea6edb55a5ab7457c2316859a6cb8175ae5e2f6","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-19 03:31:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6d0ac9cd540d717e48fe4d8f833fcb5647238f70","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-19 11:24:42.000000000","message":"Patch Set 7: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"865023ae63eef76f845ce2b3be7fe900e2ac624e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-16 10:29:20.000000000","message":"Abandoned\n\nalternative was merged https://review.opendev.org/c/openstack/swift/+/870886","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d0dfbe9a9dd07be7b8067cbe8c3c41d894d705ae","revisions":{"2d0e9a3fe7455458cd7f70c7295e15be7cd3c4c4":{"kind":"REWORK","_number":1,"created":"2022-11-03 21:30:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/863562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/1"}}},"commit":{"parents":[{"commit":"32600ee56bde7ee8bc4b8e38440d24f65b505cc0","subject":"Merge \"Refactor memcache config and MemcacheRing loading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32600ee56bde7ee8bc4b8e38440d24f65b505cc0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-03 17:30:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-03 21:19:47.000000000","tz":0},"subject":"WIP: restructure cached shard ranges","message":"WIP: restructure cached shard ranges\n\nRestructure the shard ranges that are stored in memcache to achieve\nthe following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d0e9a3fe7455458cd7f70c7295e15be7cd3c4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d0e9a3fe7455458cd7f70c7295e15be7cd3c4c4"}]},"branch":"refs/heads/master"},"bbceb820ab1d9f1d6b40f56eae98d8bc512e09c9":{"kind":"REWORK","_number":2,"created":"2022-11-04 18:35:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/863562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/2"}}},"commit":{"parents":[{"commit":"32600ee56bde7ee8bc4b8e38440d24f65b505cc0","subject":"Merge \"Refactor memcache config and MemcacheRing loading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32600ee56bde7ee8bc4b8e38440d24f65b505cc0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-03 17:30:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-04 18:35:11.000000000","tz":0},"subject":"WIP: restructure cached shard ranges","message":"WIP: restructure cached shard ranges\n\nRestructure the shard ranges that are stored in memcache to achieve\nthe following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbceb820ab1d9f1d6b40f56eae98d8bc512e09c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbceb820ab1d9f1d6b40f56eae98d8bc512e09c9"}]},"branch":"refs/heads/master"},"ddb3d144f8830017c7c58d4377356186f786f8a6":{"kind":"REWORK","_number":3,"created":"2022-11-07 17:06:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/863562/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/3"}}},"commit":{"parents":[{"commit":"32600ee56bde7ee8bc4b8e38440d24f65b505cc0","subject":"Merge \"Refactor memcache config and MemcacheRing loading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32600ee56bde7ee8bc4b8e38440d24f65b505cc0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-03 17:30:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-07 17:06:26.000000000","tz":0},"subject":"WIP: restructure cached shard ranges","message":"WIP: restructure cached shard ranges\n\nRestructure the shard ranges that are stored in memcache to achieve\nthe following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddb3d144f8830017c7c58d4377356186f786f8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddb3d144f8830017c7c58d4377356186f786f8a6"}]},"branch":"refs/heads/master"},"c69483d38b5e6e28ccebd00a601f6b1fa8da17aa":{"kind":"REWORK","_number":4,"created":"2022-11-24 18:18:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/863562/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/4"}}},"commit":{"parents":[{"commit":"e748cff5c407b61045326f235f69552ac82b773a","subject":"Merge \"proxy-server: include suppression time in error limit log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e748cff5c407b61045326f235f69552ac82b773a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-03 17:30:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-24 18:18:00.000000000","tz":0},"subject":"WIP: restructure cached shard ranges","message":"WIP: restructure cached shard ranges\n\nRestructure the shard ranges that are stored in memcache to achieve\nthe following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c69483d38b5e6e28ccebd00a601f6b1fa8da17aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c69483d38b5e6e28ccebd00a601f6b1fa8da17aa"}]},"branch":"refs/heads/master"},"ad9549b5ff824d792783e60df14214a0dc36d56d":{"kind":"REWORK","_number":5,"created":"2022-11-25 18:25:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/863562/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/5"}}},"commit":{"parents":[{"commit":"a1939cba037926400cd777059ee511481916cb08","subject":"proxy: add test for ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 16:46:48.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 18:07:00.000000000","tz":0},"subject":"WIP: restructure cached shard ranges","message":"WIP: restructure cached shard ranges\n\nTODO:\n* test obj controller with deepened tree in memcache\n* more tests for NamespaceMap\n* clean up all references to ShardRange that should now\n  be Namespace\n* move any more Shard Range methods to Namespace\n* ? make ShardRangeList extend Namespace\n\nRestructure the shard ranges that are stored in memcache for\nobject updating to achieve the following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad9549b5ff824d792783e60df14214a0dc36d56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad9549b5ff824d792783e60df14214a0dc36d56d"}]},"branch":"refs/heads/master"},"58aceb98bcb8d7bf95b8dc4be04443dc4ee9a40e":{"kind":"REWORK","_number":6,"created":"2022-11-28 18:48:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/863562/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/6"}}},"commit":{"parents":[{"commit":"a1939cba037926400cd777059ee511481916cb08","subject":"proxy: add test for ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 16:46:48.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-28 18:48:22.000000000","tz":0},"subject":"WIP: restructure cached updating shard ranges","message":"WIP: restructure cached updating shard ranges\n\nTODO:\n* unicode woes with py27 - failing unit test\n* test obj controller with deepened tree in memcache\n* more tests for NamespaceMap\n* clean up all references to ShardRange that should now\n  be Namespace\n* move any more Shard Range methods to Namespace\n* ? make ShardRangeList extend Namespace\n\nRestructure the shard ranges that are stored in memcache for\nobject updating to achieve the following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58aceb98bcb8d7bf95b8dc4be04443dc4ee9a40e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58aceb98bcb8d7bf95b8dc4be04443dc4ee9a40e"}]},"branch":"refs/heads/master"},"d0dfbe9a9dd07be7b8067cbe8c3c41d894d705ae":{"kind":"REWORK","_number":7,"created":"2022-11-30 22:34:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/863562/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/863562/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/863562/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/863562/7"}}},"commit":{"parents":[{"commit":"a1939cba037926400cd777059ee511481916cb08","subject":"proxy: add test for ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 16:46:48.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-30 22:34:14.000000000","tz":-480},"subject":"WIP: restructure cached updating shard ranges","message":"WIP: restructure cached updating shard ranges\n\nTODO:\n* test obj controller with deepened tree in memcache\n* more tests for NamespaceMap\n* clean up all references to ShardRange that should now\n  be Namespace\n* move any more Shard Range methods to Namespace\n* ? make ShardRangeList extend Namespace\n\nRestructure the shard ranges that are stored in memcache for\nobject updating to achieve the following:\n\n- break list of shards into sub-lists and store each sub-list\n  separately in memcache so that individual memcache values are\n  smaller and play more nicely with typical memcache slabs.\n\n- only persist the essential attributes of shard ranges in memcache\n  (i.e. the namespace bounds) so that the aggregate of memcache values\n  is smaller.\n\n- use a tree structure to lookup shard ranges so that 2 (or more)\n  smaller lists are bisected rather than one larger list.\n\n- only load relevant nodes of the tree from memcache so that less data\n  is deserialized for each lookup.\n\nThe trade-off is that lookups will require 2 memcache gets for smaller\nblobs vs 1 memcache get for a larger blob.\n\nChange-Id: I71adba876503cd41f3e451ec3ddd8ba87ff49ab7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0dfbe9a9dd07be7b8067cbe8c3c41d894d705ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0dfbe9a9dd07be7b8067cbe8c3c41d894d705ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
