)]}'
{"id":"openstack%2Fswift~864088","triplet_id":"openstack%2Fswift~master~I88bce3c178923457e3887790fa58aad56086b1e6","project":"openstack/swift","branch":"master","topic":"transient_overlap","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2022-11-21 12:44:19.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-18 16:06:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I88bce3c178923457e3887790fa58aad56086b1e6","subject":"sharding: avoid transient overlaps while auditing.","status":"MERGED","created":"2022-11-09 07:14:20.000000000","updated":"2022-11-21 12:45:40.000000000","submitted":"2022-11-21 12:44:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"864088-transient_overlap","meta_rev_id":"549ec8ad7e5116c64fd6d4f805516202dc5e24df","_number":864088,"virtual_id_number":864088,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-21 12:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-11-21 11:27:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2022-11-21 11:27:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 07:14:39.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-11-09 08:08:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 08:22:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62d8803d2f7207185b6cf45c67eaac68a75bb7b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-09 07:14:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3edaa70efd2ef49b9db191cbb1b7ac9c46825133","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-09 07:14:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"90142cd40bf0f0d89830007d1eab82160fc4bfe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 08:08:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cec784e22fc0437cb80d78d6cd6053a3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/582b358e7031407db496bf27b772049c : SUCCESS in 43m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1159cc1fc8304f9399945f2900b699ca : SUCCESS in 44m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5e95f5e95e54e8ab9880b030d7ea38b : SUCCESS in 47m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b16866774c7420a8210690c3de91f72 : SUCCESS in 46m 31s","accounts_in_message":[],"_revision_number":1},{"id":"947bd39dc259c6edbd463e22d9aaa24701b1648a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 08:22:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3be12f873cc84829a3d18ea3b558d59b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be15c66ca61d4b50b4a07dbe066b80ec : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/590d557ac3f74b6aafbdea9918f0a721 : SUCCESS in 44m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24d3849ea593475b80ebd4c8321a0f3a : SUCCESS in 46m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe8bdc75e388466d8d289d44cf7e2205 : SUCCESS in 59m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f402749e81d47d396f83ded9b9aaa97 : FAILURE in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d6576d2f6bb4801a0f4d23c076a9a41 : FAILURE in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d7839f35d2e444cbba41d9cab44cc0f : FAILURE in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbd33ea233ac473aa42aaa6db4f83c1a : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de0bc58ca5f746c1a1c575846cad86cd : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a05be7a739b14a8897f58189d7ca41a3 : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/251ff30d49144012a40eb2883473c1aa : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b257bb67a6b44fe9a00a0af376127f3 : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd28ffb82fb64869aec61df830bc88ab : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed785bbf2f184ca59fda16156eb90b7a : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d0d9ed2c80a4591b96b7a4b543241dc : SUCCESS in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f234823c62394b1a91f2517c98b4d82e : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d69300ff0d8d4a38b199dada5fb689a8 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3aadede704c4c93b6390423b01fec4b : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8dc722aef0ab47f99e3cc9af40255910 : SUCCESS in 13m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d62491a7810c4c27a2a3f231b26893e5 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43a905754a6f4b809972e6412e88fcbb : SUCCESS in 6m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cff327f7fe864e59bcce2968d82665eb : SUCCESS in 57m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/50bcc08dca7140c38ca797ee7f210f66 : SUCCESS in 1h 02m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10f3717729df411483678c798acfca8a : SUCCESS in 37m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e3a2e619e9a4ef48820ade2540d971d : SUCCESS in 48m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4710f6fe741499fa37a16e40d8b032f : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e05c5ea5d742dda49ebf33f890c753 : FAILURE in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14af5197892b4ec0a79c928d31feaa28 : FAILURE in 19m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6661a4033f3d4af498b8ce0318676058 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"db6394a2b768afed8d65decbc75c1e19254687e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-10 05:07:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6971ab79e829258565aac1a552a91819dca553bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 05:45:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3190a7b733284509a59c6a5f9592cf56\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cedffa0198f48fbb0d58c2feb9dc653 : SUCCESS in 29m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26049ccb6d934d9aaad6ff0602ea609e : SUCCESS in 29m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8812449e2c8a415eb363fc312bf4f0f8 : SUCCESS in 31m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3eea5b37fe5a481ca90a2576636cb6a9 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":2},{"id":"014df18a672616b8c716c8262f8207c641d4a8eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-10 06:09:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6e149d9363286c06fcbbe9554ee6e0f7e47f607","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 06:51:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/511a5ab1d1f4460ea42de57480ae79c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fd9bc2d18084c358c159498557a859a : SUCCESS in 34m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c4cdd9cc4b04ef1bd85749cca91f9fa : SUCCESS in 34m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e5f1fe8a5e74469a21b7ce8a559d084 : SUCCESS in 34m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/daf529f513d64dfe95a9640125c77f5a : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":3},{"id":"692e2a25ffa83640b1fe90d2bac43ec21c22da48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 07:09:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/addfae0acc184750b382e1286c51e828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217506084f004deea8cf5a8237561b61 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/973be3827b71436d9b3c66c36fa02c4d : SUCCESS in 48m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c804694bef5438d8572a7da5aec38e4 : SUCCESS in 37m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2a27bb95c7940a580447c73ed8e5f31 : SUCCESS in 44m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e3b333a4e30464d89a42d46bc9bfe88 : FAILURE in 21m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1f4fd2e22924bfbaf4c4f5621579124 : FAILURE in 22m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2d889eb0d0844ffb7c33df459375122 : FAILURE in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/975385ca9f344cd998ac32549e971d79 : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90bf3db37c2f44e79228af10af6278f7 : SUCCESS in 4m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a28ab16e84aa42dab943ae5cbfae4a76 : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c23b16838c74058b942ec6de335afa5 : SUCCESS in 16m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3724f8f9fffa42329139aba81f25b3cf : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00fdc53e615c42c99876f41132ec60c3 : SUCCESS in 21m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f331636c346943e7b5e840b875ca3ea6 : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c177111636b041aa90e4fbd7eee01713 : SUCCESS in 19m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8fa6cf61ffa445d9726966a5b5d4792 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9344306e25f44ccbbf38b8068db6d3c : SUCCESS in 14m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ae8558932ff4e84916fe9257d127bc3 : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca75753195564b5db2b608feb883a706 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ab3932f62b040c792dedef298023ad3 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/05ccd3ca22fc4695824eab5a9fd11c25 : SUCCESS in 6m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d8acff7192174a8696d40f7f66d2ac3c : SUCCESS in 48m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6f43a5077fa42f5896289a32da1b023 : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0cf525202fc246c9ac9c887871b47215 : SUCCESS in 38m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5475ba9458c84e17be07c0b22f9bebd1 : SUCCESS in 32m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6318c631846c413db7f9ff820b369847 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80d8c3104ad447c9c4332c71b2e7f26 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d8ff3d6f86b453c92047ef3e335bd14 : FAILURE in 15m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fc70563c99646e98dc477fe614c65ed : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":3},{"id":"3e4f60352f8f7a2bc8b476f4a2192971650baabe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-10 14:05:44.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"19b0b1428a94abaf357fb76d6c355707765443a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-15 06:29:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5058ea10fb03f4ab95a70ea917328ff9bf558d9b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-15 06:30:30.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"43ccf8f0ac4a16fd705b6340f1800ee41a9b4c66","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-15 06:30:30.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"8841b909fa4f874c3e0d91e4e7727005c12d0d74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 07:25:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d42c54773d12404da0614547c9e48402\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4225b85a001d4f99bc6c92e4e17ff282 : SUCCESS in 30m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c68d44380e3d463db90c7e56416d6f82 : SUCCESS in 31m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a05031ccd914937a1916818c54641aa : SUCCESS in 53m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d75041e93c1433f9438ab30f17be5f5 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":4},{"id":"12f5395e6f480100779a6acb1921cbb08cbd5b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 07:32:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a27f0590733745f0918140f18c673000\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5b9b1ac8ff4433a8b0594dc8f96418 : FAILURE in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/e2b2dff553734bb896020616f7b908c0 : SUCCESS in 40m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd79ed570f394ee08c9b15665b1b57a2 : SUCCESS in 34m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e219017f675d4084bfcec756a1ebc520 : SUCCESS in 38m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da8d4d1039244aba893ff24c8de8f223 : FAILURE in 14m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5fd1250dc40447595d839fb769efa4b : FAILURE in 22m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb6595d60417451c8d84c230170ae992 : FAILURE in 22m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca44e8d37a8d4aa1afd390b290e9b060 : SUCCESS in 8m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edb1925c471d404cb865beac4b1e9cbb : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bed7a144854b4ca6b826e87dfc6d8606 : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/350c5706f7354367a7bac9fff8b20daa : SUCCESS in 19m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0cc786d3d2bb4d189c96be7649c8be3d : SUCCESS in 15m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2f256e74f124d5b96c273433f779597 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d325fe508807455a85c6c8199d5c9dc8 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae0c0cc659a14b0fb735110453665814 : SUCCESS in 16m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2cd1ebaebfa434582403cfe564f9761 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/70b419caf9464f0ab51e5432d7a73dee : SUCCESS in 18m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9fa7a124af18459da3d2d6794bfb6a8b : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d56fa6a2e3da45a388a52f6aa4a33836 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8fe452b6bbcc47c38efae4a14e13cce8 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b2a0cf15034405893c8ba3903a32a95 : SUCCESS in 6m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a23d7d63edd477db10fe70edc99f3e3 : SUCCESS in 52m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/88a6c685890c4c4fa0adb8add532f08d : SUCCESS in 58m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bee5415f55148edaec821cdbad52fba : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74ee124a2cd94ee9a07bdbe21d8527c3 : SUCCESS in 41m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02bae673ac5e49538f9fb5c24d34adaa : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3a3308ef444daba9a8883dc90a5208 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbd333d342444d48be58583429e101a5 : FAILURE in 25m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1946503f26964d658cd45de69809b337 : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":4},{"id":"5d01831aa4882b42b07b8c091508c250b88c0242","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-15 18:12:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"832c0c7ecff46e4f29b0e4c414dd4826a44b34df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 19:21:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89f63e9199fa4a23a238009794807570\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588db56105084ae899f72d8948373c23 : FAILURE in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/c98092a1734b4ee98c7433a4a06d99aa : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c6edb885b4f4810adb2013212daa370 : SUCCESS in 54m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91c13534f91e45b4911db49b6e3a9874 : SUCCESS in 58m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2f05b2454354832af72bfc0def255f1 : FAILURE in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77953939ea6843369cdac25dc4098330 : FAILURE in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df10403912794129a24a9d0192b8a56c : FAILURE in 19m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48ce5c815fcb4d139c2b3646ebe27b9c : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d297cd98da9c4352a47b5b2c5d907d18 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/170a222b7586444885cfeb8952363f39 : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d177def3b7344909172da9d020b3e0b : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a5bc9428f0245fc8ca9f6ac5b02d440 : SUCCESS in 14m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a063e5e7f714b65b5ac0b76a526dd4b : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c54dafa5c1f420ba5d30da5736b7a7a : SUCCESS in 20m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5090b9604c394eeab430a390469e89fe : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9b69995a2e82495aa4b8fcb5ed97fe51 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8854e20a6fa649b4b5aba94acd7cbdfb : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/88cfdc39d46b429281dfb42d45bb4ab5 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a13547f762b46de8909baffb73e191a : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97a8ff08bd284c7bbe14083c9073dc17 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20e0dfb6dd604365b0cba9e43c90beca : SUCCESS in 7m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9f1ed52f84b1444384e2472487913cc5 : SUCCESS in 50m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bf1657a388e42e1b78601dba783b36e : SUCCESS in 1h 03m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/984ec83f61ba4d84bca88f6524169111 : SUCCESS in 47m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa59d8beb536478ea222b7316ee8763d : SUCCESS in 46m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86e93c28f6724dfd99eb70f77568cc50 : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a8672c28d36488ca6a180d362e8bf25 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f7a0e5a6e0e4851a0f736ae17c68609 : FAILURE in 21m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3008f0189d67431ca48a1f29e430ecab : SUCCESS in 44m 51s","accounts_in_message":[],"_revision_number":4},{"id":"bf50fb4ca4cd9ac04062967631e501656dc3122d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 19:21:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0515b65f73b400c8f478a20de38333d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93c1897c1fd344db8fa88c93470647fe : SUCCESS in 43m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a900611b88454e2ead413ad8907c2655 : SUCCESS in 43m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/216f72d27b7d42d1a4be2c3335205baa : SUCCESS in 49m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/56606a5b07c645428ec0c691881c644c : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":4},{"id":"752b4c7556edc29fef2a09f4c46ea952cb9c28ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-16 00:07:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d730a1a6c8035a857fc778e4c5ca5d7b249f3e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-16 00:29:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"291dbf9fbcddb3fdd9f3c39ca8f783f64183aa26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 01:07:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c95ee450cb240e69a74d55915e05ac8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c832c05451c43fba99eaa484f030c73 : SUCCESS in 31m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1184c177c0e4b938530278bd45d0548 : SUCCESS in 31m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f23edb57cf048d8b03b433311db3367 : SUCCESS in 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c885b34a7b54d9ca0055d9968cc1b1e : SUCCESS in 31m 05s","accounts_in_message":[],"_revision_number":6},{"id":"47d9f81d9bea4807bdf1637120d0bd3f481afa3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 01:27:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca928be049624014b3c2265096ed89ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f792da844454da6ad3ea3f0bb1e271c : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/9ed8885caab9488abf7fe21fa8563764 : SUCCESS in 52m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/11c3d081a4e443289befe3b9b62c8c0d : SUCCESS in 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8e1d9f3f0e140cdb87b790221f4fe58 : SUCCESS in 49m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2b085c47c81448995c910c30be1e404 : FAILURE in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bea2b6a1c2d4d17bb08e3811132ac3a : FAILURE in 18m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5154dbebc3294a9c82ffe2243beb3f5a : FAILURE in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b2adca376b846f98ce52226649b960e : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d31ecd8200d47a883ea0f2b5505bd47 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26b9b007ae9a4763ad794b8e366ce8b0 : SUCCESS in 21m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ad9b3ddd3f3406990a920c9a60716cf : SUCCESS in 19m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cd2bbb2687ca4de0aec7d444810b3bb1 : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9186ad78ebcf40549bfa2da92780bd66 : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1980c87c5fe14921aedd5e5d379b04ac : SUCCESS in 16m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/341f4851a7844e178a4026e16e3d3e6f : SUCCESS in 21m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/563f2af09ae54b3f908a8778e7828775 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f8cedf0529e492db21c842e22db20bb : SUCCESS in 13m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3bc8f78e316441c28518f00636a4b3e5 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a525dd80c314f658a23b333ea9f2359 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97d3ce77a11e433fb9629acbf59bb57f : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b57464d6a96c4e3298b180821f69bad0 : SUCCESS in 6m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b5844bf4513c47f3a9e9cd5a8f0e9113 : SUCCESS in 54m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca039a5b380744f39dd88d3a36d8a616 : SUCCESS in 53m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2277b221b65147a3a747bbeb34dfbf63 : SUCCESS in 40m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a16447414a43438f9cc4602877f86a4d : SUCCESS in 40m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c1ee982235344699a8fb1fec66caeb5 : SUCCESS in 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9574a6fb78bb4a9ab739abb8d884e532 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e6cc3a79a584896af62632912cce475 : FAILURE in 25m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5ea490e66d74b75bcc476e50584a513 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":6},{"id":"ade5a7edcb62f1cd3401dede7250f6d8236ffe1a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 13:33:53.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2db3f33ca2e4b4e43fe50ad7f297c395d711109f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 14:31:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ce3ee0b092f07aa8d766d35df61c29381fdc578d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 14:32:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7178c4adff8589816239e999e7d91d19c6277c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:50:35.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2a3ab036a5a4b4fbd72af7f3978e421\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d514924cade4fcea0db9c3a5c49024b : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/7fd93d058bfd4da69eb2744e3941b911 : SUCCESS in 1h 13m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21f5572b8ea74c2a94bc9a1cc1816b5a : SUCCESS in 59m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98dc5f5c661d4e5f986f6f90bec7524d : SUCCESS in 54m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/552d71847a424dffbfcd67637202c95e : FAILURE in 23m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5b37ad5415e4210868e17967dd00af8 : FAILURE in 17m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23da93815ed74648b48c752c7473b4cf : FAILURE in 19m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4898f3f45654e6a8fcb424af229d0d1 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6245ccf1b2e42cd9f94127bce48e5c3 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/13aa6090a6f14161befe0a4ef8945171 : SUCCESS in 14m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3eef883506df4c4a8e9e10f7c90863a6 : SUCCESS in 20m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c87d6e7d6e74b74bb3adbfc2bce1697 : SUCCESS in 14m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1709c76f54c24e5cbeff4e5167b3785a : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c5554c0b5f24e3caadde65ebc844535 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c326bd02b5614623a1516b25050b211a : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a29e7b91288460c8d57fa641e6f3853 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a83d0ac29059486aac0c3bd929393ccf : SUCCESS in 13m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3030cbe48d244941823cc88aad4103d4 : SUCCESS in 14m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/872740102f1d42b7b55649cc295a0503 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b92093ca784e40db8e975fc96fb79f5f : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/65111b6b919940d18046585b879ee6b9 : SUCCESS in 9m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3307f4b21f55422eaa0c1ad3f03cb555 : SUCCESS in 51m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab08d0f5ccab463ea2452d3dcaead80e : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcc8d44a86fa4c47970915745274d3b8 : SUCCESS in 42m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/083e3b82626d46f6ac95772ead6d74bf : SUCCESS in 42m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da3b7dd64b4f4896bfee89b5b207dd10 : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51f9de0ab4a42cc923681e427504033 : FAILURE in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee1176389bf94f2389e06c9eb7cb6929 : FAILURE in 18m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/340de115e81b405bbde6fde523cb8819 : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":7},{"id":"fa2bbeb3735196c3cbaab98ff0ec6d7a58485495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:04:43.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abea8e806a72499893c7519f965d9c0a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f18f43416e64b64af4f8ae9cbb881bd : SUCCESS in 54m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d7a536ef1664b9f94effc098e944c2b : SUCCESS in 46m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/37d0a8f9f9b84f6996c57fafc1f21af1 : TIMED_OUT in 1h 31m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9436d33e4a324fce88200ba85c8ffc5a : TIMED_OUT in 1h 31m 01s","accounts_in_message":[],"_revision_number":7},{"id":"fa6ec919e186ae2a554a2ab5ad23672ba4bf8591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 16:18:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"81a2ae4468b566143ec3460c7ab5bcba4ba0281b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 16:19:07.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"10e0448661b83440dabef47ada88ab98b5de9750","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 16:41:56.000000000","message":"Patch Set 8: Code-Review-2 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"83a17ea6272e7b800c07605c5254f53f860670f0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 16:42:33.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"675167cf06daecfdee5059bd0c5fa4ef5092c4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:23:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1e5e2a6b4544c980686881c246cbb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b93b224d15498aa98b50d48ca58a5e : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/2a591d8a2fde4cfabc5a7d08aff3a8fc : SUCCESS in 54m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65a3386758084035be9453fe2a76f40a : SUCCESS in 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b898c2148d7847c6a911412577146a58 : SUCCESS in 58m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cb2f05d1be94fdfb8539f370ea96dc2 : FAILURE in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/183b6e3508b0430093737a140814ba97 : FAILURE in 17m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e9358e5dbf64065b429b3b362d90275 : FAILURE in 23m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/92ca515e39634f348488ea4cfd8b91c0 : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b28a787e3fa4287a37e660420146b84 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9e6155b8fbc0440f863957d8a2990475 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a31024e60184af8b918f4e68a5dab40 : SUCCESS in 16m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c4e14bba3f9421e83629f17adc61d5d : SUCCESS in 15m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4a834d9400924d97857fdbd6721e183e : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/16a7feee87e64f27ae24106ff3f60df7 : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/133167264d14491880168f5566aec32c : SUCCESS in 22m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/72ce966dc29543dc905398fcecdc1d5f : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/542d91ad1fda4ba5b53dff2cac02a03c : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d77399f265674a5590ebd2619fadeef5 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d0f43328cffc47868cf768481d3984ed : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6036939d61240439c811f3117edb162 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b35c0b30a6454cb88e575ba7f7127409 : SUCCESS in 8m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93f1b86d991a48819f0efe59dd2804c6 : SUCCESS in 56m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e37524cb4af24ce4856b61dd1045eb47 : SUCCESS in 1h 04m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2f473dfa13c4f99b66467ae2fb18591 : SUCCESS in 39m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/928a7c2881c048f5a9a8ced2f94bcaa7 : SUCCESS in 43m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20abf6fcb3484ddcbafb62ef1bba67ca : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1dcc12572c4496981772daa36fcb3e : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ed19934341e4c26aab869d25e5d8b40 : FAILURE in 26m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d75382b5be5849338875edd30b20628e : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":8},{"id":"4e1c12bdb186d83b156520b0093b9951d134ef51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:45:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ab5b5c6bc547a7bd054bf96bae2ef5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9dcdaa1ca584eae88aecece4f0efbdf : SUCCESS in 1h 09m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8982a38daec048d6bace3385536d473a : SUCCESS in 1h 10m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f64ddd36e7a14fe785314d0f669b0229 : SUCCESS in 1h 19m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/677f17924fe94d2dab6fbc110d5357da : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":8},{"id":"d36983c6b25a089085cfec6d92b8ca10b835c52d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-17 06:16:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"64b7fa9857d7bc0c9e7239be64c8b5d110e4bcca","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-17 11:22:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6a3baf047a2304a597531f5c8b94b272e6ba69ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-18 07:17:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cda33a2eb5025e84dc929fa359e2d2458f110ba8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-18 07:19:27.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"69ea3cf5157c68956b0309ac0ad8b8bd13187032","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 08:12:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca13528ada84b6ea4139c9db74f5089\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a39f0aff98324f2080139e8abb00fb80 : SUCCESS in 43m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e07888196874facb551243529d01b02 : SUCCESS in 43m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7955d1519ecf40acbb98ba96bc5bf0c2 : SUCCESS in 44m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b9f6161b58f4952aaa25e5f125ca378 : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":9},{"id":"9dfde827ade4a319c48b0c6b0090370c1c260f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 08:36:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf4d773be9bf4cbb902c009c4cbeee23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491d697861384a6c85884edaf80943f5 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6d81fa6cd1224e32bd3714ff0ea5883b : SUCCESS in 47m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d99eb978bf3d4e0d935440cf60a3dd66 : SUCCESS in 40m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d1c37a7a04840a583971814b012c81c : SUCCESS in 41m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53b262385fd14a40a52bf0062a864b7c : FAILURE in 15m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83833d25c2034c43a5bba99e7e153dde : FAILURE in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e9e0653497544f08d2c036d1cee73ee : FAILURE in 17m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ad27c159a404fd4b184bf28b9f65d53 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ab47bf98b2ad442481f0b94b2b59a0e1 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/923ef9e00584452d9c54e483d61f467b : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bbec72a661464d938775db49578aa8a7 : SUCCESS in 16m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a16c7f3e440946e89b16eeb35edea467 : SUCCESS in 14m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb22b5525af24a0db61e19928e637c4c : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb17755c2a0f4222b1b956d547d7a19e : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c9c5d28aca644ffa1cb47bf0cbb0807 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e847692decd42eabe62206213872e21 : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b0f8a8d6081845df96c76e113ce99423 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca1f9bbb29dd4f06b1a9ffd26dd8393a : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1845289d32864b30996be7155e0fee11 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0ec28d2cec940d3a627dc2fb7f8c563 : SUCCESS in 13m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb482ec87f2d4dbe9d886cf4903a94dd : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d89c90636ae4d12b95a01bcea6cdc22 : SUCCESS in 52m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0170f1b0b35b420c8bb8cb76e0ebb0e6 : SUCCESS in 1h 05m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1352d5facbd94289a209786bc16f0a44 : SUCCESS in 51m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59f313f1ac5241d9a3fda0e0a2d08300 : SUCCESS in 47m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afa6aaf1384f4fe0bea7fe3c4f439762 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c398ffbceac42f2af38553270502a94 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4b0c48a9ad94fa19ba2d1a9984ebd46 : FAILURE in 25m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77e8fd7199dd4880a6da17cfbc646fac : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":9},{"id":"b59963cac06fc0f222c307af8c7998c8a90d4a18","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-18 16:06:12.000000000","message":"Patch Set 9: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"167befb4d02b6d7e16885ea7d609032298dfa2e0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-19 00:03:59.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5c2ce588e3d9cb2f93e569b0b632f61124b34223","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-19 00:04:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a125a8e0c8c5039afca780c9dc9112fd367df2ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 00:44:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c249062520b64a21af9125ea15c5fde9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37fd0a42efc94da7801cb1c5394a9c0f : SUCCESS in 30m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28c1047ffd6c44d6b59eecd62edea32f : SUCCESS in 30m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/700cd4ef65fb48deba2920d003b30e13 : SUCCESS in 29m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0592ba2636ae46a68f8680f31a4dbf38 : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":10},{"id":"448b7fcf9dba81680439eace93e6bbe4144b573b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 01:11:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b91aee0943745e7bdc07c267dabd271\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce6a7a4c72946bcb6015c8ba10114e0 : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/bda70facd07d48bc80ec2f9147ff04b9 : SUCCESS in 48m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f874f08a1514c4c914413d553f5741b : SUCCESS in 49m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8d9a9d309a1472da956e6176e6acfe7 : SUCCESS in 45m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84cf29c2a14f403ea1dd9ce7fa853e8f : FAILURE in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92df9d7e07114d2390d6590cbe344eba : FAILURE in 18m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c36413bf15143b8a1047a4c01a5c253 : FAILURE in 18m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37d0df36ca864c4aa4836b8dfcf0954e : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e963ad29bb644ce80843dc3f13500ee : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06bd3992c9024fedad5cc1897d329702 : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e49807197d5c44f3a72b757eff637711 : SUCCESS in 14m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30620803ded747b9b3f0e36acef2d2b8 : SUCCESS in 14m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df572189c6a14d36b7be319541ec7877 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2af6641f989140d78e395f4ea61bb708 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ea54cb190b94a568bc76ccedcc573c9 : SUCCESS in 17m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/832f99a2a50b4b1089f578ab252d2af2 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/16f76a6f64b548f9943c1da38f8bc066 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4b86840a62554544ac439f62d42f8b67 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/512b01aabc364cfd9582aad61d00301f : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9595e6a21454819a77cb65cac82d754 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0280850c1a5544b190b8210d659a4693 : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f137dfee677a4bc382f8692524515f93 : SUCCESS in 52m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b549dfb93e1e4b3b90c304ea0caf48e9 : SUCCESS in 55m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5dd51057a424475d8e62d958d5193e21 : SUCCESS in 39m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4058b283b8254906b79de17822609d0a : SUCCESS in 35m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52e7a850ef2940ba86f87739c571e98b : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/305741d97c0344dbace27e9c0e86e62e : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0788183c8f534f2c81fb8147f1fc85b5 : FAILURE in 23m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7efd698a9604347a8baab9c00700c47 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":10},{"id":"5fd6e10018ad46afe14b4d8184aa1678ac82f907","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-21 11:27:34.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"456f02477ab0ba45dfa63aed6164c177b02786b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 11:28:08.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"93404cfa680b7cb33a68ef8e6dab0056e78148d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 12:44:18.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f82f206d36ef4e01b6d109865e33abe9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d899de1824c745e9b869585959448a22 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/50c8a3de3cba4e10a7b1d6aaaf37d7fe : SUCCESS in 1h 04m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cff1c126062542f99a2e651e7c2402b4 : SUCCESS in 55m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/323e99f27ede47a2a62dae4f088b8a6c : SUCCESS in 1h 08m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4b45eea2cdf649c5a11475bc2d8dced6 : SUCCESS in 8m 56s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dd55c6017233414b8b894373c3febf0e : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97a7ae0190c64075884475679b6bd5e2 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/37b7076ad37f43bfb50ca42dc850906b : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1540e17e1514affabb3be6232e3ca5e : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41c6aaf6e96f451aae892c191a3c2ef7 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4740af7ea8a4928a1b35e4ba8556d2f : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14e06fbbf2674d8c9c52375ac0f3c76a : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c3cfb8ede7b64d05b8c303d67ad4addd : SUCCESS in 13m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e13650c381804bfc803d5ce2a748127f : SUCCESS in 14m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a0234e492eb43b3be39276dfeec46bc : SUCCESS in 13m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f691c8983d5b4e87a31d84a5a8a224be : SUCCESS in 6m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ebf703481e7145a197644b061a86babb : SUCCESS in 6m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4cda3af239c0469ab5e5a0ea1da0cb2b : SUCCESS in 58m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7babb7a712ed4ab693ad0c6ad757f309 : SUCCESS in 54m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eae7d7657cfb43848311fbb4825c9c2d : SUCCESS in 47m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de283556498d4042be9767d3061dffd0 : SUCCESS in 41m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6655f93cfebe401daa5c80e2195ecf1d : SUCCESS in 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6128bedff48847d58906ba83f656bf2f : SUCCESS in 4m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a5214a3889e4cdba8af152909b06fd8 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":10},{"id":"87aec09802525e3940abdfa146887d90d335bc3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 12:44:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"549ec8ad7e5116c64fd6d4f805516202dc5e24df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 12:45:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0339eb4e954c11b76268322d5e0cfb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22ef0521cb14dc4a8adae7f77f72b69 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6fe0bafbd8a6465da3034990df97d4bb : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5134b6ad11034bd9a50c228428806540 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"230b467af09af2d4df42915d1247a1304ac61b2a","revisions":{"22f3608f3763671b79e9e12cfeb42d9017a66ebf":{"kind":"REWORK","_number":1,"created":"2022-11-09 07:14:20.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/1"}}},"commit":{"parents":[{"commit":"03d8f823ef0b24d40376e6f8db995eece9215e47","subject":"Merge \"Mark rolling-upgrade job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03d8f823ef0b24d40376e6f8db995eece9215e47"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\n   NO test yet.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22f3608f3763671b79e9e12cfeb42d9017a66ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22f3608f3763671b79e9e12cfeb42d9017a66ebf"}]},"branch":"refs/heads/master"},"bb51c53222faeb8f3b26e5bc94a63788e0bd6705":{"kind":"REWORK","_number":2,"created":"2022-11-10 05:07:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/2"}}},"commit":{"parents":[{"commit":"03d8f823ef0b24d40376e6f8db995eece9215e47","subject":"Merge \"Mark rolling-upgrade job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03d8f823ef0b24d40376e6f8db995eece9215e47"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 17:13:16.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\n   NO test yet.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb51c53222faeb8f3b26e5bc94a63788e0bd6705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb51c53222faeb8f3b26e5bc94a63788e0bd6705"}]},"branch":"refs/heads/master"},"d9b280c740d0d171e297e3cd999cfa00197da779":{"kind":"REWORK","_number":3,"created":"2022-11-10 06:09:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/3"}}},"commit":{"parents":[{"commit":"03d8f823ef0b24d40376e6f8db995eece9215e47","subject":"Merge \"Mark rolling-upgrade job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03d8f823ef0b24d40376e6f8db995eece9215e47"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-10 06:09:20.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\n   NO test yet.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9b280c740d0d171e297e3cd999cfa00197da779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9b280c740d0d171e297e3cd999cfa00197da779"}]},"branch":"refs/heads/master"},"d2d5d1fb757bb12c69dd0b5d0c4f5cfca44d2080":{"kind":"REWORK","_number":4,"created":"2022-11-15 06:29:58.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/4"}}},"commit":{"parents":[{"commit":"03d8f823ef0b24d40376e6f8db995eece9215e47","subject":"Merge \"Mark rolling-upgrade job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03d8f823ef0b24d40376e6f8db995eece9215e47"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-15 06:28:55.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disapear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will only emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges don\u0027t have parent-children relationship.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2d5d1fb757bb12c69dd0b5d0c4f5cfca44d2080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2d5d1fb757bb12c69dd0b5d0c4f5cfca44d2080"}]},"branch":"refs/heads/master"},"af9692867371c381e817b728dff1f706a03c1d79":{"kind":"REWORK","_number":5,"created":"2022-11-16 00:07:00.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/5"}}},"commit":{"parents":[{"commit":"03d8f823ef0b24d40376e6f8db995eece9215e47","subject":"Merge \"Mark rolling-upgrade job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03d8f823ef0b24d40376e6f8db995eece9215e47"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-16 00:06:52.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disapear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will only emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges don\u0027t have parent-children relationship.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af9692867371c381e817b728dff1f706a03c1d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af9692867371c381e817b728dff1f706a03c1d79"}]},"branch":"refs/heads/master"},"6191fd3ec4e4fc0d4aaa3241b8c20e2afa8b3194":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-11-16 00:29:47.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/6"}}},"commit":{"parents":[{"commit":"c82c73122b4ceeca1740989bc213d7ea9ec6fdda","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c73122b4ceeca1740989bc213d7ea9ec6fdda"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-16 00:29:40.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disapear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will only emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges don\u0027t have parent-children relationship.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6191fd3ec4e4fc0d4aaa3241b8c20e2afa8b3194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6191fd3ec4e4fc0d4aaa3241b8c20e2afa8b3194"}]},"branch":"refs/heads/master"},"7495c742cd60e80243bf9ea7c10ee17db17909b5":{"kind":"REWORK","_number":7,"created":"2022-11-16 14:31:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/864088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/7"}}},"commit":{"parents":[{"commit":"c82c73122b4ceeca1740989bc213d7ea9ec6fdda","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c73122b4ceeca1740989bc213d7ea9ec6fdda"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 14:31:23.000000000","tz":0},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disappear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will only emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges don\u0027t have parent-children relationship.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7495c742cd60e80243bf9ea7c10ee17db17909b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7495c742cd60e80243bf9ea7c10ee17db17909b5"}]},"branch":"refs/heads/master"},"3515141d27acad7d8244abc04323d15534079541":{"kind":"REWORK","_number":8,"created":"2022-11-16 16:18:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/864088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/8"}}},"commit":{"parents":[{"commit":"c82c73122b4ceeca1740989bc213d7ea9ec6fdda","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c73122b4ceeca1740989bc213d7ea9ec6fdda"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 16:18:11.000000000","tz":0},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disappear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will only emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges don\u0027t have parent-children relationship.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3515141d27acad7d8244abc04323d15534079541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3515141d27acad7d8244abc04323d15534079541"}]},"branch":"refs/heads/master"},"a1989b075a3502cba6856e3f53a757368f1f15c1":{"kind":"REWORK","_number":9,"created":"2022-11-18 07:17:52.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/9"}}},"commit":{"parents":[{"commit":"5a9f5f521aca0ab23cfd8de3065fe5887469b5ea","subject":"Temporary pin the swift-dsvm-functional nodeset to Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a9f5f521aca0ab23cfd8de3065fe5887469b5ea"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-18 07:17:21.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disapear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will not emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges are parent-children during the past time period\nof reclaim_age.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1989b075a3502cba6856e3f53a757368f1f15c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1989b075a3502cba6856e3f53a757368f1f15c1"}]},"branch":"refs/heads/master"},"230b467af09af2d4df42915d1247a1304ac61b2a":{"kind":"REWORK","_number":10,"created":"2022-11-19 00:04:28.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/864088/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864088/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864088/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864088/10"}}},"commit":{"parents":[{"commit":"5a9f5f521aca0ab23cfd8de3065fe5887469b5ea","subject":"Temporary pin the swift-dsvm-functional nodeset to Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a9f5f521aca0ab23cfd8de3065fe5887469b5ea"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-09 07:13:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-11-19 00:04:10.000000000","tz":-480},"subject":"sharding: avoid transient overlaps while auditing.","message":"sharding: avoid transient overlaps while auditing.\n\nDuring normal sharding, parent shard and child shard ranges\nmay have transient overlaps which will disapear when done.\nTo avoid this transient overlap shows up in the warning log,\nwhen sharder audits root container, it will not emit\n\u0027overlapping ranges in state ...\u0027 warning when overlapping\nshard ranges are parent-children during the past time period\nof reclaim_age.\n\nChange-Id: I88bce3c178923457e3887790fa58aad56086b1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/230b467af09af2d4df42915d1247a1304ac61b2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/230b467af09af2d4df42915d1247a1304ac61b2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
