)]}'
{"id":"openstack%2Fswift~864199","triplet_id":"openstack%2Fswift~master~Icf3a1df87d52dc5f95e15e676da229a3323326a2","project":"openstack/swift","branch":"master","topic":"err_limit_metric","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-21 18:58:32.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-11-21 18:58:32.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-11-21 16:50:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Icf3a1df87d52dc5f95e15e676da229a3323326a2","subject":"proxy-server: include suppression time in error limit log","status":"MERGED","created":"2022-11-10 13:01:37.000000000","updated":"2022-11-21 19:01:00.000000000","submitted":"2022-11-21 18:58:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"864199-err_limit_metric","meta_rev_id":"5cb939f2b715ae24f06afbedc010d055a5cec433","_number":864199,"virtual_id_number":864199,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-21 18:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-21 16:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-21 16:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-10 13:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-10 14:12:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-11 01:22:00.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2022-11-14 16:22:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-11-14 17:35:10.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-11-21 16:50:17.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"cbee735905e7ed293e7beee70f913bacaf1e3c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-10 13:01:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55b98eff528dcfe9845c7bafd5cb11571cfe67a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-10 13:05:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"851900a67e0d90ddc90a547dcabf91f586561e61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 13:40:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a80edb95fb46cba0a57822928a2c38\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/876d871c0e4c4ee1a6d7475a136b2820 : SUCCESS in 28m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ebd0063d47943f5b5a8355c619b1c42 : SUCCESS in 28m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eb048d9778c4fe89ba2f249c7bc23da : SUCCESS in 28m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f26ee06f17294f7ea641acda8aa2750c : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b551e95a67ad61cd91f34d74673d0a69c93335ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 14:12:20.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a3cd548fdf440469f502e27a4d8d6b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6982bbd2910c4f8ba8a8abfe202d4f59 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/5adc89d81dca4bc5af94ae38425b19ee : SUCCESS in 44m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/095cdbb421624a96ab0e1e04586b05f0 : SUCCESS in 36m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29f1d53cdec84eeebb227f6e4bbef138 : SUCCESS in 59m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d4650d38ea145448f2378ec099282ab : FAILURE in 21m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/887602f91daa43db876f42d31d9ed06d : FAILURE in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c5317cefb084af38ccfea1e49ec7310 : FAILURE in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39cdec3f5d7343e2b6aec9a05030b406 : SUCCESS in 7m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9e1d2a691dd049b2bbed0258eff35cb2 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f71ba0cded6142f29d6e5fae98f5e74e : SUCCESS in 21m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/daebb80629474265b6cd3814f9490ef3 : SUCCESS in 17m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f3749aab9eb44c78b8344e8484c7131 : SUCCESS in 20m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d9b71476ace4974a7b61b50c6e34ec6 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e6a647e0c8a4110984d6075c8b449e5 : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e8e9f8427fb44b8a05d7cc921b40e83 : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/085910c1b0e049eca970b946db693cf2 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/98b1b606d54346e28531efcc3a39979a : SUCCESS in 21m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/510c9fa4aedb4dca9f1afc70da44693d : SUCCESS in 20m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cab0ee24b25b4aa9b1511f6bdfbfb358 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfa2d192b9f34c0b904132dc2103f01d : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6872797414e04bdd921c53edbef345cd : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e9c00a3ad554ac2be1fe28a2b504f8d : SUCCESS in 55m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc4a91c623be415a955b72ce41d2747e : SUCCESS in 55m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fef1dcca82546dfbfd918d69499e048 : SUCCESS in 42m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4898faaab5e7425499f3760e602a4590 : SUCCESS in 37m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2731aff4947432a877e9776ea34bd19 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ddc2e1e38a4cecbdd3a46052e548f4 : FAILURE in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3e0894acfec41f395b4d251b84ed809 : FAILURE in 26m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be047c108ff5434b98f646261e95ff5a : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":2},{"id":"efd6860430b66a65370eab264e637099738f8934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-10 15:31:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcbc2b3d3bad23415dc3e56564d75ebb105fbbeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 16:13:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8159257bc564bdea073917c4108f90c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d9c5e0096a0496e8a9eed384a540764 : SUCCESS in 34m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53953cf2cccb4bffa2b10f8626c87f87 : SUCCESS in 36m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c5e0700a245468e8ffb6b564f1012cf : SUCCESS in 35m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c18d670453f548ba914b4002c063f35e : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":3},{"id":"82a1d98693e62399420d673e1e15123d55cfd4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 16:26:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65543d8c26a04393b1a109a110fb1306\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75dc8d33239b4ff18b104803135cf4f0 : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/2d9b7adb00a548fd999c983eb61e84ff : SUCCESS in 50m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/820e797c422a4255926eaae4a6d9b4c5 : SUCCESS in 39m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cb2b8826f0d4133971132ec37c9b302 : SUCCESS in 50m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5a72efed4024de4a45fd85d598057c3 : FAILURE in 15m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/830371e1db114501863582183dd74a61 : FAILURE in 15m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41713b4512c047b9ac6285ae16eb5d75 : FAILURE in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9236a03ca0ea4b27a2ac90a0f92422af : SUCCESS in 7m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de05ac119ae54264a44016e45bbd3ab3 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b5fa3083d9f4a0593d051911e038ea7 : SUCCESS in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f512486a84d04344830c1cc292bf0d0b : SUCCESS in 14m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f79accc706d64df1a8c15e04bcc14ade : SUCCESS in 11m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb36ca5602654abcab9d99c824f4ab9e : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0639db0ea57d4f48a50b4e6dd02c4d96 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8cefcabe3fd64e7eb240942712c747c4 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cbe4aae01fe1494da8f173af943cd589 : SUCCESS in 11m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2302b3ed27a4e4d8424ee3e7ecbd008 : SUCCESS in 14m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f7d684969fe04ff1941d74fa2ab52ca9 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7418aa56881f42b7ad44cb4add1608cd : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6937b63f869646e7bc915806b7a13ce4 : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d12a033f2b0a4b2cbab034f095d0a14a : SUCCESS in 6m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/42554878dcf5488d9d04aabb145bcbd3 : SUCCESS in 47m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/951c32649ba84fc6ab86780e7c3c3af6 : SUCCESS in 50m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c772dc665d58428b857f954820211b70 : SUCCESS in 36m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0934078929a44a7993f9f9270ce05d9f : SUCCESS in 34m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd39d77097e5443091fda2652dc3e099 : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03de5a231985445993097ba0503c791e : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7165e92f2d3b4e80998a7dfdbe89ce6e : FAILURE in 19m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03d3b89ca8334e9396e410989eeb618d : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":3},{"id":"87caa5b42ca832916227e708add10b1205cc9ad1","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-11 01:22:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6256d0b7f6a09fd72ab9e3a97b6aa22940cb302f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-14 16:22:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d4eef0165dc1056b296434e9193da5f0c070ceb1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-14 17:35:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1786c7499348c9293b16850cc5942df9db0e0496","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-14 20:23:25.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a9d7063ad367e83870d4d8955d494dd8f52d9c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 15:02:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"665ac3ce1ea23a3270db32233180d96dbc99e069","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 15:02:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0876df003bf8397068225ba78bca4d4fb81b69e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:45:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cddfce098824089a726b96b736635de\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90df4ae7fd264db0adb9e6785d648a42 : SUCCESS in 37m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba421d0e652d462db5a50d5c50de5b22 : SUCCESS in 36m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/728c80601a664173bfcb3f37e1d68b70 : SUCCESS in 36m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f22e54d64434bdea11baffa00202d42 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c230dcc80d3b6da3f40ddd301defab393b458df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:07:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a3ac016ebe489e899553f0342f4689\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ac9b4e2b0a4cfc9e3818a20c698554 : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/702a8cb6d1d14b0eb46d5ff5cd22c4ce : SUCCESS in 46m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/371403ff55db453f9e5a3f47eb0ee725 : SUCCESS in 40m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cc537d059544c66a45c2918bb08d135 : SUCCESS in 49m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d5ef28d9cc84b91ab7cec9c841539bf : FAILURE in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c62e2af615b4741b0e65552e2201536 : FAILURE in 22m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c09f50f0119424ba14c5fb231f1fdab : FAILURE in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c888b62c37ce41dabbbe13c9eca9c1a0 : SUCCESS in 8m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac475148e08b49eb8e80c778382a33fd : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33b7e1ac7fd543f7907d12bf67b85c2a : SUCCESS in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/182957ca26e64e7981a695be344f022f : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39d65e262cd14c92a7f1283c1b348812 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e25b5179685d4910bdd1d6134b24a212 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e2753f4946545b886b1479029ce4a56 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7bc2ac000c24ee88fc3a510687c15fe : SUCCESS in 21m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/18af1a5aadc74eab8822ceaf9f6c54b7 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b781e90bdf0d4f98a9c902c32b531dc8 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5a6a339f96634a53a4c8c485138e5525 : SUCCESS in 27m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/16f5eecf70794b7fadbb166888a4a35a : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc02399b439a4a06890e0869bb9be99c : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5a8915da38e44f00a0cb283f609756eb : SUCCESS in 8m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/439be1c5d56748999fc6adc66db937e7 : SUCCESS in 55m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ae1b26017a447b88480f1fdf8cdd090 : SUCCESS in 54m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee5ae9161e314bb6ab8d827073861d00 : SUCCESS in 40m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6efa87990eb482fb64f3b584fa79fa8 : SUCCESS in 47m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cd217cf08734d6b955533cfe53695a7 : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277efc304568455194ac08b139d04c13 : SUCCESS in 8m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d06188a471754e0bba81828464a4f5e2 : FAILURE in 18m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70196d03e7ef402a9e122b293a3e3525 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d5536786ed17693b1b123263e44bc2d461c8119f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-11-17 00:00:18.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8019a5c0d9a53d5164cb79e5ec0eaad8d4ab64c0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-18 16:49:32.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cc85f8d4043163539c341949a7f95f5624f7de84","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-18 17:43:36.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6a78425cb32cac53928efc206c8a636c10eefac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-21 10:52:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1f3a3a68bd5db8e8f560cbf48cfcd5b43f855aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 11:28:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f167f7683b354caea26bca59ebeef8d4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/915963d12a5949ab8061c51fef8ee897 : SUCCESS in 28m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8d1dfccca2943d58c727bcfe1424d52 : SUCCESS in 28m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4771b92eaa145c79b6d883c63a5b5f0 : SUCCESS in 28m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/358757c72a7342f78508062a7953d07c : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":5},{"id":"8b687ddefbd173099216ddbd3ba7ebb971550838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 12:09:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c1847467b945d180f4ef95f0aa13ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d209213b83da4b70a1670877e8ac02c4 : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/c16bda7300dd4b99b854dc53d102a0e6 : SUCCESS in 1h 00m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b20a4fe8d1c04681a2271aeeac584c36 : SUCCESS in 40m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dc1885dc3404442be48824c271b69f7 : SUCCESS in 50m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a42d9bd3f897495ea2ab3d4827953076 : FAILURE in 22m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6f5742c68584e4197f96729eb441bc3 : FAILURE in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19cb381a0b9c4cc3a7a37133dee49fa1 : FAILURE in 23m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de29c173ce79464d9806e2c95832b28a : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad3dd0e5c3934e5bb14d6462b46cf3fd : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb4de039f81143b4b3939b75b0457d26 : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/694b37b4f5834c7db13a783ac0dd1208 : SUCCESS in 19m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dcdd21665b2d491dbf5f4afabd39cca3 : SUCCESS in 18m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f81f1231093940e7a5abc2e4d82b0293 : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0109659f2d3743c6bd3e9c23cc2db980 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/713a524e9d7945b39c52a9cd8ae29fba : SUCCESS in 21m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/463b82fa16124db0ae7f96737d540deb : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e36a115725e4cd299263301366d4712 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/75129d5086684c329cca24a0540a37c6 : SUCCESS in 18m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1336e299cda4ac69ff3b2ab79316f31 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ad3346be14c4babbdb4aa27a59845ab : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d12ea38de8f541faa4327d1893a3ca4e : SUCCESS in 6m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f0dec8ea9d8549a1b939c241b82ba205 : SUCCESS in 55m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b76fb86f0f7b47518a48063269eeb35a : SUCCESS in 50m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7eae816899fb47e8ba6ef9edc3671f21 : SUCCESS in 42m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29809159dc614d66b7d20a1b4f01b8f5 : SUCCESS in 45m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/564f0885b47e4b6b8d9b5663cd5fdfaf : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e7e9d819ce4f78aaa213ae3d7b3da8 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c536b892ea94e7b967ceb6a97a603dc : FAILURE in 25m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/834efe01bebc4545b20f8006e61925dc : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":5},{"id":"38bb274d1330f53df0d1e0881049a8cf208776b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-21 16:50:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"304255c4a7e94c881a16fab200a7f21454640a48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 16:51:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d454546697160c018a3ea056cf15ce1ab870956b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 18:58:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a01d93d9c74423927d2c1be78fc8fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdce693b4a4d4d948beeadad85f4eb3e : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/1212d09104b0495ca0daba7f98ea7b66 : SUCCESS in 50m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8064ed10ba934e53be5d754e23f95329 : SUCCESS in 52m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cedd34ff7354a9295f6f0aa7b2b843a : SUCCESS in 44m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5f8c728dc22a42d99d15f5a96013e06d : SUCCESS in 9m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1a4c0ce326684c1f9ea77cd1325d0528 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/834a2a2f1f5b48deaa5c2bf149ee0e73 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc5623b659d7460994db447397bb9246 : SUCCESS in 16m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3fd594f469274b31a589122772df0ec0 : SUCCESS in 18m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f67ceab6d134021a882129b4db7a115 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f56f41fee274b25a19e90253684fb44 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30011848c8e84b0c94f26fe1c8c4ee9d : SUCCESS in 17m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/070b36c711344a568a811fcf7adedc34 : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/be0f5e9fb683464d88ec516c1dc2a35b : SUCCESS in 12m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a9805718451045ea8c172c97f2dc5ced : SUCCESS in 13m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93882ccc80144386892d492f84d1dd73 : SUCCESS in 6m 57s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c47ebb9d09b41a1b209c76d3012a96c : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c32d0e5f485140c58ecc95379bbf2f49 : SUCCESS in 52m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee0669769da24304ad8b3f14ef813d1f : SUCCESS in 54m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9b51d09259684bedb65a02535b9b9f31 : SUCCESS in 42m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/523f5feb8fde445898f29b1cd28b7b04 : SUCCESS in 49m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8a6ed281ab54dcf80a676968f949849 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a6fea17654d409d8e131c31e55b1372 : SUCCESS in 4m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f0227d98f2c4feeae1eea62a27191c8 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":5},{"id":"53c580580acc9a17ae24a64c832643ea3d523db5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 18:58:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5cb939f2b715ae24f06afbedc010d055a5cec433","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 19:01:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c8f7c491d844f159e501d8ff7fd3aff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0666ff7128094eff93aa1abb5b88f7b9 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8ab1d41516f64bb9bee7cc7355656b95 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e3320e44fd3249f3b797e2757b9b97de : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7a0118fca0d5a348e940283e201c7657ca553217","revisions":{"e85da32969c808dd758278014cc04ba3bfd9dea2":{"kind":"REWORK","_number":1,"created":"2022-11-10 13:01:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/864199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/864199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/864199/1"}}},"commit":{"parents":[{"commit":"704c0254f594f42a474268b7a9b22efb2d43ad23","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/704c0254f594f42a474268b7a9b22efb2d43ad23"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:00:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:00:08.000000000","tz":0},"subject":"proxy-server: include suppression time in error limit log","message":"proxy-server: include suppression time in error limit log\n\nAlso, add some more test coverage and tidy up some tests.\n\nChange-Id: Icf3a1df87d52dc5f95e15e676da229a3323326a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e85da32969c808dd758278014cc04ba3bfd9dea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e85da32969c808dd758278014cc04ba3bfd9dea2"}]},"branch":"refs/heads/master"},"3419b7b5c606e0747fc9ed179c8ac30fcfc50aea":{"kind":"REWORK","_number":2,"created":"2022-11-10 13:05:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/864199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/864199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/864199/2"}}},"commit":{"parents":[{"commit":"704c0254f594f42a474268b7a9b22efb2d43ad23","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/704c0254f594f42a474268b7a9b22efb2d43ad23"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:00:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:05:28.000000000","tz":0},"subject":"proxy-server: include suppression time in error limit log","message":"proxy-server: include suppression time in error limit log\n\nAlso, add some more test coverage and tidy up some tests.\n\nChange-Id: Icf3a1df87d52dc5f95e15e676da229a3323326a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3419b7b5c606e0747fc9ed179c8ac30fcfc50aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3419b7b5c606e0747fc9ed179c8ac30fcfc50aea"}]},"branch":"refs/heads/master"},"d9cfd3847d8ba0a056d0e42bc035c99f8e9e3f0a":{"kind":"REWORK","_number":3,"created":"2022-11-10 15:31:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/864199/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/864199/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/864199/3"}}},"commit":{"parents":[{"commit":"704c0254f594f42a474268b7a9b22efb2d43ad23","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/704c0254f594f42a474268b7a9b22efb2d43ad23"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:00:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 15:31:24.000000000","tz":0},"subject":"proxy-server: include suppression time in error limit log","message":"proxy-server: include suppression time in error limit log\n\nAlso, add some more test coverage and tidy up some tests.\n\nChange-Id: Icf3a1df87d52dc5f95e15e676da229a3323326a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9cfd3847d8ba0a056d0e42bc035c99f8e9e3f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9cfd3847d8ba0a056d0e42bc035c99f8e9e3f0a"}]},"branch":"refs/heads/master"},"10829bc9642048dd5a9e0ce0acecd3055c2edaf2":{"kind":"REWORK","_number":4,"created":"2022-11-16 15:02:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/864199/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/864199/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/864199/4"}}},"commit":{"parents":[{"commit":"c82c73122b4ceeca1740989bc213d7ea9ec6fdda","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c73122b4ceeca1740989bc213d7ea9ec6fdda"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:00:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 15:02:10.000000000","tz":0},"subject":"proxy-server: include suppression time in error limit log","message":"proxy-server: include suppression time in error limit log\n\nAlso, add some more test coverage and tidy up some tests.\n\nChange-Id: Icf3a1df87d52dc5f95e15e676da229a3323326a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10829bc9642048dd5a9e0ce0acecd3055c2edaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10829bc9642048dd5a9e0ce0acecd3055c2edaf2"}]},"branch":"refs/heads/master"},"7a0118fca0d5a348e940283e201c7657ca553217":{"kind":"REWORK","_number":5,"created":"2022-11-21 10:52:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/864199/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/864199/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/864199/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/864199/5"}}},"commit":{"parents":[{"commit":"c82c73122b4ceeca1740989bc213d7ea9ec6fdda","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c73122b4ceeca1740989bc213d7ea9ec6fdda"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-10 13:00:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-21 10:52:11.000000000","tz":0},"subject":"proxy-server: include suppression time in error limit log","message":"proxy-server: include suppression time in error limit log\n\nAlso, add some more test coverage and tidy up some tests.\n\nChange-Id: Icf3a1df87d52dc5f95e15e676da229a3323326a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0118fca0d5a348e940283e201c7657ca553217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0118fca0d5a348e940283e201c7657ca553217"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
