)]}'
{"id":"openstack%2Fswift~864494","triplet_id":"openstack%2Fswift~master~I3d9e29db1f625081747588105342bffe76424f8a","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-11-16 00:01:05.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-03 19:11:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3d9e29db1f625081747588105342bffe76424f8a","subject":"Ring: Add some initial v2 documentation","status":"ABANDONED","created":"2022-11-15 06:32:59.000000000","updated":"2023-05-03 19:11:40.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1bfb04250719bc94176593cc03a0188ca1147e1b","_number":864494,"virtual_id_number":864494,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-16 01:04:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 07:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 08:06:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 19:46:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"18171d055207bc57f118cdb5c87287059b22823c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-15 06:32:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19c19e4a3dc0d0efbe0b1de94179e5ce610223a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-15 06:36:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f886bec3aa32bb65d7cbd7ee09716c6bdb43d420","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 07:46:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2971352d43f46b9b96b781ae5cd0a40\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83c69fd9f35b4b25926c9f57023d37d9 : SUCCESS in 45m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82db7f9fbc96461d8ddd6875856b6f43 : SUCCESS in 28m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c33433636a504f5392b3be7b5f699ed8 : SUCCESS in 1h 08m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd1d0a52340c41678754a8866ff7302a : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9cafab275db1b4a4d11fc0d8b115bbf669715b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 08:06:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f910ffa49f4e460bbebc92031c736457\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7794aab0bb94a13b307e7d7eb45d1bb : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9ce9ea8cbbd34f4f990aa139b1ae4f62 : SUCCESS in 47m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1884365c1c654c8e9e4f7e315b4a9129 : SUCCESS in 39m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/011e1883244542efb9d59cacf30260bf : SUCCESS in 54m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0dc698288cd4e11ad09c6159a0c78e8 : FAILURE in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f699d4fbe3df4defb7d62ddb0e4c51e9 : FAILURE in 17m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03e525dcdc8b4da09a14cb59d28f20d0 : FAILURE in 14m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b179cdbacbac40d290ea9abd8ba0c368 : SUCCESS in 8m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/397bdee93b2d4b25877f39fd67dca170 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b42a224b27c447c88e965510044f458a : SUCCESS in 11m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06251a82d51f47efb62684b592273b72 : SUCCESS in 15m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7522a18ab8ea4ce0bb1cc676f7f955a8 : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/943ef6e1da254b33aa0e89dfcad247e4 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3cf0556e5354077876d7675e5c87d6e : SUCCESS in 19m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bdcd825f96aa46eab25ebcd054d7f44c : SUCCESS in 15m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52e4e2e8e2e44ee1ae60cf6aae18d570 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e3ee3575dfd647bb8a8680a848638327 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c2e4fd2b6f03427eae3e81d62ccd1a73 : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ff53cde6bee4b5e8b7e34cfaf391433 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f09b9e184a2447fba4236d1a01a85b82 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1c46b65b822451688d7a2710fec98e2 : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/be4ea5ad902d4e21afa296ac71b2cad2 : SUCCESS in 46m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f16c1381ca9f4d7e98ad9f642aa422f6 : SUCCESS in 56m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afdba648a31049b3bd8ac65475733d1e : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6443e4459c944e094a864445e582247 : SUCCESS in 32m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dde29d99783d49669310d8aab8e96d3e : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49c4356307d48f0aec7e781a7f39b38 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8ac6a246cc54396b207f163d2ac6b84 : FAILURE in 26m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e9156b8f7b442b69dddc72c1704bda7 : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":2},{"id":"29ea2bf9d08d2f38f9804170ed2fe30fccea9fe9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-15 19:46:33.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b7ca4b824a4325df613400fcb40f1c2ad9f54828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-16 00:00:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4cfac566ccf90397f527663160731442945aec83","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-16 00:01:05.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f1d4eb4f5f2c0f5ee1ed255df8f821673326ee65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 00:46:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05608c1fe2d341d789446d704daeef11\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c637824ed61149c9ab4d466da07119f0 : SUCCESS in 38m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80bd593a051e4e03b3ee6aaa0a1b2aea : SUCCESS in 38m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f30fcfdab52542d5976ea9102ffc59ed : SUCCESS in 38m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/64421acaff4f4574a27e4e0779091d16 : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":3},{"id":"80297995c917fbb1a122dee8cee41113f6dbf1fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 01:04:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aa40bf9e71046deaeacc20d2ad0e9b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/453c99ac520540249015c02610d52fd9 : SUCCESS in 6m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5d1d833dfab14eefb2a36c706412fc4f : SUCCESS in 57m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b3d23a2cf9f421ea8b0bb96938336cd : SUCCESS in 32m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/799c7e4a94fa470a8215abec77793dcb : SUCCESS in 52m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f46e2b9e8804b628a7be4f398cb08b2 : FAILURE in 17m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd62dc446bd34e638508e079e4b95ba1 : FAILURE in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b615e604b0bb4fbbbf16213a401e47e2 : FAILURE in 18m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0452ccbcec0432dab58ab6a999543a9 : SUCCESS in 7m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f439588ea984aa6bd678c6c8a25acfc : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc97410ff4e54f8eb24f4da13a2cf82d : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0835141edd91432fa6a74cdafd34bb51 : SUCCESS in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ac84defda8e43cc8ef78f8db1898691 : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3135b0c985c4232956076d0d1dcff10 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ed3967e3fb0489daba718d359111b5e : SUCCESS in 16m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/82388cf837664e46a79f4c746187bc26 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2147f3c95fce4c68a6ef90d215a970df : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8651221aefc64c14b5f2f976a8609f96 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa853f4c72e6473da37a78cc2a771872 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1670a2c5a5b4f67bc6ca5d28ea0e2a1 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3cb0706f53344d578e7545286acccfec : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cbf5effbfd2486786ef59fd192832eb : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0b6138ed264b48ba87974bb21c4b2315 : SUCCESS in 47m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab453bd7a56e40769ba9e8760801bc44 : SUCCESS in 57m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9ee301962e74bd5aef320c7f26145b4 : SUCCESS in 42m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfe26ea12e4a4cbf95717ccd4476f2a4 : SUCCESS in 43m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa92bb0f0c704090acd80aded59a88ed : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a8809bfb334ecd97aba1bb5bd679a8 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ddf1d6bef464d96bed24778b0af592b : FAILURE in 15m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfd0ff1d0503486d96126c086051b5a2 : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":3},{"id":"1bfb04250719bc94176593cc03a0188ca1147e1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-03 19:11:40.000000000","message":"Abandoned\n\nSquashed into the two parents -- thanks, Matt!","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"426ec02ed67c669a8d3e09c8ca141301412d81eb","revisions":{"aad82950403d5c5f3f83268bc014740f86688cef":{"kind":"REWORK","_number":1,"created":"2022-11-15 06:32:59.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/94/864494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/864494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/864494/1"}}},"commit":{"parents":[{"commit":"16d00bd570b708befc8886b008d29fff58d8efcf","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d00bd570b708befc8886b008d29fff58d8efcf"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-15 06:32:02.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-15 06:32:02.000000000","tz":660},"subject":"Ring: Add some initial v2 documentation","message":"Ring: Add some initial v2 documentation\n\nChange-Id: I3d9e29db1f625081747588105342bffe76424f8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aad82950403d5c5f3f83268bc014740f86688cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aad82950403d5c5f3f83268bc014740f86688cef"}]},"branch":"refs/heads/master"},"3fb719739576f4886582331d0025a53e0cc0a0e1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-15 06:36:31.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/94/864494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/864494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/864494/2"}}},"commit":{"parents":[{"commit":"16d00bd570b708befc8886b008d29fff58d8efcf","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d00bd570b708befc8886b008d29fff58d8efcf"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-15 06:32:02.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-15 06:36:08.000000000","tz":660},"subject":"Ring: Add some initial v2 documentation","message":"Ring: Add some initial v2 documentation\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I3d9e29db1f625081747588105342bffe76424f8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fb719739576f4886582331d0025a53e0cc0a0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fb719739576f4886582331d0025a53e0cc0a0e1"}]},"branch":"refs/heads/master"},"426ec02ed67c669a8d3e09c8ca141301412d81eb":{"kind":"REWORK","_number":3,"created":"2022-11-16 00:00:48.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/94/864494/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/864494/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/864494/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/864494/3"}}},"commit":{"parents":[{"commit":"16d00bd570b708befc8886b008d29fff58d8efcf","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d00bd570b708befc8886b008d29fff58d8efcf"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-15 06:32:02.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-15 23:59:53.000000000","tz":660},"subject":"Ring: Add some initial v2 documentation","message":"Ring: Add some initial v2 documentation\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I3d9e29db1f625081747588105342bffe76424f8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/426ec02ed67c669a8d3e09c8ca141301412d81eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/426ec02ed67c669a8d3e09c8ca141301412d81eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
