)]}'
{"id":"openstack%2Fswift~864775","triplet_id":"openstack%2Fswift~master~Idc3689755b635a8492dff9c1e7f3ddf7eb3946d9","project":"openstack/swift","branch":"master","topic":"p-account-auditor-log","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-18 17:56:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idc3689755b635a8492dff9c1e7f3ddf7eb3946d9","subject":"db_auditor: remove logging translation","status":"MERGED","created":"2022-11-16 18:33:22.000000000","updated":"2022-11-18 17:57:36.000000000","submitted":"2022-11-18 17:56:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864775-p-account-auditor-log","meta_rev_id":"5d572c98b32845520339aa32b713629c40beec7d","_number":864775,"virtual_id_number":864775,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-18 17:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-11-18 16:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-11-18 16:41:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 19:27:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-16 19:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-18 16:41:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ea6d021f190c9d736057135f606cf3ebc3ae8e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 18:33:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cda44ca3316fb6b5b7107bb5f86208f78757259","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:27:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9611859db07a45c59f7a9249a7b2b273\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20ae542f1f2d484d9609a8ace11fb44d : SUCCESS in 48m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd390c6dcba246cf9ef6bbd4a9e67bd0 : SUCCESS in 44m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eecf4af8a5b4ef58f3aa70a0d0aaa4b : SUCCESS in 45m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/436523ecc10f4f97914bdab29105f795 : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":1},{"id":"07c0752bde1fa2239a01e414e6da291f2ce391d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:37:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d608ea1eabf4ef59d0903711918467f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f0c9282fca4fd289a4dfc9fb73183d : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/66d5f67f4c99464388d370494d5410fe : SUCCESS in 47m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d203a81f0edb4c59b7242fbd847a3711 : SUCCESS in 40m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e074a78ffc34b7e9caeb0a6949c3a3d : SUCCESS in 45m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fcde592c0714a459a13fafaaf418eb9 : FAILURE in 16m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7612ef9155e2426abeeb576ce4f993c8 : FAILURE in 15m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eda6961d628a4eaf95b5e91b06e58210 : FAILURE in 20m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9791ebe566a546bfbc5174ca6e7c90a2 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7269249aaca245c0a1327c803f5e189e : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2aa1c8a325354d4082870be1333b07f3 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e6bfed938e54317b158b24dad6b54fb : SUCCESS in 16m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/896abb9ba9e74c46806f82cccfeba38a : SUCCESS in 11m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd6a208291e74cf98897e78f42e995d9 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbc4bb65c07a442cba1769e6037d6099 : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f65c836e20b4b478c161c8979c9e981 : SUCCESS in 15m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/94882c604c0742c5b363a7b99dc46e80 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6d046644a6094de9ae41efefda1dedb1 : SUCCESS in 13m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/47ff6ee5ac814206ba9c504a6ea52d25 : SUCCESS in 19m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2be128c8cff14d8c8f3af0d86eb727f0 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b608f592984648888bdd9719a069a5d0 : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/55dd1e573fad4a8c91ca21bbc811c79d : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b3ce13681c74893896bc2ce350a2898 : SUCCESS in 46m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7317752b39142ce90d43d2f47417eed : SUCCESS in 1h 02m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3a9c6291e3e4ac28e09e25c8c689d31 : SUCCESS in 40m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ae2786e49bb44478c493bc24398905b : SUCCESS in 39m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/221d9f5edad74667891486a85ad6e30e : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82aafc65a034971bedcfd7d08784158 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6965b2a33d7345b8a129ae9545d2bd5f : FAILURE in 18m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da59f28d6e6493b957f182290c143a7 : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"331bf0c2088c8fab3f0df9c7fd91969ccaca3371","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-18 16:41:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"838a7a1579725cba11cfe26c455c8d16734e5406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 16:42:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c8666cb89a5ce2117e600da91e8a51206a533dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 17:56:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d070ab0b2714d29bd0856e661be6ea4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1de8d8d053942668feb02c1ec4a7cf6 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/d454c3c221f34d24852a1acf96430b91 : SUCCESS in 54m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53eb652f336d47c9a92dbd5f984d440f : SUCCESS in 43m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0eb10b976cc4556ab0548b133a81f59 : SUCCESS in 51m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f16f578c5de24359966ac122267e3769 : SUCCESS in 9m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4da843ec4f79428e87d97e0d1128dd4d : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28a006d1153241498ba9e3ae35289812 : SUCCESS in 17m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/704fc63d0ca942239ee933055a61b41d : SUCCESS in 16m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4909a4aac31f4e9aa131040031f0a453 : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/897c5afe27b94570a34f6d3a058762b1 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/509bf10dab984e4291303a637281af85 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8bb0c55e50394b019b3c654117716f11 : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7c6fecc636c04cf2bbdae60d6200e8fb : SUCCESS in 21m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f8c18dd625934f4cb0c193e6f811cf21 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c56b08f0762c4560884e44cf9a79ad28 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/208391f3b79b4142a26b97913b0ab25e : SUCCESS in 6m 53s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a980d476dab41bcbb1a106fe8ce488d : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b195f570dc3e43dab50eb7dfdbde31dc : SUCCESS in 53m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/260404c7b5204c6eb66aa394e263eed5 : SUCCESS in 1h 07m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1831af4162c49b78e27e14d736a33b5 : SUCCESS in 45m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca856ac8093241ff829c9a8b0bc625fc : SUCCESS in 47m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010ec3f11cb047f59ffae0fa5f177479 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652f71d1e8c84a8aa3a3a8eaf3bada20 : SUCCESS in 5m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09fae36567d140259fef5cd18a1dc183 : SUCCESS in 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f309f63915e005bb679a0ebf35cd1c2cbc4d9b19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 17:56:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d572c98b32845520339aa32b713629c40beec7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 17:57:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273af883e9bd443a81ca4d6d79e0bb9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19fcc3db08ca466884655e9d7f92cc59 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fabb9b272d684a5da9410c1b4f003431 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0faf9114862e4a689ad8087cb5243eb8 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26c86c0e4103a6bb9d33b47c6a566d5e8230f9bd","revisions":{"26c86c0e4103a6bb9d33b47c6a566d5e8230f9bd":{"kind":"REWORK","_number":1,"created":"2022-11-16 18:33:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/864775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/864775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/864775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/864775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/864775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/864775/1"}}},"commit":{"parents":[{"commit":"c82c73122b4ceeca1740989bc213d7ea9ec6fdda","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c73122b4ceeca1740989bc213d7ea9ec6fdda"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 18:19:51.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 18:26:25.000000000","tz":0},"subject":"db_auditor: remove logging translation","message":"db_auditor: remove logging translation\n\nChange-Id: Idc3689755b635a8492dff9c1e7f3ddf7eb3946d9\nPartial-Bug: #1674543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c86c0e4103a6bb9d33b47c6a566d5e8230f9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c86c0e4103a6bb9d33b47c6a566d5e8230f9bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[]}
