)]}'
{"id":"openstack%2Fswift~864776","triplet_id":"openstack%2Fswift~master~I6e7058eb9f5ed4298b5055d46dafb66d24ec5d09","project":"openstack/swift","branch":"master","topic":"p-account-auditor-log","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-18 18:51:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e7058eb9f5ed4298b5055d46dafb66d24ec5d09","subject":"account_auditor: fix warning string","status":"MERGED","created":"2022-11-16 18:33:22.000000000","updated":"2022-11-18 18:53:08.000000000","submitted":"2022-11-18 18:51:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"864776-p-account-auditor-log","meta_rev_id":"8a7908aac47d08fa8c3220b3089ff219307e996b","_number":864776,"virtual_id_number":864776,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-18 18:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-11-18 16:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-11-18 16:42:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 19:14:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-16 19:33:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-18 16:42:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"9377f1ee535199f2bd8f2193d059419c4f0aa5d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-16 18:33:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbc31fbb3b0966fbeaefff798f28a96e2014178f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:14:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7af5bcb7584a14b37993eb597c4c98\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cd83b48605f49fbb0170ff723bd1c39 : SUCCESS in 34m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/107508386cc9486fa66b27d2a3a2d26d : SUCCESS in 33m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f812e375b82a4f96a1c1e8ae6efa0402 : SUCCESS in 34m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d81692c3e7fc48159e7fd2f8d23ce668 : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2dcfcf06ab9aa1fea4bb1ea8286b641842d1e0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:33:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fde4cdceb0bf4d2bb4ff93df5758a068\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393de0c2ef5e4fba997762f0ed28c5ff : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/ca7877d803604de68dbdeb4fed372921 : SUCCESS in 45m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3f3f6cb8758c43e6924da294a3b7f23e : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05c708190f8e45f9b0a3426ef2bbd840 : SUCCESS in 54m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32036aedb46f494fadb9555ba516c09e : FAILURE in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59a15b778630478198846676d71afce0 : FAILURE in 17m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dab998ea79040a6a78a95c5ef5d71e7 : FAILURE in 17m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/957efb34d12c4a0fba20d69f3a8c6580 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/465b3f6fb083451f818cf2c610ad8924 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/746203d2472246169c89a5a15d08376e : SUCCESS in 14m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a400f132f23d4b12a75c8c8aa2603dd3 : SUCCESS in 13m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9bcdb121de142c1b79c5ee87397b6af : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8af6089258e14b4a81c9cd884b73b49d : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5d57773a8a343d29980bc69cb08c404 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b10e4a58d22947fdab65b24fac8b8cc0 : SUCCESS in 18m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ddd89ede7f3d431e90715f0d1e787dcb : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c422a24f0ecf4759845786590ae9e771 : SUCCESS in 13m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/580acad5cf904cf49e6eab3538de5ad1 : SUCCESS in 14m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c7b3bc446374ee5b7bef814ee7d93a5 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/436fe60a5da34ffab9f3604128a1e543 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db818b1094f64ff7a747c032fe38ea00 : SUCCESS in 6m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5ec1c1542f5142399c05e4217840bda1 : SUCCESS in 54m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee59876c851841c2a43e7863a6a300e8 : SUCCESS in 53m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59aa069e48c94faaa06be01fee51ea22 : SUCCESS in 40m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7698fee26e6f40528c1bf5f8fb6b56c6 : SUCCESS in 45m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f14076a3b5f04e1db31f28d6a5b68326 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54283686d0fb4994b5525ce650eec35b : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6402ef8724cb4c3d9d23412f2cd46518 : FAILURE in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b98c0f988ea34d70946ec4b4ffe5d057 : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":1},{"id":"60f872facbc6fcb321a99758be8b6248b72a4c53","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-18 16:42:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"10af57ae0ef8d6e282d094aa413c4139679bdd74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 16:43:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bf75b1f9ac5d2063ae8010cc168e592f256298d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 18:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82dfd0919d844f00a7976e1e1b073714\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48614c36ce774e3ead32f19705dad860 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/785fea2dd19e45c3a86b8dbc1eb26472 : SUCCESS in 48m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/602bb01a91464c848c7e03b6640094ae : SUCCESS in 40m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39112b3bbf4a4d32a9f5df10ecb1650d : SUCCESS in 59m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f549b9fec005469c98b9a722b72b62fd : SUCCESS in 8m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/27a214e14398421ab729b9f4d83bcbb3 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7dea1683b8264298a22b0f3ebcc42dea : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/287084c83ee34255843c2cd4787cb804 : SUCCESS in 16m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4461cbbad92143a582c2f8bf416e2c16 : SUCCESS in 14m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/219a735d76a348ad94f51fd31208ce5f : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25c68418e7464d28881dd384be426799 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47a8b844c33940568483e68b846533c0 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/025abbffa311473aa6242c2410dc4d06 : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/75201827c5014bb59d4f99efe949743f : SUCCESS in 17m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6c4e751f4804ec89fdedfc82c5161d2 : SUCCESS in 12m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9298890961d45a1b213200bdc3acff9 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/735b39f8c5c74e02863c8c71da439311 : SUCCESS in 8m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6ddf21bead344a06a83bf77e19810c9d : SUCCESS in 53m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebcce5fa39314a9f8c37db014f5038d9 : SUCCESS in 56m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f0546c9baff459aa53aa80943682409 : SUCCESS in 49m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b57c932771654b049568b5cb3511ee4d : SUCCESS in 42m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd547b6f72f74aa2b69d2dc58469db3b : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e9352e87cf4f3eb91f7ed3b8051c6a : SUCCESS in 6m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/978a1287e643446daa501004df21f425 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f249875dd0c7de3e0d7af3466e11e0eca2ed2d8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 18:51:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a7908aac47d08fa8c3220b3089ff219307e996b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 18:53:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5315c5293124d89af85fd9ddaf15865\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9c344874ae4ccc8454d33278d86a67 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/56eef63aa447445fa2cc10d288a050a8 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8b15c96974e449ac8c6ae522579f548e : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"983622d51203161681e4741a5815a623a35166a6","revisions":{"983622d51203161681e4741a5815a623a35166a6":{"kind":"REWORK","_number":1,"created":"2022-11-16 18:33:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/864776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/864776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/864776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/864776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/864776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/864776/1"}}},"commit":{"parents":[{"commit":"26c86c0e4103a6bb9d33b47c6a566d5e8230f9bd","subject":"db_auditor: remove logging translation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c86c0e4103a6bb9d33b47c6a566d5e8230f9bd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 18:20:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-16 18:32:53.000000000","tz":0},"subject":"account_auditor: fix warning string","message":"account_auditor: fix warning string\n\nThe warning string should refer to account not container.\n\nAlso, add the account name to the warning string.\n\nChange-Id: I6e7058eb9f5ed4298b5055d46dafb66d24ec5d09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/983622d51203161681e4741a5815a623a35166a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/983622d51203161681e4741a5815a623a35166a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[]}
