)]}'
{"id":"openstack%2Fswift~864788","triplet_id":"openstack%2Fswift~master~I2c5b18e52da7f33b31ba386cdbd042f90b69ef97","project":"openstack/swift","branch":"master","topic":"s3api-traceback","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-11-18 16:41:15.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2022-12-05 20:03:58.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-12-05 18:59:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I2c5b18e52da7f33b31ba386cdbd042f90b69ef97","subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","status":"MERGED","created":"2022-11-16 20:40:26.000000000","updated":"2022-12-05 20:05:01.000000000","submitted":"2022-12-05 20:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"864788-s3api-traceback","meta_rev_id":"d11cc624cfbde48b536d37f28c06a079d76f7be5","_number":864788,"virtual_id_number":864788,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-05 20:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-05 17:36:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":2,"date":"2022-12-05 18:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":1,"date":"2022-12-05 18:59:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 21:12:58.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-11-16 21:12:58.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-11-16 22:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fd7c8284a435e316049d8ebfd4a33d734862fa1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-16 20:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19b1f78a88317df7c7200abdb10ee4dd962f64c7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-16 21:12:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"f3addb8164e8b5927219871295ca0b1c02860201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-16 21:14:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1887773efb9efee6cc7da935e6b8bb3325e1e76c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 22:16:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22639db0b19e42c2b6225ee1e0932da6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e52080702f43bd9279ad05ab835868 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/8ba7758547b247b59fc8b04d1e18b9ee : SUCCESS in 44m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afceee5bac2a4622878f26476cc2fec5 : FAILURE in 17m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf0681bc5d7645f7b235eeb3414ce855 : SUCCESS in 36m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07974bc49aef4bebb27ca5746c280ccb : FAILURE in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb1eb2cf52534be1bacb0d439dfe3665 : FAILURE in 17m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/938f7896d054447db8d5462b65c4e167 : FAILURE in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a81058440a834348bc0a413e7aa8694b : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5f2775cd3614ada922321cacd82d815 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fb949b6f02246899fa3fcec1e0cfc5b : SUCCESS in 23m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e3bf56eb856c43598b7fe0721b8dd44e : SUCCESS in 13m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af0c3285e6824986b02ef6e4164ab85f : SUCCESS in 14m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3dc1171565644ffcaeff037b819903c1 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c2fe6f41e724c55a0fe8cea8257ad3b : SUCCESS in 15m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/77128cc90cd349c58df5382f9b7a707c : SUCCESS in 16m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/adeaa60517604ea78338fea6275d85a7 : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4bf60b233a07489aadfc2ff9a5bc232a : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/97116185ae0c4fabb8c7026ac87f556b : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9098390211b49d68bb5349e44577cb9 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f4d27f51e2544f6a9f154afc6728f24f : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5773c1fcca342a2af74024018e17da5 : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c8bc60d14a44be6ad4a45a2eb3496a2 : SUCCESS in 45m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/174d91d127604cb29baf4e39eec93654 : SUCCESS in 54m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1845e3aa3844f50907ad275eb197054 : SUCCESS in 41m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5fa591596a54491f9985908fe1225182 : SUCCESS in 36m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04419476b0714013a98abf763d126332 : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c3d484fb1f94676b20fb5dff9199a9f : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f473439a7a5b4710ad5ad829cf7e8526 : FAILURE in 19m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9feb70cb87df40eda7796e239fc70371 : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4d8194b8a58d89d5b29cb6e0342d20e0cf3bd678","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 22:19:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29bf773449cc40fb9d720faa5202924f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b131d755f84d40fe956859398fc20faa : SUCCESS in 1h 02m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bc878e711d24958af169417b97c4e9f : SUCCESS in 54m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/90a68f48be65456cb915199657213112 : SUCCESS in 55m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1cba8f2d2134d60a5ee52c770484e43 : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":2},{"id":"68ff2ca4dadb90f84e34373adefdd72eab5b8863","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-18 16:41:15.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8ed07425cea1ae36ac19fa01125068e406aa2358","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-21 20:22:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e5824b179248f1a71bef320efc50bddfc37063c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 21:10:19.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7bb754af00745e88e6fa6d6aaed1674\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc48443c3f9a4313a2264d0e2f5a8e74 : FAILURE in 44m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e575fa0efc3847678043249966074e26 : FAILURE in 43m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6806abdcd4fa4d069efb69c9ae5964f6 : FAILURE in 42m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a94d8fffc994d7b8dcf9bcf6f9e690b : FAILURE in 45m 18s","accounts_in_message":[],"_revision_number":3},{"id":"01885e2695cc91ea70dfa5f63458b8274181aa2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 21:21:15.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/449c545cdfbe48489c629560562d51b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3597b5e8cd2945979ef5cad413fea95d : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/f8156b8c1ebf4d12a3964ca9a6929964 : SUCCESS in 43m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3047a53e174a4e63a6bd79cb79fbbbae : SUCCESS in 48m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9f28cd46e334e20aeede5ec19a93d9d : SUCCESS in 50m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42ba04da93a649ddb714d18608e00c52 : FAILURE in 15m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1861039408864e01b6c14f149169ddae : FAILURE in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81f2f055070f4d5b8ce54b1e6f730abe : FAILURE in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6a44da13a93b4a8bac3d7303c0f7a2be : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d48a5efb38e44dd0905166e318ea3290 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56605ea1415f410f916e8bf679db6e78 : FAILURE in 15m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5156031da73486f85d825c7bd394a02 : FAILURE in 15m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d9d6227ad3645d095676e09ffbf4a23 : FAILURE in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f5c6857922d0498d8ecaad333d705883 : FAILURE in 14m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6fc5489cf654fa3837f3e97fb0f1948 : FAILURE in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a692547c6cf4240be6acec07c46e9f9 : FAILURE in 19m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb4a4da634c04a6e81742e1c92c4c036 : FAILURE in 15m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c14a01c7b84a4f50bb612e24f78c0900 : FAILURE in 14m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d27cd8288e9f4a59b3e67d14a063db48 : FAILURE in 13m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2310eb410f094c7fa8a798d313388c4c : FAILURE in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b40b1615c403498aa841c72a899d7b98 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/481f80a02e2c45ef814640cb6c361958 : SUCCESS in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/28f1d1885b14483fb4f1cd17ec36bed4 : SUCCESS in 58m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b61bc46810a47b79947cbabdae3ffea : SUCCESS in 53m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64f44bc6b3324044904c02bc2ecc53e7 : FAILURE in 36m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ddf632a6f4e74ca2a98f74ae121cd2e6 : FAILURE in 36m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cc7ed3cd0ac47d6aa65b44777671729 : FAILURE in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da9ef6d085547b198b72f249b8d7cf4 : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4314cf842c2940879e39f065d192f3fa : FAILURE in 20m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50760f2ed15c40038c258671f182be35 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3138c7f3288f4a514ac756f196b868858962decb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-22 00:53:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fee1e59b791071ea725197977e0c83ca89aa1202","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 01:30:23.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee655d8a42344e17872d5e2c9fbef131\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84d8de86b356478b9465afc8a448fd33 : FAILURE in 33m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d76e7de884f74eb8b1113543c047ffd9 : FAILURE in 30m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/041eb3b125e34166b92b200ff8789a8f : FAILURE in 29m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b559fbea29f491fa2c7ad2bf5c7e0f4 : FAILURE in 28m 57s","accounts_in_message":[],"_revision_number":4},{"id":"326d373910b272da66f6b7dcbe7669f5076dddac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 02:02:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2f0fb53291541eb85583c7f63ee67d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ac3c7775d8449baf96342182f74049 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/512b769bcb63481c9581972c5c6c9e82 : SUCCESS in 49m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a72b749c97694d0a99c4a7c57403af1d : SUCCESS in 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b22c2881e5454d41887e709eec4fa304 : SUCCESS in 50m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff6cb798129b4acf913a15c41071f134 : FAILURE in 22m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/993991c90b81401eadddb86dbab2f164 : FAILURE in 21m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd0bc5c8047e4565bb750699718767bf : FAILURE in 22m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/adb9c1164a6142308cd3bac4e9061d2b : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c9213a75bc444938bb74230f908fe1d : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a93ab0cc1f94fb89f1bd642685031cd : FAILURE in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0af0c4608eab40bca402a8b988dc29cc : FAILURE in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f72b6140852d4e5bad81e1188ea5b914 : FAILURE in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/049d15c4889a41cb85d56f45b59c5074 : FAILURE in 13m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb4cf33b5e844ef4aeb055f9ed688f17 : FAILURE in 14m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5513117244f442cbad48009f86671edc : FAILURE in 15m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1291bafabb7f446eb7f39ef63b9573d9 : FAILURE in 11m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f3125cd9cb241589f6bfd8b367912e0 : FAILURE in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bc084e88936f4819aff21709aaf00ddd : FAILURE in 12m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b654df0583574b8895d23f761b9e630c : FAILURE in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab5d9514bf58410789c33e30d9263c5a : SUCCESS in 12m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83657103e392431a8ab4777ad10d211a : SUCCESS in 6m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ff0d27a205be4c6ca7f876059e0a6ad7 : SUCCESS in 41m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/47e04f9bcb6e421c9fbe83a7e45d6094 : SUCCESS in 47m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91edc1a16e994ca5bc0ecd9ceb048fbc : FAILURE in 28m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e58fbd7f52554a1db1dafc486e5f534d : FAILURE in 22m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74b7d9ae01d64467b48c609aba9510cf : FAILURE in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14b2bd457b64a52bc12f1d5e2be28c6 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01e72a0ad414445f9c53ff9fe7dde3f2 : FAILURE in 18m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8d16b4fe2f94be791ad9f819bd0d0a4 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":4},{"id":"e2270a31bd32e9eb33ccff34d11a1528f262cee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-22 21:28:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5b329cfab1352f321357ddba4865a8081d728c06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-22 21:29:30.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"57c9c83e73fa1d9ee5cc3529149a8e938229b308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 22:39:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3938a2c4d62846d1817f9eecc9de2b02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ebb6c4d46f4249b963e562937bda30 : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/375abab32bc54703b9182fea12e8139c : SUCCESS in 1h 01m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a4ba2128f884080aeb804f13235c98b : SUCCESS in 39m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a14c46a4c93e493b8ba82f515d20543e : SUCCESS in 1h 01m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c816329f49a4aacb3dbbf586703d436 : FAILURE in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9b57a18c87248169afad0ac89f4590a : FAILURE in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef1aa1bf76f441e0aa44d218286d06b2 : FAILURE in 20m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef2eaa541a374c85bf407bf62030300c : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/404ff8557cfb4aedaebb8327212f54ed : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bb0d089181ca44a0bdfa0eeddef565b0 : SUCCESS in 21m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c576533197bd440381ff70685d571e40 : SUCCESS in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ee062f1995a9453d8c9b0e73c93321c7 : SUCCESS in 14m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92a30f25ad8546ac8526f7efb7b9c75f : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88737a9891dd4f8da85be512f8f96fca : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76f7ee2f223d418f937f3e8702318737 : SUCCESS in 20m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4b373776123f4c4187193cede6c07c6c : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/11662e0ea92047ea9b36f74aa71b505c : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/84eaf5fb542844b28dc99ebb0e4a7c18 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/40d8c57c7f394e7d838b8cf970dc1b86 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ff04edba8da47ebaafce428fe5cacf1 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/548cc460846546f3b2e35bf0299bb202 : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96667999ef414194860ce30d353f138a : SUCCESS in 55m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a925aeed36140f2b1ff54099272c50e : SUCCESS in 1h 03m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/289408cc936e4f4f86ab24d283045a74 : SUCCESS in 48m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0dc1cb93b4ae44a79af7d773b50b823b : SUCCESS in 41m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa14f16c370484e8dd0babdbfa9c0f5 : SUCCESS in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc8ec71fec34dcd9b95336dfff25e3b : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f70c7f069624bdb8d9b44625fae3207 : FAILURE in 19m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db75d2d27fb04317818f231b7608e607 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":6},{"id":"4c530f29b41a00b2c84028a94bf4a3fd984b37d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 23:10:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d708723a210b48419d8918c17192f6dd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99bd6411558a46909ecdcaf2621ad2da : SUCCESS in 42m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01a74532f2164e00b9becdf2fb888377 : SUCCESS in 41m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/67811ab1ba494dcaa06f39aaf7e462c7 : SUCCESS in 48m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d33e04e346af485da451d5cd8072d312 : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":6},{"id":"63b79691e00f6d201fae036f2bd22995db31af04","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-28 18:31:39.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e194db91b6b23fe2e1aa5e861032adaf4aec5438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-30 01:25:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"874818c37614651a37f8efecfa2626dcb83ef2f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 02:31:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28264bd3b0f040f4916f4f0da2a1b01f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5657f726453f4045bd358a03d723e213 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/25ba7a7223184cf4ac74c9ac8618da93 : SUCCESS in 52m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2afe38668522422da425d47857d94ebb : SUCCESS in 39m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4902cee8f7824125882b3df91ab9d829 : SUCCESS in 1h 01m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d501e0a392a4d69915db277ef8876ae : FAILURE in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de40e9e7ecda44f2ae59a95d65674efa : FAILURE in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05c2876edd3944c99deb4898d8dc73ff : FAILURE in 18m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0e5c90cdfa041e4ab349a9c5688a25e : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f13f3ee8ec54bfd8fda5dc41fa06de0 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df28fb868279479fab1d651abc4ee34f : FAILURE in 16m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5367ee40edc04d3cb9dd5ee9ee2ed92a : FAILURE in 15m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/615e24c10ffb4ff09b4f3183e2cad211 : FAILURE in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f48794869064fa8873e82abb32332c7 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e9f2f0ad49c490e9043d9f20665d35c : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8dfb219c143e4f0f92bb204f08b30aad : SUCCESS in 15m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/92a8ec7b725b4dc498f3df62f1008fb2 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3a53fe3638f147b8b1e0c3a1165decc3 : SUCCESS in 12m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ee93c04e9324913a43645c40c6b340e : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aeb1c0815d334bd3941cf9bc5b468146 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa9ecb9aae7f4b248893af8cc14965dd : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/745c9bfd36914e97b9318e192ce760b5 : SUCCESS in 7m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f4cacd79a0864764bb0b4cf9eb197d4c : SUCCESS in 50m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b2b1452b32f48c4b25c3a53123d19ba : SUCCESS in 1h 03m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1ed034b2c7a4ccb993dc703e1e55af3 : SUCCESS in 46m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5543ea92e2a04b1fbd61677a6a384d87 : SUCCESS in 27m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf35080565944ddf83337d63de09e3bc : FAILURE in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334640a6fb5d4af8a1b0f66ebf8b3d60 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72dd063f738147178a240d01a5128c63 : FAILURE in 26m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9439d2a39dfa42bda3198be6ebe4439d : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":7},{"id":"a653a79d1ef499bad27840970f16c70f37e95840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 02:59:00.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a701d719dd284c09abc7e73156f45957\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8ffb427cb8a4385962fbe0fe918bdbc : FAILURE in 45m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37b1346b48144eff9c60b5842eb2cfbb : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/79e0c12925624e83922fc38999deedc5 : SUCCESS in 1h 04m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99a5529e7374abba2e4b018c36a5071 : FAILURE in 1h 06m 40s","accounts_in_message":[],"_revision_number":7},{"id":"d19404daacb0395c3c8b9cf1cbc52be4f9953aca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 18:45:12.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"cea427948b40ff7bb67aecf00678dee07d5ff28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-01 21:21:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"be676ed971e5e578c3bb655c8950b955dbd285a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 22:31:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb2148129ac5469ebc042847d774223d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52901538fd2d4d11a45f39f51e5bc5ea : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/67a4ec216e2740139f9c041894e7c25d : SUCCESS in 40m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ccb46f5ee2114167959e237a0d8d1576 : SUCCESS in 39m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b18321d28438488c81d0d516b73749a8 : SUCCESS in 42m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b9373457cc5472c9b48cc76af2600af : FAILURE in 15m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89deb876a835472aadcac190e7ff1433 : FAILURE in 20m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bf829c7b96b4fdab8f411ddd10ed872 : FAILURE in 19m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/171bc9040ef54e2f97f9163c82a018e5 : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad3ea4ff0469459f867d1bf44128a9e2 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7899161078354534929b6152929c5ec2 : SUCCESS in 14m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9b6e480a19d4fd293485f4ed8c5454b : SUCCESS in 15m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f20cb7b41d841ddb96f58855ddd9989 : SUCCESS in 15m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c871ff3d348b407a839afadc829ff3b4 : SUCCESS in 23m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1da346fbb9c64b5c98cb56d9e2efe040 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef978c4094004373be7df181a5ed4cef : SUCCESS in 24m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f37bef482dba4c619d1368aee4eeb904 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7fc7ef8551c74d45b6ad6fc5dcbf6b3f : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e67551770e81487cade6e1ec9477b2f1 : SUCCESS in 25m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f01f0c548828436792aa3b4a64131da5 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7b20f0b75574e63aa2b1dfd6da819e7 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0a2ad981f46b42ada66795f0c8777b17 : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f1124ac20c74095b90bf85200e12d77 : FAILURE in 50m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d2d2fab24ec4247a1fd6d5db54ced57 : SUCCESS in 1h 08m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ccf3150f5094d7f83f1947b14edc2a8 : SUCCESS in 32m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/edf5bdaed27b4d6f9824b29a6d73e63d : SUCCESS in 51m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05805a4981564d458a994153e073cebc : SUCCESS in 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca98ce2bcb1849f49e874f421e514833 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d77636d0ea1463e8d7e38774dc03e92 : FAILURE in 18m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ed460a5a90b4be18e0262e2daa4c4bb : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":8},{"id":"fd0683f3e354d5cdf3e9b6ff5f7a05021cde27af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 22:31:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02737ad045274b6fb3ff397e674a2f78\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/609de7f8e7d1449185aa2dbd876e5ef2 : SUCCESS in 43m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad9927e3a4e4cafbeec972baa194ee1 : SUCCESS in 52m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b213b0cf7ad497c8f42bbfc075536dd : SUCCESS in 58m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/289a22cbfc8842a4be99c4a10707a33b : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":8},{"id":"605d600922265b20cd89af048449af7100a73217","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-02 16:27:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3f1e7cc471361b974614aad08ed6e442a5aa9495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 17:51:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40800e072aeb4409a2590d5e3b5809ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f0542bf43e4dfcbddd6e71779af0d5 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/1a725512b59944db9010f034155f7b2c : SUCCESS in 51m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98b87b3524284f0f9b3c07536c354d41 : SUCCESS in 48m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c1f79141c104a278f36a4536f30daab : SUCCESS in 49m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4a7bc957358400b95ee5600cc999338 : FAILURE in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6194c9202f14fe78fd5cdc1677cb94a : FAILURE in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43a674cd8e0347e89fc9717cbab06b17 : FAILURE in 22m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c088fa188ac4a9685b576712d175309 : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f1a3cea9cf344d039ac0e4ef4665ca40 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a22c27d07407415e919c96172aecbefb : SUCCESS in 17m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ab79d3789984d1f95aa1f3ba6925f82 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9318715d2e4adc918c21bf2fef4a14 : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5643bf73f1604a6dbf25ed58d7c798dc : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a50ccbb4f8d43548ff93c5a2de11cfd : SUCCESS in 22m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/11465b3bb5264b5eac21e2f78ce30634 : SUCCESS in 23m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3e22137c0223454bb2775b651d02e922 : SUCCESS in 10m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d6da0bf1120418faaffe51f03256cfa : SUCCESS in 18m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/674c1a2cb20c46f68184b90cdc88654f : SUCCESS in 11m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7c20d7996a5474cb3e0adf6628925cb : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c0f5df63f1249cfb55d0a8327dbdc55 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3f4aa9e1a8624233896449ad19f37653 : SUCCESS in 9m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/307489246c404a8db6ffefc82bdcec7a : SUCCESS in 53m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e045a591e8864d588e572a014afc9585 : SUCCESS in 45m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69e14d556cd14497a3d9b644b3888605 : SUCCESS in 42m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/496244e5611244baa9fa797f6650d40b : SUCCESS in 42m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/838e8229a607408d80ba04c2451ffc1f : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cafe1f2186944e58d5a60ad1c6a583d : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34eeeeabc0914c58b2a9ce28ab10bfb6 : FAILURE in 26m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/277f366e64c046c29c60af28e2a91cb0 : SUCCESS in 42m 29s","accounts_in_message":[],"_revision_number":8},{"id":"5df6c45fd2f353df89941ccb4822d40bb2edd7b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 17:54:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ef9d35905a43ddb72b5bbd55560206\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46695dce969546ab88005893e0202684 : SUCCESS in 1h 08m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b349ee885c7b4277ac1d3eb9b129c780 : SUCCESS in 1h 10m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/47b6d99ab23e4c7db209afd80685293c : SUCCESS in 1h 24m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e2fe5814ecf484fb8597b759f779b36 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":8},{"id":"6efe8a9877062f06cc709c7481acd69dcf7f6c54","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-12-05 17:36:26.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d4d9780eaf6009c5d925a7cd7aa4f263e7004d9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-05 18:59:14.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"334933e3fbbf54e7bc0791a7568a38ab3151d124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 18:59:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dbc70366784f8af2243313afa0f15fbfc6dfe4cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 20:03:58.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8d464d94c8445ba7077f181dcffad0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c872f54086849d994c91f0fe841e6d4 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/949bbaa1cdf3438996aca9deab1734d5 : SUCCESS in 50m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e202eb0721c4400682399546bf163ec9 : SUCCESS in 33m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac7aa2c58d94afcbb486160931573b0 : SUCCESS in 48m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0e056eabb8b04798ab8e91dc35a0e2f1 : SUCCESS in 8m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1e1a39a0580041229921e33978043277 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05473bb651ac4504acd8b8f062a180a8 : SUCCESS in 13m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41e8c9a49da34008be5f2a9b6cf87b03 : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b243d795b02c4d4eace4964e85f7aaf6 : SUCCESS in 12m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1a4a0f1d0664963a4a2c5f462089a87 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb726a63c45c43ddab2105a467fb2d81 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9eab1a064834b5bb80be3cdbf683bd3 : SUCCESS in 14m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b1af4a7148374dc0a1ade47ecff0df60 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/17fbc929cf79414686e30917ee22bd11 : SUCCESS in 11m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5884862156e14ada9ba8d1f61c170d44 : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b0c29c9860f464495755cba22585382 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6dfb1801df744c44affaf4cc00024f2a : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/44f6447bff21431e9c9396697d2c6e16 : SUCCESS in 41m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/05bfa18fb91d442c90ba5d668c0ad337 : SUCCESS in 1h 03m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a923d2f334d4b57b34000cc6b86a016 : SUCCESS in 29m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2703b1306337437389e4e11b6851f200 : SUCCESS in 41m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38223f94765a4e3e97c2dff353cb047d : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2adeeef34ad460f8d0b1ba3ca02fdd3 : SUCCESS in 4m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe1c1e221382493788e9428e9efbea6a : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":8},{"id":"0b0490ea44512aecd1eb96c5f09023255bd18ec8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 20:03:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"d11cc624cfbde48b536d37f28c06a079d76f7be5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 20:05:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d538925882d644acb54cb92aab861d34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166352f6c17946dba49970a76f737d15 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5e03dcf36b914d778abdc55566495bf7 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9bd53218e27842749c2a77f11984e251 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d363236a245ed9112c19fdd5df5fff31fa24b865","revisions":{"5cb2717707bc97ee30d14436ddf37ac28e339fd8":{"kind":"REWORK","_number":1,"created":"2022-11-16 20:40:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/1"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:40:18.000000000","tz":-480},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe do not need to support the aforementioned headers in our s3api\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb2717707bc97ee30d14436ddf37ac28e339fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb2717707bc97ee30d14436ddf37ac28e339fd8"}]},"branch":"refs/heads/master"},"e1c13ead9c0fcbf890545c8e030597f8332e7c7d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-16 21:14:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/2"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2022-11-16 21:14:58.000000000","tz":0},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nA s3api PUT with a \"bad\" x-delete-at should get the 4XX response from\nswift and emit an InvalidArgumentError for the client\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1c13ead9c0fcbf890545c8e030597f8332e7c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1c13ead9c0fcbf890545c8e030597f8332e7c7d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d7f82b7b0537303e5540b6791a50fe895401c953":{"kind":"REWORK","_number":3,"created":"2022-11-21 20:22:25.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/3"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-21 20:22:18.000000000","tz":-480},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe do not need to support the aforementioned headers in our s3api\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7f82b7b0537303e5540b6791a50fe895401c953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7f82b7b0537303e5540b6791a50fe895401c953"}]},"branch":"refs/heads/master"},"c45f6edaa597910f1ac0e9dc2805e0d4e3973b3b":{"kind":"REWORK","_number":4,"created":"2022-11-22 00:53:35.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/4"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-22 00:53:28.000000000","tz":-480},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe do not need to support the aforementioned headers in our s3api\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c45f6edaa597910f1ac0e9dc2805e0d4e3973b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c45f6edaa597910f1ac0e9dc2805e0d4e3973b3b"}]},"branch":"refs/heads/master"},"f586488b576423a649ee5fe0d8df99c2c5b06f76":{"kind":"REWORK","_number":5,"created":"2022-11-22 21:28:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/5"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-22 21:28:31.000000000","tz":-480},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe do not need to support the aforementioned headers in our s3api\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f586488b576423a649ee5fe0d8df99c2c5b06f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f586488b576423a649ee5fe0d8df99c2c5b06f76"}]},"branch":"refs/heads/master"},"4a0b1751b3e867980542955930b9684f4cb449af":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-11-22 21:29:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/6"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2022-11-22 21:29:30.000000000","tz":0},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe need to support the aforementioned headers in our s3 apis\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a0b1751b3e867980542955930b9684f4cb449af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a0b1751b3e867980542955930b9684f4cb449af"}]},"branch":"refs/heads/master","description":"Edit commit message"},"316563f6d2b07bfaf8a91d6fcfedafb6aa520d21":{"kind":"REWORK","_number":7,"created":"2022-11-30 01:25:25.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/88/864788/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/7"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-30 01:25:18.000000000","tz":-480},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe need to support the aforementioned headers in our s3 apis\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/316563f6d2b07bfaf8a91d6fcfedafb6aa520d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/316563f6d2b07bfaf8a91d6fcfedafb6aa520d21"}]},"branch":"refs/heads/master"},"d363236a245ed9112c19fdd5df5fff31fa24b865":{"kind":"REWORK","_number":8,"created":"2022-12-01 21:21:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/864788/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/864788/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/864788/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/864788/8"}}},"commit":{"parents":[{"commit":"38124221d7bbce8459fd544a15fb5e57dfc7213e","subject":"Proxy: add metrics related to error limiter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38124221d7bbce8459fd544a15fb5e57dfc7213e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-16 20:03:07.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-01 21:21:01.000000000","tz":-480},"subject":"s3api errors for unsupported headers x-delete-at, x-delete-after","message":"s3api errors for unsupported headers x-delete-at, x-delete-after\n\nWe need to support the aforementioned headers in our s3 apis\nand raise an InvalidArgumentError if a s3 client makes a request\n\nChange-Id: I2c5b18e52da7f33b31ba386cdbd042f90b69ef97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d363236a245ed9112c19fdd5df5fff31fa24b865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d363236a245ed9112c19fdd5df5fff31fa24b865"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
