)]}'
{"id":"openstack%2Fswift~865016","triplet_id":"openstack%2Fswift~master~I749e8faabc8e7ecd028c43225ada2d73fdc8cd09","project":"openstack/swift","branch":"master","topic":"p-sharder-update-ranges-from-root","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-18 12:10:16.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I749e8faabc8e7ecd028c43225ada2d73fdc8cd09","subject":"s-m-s-r: use sharding epoch in shard range names","status":"NEW","created":"2022-11-18 10:59:53.000000000","updated":"2022-11-18 14:15:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8217fbf8ee6bb6502276e1f3c48f23af26669bc3","_number":865016,"virtual_id_number":865016,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-18 14:15:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 12:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"026c206072f876b3688272373d51785aad945459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-18 10:59:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d70ee7b295bbfb099866a48804e01918d7061801","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-18 11:53:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be2cb86ed9770a83d5eba6545dbe8e7d3390b982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 12:10:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cacff335ec044a888f082c2901135ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7e82823f6e4f34bb77453497d87789 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/929d8b3a8d094d11a9f54527396e1624 : SUCCESS in 52m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a0f0fc461274fd4a17bf8cc36abc16e : SUCCESS in 52m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/604c11436d8c4822a9bdd7d0de84305f : SUCCESS in 1h 00m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd61f98193184febb4668b020f533d88 : FAILURE in 16m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21d8da494cd74a82b60b9b061ed54059 : FAILURE in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9aa9126ecb95400091978e54ce13ca61 : FAILURE in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c64c652dd76e4b0099684478247a6546 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c6fb5815d6ee4fad83a9a63436426832 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/facf9e6cfdd34321b80d886db56f4def : FAILURE in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/acbdd67236014676bc7b84a52563daf1 : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c48653258ab46e1bd724a48b66fc160 : SUCCESS in 15m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e15048f756054d7a9d8df2fc64361624 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0884e2d2fb73404b996527440941b804 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/586f2ac0ff5b488f90c6a3d5772c5b81 : SUCCESS in 15m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/72d896d3ba0842809beeba0d0145f059 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/69f12110eab84df7a967fa744bc7d641 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a99227fda26f4bac83d9769e59e9cd47 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a935aae04c37441ea871da44fe96f699 : SUCCESS in 7m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4dffb2668f684688b1a30848603b0fc2 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3b036cfe2c22440aa825c895035d6576 : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cd8d2d460c8041bbb089a1c7a42d0418 : SUCCESS in 1h 00m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/94b4b6fb16c54a5884f96b353898330f : SUCCESS in 54m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98b3f87756414fb38b07abc0fd910337 : SUCCESS in 45m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3b953bf5a044077ad7a36d8e7d21dc1 : SUCCESS in 51m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4b307ebafd64682b86eab43ab066a82 : FAILURE in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2d9503102d4a4aba75d7196231c3df : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11aa6db9f657499bad7201d8c9a65e76 : FAILURE in 19m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96b4d73b940645a392ea21cb57b2e990 : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":1},{"id":"46c1f3e7a9f755f6cdc7cf8b55ad626da4938110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 12:30:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ed96849f8494afcb638b55ef1ba6f54\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ca50f72c8c64670b27672d8ccc6af69 : SUCCESS in 1h 27m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c0aa1e5afba45ed91cf0e9fb8d50be0 : SUCCESS in 1h 06m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/494ae76a46fc4908b781f35431bb5fb3 : SUCCESS in 1h 04m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcf0cb9ecba14f19813564f8ece79872 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a443f0638dc80e8f6ed04cf548ee4a591510894b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-18 13:16:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2ecd480bf0bf27e304d28c72c0cb264e312154d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 14:04:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e2872010e3f4407bdf27e4687c324bb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e75aad6566e44bd7b83396ab77b56d34 : SUCCESS in 38m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40742f5a914b4310ad19611e6447a011 : SUCCESS in 37m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8a0d450c51d4feab5d39b8234dd6f62 : SUCCESS in 38m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d00432866dd4541908d461390f4ceb3 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8217fbf8ee6bb6502276e1f3c48f23af26669bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 14:15:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/165eafe9fea9411abdc69170d97f9cf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e78af67ab3414faa172a088371c8dd : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/0169744af7a94ca98c6f56146774720d : FAILURE in 33m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcfa4863a4054a9bb192135d77970074 : SUCCESS in 41m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19dbbed1855742d0b1bd88ca1589f74a : SUCCESS in 48m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae100d6c89f64aa5899c864db64dc3c0 : FAILURE in 22m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/661f5826154049919794d47035c4a82e : FAILURE in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c4cc9813a8f4b4ca2b5a62cbb3b7739 : FAILURE in 21m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77154f40c8df456ab64d99bb2f2d745c : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/318f980b9a8240ffb76250671eeceace : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/711a4d89963647e0b7cafcd049ea56a4 : SUCCESS in 17m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57cc30631b59461f8767c314b020b01b : SUCCESS in 18m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3965d3c4d6c84300a48b0b2523312cc3 : SUCCESS in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55d91d72fb5a40548f2af35ba392022e : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5760ed0b25f64f5fa6920eedb5291679 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa29921eabeb446c9741f2c8b824378b : SUCCESS in 21m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/24786f0c16a340b88c28a42c64edceda : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dcf4e7a353c24f5d944460fdcf4ef5c9 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/433e6fd30f3d4dab9fccc489c0bb6311 : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e95f797e9ee4efba00ccb06b5db368c : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea257205dfe74474b18da167bf9155ff : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f25df24c789d49ecb1b64d75f71a142d : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7dd0e91b96e5429689974a3211c07b5c : SUCCESS in 53m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5213bc69e9c644a7922435b561a77277 : SUCCESS in 57m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf28605d26c54eac8f6f1ce7ed5cf1ba : SUCCESS in 41m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8910aa8a32bb4b8d86b203d09e551530 : SUCCESS in 41m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1cd41ad104f40d69448cf37e95aa4c7 : SUCCESS in 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92527afc976b4a979a0947c5bc15b658 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8010d0b7c3a5422f9752c653da7f7241 : FAILURE in 16m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e77179a6336546658036689801421487 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8ed89454e868fb01a32c0c35e92cb99c3daca8d","revisions":{"8f78275774ddeab4678ec019cd236b4b0304c4d1":{"kind":"REWORK","_number":1,"created":"2022-11-18 10:59:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/865016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/865016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/865016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/865016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/865016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/865016/1"}}},"commit":{"parents":[{"commit":"2bcf3d1a8eb2e465b3d7afea7a6efb3b4a7d43e8","subject":"sharder: merge shard shard_ranges from root while sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bcf3d1a8eb2e465b3d7afea7a6efb3b4a7d43e8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-17 14:31:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-17 15:16:16.000000000","tz":0},"subject":"s-m-s-r: use sharding epoch in shard range names","message":"s-m-s-r: use sharding epoch in shard range names\n\nThe \u0027swift-manage-shard-ranges find_and_replace --enable\u0027 command\ncreates shard ranges, merges them and also creates an own_shard_range\nfor the sharding container. Previously the timestamp used when\ncreating shard ranges was close to, but not exactly equal to, the\ntimestamp used as an epoch in the container\u0027s own_shard_range.\n\nWith this patch, the same timestamp is used for the shard ranges\u0027\ntimestamp, for the timestamp component of the shard ranges\u0027 names, and\nfor the epoch for the sharding container and therefore the timestamp\ncomponent of its fresh DB\u0027s filename.\n\nThis change enables the ancestry of shard ranges to be verified across\na generational gap, i.e. given shard ranges X and Z we can establish\nif X is a grandchild of Z without having knowledge of the parent Y:\n\n  * the name of each of Z\u0027s children, Yi, can now be calculated for all\n    indexes i using Z\u0027s container name and epoch.\n\n  * for X to be a child of Yi, the parent_container_hash part of X\u0027s\n    name must match the hash of Yi\u0027s container.\n\nChange-Id: I749e8faabc8e7ecd028c43225ada2d73fdc8cd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f78275774ddeab4678ec019cd236b4b0304c4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f78275774ddeab4678ec019cd236b4b0304c4d1"}]},"branch":"refs/heads/master"},"c8ed89454e868fb01a32c0c35e92cb99c3daca8d":{"kind":"REWORK","_number":2,"created":"2022-11-18 13:16:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/865016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/865016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/865016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/865016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/865016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/865016/2"}}},"commit":{"parents":[{"commit":"2bcf3d1a8eb2e465b3d7afea7a6efb3b4a7d43e8","subject":"sharder: merge shard shard_ranges from root while sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bcf3d1a8eb2e465b3d7afea7a6efb3b4a7d43e8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-17 14:31:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-18 13:15:35.000000000","tz":0},"subject":"s-m-s-r: use sharding epoch in shard range names","message":"s-m-s-r: use sharding epoch in shard range names\n\nThe \u0027swift-manage-shard-ranges find_and_replace --enable\u0027 command\ncreates shard ranges, merges them and also creates an own_shard_range\nfor the sharding container. Previously the timestamp used when\ncreating shard ranges was close to, but not exactly equal to, the\ntimestamp used as an epoch in the container\u0027s own_shard_range.\n\nWith this patch, the same timestamp is used for the shard ranges\u0027\ntimestamp, for the timestamp component of the shard ranges\u0027 names, and\nfor the epoch for the sharding container and therefore the timestamp\ncomponent of its fresh DB\u0027s filename.\n\nThis change enables the ancestry of shard ranges to be verified across\na generational gap, i.e. given shard ranges X and Z we can establish\nif X is a grandchild of Z without having knowledge of the parent Y:\n\n  * the name of each of Z\u0027s children, Yi, can now be calculated for all\n    indexes i using Z\u0027s container name and epoch.\n\n  * for X to be a child of Yi, the parent_container_hash part of X\u0027s\n    name must match the hash of Yi\u0027s container.\n\nChange-Id: I749e8faabc8e7ecd028c43225ada2d73fdc8cd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8ed89454e868fb01a32c0c35e92cb99c3daca8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8ed89454e868fb01a32c0c35e92cb99c3daca8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
