)]}'
{"id":"openstack%2Fswift~865066","triplet_id":"openstack%2Fswift~master~I75a92ef258d577659eb4716a20eeeef2fc4701a6","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-06-11 22:50:59.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-05 19:32:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I75a92ef258d577659eb4716a20eeeef2fc4701a6","subject":"WIP: Rings as zipfiles","status":"ABANDONED","created":"2022-11-18 23:59:20.000000000","updated":"2025-06-11 22:50:59.000000000","total_comment_count":13,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"4b134d1e266fe7b763d6aa06dcc53c17d34a4605","_number":865066,"virtual_id_number":865066,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-23 02:17:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-19 01:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-21 15:57:34.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"1587a82971e4554f07ec7cdb2c26fd4ee42c26a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-18 23:59:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be0392b5b02e0d5370467ec41abc798a1762ba13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 01:07:19.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/843ee7e0664547c683eda05910834841\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af39d870a7fe4b3485b54fce2ffb87c6 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/35819479334b4c5db292011a4293a189 : SUCCESS in 43m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/146680aba09f412184018a0b2fc5dd01 : SUCCESS in 36m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/905db033fea94073a4055f9b32028955 : SUCCESS in 40m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1c40abfef644e2998c75865fbe835dd : FAILURE in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3dd607d8b534518a778ae20a8c0fee9 : FAILURE in 20m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d919bf61618649c8bf2b049cb3d76730 : FAILURE in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f46a39e4f6748f18119a47929a4d1b8 : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bf7243f3b3e94a9290bff644a538b0a1 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f898c83982af4bcda0ab94ff44970a59 : FAILURE in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bdaa94eaa54aa3bed2771cca1b6958 : FAILURE in 18m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53e601424c5045fdaffd3865eade46b9 : FAILURE in 19m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e527809500a245c2bbd90451586aa8d1 : FAILURE in 3m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/565fc56dd16247219b7f5d994981545b : FAILURE in 4m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ceff9a3c5f6e49d6bd5b80c348d76ba1 : FAILURE in 3m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/60f71a900abf41b3bdfcc61e23795fab : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2702fad4f6b14c38b263e1ab7955d329 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5cc9eaa2f9db4f71b5cb691b6326931f : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69446ddc96c7416e971d6dec4811b504 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b87cb7273ff14350a531fb4e81ecc16d : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/943c526dcb1e4311b64ae5e68633e215 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/688ca2f6606247c383be6a8d44765766 : RETRY_LIMIT in 3m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/533ff887715949859aa0b832da662f7b : SUCCESS in 1h 01m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1189f1db62ae4f63aef803b55a9d3a4f : SUCCESS in 43m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/817f790bfbd2496f9f942e068259947a : SUCCESS in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7392626f0604624834a8b4e61e6c8f3 : FAILURE in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e8ae7262054a43b00f7457c5707881 : FAILURE in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee19e5ad16f84266befb002a0762c1d6 : FAILURE in 24m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65763b0407d24e5697d1aa5468edaaec : SUCCESS in 1h 01m 56s\n\nWarning:\n  Comments left for invalid file test/unit/common/ring/test_builder.py","accounts_in_message":[],"_revision_number":1},{"id":"95c0a58751557a9275a4c6261fe80e13538c2568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 01:07:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1755a14ef549258ff9aaa185877377\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/614f0e2dc4364cbbbc6b842441a6b054 : FAILURE in 40m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5a9e0c39d694aa5847af68d95e5eea7 : FAILURE in 38m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd700bfeb0b94924b4a9b27ad4545bfd : SUCCESS in 1h 06m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/08f4ba288f1c48f5990266fbe93ded30 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8c615845acdef599809342ff4ec72eade4984882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-21 05:48:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6560687d05ed7f70e3a51051c864497dd999e6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 06:50:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20db6cd826ae4b0082278e1ba56cb29f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f2a96683704086ad2f5584eca46e49 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/eb932aebf91242cd92e7886d4869ab39 : SUCCESS in 50m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d50013f68ea4bc2b1a9b8a25f2b0613 : SUCCESS in 54m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5fffade71d542dfa405159acbfa0b5d : SUCCESS in 49m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57adffa8847f486b833c85544cdf7237 : FAILURE in 19m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e3c01c48ba9418da88de232eb32121f : FAILURE in 18m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7544176c48074ca5820eafbeed9a1807 : FAILURE in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca9a7536801849e189cc6b3be3e93797 : SUCCESS in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b820b262a2c841648de2668d341fec67 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c1e2b39e8da44b46bb5e0f530415b2c2 : FAILURE in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7f9d5f6e8b54f418f4bf2f2241f2873 : SUCCESS in 14m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/525070f347314d2e870f137435184471 : SUCCESS in 16m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ea72082134f4c59b4bcac7878616ee5 : FAILURE in 4m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a3632df2ff54b4d8047e8bbfa52d0e6 : FAILURE in 4m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3852192fb9b14b888fac2ccb02b6499b : FAILURE in 4m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87c1213372704d01bf898c99bfcf8ca5 : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea69d85720074948bcf3b840a72822fa : SUCCESS in 12m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b4a6da5edb44735b7858eed55d328d5 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/223a976f1b75421d92b8b0ff48873036 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0b9aaadf21f44da876a97ff44ba8807 : SUCCESS in 12m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13206bfa04e64aab8912ead2fca2e698 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c28c4cadc6d14a4fa646014f94b2db60 : RETRY_LIMIT in 5m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/86784f6e8b0d4f5abd90bad1c1018850 : SUCCESS in 47m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/87333eed37764b97b642dd2eb9ef96cf : SUCCESS in 39m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4525b1f5ba8b4b0389fe8b98eff486c9 : SUCCESS in 42m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77438f8d8deb459fb624da8b8195a4ae : FAILURE in 15m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e664a5c495bc4aa4b3c9db62431be886 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5664d7c66ec441d9074d06c5752eaa8 : FAILURE in 23m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c22ffddca064fb895d5aebfbeca9d40 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bf78ec8cc47e2e8d4b541b993b83809b7dc7b4ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 06:57:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5b62600d7e43d7ad361184de22bb91\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d6442abb8a841eaaebfe0cbe90e4a93 : SUCCESS in 1h 00m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbca0673201f484692f1e4d35496864c : SUCCESS in 1h 02m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/424760473ba94cfdb9641f3acf2d06b1 : SUCCESS in 1h 02m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7faa82ce42f94a12a3408196f8c1f9b6 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":2},{"id":"98ea373016c61b910e438651c914ce60d1b787cb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-21 15:57:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"407d2661486e89680a70bd2c97cf734770f77d19","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-21 19:23:19.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d85e61985bf6362f3d5afa089c566189b57de70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-21 22:20:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"424cb967af16d0a00a554932bc9995276fd39e22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 23:03:21.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b9eae1e8a64b489307497e6c9e6fcb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d3d47405c0f45deb44aebba1c09ed67 : FAILURE in 37m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27215871502847ffabcf261683a443de : FAILURE in 37m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/55a82fcc00204f09a41373e37505eb27 : SUCCESS in 36m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6ec745b54d542f59f48747e2f305072 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9afa0e7d2cab610d8f0a76a094e348c503c71e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 23:47:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab382695ad1641d28d0d345af42735a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ee8f52ca1b47929a154c5ea18b7d24 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/04a8c1836e9a40bbba4b7a5c605e9212 : SUCCESS in 59m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91ca5778cce04ca4846de4c1557fcbd2 : SUCCESS in 51m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b68b64ce6444e13a6cf3d4fc3719dc2 : SUCCESS in 1h 21m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e653e958cbd34cc991dd98f7f2e551f2 : FAILURE in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1b2ff120c0c4adf9379db5927854e7e : FAILURE in 18m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbde69ac71a2413b9f5023f04f83ba3b : FAILURE in 19m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87d1353b695e47988a0ee7d7ef8c79d9 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a04bc122bba14beb9729f995eabc3b67 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5322bfc26a4c44de8fad4de9e0df68b9 : FAILURE in 19m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de16887c8bda4756bd715f78155e6d21 : FAILURE in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f96cfc9b19884996bd2e2d2b78e6fe33 : FAILURE in 15m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77ef6ba83e424beb82cf068a86c8b69f : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/93abd728ff814bf7b3fca9c07ad6916c : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc429e6941f7456d97ff7648bca4e369 : SUCCESS in 15m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4769b3f243404f8985ec3d10dbf86434 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e9b5a42abc848a889189d6059fdb422 : SUCCESS in 13m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1d2f90f4eeca41759e497300b94d0ed2 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce1f3b17b7fd48239e8e0a726190daf5 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8ebea8162fd40118644a2e83d521306 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99661798fa8843a09a356f8af9c1d562 : SUCCESS in 7m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5a07f345616543cb87d2f6885df8a26c : SUCCESS in 50m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9277477f9ff94a35a9351fc0258e531c : SUCCESS in 1h 06m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/421f2b81d75141e498411bc1a213c848 : SUCCESS in 49m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1647dc6356eb4d88a870c6e69da95ecb : SUCCESS in 41m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2d498ed74f54bb9809ae4b4d8d50a9c : FAILURE in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e024b9f9de7a4f7b8e8107b85c1e0c82 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32ebd443400b435ebc388cdec4a32b0d : FAILURE in 19m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16052d8303ce43d2bd41e8470ac1b1a5 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":3},{"id":"e41b38664ec9dfc9b55d47d0244aff792111951a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-22 17:21:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"31d3cedeb0752b75e430fe1c5aaf5d5993bd74bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 18:23:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9b3c006a09e4b86998d939f4479c57b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfcac4d573742739083f9554ced82b6 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/bf416a9052c84803b1a362f57c495670 : SUCCESS in 50m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec0e4054fa514880bfcfa05f0114f004 : SUCCESS in 41m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e60e1fe73c9b4e059fd7c0e20bc209f9 : SUCCESS in 49m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94ec75811f794363b797fd6f02f2ccc9 : FAILURE in 22m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c4e49ed1bbd4fc397e84c5162a37da7 : FAILURE in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e249604bdc5410dba7e34216f98c8a4 : FAILURE in 22m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ed915888c234c7195960894e866ebc3 : SUCCESS in 9m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d4d58b21a3204fee8c15c3f71a4731d2 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5a738eb3ce84de09c90084b94ae15bf : FAILURE in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba398b7c647d48fca619cc9653deb052 : SUCCESS in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1dc2e170434355a08bc4c28fb22840 : SUCCESS in 19m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2af271e183854ed7be44fa1cc0c3ba0b : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/01f9fd0e7a6b459b8118f109dc84f295 : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7bfe51b6f5942e9a8ff9d7a3b6c292b : SUCCESS in 21m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/24d7fe1de3f54aafa1c81abb592aff5d : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d127ab9d050849d9b3a5655fa208e3f0 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/26e0f578a9ff47dfb733ebd3c6124340 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93a5bd20522d4c30a5ca69f324c41b9d : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f55f414ad236498e82e5714e99719fbf : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f98864646b1d4f85bb633c4fabdc0b22 : SUCCESS in 8m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1bbc4b4a70b447a68400ba000a0e054d : SUCCESS in 45m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd701505598644fe9b0601f7d7cdbb89 : SUCCESS in 55m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2e92315713149dd886a48d94fbc38b2 : SUCCESS in 39m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e773159f631447589b028914b174cf0d : SUCCESS in 32m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d5e5b3e10c44203abf7adb3a7bd0647 : FAILURE in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9afe0f16c8c4fce8e92c08d476d0308 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0812a2093104497fa9df35dfd10bc65d : FAILURE in 16m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd12e941195e4f70936cc857e56121f6 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":4},{"id":"53a5dd6f87f121d2d63a55a7f3e4e4cc2e9739a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 18:29:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be8d6fa5136343ac873216f7b51ef1b4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88435ff6e63246fe81c11d5e41cf478d : SUCCESS in 43m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fac8454fb18b4aedbf76a2619f216428 : SUCCESS in 39m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a25f6f658b90449c9eb4a66645b27eea : SUCCESS in 1h 05m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0b7535793fd43eb87a6e2d3215d8838 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5860f0b4920ff1019b495f68283e8c8ab5a2d656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-23 00:42:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"101a39b7f86ac12867cef4b74147ca25d3da4f60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 01:25:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fca56ff3632646b8944a1b4ea9e41e05\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7b902267e3a4cdd957c1b7f4204ce81 : SUCCESS in 36m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcdc5bff23034d689af30cab47ade1a7 : SUCCESS in 35m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/07f7d7c1f9e445abb0fc33aeecf1bd13 : SUCCESS in 35m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc5f6ad3b16c4d5e97ab468e78129b79 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":5},{"id":"48e6d994f9ad376e559683306124a8c8076b03b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 02:17:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ab18e0482c44868acf19abad40a65b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/811f0673d5a04320b31bbe6c27a176ce : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/13f1c4f3f14b4aca9fcd16abb4fcb176 : SUCCESS in 52m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70130056dea84ffeb21922b09bcd77cf : SUCCESS in 58m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d34b894bbddf4792b259593c46313b98 : SUCCESS in 55m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60b3a30bb4924363840acf9db9f1ea14 : FAILURE in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b71e25665c3f4feda4c8ce9763d41ba4 : FAILURE in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/010f1a2dc2de44a8af3e029fae21537f : FAILURE in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/500280ea549349ea9f99f0fc1927fba4 : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c3802de863d240adbcf4b6c9d00b12de : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ffd1863f5e1438e8ecbb932f4e57734 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6de096bdd24b4744b4bb76009a969d48 : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aa610fdcb30e40109585f90138b41c28 : SUCCESS in 15m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25d8ab664cc640dba12a4da51b233352 : SUCCESS in 19m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/481103af490449799f12a9fc04de7d03 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33c4b782e1e241ada098a2632f944aea : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c386b6a2bdfd473ebbb3b7f1b32f13a6 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a9dd6383cfb4dce9e1968416b50d6c4 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5de5f74f81954baf992104e96ef8763e : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05d7ef7f0d2b4a5692de7bb17a3a4358 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b603ab302c424b7d99e54eef109b154e : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dbed50d3d8db404690021bad9eae1784 : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6aaef0a0d0c9486f9b7df204866649a8 : SUCCESS in 55m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e9ccf99b58049949af0303dc06f4bfc : SUCCESS in 59m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d87f3c2bc2b9406899afb33b1ba03566 : SUCCESS in 44m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ba7fb1baace4c12aaf527ba25a53433 : SUCCESS in 44m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66e3749b1d7d43d69b3c85607390109e : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffbda476d4c4b96bee8490b17edf68d : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6d6eeb002f143ee8cf5f30a0e2112a1 : FAILURE in 27m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e4d89656132427fbf660eb9366ef863 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":5},{"id":"622c530cc70e4d54384bfdbd1b029bb628dd5b9c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-12-01 15:58:15.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7b49ee093e2961dcd731e02f6c5022d0470e2332","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 19:32:12.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4b134d1e266fe7b763d6aa06dcc53c17d34a4605","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-11 22:50:59.000000000","message":"Abandoned\n\nIf we\u0027re doing this, it\u0027ll be in https://review.opendev.org/c/openstack/swift/+/834261","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2121a3cae7faea5c8e71d40c4e2caded6b6bad4c","revisions":{"9f89f76d807791c0696432527b9f2f978253919a":{"kind":"REWORK","_number":1,"created":"2022-11-18 23:59:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/865066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/865066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/865066/1"}}},"commit":{"parents":[{"commit":"729b7324a3fe897b9e8e834055aab589689c6bc9","subject":"ring: Create ReplicaToPartToDev helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/729b7324a3fe897b9e8e834055aab589689c6bc9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-18 23:36:12.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-18 23:36:12.000000000","tz":-480},"subject":"WIP: Rings as zipfiles","message":"WIP: Rings as zipfiles\n\nChange-Id: I75a92ef258d577659eb4716a20eeeef2fc4701a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f89f76d807791c0696432527b9f2f978253919a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f89f76d807791c0696432527b9f2f978253919a"}]},"branch":"refs/heads/master"},"9fea3830637e1b42d76bb10ef8b5202cc009e7a3":{"kind":"REWORK","_number":2,"created":"2022-11-21 05:48:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/865066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/865066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/865066/2"}}},"commit":{"parents":[{"commit":"549f60081bb1286d5cf3c17e1380495396a4b948","subject":"ring: Create ReplicaToPartToDev helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/549f60081bb1286d5cf3c17e1380495396a4b948"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-18 23:36:12.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-21 05:47:56.000000000","tz":-480},"subject":"WIP: Rings as zipfiles","message":"WIP: Rings as zipfiles\n\nChange-Id: I75a92ef258d577659eb4716a20eeeef2fc4701a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fea3830637e1b42d76bb10ef8b5202cc009e7a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fea3830637e1b42d76bb10ef8b5202cc009e7a3"}]},"branch":"refs/heads/master"},"fd421cca4d95c4f6f33c98aec02f3ec231c5c6f8":{"kind":"REWORK","_number":3,"created":"2022-11-21 22:20:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/865066/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/865066/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/865066/3"}}},"commit":{"parents":[{"commit":"280f460c80d2170f5e0c04e03605e3eebae2329c","subject":"Merge \"account_auditor: fix warning string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/280f460c80d2170f5e0c04e03605e3eebae2329c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-17 23:14:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-21 22:20:23.000000000","tz":-480},"subject":"WIP: Rings as zipfiles","message":"WIP: Rings as zipfiles\n\nChange-Id: I75a92ef258d577659eb4716a20eeeef2fc4701a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd421cca4d95c4f6f33c98aec02f3ec231c5c6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd421cca4d95c4f6f33c98aec02f3ec231c5c6f8"}]},"branch":"refs/heads/master"},"462bec4c3d7833db93ff9f16c2f6ca989a8ba72a":{"kind":"REWORK","_number":4,"created":"2022-11-22 17:21:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/865066/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/865066/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/865066/4"}}},"commit":{"parents":[{"commit":"280f460c80d2170f5e0c04e03605e3eebae2329c","subject":"Merge \"account_auditor: fix warning string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/280f460c80d2170f5e0c04e03605e3eebae2329c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-17 23:14:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-22 17:20:59.000000000","tz":-480},"subject":"WIP: Rings as zipfiles","message":"WIP: Rings as zipfiles\n\nChange-Id: I75a92ef258d577659eb4716a20eeeef2fc4701a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462bec4c3d7833db93ff9f16c2f6ca989a8ba72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462bec4c3d7833db93ff9f16c2f6ca989a8ba72a"}]},"branch":"refs/heads/master"},"2121a3cae7faea5c8e71d40c4e2caded6b6bad4c":{"kind":"REWORK","_number":5,"created":"2022-11-23 00:42:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/865066/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/865066/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/865066/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/865066/5"}}},"commit":{"parents":[{"commit":"280f460c80d2170f5e0c04e03605e3eebae2329c","subject":"Merge \"account_auditor: fix warning string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/280f460c80d2170f5e0c04e03605e3eebae2329c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-17 23:14:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-23 00:41:26.000000000","tz":-480},"subject":"WIP: Rings as zipfiles","message":"WIP: Rings as zipfiles\n\nChange-Id: I75a92ef258d577659eb4716a20eeeef2fc4701a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2121a3cae7faea5c8e71d40c4e2caded6b6bad4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2121a3cae7faea5c8e71d40c4e2caded6b6bad4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
