)]}'
{"id":"openstack%2Fswift~865706","triplet_id":"openstack%2Fswift~master~Ibb7cd761be4a5b1ec53dd16b7c5d256ed7666a88","project":"openstack/swift","branch":"master","topic":"p-namespace-map-2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-28 10:56:25.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ibb7cd761be4a5b1ec53dd16b7c5d256ed7666a88","subject":"proxy: refactor ContainerController._GET_using_cache","status":"MERGED","created":"2022-11-25 18:25:15.000000000","updated":"2022-11-28 13:10:24.000000000","submitted":"2022-11-28 13:09:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865706-p-namespace-map-2","meta_rev_id":"de2a030c05afe868838855b0f6a13d741af3d5c9","_number":865706,"virtual_id_number":865706,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-28 13:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-11-28 00:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-11-28 00:03:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-25 19:32:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 00:03:41.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ecf7a87788e6f9926f2adbf0e2c0d1eeb993c244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-25 18:25:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd3ff5a6fcf368f6cb65a67bf2769c3ce2900e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 19:32:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/300143c2779d433a92bdbd1073e248b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6392ae0e230742c4a620600fa679d465 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/6f7c416625f84d9698647126212e982d : SUCCESS in 55m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec502ba6cea0486183a2c57290301ea9 : SUCCESS in 36m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb57ee7a61c14a6e81dfd50348ff0f56 : SUCCESS in 59m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8497bb271fa74ac284dcca748d7fd23e : FAILURE in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2aa9bc1f4fff4c1586ee9ea17fbde0b4 : FAILURE in 17m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3405f6326a6b4ae9833d36f134a8f535 : FAILURE in 23m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c51def76be144422a2ee26d0fa64caf5 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b79b8edf9394ef380a4a93cac6a9d91 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/814d67dbc44a467189fa8437f9e27e57 : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19abffb4abe54e0f9d707fe6205012e3 : SUCCESS in 15m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e0d208106c243c89db555fc1eb50960 : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e110ca2640c0401baef1138caee8fa3d : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d07a313b2572403ba335aa12e0b93a8c : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d9d042a1fbd42209c0d0f5227eddeab : SUCCESS in 15m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9450f922833949b79719d28890cd56c1 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/820cf94b105e4f8c9e7de2b84206c3e4 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc6e3f0edc1849a1aec56b2309c21dff : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1679478f96441cbbd9d5c93e8646f29 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d51f17e3196436095f59580ce944dd3 : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/461f9825a90c4a40b63e41c727bf228c : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd0fed67e01b4fddbdcb731c415960c9 : SUCCESS in 52m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6150d4dd6824102bb1432dd12209a6d : SUCCESS in 54m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/774076eaee00442c9cc5d273b3078e86 : SUCCESS in 42m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad20436c05304a2babb339c266167d33 : SUCCESS in 46m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c17c10acd7249bbb8d10caf333bc93e : SUCCESS in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331952fb60bb43f28ad9b4031b801a96 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04cd89877f244028a67e5040d6a22111 : FAILURE in 25m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a382d84381b464591ef0657ec7547a2 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"08d66cf076874308bf4184de2ce41b4e8b2d740d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 19:40:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2958f8e109b34e53844ef9e41239fb35\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/678efb1d9c8a4df091204db88848d3d7 : SUCCESS in 1h 00m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fd6f9156ad8416fba2be12d1b6ac562 : SUCCESS in 1h 04m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/28591fccd5dc41a0bf416ea37f2c3735 : SUCCESS in 1h 12m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/690cbe164fa44c5a80868b434cfae3d0 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0e1fb01b34387b787d90d92147b5365c46b40fe2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-28 00:03:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d87cbf064cedaa521d43893b534f99f824e3f3ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 00:04:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d11bc768d4782a445058608146b52169105d4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 01:16:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84d46d70c7b94e77b054d420e42f75a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4cd63b0977c42eb806eb7d5f64f7e18 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/da73f034da4a4009b9cb3fa95b9679de : FAILURE in 46m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c867f0567db246ccb41918863c7659db : SUCCESS in 45m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3c4ede6c8d8449db3f85607414cafbc : SUCCESS in 51m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/355a5e0f3fff437cbbe2632b7471acca : SUCCESS in 8m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/902de2ad45c447f0b348a4dab0ca161b : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b168df7c01049529ff32de328a484b6 : SUCCESS in 20m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ecd8fa4dd8f044c78b698e5770b45fda : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4b9fa138834a1eac3cf48035461f84 : SUCCESS in 20m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/40948c27c56a48f28a4dd29f6acbbb68 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d70a3e2c3de34a88ae17dc5a2a12eb58 : SUCCESS in 21m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d9f912f58644412b19b88496703a728 : SUCCESS in 15m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9e8447770f7848b6a9cf92420fb37d7a : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb7c87de3f2d42c8a2e25bd8c048fa17 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b1aed2e7d1d462dab3f0807b26148ce : SUCCESS in 21m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/faeb0537b23b40da8dcf3709b3540eed : SUCCESS in 6m 50s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7fab82fdf704394a522e4dc731e82ac : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d06333a582ad492fa860860c749e1393 : SUCCESS in 51m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c17880f424b4bdaa504812795e663fa : SUCCESS in 1h 07m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51358e667d744e15b215ca9458e57fb1 : SUCCESS in 48m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a889efd99d984668a9f077e57e3c6f4b : SUCCESS in 43m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1044ea90239943a09bf910b7ac3e5599 : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16f960762ff4da6bd44d46a9b6ee74d : SUCCESS in 4m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19b425cc91cc485e822f3fdca75efd06 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5e9fd816cb537358b1f4819f1776ba042c29d7f6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-28 10:56:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ba6559326237820c38d7b26dde3a0cab3100ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fbd83a4c1d146ba898974f025cfd03f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb300a0a1824cc6870b58703bbe9398 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/658afe59b9954664bcac0013dc95e7ce : SUCCESS in 55m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c15a64a0469b40b0872f99d0b08fad20 : SUCCESS in 44m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b32a4631577c45c2b2114dd5c4eaea4d : SUCCESS in 53m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a01f484b8ed44e098cfa22aeb9eb3468 : FAILURE in 24m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b210e0e5f279410a86e505597579ca6b : FAILURE in 26m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b01dea99c8142b28d93314bc1ef57ab : FAILURE in 19m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/363c6b0697a7433891afd7f18c41f7f2 : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f8d6804f8774466889115589595c6174 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6d7c13b18254b149dbe60aa8415567e : SUCCESS in 18m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa4da71805dd4cacb2a607d38e5bd840 : SUCCESS in 17m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a784b01f5b0645c9affe7faed955d0e7 : SUCCESS in 19m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d76f761ab22a43fa9ec5e5e2019a1821 : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb27e7f987be4050bc2af3f2f07b26d4 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b74d3de1ffc4fba8a8985db1a5e9d16 : SUCCESS in 19m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/703c44d8ebea418a9c329cc69b1aa14d : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/73cbb5405d694173a42147f5f1296ef6 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ba61a34abe04c9aa08ec092e14c3726 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9e865f7cb3c49728073e839f4fcf689 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d69cbd132ee4dcd9aeba2cc1287f818 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b6d54baabac40aaa60399844e6caee4 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7a1cee40a58143fdb49b3f7114605ee9 : SUCCESS in 52m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/32d4e404af3a49619067cca2036a6c39 : SUCCESS in 1h 01m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86c8c1888bb941a99c6060e70c7e07c9 : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff5ea21c978c4db283035f18e59a1bf5 : SUCCESS in 45m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d337bc42cdde494aacc8f7fa741d6bf1 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3f10d162b540e8a373be78c33c08dd : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55ba9c4fb469412583ce3caa567d71b0 : FAILURE in 25m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/859168a545a24a84aa861a58a1f8159d : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3ad1e69a8f9335011a44307fc37887af48dff1c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:08:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdb5c53f6a530bea475aa224b4b247cb222dbed0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:23:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39d4ea515b7b47b18ae5085f9a389d72\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29d90ac9a98f4c648b71677d060333cc : SUCCESS in 46m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22a542432d124b5ea8b18acba4c05403 : SUCCESS in 44m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/44c5b89446f44296be3953352c13ab63 : SUCCESS in 52m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0b3b8d3ec7247bdb309e4d967b52730 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":1},{"id":"77ef538bd5969e5250fb44a3a4fb0b562337ad40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:09:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea4515b9715f43cc9211daf1e59c4026\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad5d370ae834211b9dbbfd325413203 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/51bef7f564fb499fae7f472b69ca7dcf : SUCCESS in 54m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d5f31ff8538481fa17dda719f13777b : SUCCESS in 41m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0556dd553224729a292cd0d4acf86e2 : SUCCESS in 55m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/28d6c6b907ec460385a35ce6abffdeea : SUCCESS in 8m 42s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d6807de3b161499b92a41cce6528876a : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed3b49c7044b413381c59b7ab601d3ad : SUCCESS in 18m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c35169ccb73f4a1384f1843691ebd2c8 : SUCCESS in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9437bd5fe22a4b229b78c4d6d75c6b60 : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a24be9a700614dbb8d20098d58bb3de2 : SUCCESS in 32m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92c516dc514e444abdf8c0dec5e59bf2 : SUCCESS in 21m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba35bc19db854c7fb8982c37bcd0c290 : SUCCESS in 33m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a90e0f47923d4f1cb856fba785d9d37d : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/178beece18214e16be56a783c482ba43 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7de16a8d4a7c4c90ab4700db0276f706 : SUCCESS in 32m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6348f2af1bc43bdb3a39d206fbe06e0 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17b9b0e0890f4cd6824d2e32c232b2da : SUCCESS in 9m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30adc2480e4a4784b4c5f34c2825761b : SUCCESS in 59m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6334691287424267b07c280c3e6a30d9 : SUCCESS in 59m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe3679a99df2422a93d39571e521287d : SUCCESS in 41m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f175d90880e4682874ad1348e6a2777 : SUCCESS in 49m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80596d6f577b4d78a62aeeb2ecc56492 : SUCCESS in 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782a7702d3fd40298b71b80636369779 : SUCCESS in 5m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bac36a674b1400e89fec50b1bcd4e8f : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c9c1bf92a84ba18222bf17acb102543410926179","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:09:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de2a030c05afe868838855b0f6a13d741af3d5c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:10:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f000e9972e4ce2a89b2d4bcb52e79f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642d164f7c564c5ca5b12d6359e04825 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e3811704d83046dfb6c0cda17948f81e : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/95faf05e47854ff484814f7b00a0fa69 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd5d099035a50c419c4c257f41de15966ff3f5b3","revisions":{"bd5d099035a50c419c4c257f41de15966ff3f5b3":{"kind":"REWORK","_number":1,"created":"2022-11-25 18:25:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/865706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/865706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/865706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/865706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/865706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/865706/1"}}},"commit":{"parents":[{"commit":"e748cff5c407b61045326f235f69552ac82b773a","subject":"Merge \"proxy-server: include suppression time in error limit log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e748cff5c407b61045326f235f69552ac82b773a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 10:03:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 10:03:02.000000000","tz":0},"subject":"proxy: refactor ContainerController._GET_using_cache","message":"proxy: refactor ContainerController._GET_using_cache\n\nRefactor the ContainerController._GET_using_cache. No behavioral\nchanges.\n\n* Make the top-level method shorter.\n\n* Make the flow easier to follow.\n\n* Make various return points more obvious.\n\n* Change variable names to clarify those that are lists of ShardRange\n  objects from those that are lists of dicts representing shard\n  ranges.\n\nChange-Id: Ibb7cd761be4a5b1ec53dd16b7c5d256ed7666a88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd5d099035a50c419c4c257f41de15966ff3f5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd5d099035a50c419c4c257f41de15966ff3f5b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
