)]}'
{"id":"openstack%2Fswift~865707","triplet_id":"openstack%2Fswift~master~Ic0454c5f1d37a84258e43427cfe7cd6dfced285b","project":"openstack/swift","branch":"master","topic":"p-namespace-map-2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-11-28 15:51:30.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ic0454c5f1d37a84258e43427cfe7cd6dfced285b","subject":"proxy: add test for ContainerController._GET_using_cache","status":"MERGED","created":"2022-11-25 18:25:15.000000000","updated":"2022-11-28 17:45:50.000000000","submitted":"2022-11-28 17:44:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"865707-p-namespace-map-2","meta_rev_id":"caa8831d0075a79420c1d4355838250969a4298f","_number":865707,"virtual_id_number":865707,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-28 17:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-11-28 00:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-11-28 00:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-25 19:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 00:15:02.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"8cbabfb9b957ad78a70aab43b88addc35ff62d07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-25 18:25:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71f50a9e294f511fca6ac2ca526a576565f6e969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 19:08:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e229e1ae218426da85616b196b27e19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c576815061194487ab146320b60c05a0 : SUCCESS in 8m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8538a0d5cef94539ae68479043b94faa : FAILURE in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5b26912db2a401ba6d94dc140976da1 : FAILURE in 21m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f03177a8813b4752a49cf471f7cfd616 : FAILURE in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/88ed3f813a1340ac9caed941d82b8330 : SUCCESS in 8m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3f316c5c899a4fa4b5d56a72e50d77bf : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eae0b4a0bb28472780f0fd368569c985 : SUCCESS in 16m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf39146ed00847b5adddc7611d449364 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b2ac220c734d46904036c702ccb4b9 : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b371dd6b8ba049398faba29522761206 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9797f4d9860b4320b55e1e58c9148115 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6182879f0d34f4fa8948ef3585ae7ac : SUCCESS in 14m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3a2c6feed0364077ba712fdd08fce77b : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/51941ad0a469480a9857338b85022111 : SUCCESS in 13m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6e08c3f9f9d1449a87f848adb93888c4 : SUCCESS in 13m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3458a100b3e747ad8ad0c11a9cf10ccf : SUCCESS in 40m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c78d21d61a744cfba1babe8c42ff2f1b : SUCCESS in 28m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11d9fa4ad9214772b41af59394f8a143 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2a2af5a5cf40e09d7f52e90602ecf1 : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18732616993c4553bb83e730d4377478 : FAILURE in 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ccd5319679ac82953a03e50f110392232c549bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 19:32:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a119e60317514062a04d8ca0493bccfe\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23dcde8e07434902b9b569ce7b081178 : SUCCESS in 1h 04m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8582ed5d67314ab99dd0fd880c444d8c : SUCCESS in 53m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5253da76d06d409c94407dd1b4b40aa0 : SUCCESS in 1h 01m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d55b5b3f42034a96a201f32bf5005c96 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ff7c9e494d7749388090d64e383f10591d55666f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-28 00:15:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9d506b91adf34453853019af876c4f9268bbb6d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 00:15:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61444b21883ec437678bf0cfe222a54e93a2b145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 01:16:21.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/865706 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"fafcd32072ca068ffb34e31a113804f6645c004f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-11-28 15:51:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db889cb0d14fb16691b933c998cc92da42f76b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 16:36:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8822c4401914bc7a04eb750ed57c6d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a2abbfb7934287843d312612c1f681 : SUCCESS in 10m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77b0c43df2dd4fec9b76df1c2fbcbf01 : FAILURE in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d76c25e4f77249b29c1bab6df5880dd8 : FAILURE in 17m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f804200bdcdb470484371b3dfdd506ca : FAILURE in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01364c8c33064b418e1b08dc3a01a57a : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a5875d1287e4495a08ba3c70ceddd12 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/edf94f35cf4d4cd0a6e42184b12c40ca : SUCCESS in 15m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff7dfa64a6354cfaa6a8421507857fe0 : SUCCESS in 17m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3d072ed2d4c46c193fd181a12936995 : SUCCESS in 14m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0fa544501474cb2b52fe43cf4de2aa7 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c03992302b504728860e6e1678d273ca : SUCCESS in 20m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d7b1416639b48a89bafbf96cda48fb2 : SUCCESS in 21m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/78878982dcb34daca992d4618807f8cb : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c0757e7b1f114c339405026fb0351a34 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d0b29d3c986848bb9f89407b36464648 : SUCCESS in 19m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c1308db2c014facaf089e33fd39d96c : SUCCESS in 42m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36b0d96ff4b94387a75ea22e7e1096e8 : SUCCESS in 42m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ed135bff9414cdbab235834015144a5 : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38180a4189743138f4c0053573e2470 : SUCCESS in 7m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee7245f97e884bfc9f2d02eac7a5f14e : FAILURE in 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20432a736b6c2647012ab6aa55f38939b2db420e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 16:37:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"797d9378966e85d69b84a2fd91bf36a5dec8d227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 17:31:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cfb7afab20c4cdebd93dfc8603f5193\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28fb5e2fbf714f988f6598467cdc481b : SUCCESS in 1h 01m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1da64cd502574278a382863263f791af : SUCCESS in 58m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/185103bafbdd469ab6db75edbfdbd0b6 : SUCCESS in 1h 16m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/77cc3f9c63ce46ccbc53070c6e3a18d5 : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"353b5dfc7c4a534e9b5cebd1f42b55983214ae41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 17:44:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf58d0eaf2834331bfbed634eb557875\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f015efa3e3d64c49acaa4db26bf60a14 : SUCCESS in 9m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/87866897de2f4935bd94033272b47744 : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ec17e432251e4676a802cc6f12ce6586 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18f783865d0948c6a833d54bd6699d47 : SUCCESS in 14m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a786bdb8d4e4505ae0126c6ffa02a84 : SUCCESS in 14m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0243170880d1433ca0cb4da0781b4324 : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c3c48192aec435391f7e769227c119c : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85b18ccd07604721978287413c7efd56 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/952f8c1c48ec4a0ba7f8910498ae95b4 : SUCCESS in 16m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f28fe460b74f445294a9d5254a50e3e6 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ecb3aa63d4644164a0e232aebae6ae1d : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3f8fdd6687dc4d9dbd8a27e3d1dfe3ed : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8cc512fe71be4ce0b9abe58b24d90f1f : SUCCESS in 8m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/580cbbb9e315431c8926fa1092cbbf9d : SUCCESS in 9m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10237539d8024cdfafed4cf504e156e4 : SUCCESS in 40m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fa3304a9b0644a3ba0642b8d24293f6 : SUCCESS in 40m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66931ec4a49e4a2dbaaa7e3e4bf489d6 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f2962809e045a5a4d7c72854b9bcf0 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b9b1a5ee0f3041383a321165dd517341e56a6923","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 17:44:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caa8831d0075a79420c1d4355838250969a4298f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 17:45:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aaa9af639df4ae3bf63e4c198cf938e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af25c818e8e4447c96dfb3b1a552c2c3 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/40f9d48c143d406c84d683b199e29746 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/62c279f79e5f45d780bcd52225affdd2 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1939cba037926400cd777059ee511481916cb08","revisions":{"a1939cba037926400cd777059ee511481916cb08":{"kind":"REWORK","_number":1,"created":"2022-11-25 18:25:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/07/865707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/865707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/865707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/865707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/865707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/865707/1"}}},"commit":{"parents":[{"commit":"bd5d099035a50c419c4c257f41de15966ff3f5b3","subject":"proxy: refactor ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd5d099035a50c419c4c257f41de15966ff3f5b3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 15:12:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-11-25 15:46:43.000000000","tz":0},"subject":"proxy: add test for ContainerController._GET_using_cache","message":"proxy: add test for ContainerController._GET_using_cache\n\nAdd coverage for container the GET path that attempts to use memcache.\n\nWhen memcache is not available the backend requests will reverse shard\nranges if required by the request params. When memcache is available\nthe proxy sends a \u0027X-Backend-Override-Shard-Name-Filter\u0027 header to\ninstruct the backend to always return ShardRanges in their natural\norder. Add test coverage for the latter case.\n\nAlso add a last_modified item to the shard range dicts in mock\nresponse bodies to be consistent with the real server response.\n\nChange-Id: Ic0454c5f1d37a84258e43427cfe7cd6dfced285b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
