)]}'
{"id":"openstack%2Fswift~866051","triplet_id":"openstack%2Fswift~master~Ic2695e2e836da5607f4f5c016c660496e2821e07","project":"openstack/swift","branch":"master","topic":"py311-support","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-22 09:54:51.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-06-22 22:27:28.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-22 16:16:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic2695e2e836da5607f4f5c016c660496e2821e07","subject":"Green GreenDBConnection.execute","status":"MERGED","created":"2022-11-29 17:44:00.000000000","updated":"2023-06-22 22:28:42.000000000","submitted":"2023-06-22 22:27:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"866051-py311-support","meta_rev_id":"ec230e596c931b69dd350247f7df1df9bdf25d81","_number":866051,"virtual_id_number":866051,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 22:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-06-22 09:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-06-22 09:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-29 18:43:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2022-11-29 18:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-09 13:00:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-06-22 09:54:51.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"1dc835178b7061a942f0831903e060f1dcf9506f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-29 17:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0aee80e3e257bb04015b058e3b710d982e463f41","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-29 18:43:59.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7aa82d5cc93886fc65c719e5ac00b7d50d3054a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:49:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08cacd50a7704de7bfba97d48e5b1284\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36833722e75b41b9b09bc55594dbbbe0 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/b36c54a905a84d7bbea0a3baca58b311 : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f0c3f94487144f990fa6fb68c1bf923 : SUCCESS in 48m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07f551137ab9402ba91afdabb27f90ff : SUCCESS in 56m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1f4e2d3dab74a378d4d49f775b2dfe0 : FAILURE in 15m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98107ca8825043629b5ed2b3bab3916c : FAILURE in 15m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3b1bfd08c394ab2a8c82a43f34149be : FAILURE in 15m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4b3c331d55064ed19c287623f7868bcd : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a676272e969849d3b0b94652c9345dc6 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5cafed80b30f42438f81afcb8ea790d7 : FAILURE in 14m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4716f0d69ea247a7aa91d6d0564f4cbf : FAILURE in 16m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/171373bd009545b2a5e9af307110f32b : FAILURE in 12m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0192f62f381044bc8a62993cdea2e583 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/39b1e1d7499c4009b539f9589ef7c1b7 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56a1cdbbb6414a788ca9745d37ef9b3f : SUCCESS in 15m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc6051f1fb2146719c44699f0f47142e : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2c196e86f6e46ac9e817bb3041f9175 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ed7b158c4b24e31a7d02c3db5798f8e : SUCCESS in 13m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54e1cc5524284f14b33dcd030570d4a6 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b614e4a9097b4862a19c42d7ebbc11e6 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/93e92602469e40989c88838ff0a16d4e : SUCCESS in 7m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/926a1131ee924024b5d7abb07d367e80 : SUCCESS in 40m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb815c44ccc644f987e08fbb71ea7c6d : SUCCESS in 1h 03m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/888cffd8b23e4f1fb108ea3819ea14f7 : SUCCESS in 29m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9870042957c4d3b8ba818d667cfa843 : SUCCESS in 48m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d9c9cb5f7e48248cab5ce2fa3ae8fd : FAILURE in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013d6065104942748bb7375bb84136f5 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc631b4f1bdf479286b354f172db5920 : FAILURE in 16m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f287147b5d914fc78d58fa46f888546d : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dcefb1423caba4ece6121ac02ab253db797f9965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-29 19:02:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99125f7d6bd0cd1b762426ad856daa4a36d3426a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 20:14:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/481a81a935f648d0846efd8cf1a33bde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22d0be4558140bd90f55215e1d7f5ed : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/d6111dfea668438f80d6bca9d0dd98c3 : SUCCESS in 51m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e571c42aa26443648d6e9591638c06d0 : SUCCESS in 51m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/631391ae9264406a919216e215b836b4 : SUCCESS in 1h 06m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94b455263ee3404997a3f5f965e23abe : FAILURE in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e9d8c47b1454e7598c4a439dc17f9ce : FAILURE in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f66d0b4b1b51473f8c86e06792a2b233 : FAILURE in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/426cb9624ea24caa8013d71eb2a0a5ec : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a252ee82998641a9a347cf94a3b19c37 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c44ccc211afb4e30b78c75648fe12e42 : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67605746656f423db9a8589edf5b3fe5 : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b861dcfc1e494be0b6434ddd14d5cc5a : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43fc380d7525489a8dd2c8cdbf0fd147 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/735689d621b9494e82aee40eb95699bd : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ca3fd38aad5e4597b610215b6c639caf : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75d492cd517f4c9caf1607b05588656a : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c79f4314e6948f180cc8711ff6a2c84 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6fbdc709cb5d4865b002dcc8f22962af : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20ed72bbe89b4579a106860991a88fe4 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/48819d7e653d45e5a968ea7eeb3140c9 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d50ebfc059f4662b064c072978646c3 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ee3f88c263fb40a88b1f7daeff2e3ab5 : SUCCESS in 56m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4186c280d954381ade7c9def2736751 : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/431a0022926641dbb84cd94ffb1bcd02 : SUCCESS in 48m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c6bf498f428469aaa2fde7d37c80762 : SUCCESS in 45m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/757782c1412f4b41b5a6c2405be001c0 : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/653dd7c8a7954f63853c943b18177e73 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/660c8445e3b74cd0939837b73945fc37 : FAILURE in 19m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fa6c084242645c480bb41562a79d8bb : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":2},{"id":"80152292c8538a94b34ccea814ba563048833fb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 20:19:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf4e07512a94f3a868864d5ffd34956\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bd64c8ab05f419a91627915d81545f5 : SUCCESS in 50m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b57aa070116d48509b2c32ef00c3118c : SUCCESS in 1h 10m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7ce243a342c4413b8f62b0e5a34b6e3 : SUCCESS in 48m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ab15a6225a84f0e8e543c41d4eebe06 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e9f1c6a4cb4a8193afd306ee770c04e102c5b2f2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-29 21:38:03.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8562692d47f9bc6fc451965f2bd67a6c8bc245a3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 22:11:04.000000000","message":"Patch Set 2: Code-Review-2\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"304e4b1e045afb34a7dd6dc659baca9564168980","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 23:52:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9da85f2b033fc36c14eabca1e91c359a3a0735b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-05 22:46:32.000000000","message":"Patch Set 2: -Code-Review Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c9750fcdbef207584263fbfad4b6aa066db15541","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-07 22:04:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87d3c9619cb3c31d046dd26eaf5d381543a4651a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-09 18:44:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2dc40014f1b002649e40fd201c110dd3dcf9125f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-09 18:49:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"069ca414949c3922dc299369b2dc0a56c6ee55d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 19:41:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f08dfd787b458d86342909072d595d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e0526876ac847688db2583ea6d171d4 : SUCCESS in 47m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f17223bf9e1d428badae02beacfbec0e : SUCCESS in 51m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0f99190d16d4905a29e0ba19a3e2261 : SUCCESS in 46m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44c71a70de0849a386d9a2a9e1bb9d5a : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":3},{"id":"bf0eb3411060dee1c61cc89a669f64679bba7ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 19:57:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a12f64b0e8a046eca72eaee8dc926e1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f4d7a591694ef882a3358e193e04f4 : SUCCESS in 7m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/f0776b3cf9404fc096d0ff4a9dadb724 : SUCCESS in 1h 05m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee9f5d5b3bc64c52a2c93796f2a224ba : SUCCESS in 36m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bf57b148b3d45548f457234bbc2373b : SUCCESS in 57m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af41d15f53bb4e9988ef1040da9bc6e7 : FAILURE in 21m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/931d61c6a9e64d91b15e19b6c894ae4a : FAILURE in 14m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b17287ee0cdd44c8a1177a8f788581e5 : FAILURE in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6fc6a291f534aebb9568b74887267fc : SUCCESS in 6m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/59f03710915e46e59fc81669065df396 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c33df973a65f40aea96c2f5e9773ffaf : SUCCESS in 14m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b3dfcf1059744884af8d86294534b0bb : SUCCESS in 14m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1f3e5cfea3b466faacb98d1a91491c0 : SUCCESS in 10m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28e9e448573d44028e70b2b5a570ef58 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3924d7cc5a6e417caa50148c7cc61577 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e7ff273ec5744f587bad7c3696ee7b5 : SUCCESS in 19m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8e348ad2187f47798dce9bcc18844cc6 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/071375e7d3c147518bd0ddff90cf770c : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bf896d90898448efa088791eedf74aff : SUCCESS in 13m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e0a2e59eb314090b3138a1eb62561ea : SUCCESS in 9m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/476036263a724283ba15e5b33eb24ae7 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2bb3a0b3c68b4f49a376ad711c8c5961 : SUCCESS in 9m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6dbbbb55beb84332869627f6fbc6bf39 : SUCCESS in 48m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbc52a5f57564105ad05da9a8c17fe45 : SUCCESS in 1h 07m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f0ffd75385342ddb236a9c40d98be57 : SUCCESS in 45m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aaca37fb4fe24cb9ac44e3220c55c057 : SUCCESS in 42m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45a9216647ed41e28a89dc40080ea264 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50d62d53c9a4b2f91dd0ab1116a249b : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7624bd6a06ec425f81ff6c1a81218815 : FAILURE in 26m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50ff5b5c2c194141ba5bc39186ac4ba6 : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c14cd5bea8ccee285f0d2a39b1618dd92b91a432","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-09 13:00:59.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6b5ef5b2446c752451cc843347124c5bbb42b179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-20 23:42:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2c7827de5c69d939f9cb19766236ff45526de47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 00:48:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b79af4114104076a45e0d93fa522c36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b28fa1b33d405b8fd6a1ffc5156b14 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/01d17196176145328f218ad6f823d500 : SUCCESS in 47m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4376e2034e04e98b4689f0f5eb7f857 : SUCCESS in 38m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cdef4af13cf4dd88e627d7b33bc0b23 : SUCCESS in 50m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5e5686ac096425eba4f9ab1cda8fdd5 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cda2424bd368429ab2b63318e8d509bb : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6ded2e68f1247eeb5661d9d00b4a1a5 : SUCCESS in 22m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14627309458341509253da01e5c94687 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/31097a1ac7344429a4ae6dc4a812fe57 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0c561bfdffc4ee895ac7712f45c39cc : FAILURE in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b64776ac28b417b819a1141ed715573 : FAILURE in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54a291d927fe4ffb8f1d9fa8d80df39c : FAILURE in 16m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/01d7e31c936243f0a047114ffb119696 : FAILURE in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b795056f4da241e0ba6ced8dd99d692d : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c5df7afac234259a676b6a478a276fc : SUCCESS in 13m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50c1c2e408414ab6ae576cee61dd3968 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/21560671287e4b288d327e87d6ca54d2 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c4a0685902f405687e9661e3f6021e1 : SUCCESS in 14m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/203c6a5c312f44dab984b28c65f06a2e : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d67d80ef7de64475b6f014ce19e645dc : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a7785aa4ca9c4d60a2e5ea86008b49c8 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/34a5c9205f064214bc2d9814bbc21995 : SUCCESS in 8m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4291b8624b8b4710869a1e7a74ff1100 : SUCCESS in 1h 03m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1feaff5333984224a8c254f74b7346e3 : SUCCESS in 1h 04m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af7c3ea78c94465d93179d2e24151594 : SUCCESS in 44m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c671399cd2b46fa8f1ee4acfd978768 : SUCCESS in 46m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1fcf61e847d4db59dc0199157634a14 : FAILURE in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5d3fdbc6cd4d4188d8899cdb47838c : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b428aaf1dc7d4879b351a9267a09f69c : SUCCESS in 23m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/611c1aadfdd94337abf4472591b00ff4 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":4},{"id":"3de63efe1a02aa0527f634424ab62d93b201afe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:02:44.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73f791b745ae4645bfdd69603ce1510f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2805ff44dba489bb1afa1941834d1c7 : FAILURE in 16m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f477bba4c404da2b060a1e737290871 : FAILURE in 17m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/630223d97c964068af70005cbcb29853 : SUCCESS in 21m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/11e7648f1ff34fbab902323a1b2b52e4 : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a9dc06c8ec01d14ead8a6a8cc16a0732a3a009db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 01:16:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f5c57d3ac3dc21b268ecf340e6d06b7595fecd16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 02:39:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c86425e1af34fbdb145a913cd971daa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159b72d9a65d43adb1410ffc7cec13d0 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/56e1e9d6773d49ccbff22c6d1a84534d : SUCCESS in 1h 12m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d7ca7428b5b4173bf2372c422adeca8 : SUCCESS in 59m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff81a0bbd6fc44548611febf4524f7ac : SUCCESS in 53m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b4520585fc64db7b2962e546835cca8 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af93ae6dfb404b9984710b0b60d7f2de : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d05728ef07504bb7b2e46a676b9cf036 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da0ebda2d4e74b41bbe91cc249197751 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edb9f8f5582b4a47bded6898b5d7cb17 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49e7d6f4cc5b4f1e86ecf8f242dde649 : SUCCESS in 18m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47e18ce785664936b99b80fd4daf3793 : SUCCESS in 14m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d38f0d824b4a42008a1f8be4bc100e81 : SUCCESS in 15m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/002750f440d9483781f5c373abc39fbc : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9947730c15f34f02acb082e88280d0b5 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ce8d40d2ab743b5a6d226a753a3da4b : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/00e4b726a9b741008a6bd247591b066a : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b1b28d640b7c459eb68b51cebf5e7aa2 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7b19d64338f46df93c9835fac6814f2 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4fa6c98a674a4243a65c9d1e1c4b9f9f : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca8e58118134410882c264440d2c11fe : SUCCESS in 7m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe3d2fa929cc45c6b5bd98121bea99d6 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb26d65ec2e34dc7a4d58352397ed201 : SUCCESS in 7m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b6d56124af8a4c0489f4cd837de6d200 : SUCCESS in 1h 17m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6014ee7cbccd424f90bf9de08fd69cc7 : SUCCESS in 1h 03m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eebeb36a87d0453da75532a6c4992f3b : SUCCESS in 47m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8aa5957790304f8dbc535630451fc523 : SUCCESS in 40m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1497e239cbc04735ba94dd5f4cdfcef0 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fecdbed15cf9466d85ca04d7ae5de440 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/566e1d508c8a4200b5cc6980b5dab655 : SUCCESS in 19m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5b7c855051b49be8bf3431fe438d1ad : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":5},{"id":"c3d795907948fceb7d49a94162f4a446eff65a4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 02:49:15.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88dbef8905c4569a7c0fea3b596321e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a11d6321cce4cad8813b1cadc2f5481 : SUCCESS in 17m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27d8d68a78a84651a4ca3410c2e0629e : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/06b8059944e646a1bbb20e3bdcb56362 : SUCCESS in 21m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b805af1ccf84c73aefbb8f383ce2681 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":5},{"id":"221ce6d38d7d401cb9597499ce3bd2ec9d2e1d48","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 04:08:59.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"361bd48793171d15c010e6a446211416f7bbbe86","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-22 09:54:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4ae67102538269a88ba48e4f0ced74c1eca50629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 09:55:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fd55913207d20f069fbeff3ad29dd59f84da1aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:51:35.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75e161d1036d4fc799701d359145a47a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6681b71088d14c9f871dcbcea93e6cfc : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/b10b19ff796240b2bcd8fc369dbd49e9 : SUCCESS in 1h 22m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/003cc37a12604f76a050a6fc18c7e2d6 : SUCCESS in 47m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/506e46c21f274a2188ba8a12ae082f71 : SUCCESS in 1h 01m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c2cf238a10f54b58963f9e754c65fb1f : POST_FAILURE in 40m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9312ac3dfc2545a99c6d46b0b5f11ba1 : POST_FAILURE in 35m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94b8359228734b3b99163cea1c1ac5fb : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/512990d117ea454d8ef2ead811dd7372 : SUCCESS in 18m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8213133f571440ebcfb4425dc508964 : SUCCESS in 16m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ba897c58f44241898aabfd7608e2ab71 : SUCCESS in 17m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b0a1cde52db4517889ee589714ab874 : POST_FAILURE in 20m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/452d3c4a97b147b9ab35c239dd4735e9 : SUCCESS in 20m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83d7c455508f4a03a38edd0e5c2d266e : SUCCESS in 23m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f4646ba581c4cac9bfacac4238be1a4 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/311a75c3efbd48979b9a9de6ea2c8c96 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/61a1862bde50498b97b162e05119ea6e : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f79780986914a8694a46d206c41d007 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1858cd089f564d7a8e1b0a261cc40b89 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d5cc1c809ab463d9a2dc93d62650213 : SUCCESS in 1h 26m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a530b2e3854a48d590f2cf375bd226dd : SUCCESS in 1h 04m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f6d029237d94b279cf8805d83a7fb89 : SUCCESS in 48m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a8321f1cdda47be9ff77b963fd17874 : SUCCESS in 55m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c79ddf03d64eb98ad56c335f99a5d1 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1c7450de454e758e944e619d9912f6 : SUCCESS in 6m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fce1b94f1ed6412385312278c75da387 : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":5},{"id":"a73d09b75962a428a5845051954a04565f6dd906","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-22 16:16:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2500e187df00d3c531567515b53ea55bf28d6c8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:43:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a666429a624405bc96c33d09efd5ad\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c68b692a986240caa48d87b3210d4ebe : SUCCESS in 19m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eeed1763ba194fcabed135219dc57c52 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/02ef7262daf74311a5cbf5dc573a3a77 : SUCCESS in 21m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e1da7987d7e4aeab871e9d4ab9aa685 : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":5},{"id":"68e92ce2b5228e57a93e4c0f204e76057b3a44e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:43:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a154c58fa9d04b6ea1141267478a1071\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3970e7a3b147d781269e5490b71853 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0bb731aaf1d941c08124bc65d7e6a17b : SUCCESS in 1h 05m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a215505e3cfa4c6b97c627567d0337e3 : SUCCESS in 44m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfd989a4ff6b4b498e437cc94ceebf6c : SUCCESS in 1h 07m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0609d6689534f50804748d1b540c3ab : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f762f63ff3742f3bd45bcda13fa8944 : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96171549d4d94efba41da58628636b21 : SUCCESS in 17m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/531fa840e0c841b48d56812844e8ffb8 : POST_FAILURE in 39m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/32627fb3906d40d9ab3024b74ef234f9 : POST_FAILURE in 35m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d272dca953c4fb28d31c181101eef41 : SUCCESS in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/488f157278fe44f8bdd1cd75ed292400 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0c44b595ff0441c8af4bf2deaa87d57 : SUCCESS in 15m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3b4c3e66cff4436eb7bf41639212b89e : SUCCESS in 18m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c266536861144c53be6d86e4e9404699 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea5b0d35fc7247518f1d63f3c99cc7cb : SUCCESS in 21m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/68da8bb088d2429e8280843d48b0a474 : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1086d2e2c1a7467c858b9e08ed36c1b7 : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a35dfcdcbb3431082124f7bda5b22d4 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/13b0a3a199d34ff5829d310c9a201781 : SUCCESS in 23m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bd6e0dfb87d4c88aae988ed802695ce : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6647591de0b741de9d44402c891c95dd : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7e1e7bc8434494e8f7db3879a0c2059 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b57a2b3ab994661b53209b8cf0ab7f6 : SUCCESS in 55m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3de41da961234b629270e6c7806a1a0a : SUCCESS in 1h 07m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f86c6dd3ef8748999b8a8a19465fa44d : SUCCESS in 1h 07m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f2125c9cb6b4132b5d07bfe4d35a4d7 : SUCCESS in 59m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d26f8295f7154910a35c095b609dc6d6 : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce9f82207ab4065929bfe7ebfb62c3a : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9aae2146e1674acda4d42dbe44e66efe : SUCCESS in 18m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0abe071497284fba9db6c6e446d2ffae : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":5},{"id":"1b74f2250fe2f6a4e3d468b51c985beea04990d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:44:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"db30e5af7cc1f1429728f429cb3cd2be6bd286df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 22:27:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cceb641af3d44a04868f121f639b0475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5baf7cd968c47e7888378e410b009b5 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/1002aca740f147509999772b1a60b17d : SUCCESS in 56m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e513bcf9bac4e3399b9570827305d87 : SUCCESS in 1h 00m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/031627df2fd24cd6a31508bc1d911874 : SUCCESS in 1h 00m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/78d0c3a8cf0e402bb9a0d338de860a70 : POST_FAILURE in 39m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fb006452870147e3a89fc232e1e24d40 : POST_FAILURE in 35m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d74f8d7905e346aaa58e3eb2856f37f9 : SUCCESS in 18m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19d2ab0792324c8fb0812c524c380e87 : SUCCESS in 15m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b563ebb24464d819674aba3f13fa81c : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ca967318dbb8478083aeab157cbb2281 : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b1df9afea1f4318823cdaa495e03bf6 : SUCCESS in 22m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8eb5e58dadb74b81aeec7b3429173eb3 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9deca13249144276beed47bae20cd2c6 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/da8263cc86854ddfa898cff60a6e383b : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9e192efe9c1d41dea1a09fa6537e5acc : SUCCESS in 13m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b0a719e9f786454cb6590b7a9cef7f92 : SUCCESS in 20m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1dd42545c9b428f99ed9e5b33150893 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a3b021006b74a1085a37b7b2b9a9062 : SUCCESS in 7m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bab66e6b38ed447fb9e3b7dd892c3d41 : SUCCESS in 46m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6ef7edb10794ebd8f66cfd8ff2d1fa6 : SUCCESS in 1h 00m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0efd892d59e1493c93b4c899d5197067 : SUCCESS in 39m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e286b546d2d44a8b8172525317af6481 : SUCCESS in 41m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95488cc273eb4716ac8f4531c6188a3c : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a78b211fc2ca4261b9f36d13293c0b71 : SUCCESS in 5m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/909937f4779e4592b910655c1c602cae : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":5},{"id":"b5ab7e47cb73ca61c87eb678b3d9abf30c9c15a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 22:27:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ec230e596c931b69dd350247f7df1df9bdf25d81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 22:28:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee4545ec65db41be81eb3f7e31dda01f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40fd7c6b330c4114ad49fe007d1c1bac : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/079c5761317e4c758909aa3485635c67 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/63f63a7f3d934faba9ca1d37818b9bf7 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c5a3354fea345566010420801e8d27326920db85","revisions":{"71ebbcc3661580e594bdcf1e5df8c5083ab61015":{"kind":"REWORK","_number":1,"created":"2022-11-29 17:44:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/866051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/866051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/866051/1"}}},"commit":{"parents":[{"commit":"a1939cba037926400cd777059ee511481916cb08","subject":"proxy: add test for ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 17:36:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 17:43:53.000000000","tz":-480},"subject":"db: Declare slots for GreenDBConnection/GreenDBCursor","message":"db: Declare slots for GreenDBConnection/GreenDBCursor\n\nThis seems to help with some segfaults under py311.\n\nChange-Id: Ic2695e2e836da5607f4f5c016c660496e2821e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71ebbcc3661580e594bdcf1e5df8c5083ab61015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71ebbcc3661580e594bdcf1e5df8c5083ab61015"}]},"branch":"refs/heads/master"},"03eb2ac85a619e68976d30a6b4ef9560d4960f28":{"kind":"REWORK","_number":2,"created":"2022-11-29 19:02:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/866051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/866051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/866051/2"}}},"commit":{"parents":[{"commit":"a1939cba037926400cd777059ee511481916cb08","subject":"proxy: add test for ContainerController._GET_using_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1939cba037926400cd777059ee511481916cb08"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 17:36:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 19:01:31.000000000","tz":-480},"subject":"Fix DB tests on py311","message":"Fix DB tests on py311\n\nChange-Id: Ic2695e2e836da5607f4f5c016c660496e2821e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03eb2ac85a619e68976d30a6b4ef9560d4960f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03eb2ac85a619e68976d30a6b4ef9560d4960f28"}]},"branch":"refs/heads/master"},"90f8255f65e8c41b0b5079dd25aac0d7df9c858d":{"kind":"REWORK","_number":3,"created":"2022-12-09 18:44:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/866051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/866051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/866051/3"}}},"commit":{"parents":[{"commit":"745dfec35e50fb4e3fe5d16da431215bf88a835c","subject":"Merge \"s3api errors for unsupported headers x-delete-at, x-delete-after\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/745dfec35e50fb4e3fe5d16da431215bf88a835c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 17:36:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-07 22:04:51.000000000","tz":-480},"subject":"Green GreenDBConnection.execute","message":"Green GreenDBConnection.execute\n\nChange-Id: Ic2695e2e836da5607f4f5c016c660496e2821e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f8255f65e8c41b0b5079dd25aac0d7df9c858d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f8255f65e8c41b0b5079dd25aac0d7df9c858d"}]},"branch":"refs/heads/master"},"e59de0fdfd66543228dc6d9ee37dda0c2f6b8229":{"kind":"REWORK","_number":4,"created":"2023-06-20 23:42:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/866051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/866051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/866051/4"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 17:36:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-20 23:21:48.000000000","tz":-420},"subject":"Green GreenDBConnection.execute","message":"Green GreenDBConnection.execute\n\nChange-Id: Ic2695e2e836da5607f4f5c016c660496e2821e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e59de0fdfd66543228dc6d9ee37dda0c2f6b8229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e59de0fdfd66543228dc6d9ee37dda0c2f6b8229"}]},"branch":"refs/heads/master"},"c5a3354fea345566010420801e8d27326920db85":{"kind":"REWORK","_number":5,"created":"2023-06-21 01:16:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/866051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/866051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/866051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/866051/5"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 17:36:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 01:12:29.000000000","tz":-420},"subject":"Green GreenDBConnection.execute","message":"Green GreenDBConnection.execute\n\nChange-Id: Ic2695e2e836da5607f4f5c016c660496e2821e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5a3354fea345566010420801e8d27326920db85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5a3354fea345566010420801e8d27326920db85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
