)]}'
{"id":"openstack%2Fswift~866082","triplet_id":"openstack%2Fswift~master~I81e445234a1cbfddc374af449ab68e4c773bdde6","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-06-10 15:11:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I81e445234a1cbfddc374af449ab68e4c773bdde6","subject":"Add swift-ring-packer utility","status":"NEW","created":"2022-11-29 22:47:08.000000000","updated":"2026-04-20 23:15:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"20246f7d64042793efbc78cf2571d297be955cc8","_number":866082,"virtual_id_number":866082,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-20 23:14:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-29 23:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12864a22d3bdb5f9fad9b8883f52ea01943a21aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-29 22:47:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67ced02c23701381dee5b33146daa4e17fc1b7fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 23:52:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/055274967da2431e98e911300c3c3b8d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b98387bf20fa46a7b10204e7ae875f74 : SUCCESS in 1m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831a85ef7a774d2c8c7998d31a04836b : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/cbebcd7d51464db191109a61b93587e6 : SUCCESS in 54m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91773c1a6342419fa20835c71481b1e7 : SUCCESS in 48m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b40a5f1732d5496daa504bbbc21a18a8 : SUCCESS in 58m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0b8a8dc68514d7285dc4905c81f729c : FAILURE in 16m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57e8e433e0ae415687a012795def3c83 : FAILURE in 17m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/324e836b570c4207922defb1df45a7b1 : FAILURE in 17m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cdae9d2856164b1eaf3eebcb6b88248e : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1132ac3737f43f18560194ec0be61a1 : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a7bb14ac7f84b18b5906a7f8d41ca10 : FAILURE in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7038d53112004edb93e9a356ae9fbeac : FAILURE in 14m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b59633a223446f88ae9e029a4a6a328 : FAILURE in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6330cb16c524ba6a753f0b324fb6b51 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc1412ffa301400392bdf1a380402252 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78865b8076344167bc896918cb9a051c : SUCCESS in 15m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4ac005fee1a4409f99f4bb1ca605f77f : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0187b6e62cc84d83989499f765f6b079 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bae7ce366c634376a5c2cb639df2669f : SUCCESS in 13m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f90178e27eac4879858c27b13f7003d5 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/28785b723a694bf9a61e36b4d67db64a : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1747a80173dd4bdd82cedcae9b80d26e : SUCCESS in 6m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f541b5422a45467f82c6b4c7e86c984b : SUCCESS in 50m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8979137a29724701b500c8cb8eddd4d0 : SUCCESS in 54m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40d450fc430646b8a7ffeac28eb3b6d5 : SUCCESS in 45m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/073a7f7635ad416e8d3d55e9664a2986 : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a4ba158b28e47ae812017a8efd87739 : FAILURE in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/211b2c45545446dd9f73b25d4632e7b3 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e39e369de69420590328e016a9aad59 : FAILURE in 18m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8ecf6be44694ab48a589358785fab67 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4435e55a9898d4551b25f9f728944a9918c7dd61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 23:53:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/462ba770b049407986fcc6371cc72bd6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e157cf5f5c34e81bee5f0f2b829466d : FAILURE in 1h 03m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5cb8a0ba5bcd4145b20a2eab2752fa10 : FAILURE in 45m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/698463b659c84a2e8807d88faed16e58 : SUCCESS in 49m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/191f304afc2049418fb6b4ebb96013c5 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7c4eeed9cff9fbde6805ada786bd1c69ec93e480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-01 00:11:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2f245e54297674b72951c1301e291f1e145a3bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 01:21:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37214ec6fff142a3948ea8fd344879db\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/579cacb265ee4172bfe09aae6004e9b6 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d91bf17f6814072ba78eeac93a4bf7a : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c0aefca4d16f4f60aefbb030810263fa : SUCCESS in 50m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c262ebed0d14a94a7e63ab937538649 : SUCCESS in 56m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdd77f96483446ce8b2a80fc94e7c620 : SUCCESS in 1h 02m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a47a366592f45eab499000c3d726196 : FAILURE in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0469ec79caca41b0b7c914895862bc52 : FAILURE in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff67bc697d7948518c54bb3f74f93d9b : FAILURE in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4127756f77df4a6080dd2fe1e583cffd : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1501e2bd5244aba81782374ddcbe07f : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3502073109194b6b8ecd4efa73856097 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29149a3f678f4233a26a40d2b57af99f : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6bcd53c56c4689b9cb892eeac23800 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc69b515711a449f9068261acbf81f4a : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94c39ab38ad24269997c4ee2beb97d1b : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d9eb7790dcd460ea5af9fed3cec8000 : SUCCESS in 17m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f6f263e7c9044c728f4ffc8d0727125b : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ace731932bd4e6096ebdeb88cbe102d : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8887ab0ca16a4ce3ae5692763f0f7d31 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9623deee67f4972b0386e65250dc474 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/115ee71a4fce477fb18624d2ab41d2ef : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8103776358be453189de7387ebbde582 : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c9d65a025a874e23b989cab05f311729 : SUCCESS in 50m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d24b4e57f66344739a68b3dfca7fc2a4 : SUCCESS in 56m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64ccc023fb444c98a803cc99ea577b57 : SUCCESS in 42m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8549a2559eee4cadb61fe00d19fe0f8e : SUCCESS in 50m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c10dc1a61bb0487d948ac3d2cb3a0afd : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51813c467f2d4c0998457e6cd4700c69 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7718bb0102a414d809fbedd068b1488 : FAILURE in 19m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1749a111dbb64900abcdf3dc41a6b1d0 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1d059404e895c0d679d4539e771d7c82b27270d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 03:26:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a19f820254468fa05c9935c1378fec\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cd75f638ea34ffd94358df48d217b37 : SUCCESS in 1h 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1db260b2bd5d4eae9d16637fe9812034 : SUCCESS in 1h 15m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b61519ef8d447389102d52858991967 : SUCCESS in 1h 07m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0fb2d3b31fa44d39abf434b958baf55 : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3f9c04a8aaa4bf081bf21322ecb496ef964d63f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-10 15:11:54.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a05d66602543d54d448e5f9f9c344fe7c7b3cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-20 22:12:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"92d02dc7e759d56ea9c30624060de297fd6f5533","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-20 22:13:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d7e001bc8a9f0d964854de9e01e55b4b8163880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:14:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb41a491ae694782a1c5e1c09972ee1e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/79fef57a37594ebf80c7204bc1948023 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878846a091ef428280e6af8635710284 : SUCCESS in 5m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/93be839affd94344b787e29f2e2d5429 : SUCCESS in 29m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f51b3cc86bab434f86af85769d5dd201 : SUCCESS in 49m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2d25275e0f94dcdb65dc83bc479903c : SUCCESS in 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a03407e17e2742cab7ec983ddb5ac9c9 : SUCCESS in 34m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ca80fb3962942f09b3f8987a5d66412 : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3dd9481457148c8a358b6442b698657 : SUCCESS in 11m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4612920a388841b18fa4d51658568cac : SUCCESS in 12m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b87690656e234a7dacf7e6c8cfb76505 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e61e87a27784b48bb9efdfa00faa8d8 : SUCCESS in 26m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4a8f69622e284125baec902e14cdbcf8 : SUCCESS in 20m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b840b32b8d8446e38b5ef1e8c5bba08b : SUCCESS in 31m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/64ce0a61eb814ac591b44e0dff443391 : SUCCESS in 21m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d3d1cf612ec641f594fad0bca7760898 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ab363364f8548d2870c1c15f8dfff7e : SUCCESS in 8m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6a3737d8d8614b909aca2fa5da967c05 : SUCCESS in 9m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b086344ffdff4cccba79a158cae90d8a : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ec36b9b36014c369ffc547f0224d0cf : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5732f0aae1da462588c272095bc594b2 : SUCCESS in 4m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e31661cd4bd4b3b958179e481560ade : SUCCESS in 34m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bb0f00035e9453fbd182eb2ffb619e3 : SUCCESS in 56m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6a5668f3e544ead812e14e56ae14896 : SUCCESS in 49m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad7da42253d04dabbefef6933e30e998 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8471c750a69c4482985a401c14d25e8a : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/acc5176f694b4b709e0ab9df32033283 : SUCCESS in 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fab8458a64c847b584a821d5cd699434 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"20246f7d64042793efbc78cf2571d297be955cc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:15:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31d5364f0e8041e3b77920c71209de0b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e7b809041c641d2bc7a5c600d607b7e : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fbc5718f5ac67c9cca3838e18b889ad2c606d524","revisions":{"bddd0eb090b0db19cda5b10655962e50601a4adb":{"kind":"REWORK","_number":1,"created":"2022-11-29 22:47:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/866082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/866082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/866082/1"}}},"commit":{"parents":[{"commit":"93b38b8a3c8940012727af6bd0bd33f40751accd","subject":"Add swift-ring-info utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93b38b8a3c8940012727af6bd0bd33f40751accd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 22:46:46.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 22:46:46.000000000","tz":-480},"subject":"Add swift-ring-packer utility","message":"Add swift-ring-packer utility\n\nAdds ability to list, unpack, and repack v2 rings.\n\nChange-Id: I81e445234a1cbfddc374af449ab68e4c773bdde6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bddd0eb090b0db19cda5b10655962e50601a4adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bddd0eb090b0db19cda5b10655962e50601a4adb"}]},"branch":"refs/heads/master"},"c08aa1085e8d23684495587e26d2f176718841d5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-01 00:11:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/866082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/866082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/866082/2"}}},"commit":{"parents":[{"commit":"9427b96d81e7fa3b58ebba37ffeb084c62a6b3b0","subject":"Add swift-ring-info utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9427b96d81e7fa3b58ebba37ffeb084c62a6b3b0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 22:46:46.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-01 00:10:07.000000000","tz":-480},"subject":"Add swift-ring-packer utility","message":"Add swift-ring-packer utility\n\nAdds ability to list, unpack, and repack v2 rings.\n\nChange-Id: I81e445234a1cbfddc374af449ab68e4c773bdde6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c08aa1085e8d23684495587e26d2f176718841d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c08aa1085e8d23684495587e26d2f176718841d5"}]},"branch":"refs/heads/master"},"fbc5718f5ac67c9cca3838e18b889ad2c606d524":{"kind":"REWORK","_number":3,"created":"2026-04-20 22:12:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/866082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/866082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/866082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/866082/3"}}},"commit":{"parents":[{"commit":"ceefd82d5b110a6e7d6c26f29b0c130065860df3","subject":"AUTHORS/CHANGELOG for 2.37.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceefd82d5b110a6e7d6c26f29b0c130065860df3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-29 22:46:46.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-20 22:12:26.000000000","tz":-420},"subject":"Add swift-ring-packer utility","message":"Add swift-ring-packer utility\n\nAdds ability to list, unpack, and repack v2 rings.\n\nChange-Id: I81e445234a1cbfddc374af449ab68e4c773bdde6\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbc5718f5ac67c9cca3838e18b889ad2c606d524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbc5718f5ac67c9cca3838e18b889ad2c606d524"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
