)]}'
{"id":"openstack%2Fswift~866209","triplet_id":"openstack%2Fswift~master~I2e2150e1a4e91d610b2fc307635af31e8a1dbf01","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-28 20:25:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2e2150e1a4e91d610b2fc307635af31e8a1dbf01","subject":"ring-builder: Tolerate EOFError and KeyboardInterrupt on input()","status":"ABANDONED","created":"2022-11-30 19:59:01.000000000","updated":"2023-04-28 20:25:29.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"860d7396d93bc2e9b7cb08b00ced4f8f11d06f31","_number":866209,"virtual_id_number":866209,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-05 02:23:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 21:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 11:08:13.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-12-01 16:58:33.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"a032d0deef52dfee6d790b5d9ce212783afa25c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 19:59:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"568d4ef90cd6a9a8bd82f4880ed4f053830204d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 21:45:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e40d80e2f814a70857bdfb3aa63d71a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9df3b77c6f04150983e9a1e86c7e077 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e7a20ab4973a402cbe9758f47ab11c0e : SUCCESS in 49m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47af6be4965f4a009ba0ffacfc55b177 : SUCCESS in 41m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bb97c2256264384b5f08334efc9db89 : SUCCESS in 48m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f0b240188aa4445b385509b2ce3accc : FAILURE in 20m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddfccba257d346e28f46a6dd63a8972f : FAILURE in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7f49084fe584658bb7603bf88aeda18 : FAILURE in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e89b63f777954ade8f61860ee60d73e4 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/436feb6739f740fc824491883c58997f : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4a5a8fa2a52a48b58aab6313b1c86763 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0559b269540641ceb4311718e399c2ce : SUCCESS in 15m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9b2e74a2b224f1d96dca49505c9136e : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c08b7df0c2c46e8af741131e7844e4b : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adffa02af3a9481a99cc2fb4a83d3d03 : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e2b80c7278a413da3148f91689c3185 : SUCCESS in 18m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3ed4655df034451bb73d227d68b62025 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5612ff07c7b24157a3b520335919fd5e : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4dca615bf44b4342819b034beae93c48 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d10f747a7ec4949b1424ce9bdc53a99 : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb7195f167c94bdcacd19fba12b769bc : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/168bf3bc579e4edfb58a6ceb12d64967 : SUCCESS in 7m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/905c4e25ccf241328167e7b4fb628f08 : SUCCESS in 50m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/576475d125bb444bb22f024737583ae8 : SUCCESS in 1h 01m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b05c14b5b0d4bb1af399aa84595e554 : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c358b9e1b8d3471f9c63ec33c1b4ff52 : SUCCESS in 50m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b69946056fa4811ae63a050cad72f38 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f057b84b63e04561aa5cc015dc7048f4 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd30e3fd23ab41bbafa86872533f3347 : FAILURE in 20m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a558eebf256541f9a37aea63603d59d3 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d65bb1940129e3bfda38776a8282d0fbd29b22d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-30 22:12:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08b0b819ef0bdc20ebe6fdc08542079472dd8353","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-01 11:08:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8a09ba9f5a9d6d4c52e0dd41edd416c26e499ea","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-12-01 16:58:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adeff35a06789175922c3388f6f634c54a559603","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 01:21:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/887b4432a2f6408e8537f5d13cdc29d1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff67b3d6990c4c0caa36b5294acd1880 : FAILURE in 6m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8f4381c5b614ab49db7cefbb6f2f996 : SUCCESS in 1h 01m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/60a1b7951f0b4bbfa3fcb2488c419c31 : SUCCESS in 1h 23m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/df6a22bca4b44044b80d507745fa31e9 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"637485796a402b6c4e89c3a959d1201ca37b6413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 00:21:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbadd86f307f29e5635314237badecfe48ab2dc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 00:49:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d1f2810aff98c29b85e3c86868faf64841a1ee0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 02:08:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/625bce97c1414043a2b602124029e975\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/073e668a3822419ca0a51b9546c2577f : SUCCESS in 1h 16m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab702b76cc504a9aa03bbec036b2a836 : SUCCESS in 1h 06m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7890e3c1a6284d3998550cf927e58192 : SUCCESS in 1h 11m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ab0029254434fd895b801b2f5df94b3 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c946994fb7cdef0d35236c30f4b5e745319e8890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 02:23:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3958af0c794e40e3ab59b808fed4479a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d618b34d750b488a8e0fb6a7910d1ee6 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/256cad8660ca4b1c9a4f8ab90326c3cf : SUCCESS in 1h 00m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a6eed771063148dca603d402d56f3780 : SUCCESS in 53m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bd3a95ba4d648ae81872a4cb63782cb : SUCCESS in 57m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eed8cf36c608416eb505f7fbfe98244f : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00d81aa528dd48dea864bafb4500cd05 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36d7ee7580b642cfa03ab572f085c216 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68edfbf2414d4b80b0ca4e5a858b4810 : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b29266396a73481087da04872d72e6a4 : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57cd15f2964d4167bc21e2b64a467843 : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/930cc8c057564d0597a65cc43cab2e64 : SUCCESS in 15m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae384cc90ada43018dd667e3bc9d395f : SUCCESS in 16m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1fc18f73e9f44afe9d74b9bed8603495 : SUCCESS in 16m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be515b245af64315bdf558065a5b18d0 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d4d9612e5684286b229932e7f453901 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1789d0b94fae48debba31a9058507a6d : SUCCESS in 19m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6744a50180304f098edbf8c85e4f364e : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ffaaa9435662469ba44fd76609475bfc : SUCCESS in 18m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/723dce67e10044549bcc095b4e702119 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/273f431b4fcd4a08b480115d6be05f89 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d54ac717845844e9b184c54792a8f9db : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f7fbae8162b742a7bb6fa1c88f79c403 : SUCCESS in 7m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cb868157fa9248fca9f2ad3a6c4ed1ef : SUCCESS in 1h 21m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d3a1ca5dc9449eaa26aa7d2776ab880 : SUCCESS in 1h 26m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b4f990d6ba340f0b5780873edcd5f4d : SUCCESS in 48m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e3d663bf24540a7a742e123fcf395a9 : SUCCESS in 49m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f9aa0b9f86a45bc89cf411b6a915b9c : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53273c635235446e9a7ca3d0c10b8522 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd398eb6c25f4e57b7d002f3c03f4b07 : FAILURE in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b0499aa7b29431ab1f85e538999bb47 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":3},{"id":"58fb8d7d910674f68cf30abc09ed561d4463c6b6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-18 09:44:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"860d7396d93bc2e9b7cb08b00ced4f8f11d06f31","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 20:25:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"82721d8011f07979c7fdac899e3edcfbb66cf59a","revisions":{"44a08047d0d3a2ca2087312d2ea8d7f2f53d2499":{"kind":"REWORK","_number":1,"created":"2022-11-30 19:59:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/866209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/866209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/866209/1"}}},"commit":{"parents":[{"commit":"662ef280836178bfd1030cffbf686db14882a6f8","subject":"trivial: fix flakey account/test_auditor.py assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/662ef280836178bfd1030cffbf686db14882a6f8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-30 19:58:41.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-30 19:58:41.000000000","tz":-480},"subject":"Tolerate EOFError on input()","message":"Tolerate EOFError on input()\n\nChange-Id: I2e2150e1a4e91d610b2fc307635af31e8a1dbf01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44a08047d0d3a2ca2087312d2ea8d7f2f53d2499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44a08047d0d3a2ca2087312d2ea8d7f2f53d2499"}]},"branch":"refs/heads/master"},"475ab5a99eaf1ac9484772ab19f921758d136268":{"kind":"REWORK","_number":2,"created":"2023-01-05 00:21:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/866209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/866209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/866209/2"}}},"commit":{"parents":[{"commit":"f77172e2ad6fef4b226637b8ae099a6faf810978","subject":"CI: pin tox at the project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f77172e2ad6fef4b226637b8ae099a6faf810978"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-30 19:58:41.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-05 00:20:28.000000000","tz":-480},"subject":"ring-builder: Tolerate EOFError on input()","message":"ring-builder: Tolerate EOFError on input()\n\nChange-Id: I2e2150e1a4e91d610b2fc307635af31e8a1dbf01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/475ab5a99eaf1ac9484772ab19f921758d136268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/475ab5a99eaf1ac9484772ab19f921758d136268"}]},"branch":"refs/heads/master"},"82721d8011f07979c7fdac899e3edcfbb66cf59a":{"kind":"REWORK","_number":3,"created":"2023-01-05 00:49:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/866209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/866209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/866209/3"}}},"commit":{"parents":[{"commit":"f77172e2ad6fef4b226637b8ae099a6faf810978","subject":"CI: pin tox at the project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f77172e2ad6fef4b226637b8ae099a6faf810978"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-30 19:58:41.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-05 00:49:17.000000000","tz":-480},"subject":"ring-builder: Tolerate EOFError and KeyboardInterrupt on input()","message":"ring-builder: Tolerate EOFError and KeyboardInterrupt on input()\n\nChange-Id: I2e2150e1a4e91d610b2fc307635af31e8a1dbf01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82721d8011f07979c7fdac899e3edcfbb66cf59a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82721d8011f07979c7fdac899e3edcfbb66cf59a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
