)]}'
{"id":"openstack%2Fswift~866509","triplet_id":"openstack%2Fswift~master~I124afa825a3158eef7023abd5ec87fad54c0beab","project":"openstack/swift","branch":"master","topic":"mix-policy-mpu","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-10-03 22:27:52.000000000","reason":"Votes got outdated and were removed: Code-Review-1, Workflow-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I124afa825a3158eef7023abd5ec87fad54c0beab","subject":"fix backend policy leak","status":"NEW","created":"2022-12-03 01:15:32.000000000","updated":"2023-10-04 00:30:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":9,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"d3e3aa1914040e7390f4ffc5e4bf2ec8f9f119b2","_number":866509,"virtual_id_number":866509,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-03 23:34:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-03 02:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-03 02:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d4c7ec7bf9fa5f8e7471adbc5e804a2396cb9cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-12-03 01:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fd6ac1d8e908917328f7dc26ec66a1425b4b619","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-03 02:38:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7c1321dbb44b6fbd390af255ba915c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3cf4ddbd92954dbd8770c30ff6a467a9 : SUCCESS in 1h 08m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2004deb51d414453b79648c859d2ee6a : SUCCESS in 59m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3117ea5e82554694a30af31135428fc6 : SUCCESS in 1h 10m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/41a8dd2faaf44c4d9814d5d47cbdcc5e : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"55ce5e5afe21538b5ce3e2a29c301f9552f9bb8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-03 02:47:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/384cac2eeb3b490e895625fb2323d549\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7492dcb28d47149e502a1abeb92330 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/fddd7535ec16470cbea9c738a8465d6d : SUCCESS in 56m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e3ff8052adc14924be89e35ec917f02a : SUCCESS in 43m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e988709d40843dc81b04dee2957a64d : SUCCESS in 54m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93f1c0cb3a3f4e7681505fdae70f588a : FAILURE in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72ca30cfabab4a10af84c6a4c06d4d36 : FAILURE in 18m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b452e0d900f414d82998a8ac1e334d5 : FAILURE in 29m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1c0ef29438b46a7ba332ae8891bd96f : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e277d4c5ff7490cb0166f577fd6fa4c : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/441cb015b80c4cfb82f352fff48f05dc : SUCCESS in 15m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9918bc7c1a6046f59b3cd90980a08b11 : SUCCESS in 20m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed5e984b0bb8405bba9c2730eadeb79d : SUCCESS in 13m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2137afcbbe0f486a9fe897d412905f6d : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c087a3718d7347f4a4910f03d7464f52 : SUCCESS in 20m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d3854dbceaa743a7bf6683e58561eed1 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/372e6479f5824b8283472858d6c6b329 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6adc7e7bd5634067988b5c2a44e18c7e : SUCCESS in 14m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3339dec9ba1043ad8c1cf404c1038526 : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f694da0f468e4295bc451c7927c44c49 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e614b752083e4ecbaa9085bf4bd3fe44 : SUCCESS in 12m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3c0fa48568c845e3b51d111a6a0ce031 : SUCCESS in 6m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a92a62fd62014756b2308d9d30af38fe : FAILURE in 54m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9ab208878874292946d684753515dbf : FAILURE in 1h 08m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/421746d2cd31436183caa574b180c577 : SUCCESS in 39m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a312d6eb836641569fe88110d4221ed0 : SUCCESS in 42m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d547c5d43964cbca207ae8a1976f89f : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3435ff574534b51b631508dabf0a6a1 : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f945d47405e14df49b15a8df3e564043 : FAILURE in 30m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/614ff22f575c43c6ae3a5d29d4337ef1 : SUCCESS in 53m 59s\n\nWarning:\n  Comments left for invalid file test/probe/__init__.py","accounts_in_message":[],"_revision_number":1},{"id":"4dc23bc4cb8dd35b27ed519ca661e3f3bcb84243","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-20 21:38:33.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"eb56390f0e679be98c7ba88723821197643a3fc3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-31 18:34:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad3e284c736a9d7b40ff0fa9ff4a0f2a173babbd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-03 22:27:44.000000000","message":"Patch Set 1: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f4eaf2154e004866e561b6f86b385a996c5857a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-03 22:27:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d39155a491666fd817d30eb8b194ca49054ae411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 23:34:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8743fa7a804320aa574d91a7bb15fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d76e6c71d14f8ea252cee8f1180f2f : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/6bdd981097e041ba8c05c87a1fc19d8a : SUCCESS in 40m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e34dc90f4e5145449ce070448eccb604 : SUCCESS in 51m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5244fde729a4ec4bbafaf4353a6f52c : SUCCESS in 59m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/206d34bd788e46988b2bf06dcb608aa6 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08749f1e2d144eec8876c6f3688bfab2 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a2293abd9d649019092b285796b41a1 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9bc5432ec4184130a785a3986c0d3f2a : SUCCESS in 9m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a6d8e36b7b074b1f98634149364ca701 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3aea21104b9b4f238a540874996ebb05 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1dd89012d140c78c818a84e32f3f83 : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/688776459a084bcab5e8dbb38f588a51 : SUCCESS in 15m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3f993d102478417a979b11d95ae61852 : SUCCESS in 16m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/04c8dfbdc3fd439c804de63c579680b2 : SUCCESS in 11m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e0122ef59c014f84ab9fb17dee460ec3 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a32ade56a074f048a19ded7ccb8263b : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1bbc9dc358d49e89691e2ed654847d4 : SUCCESS in 19m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c704ea7474ad46458173418673acbf9d : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2934ebba6ca749b48978fd2e37ed1a09 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/602531351d474884b1532ff45a45ed3b : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60d69ec6aa034a0aa5c7341e1f629098 : SUCCESS in 8m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8fba18ff4624e55ba09b4d1fa2263c4 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd3f235bebad47e68158da4b069a2ac2 : SUCCESS in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37da4c17111c411eaa473409e68a1440 : SUCCESS in 49m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1221465f35b142ce9041c849c6791b31 : SUCCESS in 1h 01m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3096a29e4c5849adbbe22929f608e141 : SUCCESS in 39m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f91e508f525c409d9afa6e271a0842a6 : SUCCESS in 54m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/155fd13b702148539a17b44ccbace694 : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a73d40d98b34dbeb6e4598176d4c116 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/adfe887ac3124ca59735c22dbbb565e3 : SUCCESS in 19m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d843852ff474ce2afaa000a74c66f62 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d3e3aa1914040e7390f4ffc5e4bf2ec8f9f119b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 00:30:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0513a975b6c54bf2b6bee8d73041b1d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eda94fed7b149a08a6251cd2b49e9bc : SUCCESS in 1h 14m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67677ea95709432fb2f35645c0ec793b : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9c854c5ba1e42f3a24c41588264655a : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3a2af90a2854bfd9ade4101f2bda1c2 : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0be8377ec4d56c50f75553d1bda1e90b370a362","revisions":{"8d8e8e2112eee528eb6ba1eaf3a2139b2f788c48":{"kind":"REWORK","_number":1,"created":"2022-12-03 01:15:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/09/866509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/866509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/866509/1"}}},"commit":{"parents":[{"commit":"9d370c9057961e44122ba0d45b06b77abc20bd4e","subject":"wip: passing probe with bad behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d370c9057961e44122ba0d45b06b77abc20bd4e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-12-03 00:34:35.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-12-03 00:34:35.000000000","tz":-360},"subject":"fix backend policy leak","message":"fix backend policy leak\n\nChange-Id: I124afa825a3158eef7023abd5ec87fad54c0beab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d8e8e2112eee528eb6ba1eaf3a2139b2f788c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d8e8e2112eee528eb6ba1eaf3a2139b2f788c48"}]},"branch":"refs/heads/master"},"d0be8377ec4d56c50f75553d1bda1e90b370a362":{"kind":"REWORK","_number":2,"created":"2023-10-03 22:27:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/09/866509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/866509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/866509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/866509/2"}}},"commit":{"parents":[{"commit":"ef84de9a742637bb88db02883f896ee767d37622","subject":"Passing probe with bad behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef84de9a742637bb88db02883f896ee767d37622"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-12-03 00:34:35.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-10-03 22:21:29.000000000","tz":-300},"subject":"fix backend policy leak","message":"fix backend policy leak\n\nChange-Id: I124afa825a3158eef7023abd5ec87fad54c0beab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0be8377ec4d56c50f75553d1bda1e90b370a362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0be8377ec4d56c50f75553d1bda1e90b370a362"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
