)]}'
{"id":"openstack%2Fswift~866654","triplet_id":"openstack%2Fswift~stable%2Fwallaby~Ic099fdc3925bf99f8528c7aa763b651c19df1322","project":"openstack/swift","branch":"stable/wallaby","topic":"p-syslog-handler-stable/wallaby","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2022-12-19 22:14:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic099fdc3925bf99f8528c7aa763b651c19df1322","subject":"py36: Fix syslog fallback to UDP","status":"MERGED","created":"2022-12-06 14:01:26.000000000","updated":"2022-12-19 22:15:53.000000000","submitted":"2022-12-19 22:14:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866654-p-syslog-handler-stable/wallaby","cherry_pick_of_change":780939,"cherry_pick_of_patch_set":1,"meta_rev_id":"32a3ec089da663c4241be902a3244e9737739cd7","_number":866654,"virtual_id_number":866654,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 22:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-16 23:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-16 23:05:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 14:01:26.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-12-06 15:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 21:03:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-12-16 23:05:17.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"95496b86355afa39c505617b7f6faecc57b6342a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-12-06 14:01:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0c3073b78d3b213aa16942b748eb37e77941d9dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 15:03:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e910e133349443486550354560015d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21b622e118a490e8e9ad9d5ff94f275 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/54fd812f151a466488a45b7590282602 : SUCCESS in 59m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54cc0d0485314001b70b5aa3e777825f : SUCCESS in 55m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f60d15abbd045c1834c0493c7438672 : SUCCESS in 54m 49s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7d84bbf086a74113a691bc8934ee7f7a : RETRY_LIMIT in 4m 40s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3d9e756ab7a842809df5430da13437e5 : RETRY_LIMIT in 3m 43s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eab7137ad5a643869caa3f5e46971389 : RETRY_LIMIT in 4m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa96209b0e5245f28601ed7f3b28e2af : FAILURE in 3m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0039383eb534e2da5a32eee7fae2140 : FAILURE in 4m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b2a03e2e4eb42b0a4d7f0b191dcf5f9 : FAILURE in 4m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3469324472454531a658978cd2ed1948 : SUCCESS in 17m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45cbcaed675b4f4a9fcd263a91ea1607 : SUCCESS in 11m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0b348edb91cc497ca44a0720a8ba9357 : SUCCESS in 16m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d546f9fec914974a86ca7a2fcc16581 : SUCCESS in 14m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17dc498e41a14217b79b3c0ff4ae9974 : FAILURE in 4m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b32078fd00040818c241a966e4ff358 : FAILURE in 3m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/469565e1eb224f1a82536971114af059 : FAILURE in 3m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9213970c97ea4521b014333ef8a324ef : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fb6d7e2cfefd4c8cbbbd72b39b613e25 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c21c4f016bd04e75b9fb428bec436601 : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f52918cbdeab4e9c89b70760cfad394b : FAILURE in 5m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/abdf33b1080746f8bcde5affbd09de2e : FAILURE in 4m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0e62519a6934fcd915e4ca060500be0 : FAILURE in 6m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/940a6e70c8b841f6a0076bc60131fc17 : SUCCESS in 54m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/912f75a953114cdeb221acc5899084ad : SUCCESS in 39m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8252e90ccfa4395a6f6080d5250ca99 : SUCCESS in 37m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/262e5ca6ecc541e1bdfe92297b14eeb9 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad64768358b4dddbe9d7da2ee198a94 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8781b4cbba7747559a096afc6757c2fc : RETRY_LIMIT in 5m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faf05a9c2732409fac80d1f13f8d4506 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e060d1b512e95b60ac4b25414c3bea7ff135a5fb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-16 21:03:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad798804ceb68d64b697674514668ae54693f81d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 22:19:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f96ba0f236db4dce8792d7ae2be6d11a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1369e0159541469d7ba949c46b906f : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/f3ba12051ef54639a2c760f5481a63aa : SUCCESS in 1h 15m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19b7d36285954520b2206f9e337a75ff : SUCCESS in 45m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/072935db01944b7e87f403de99f1fa0d : SUCCESS in 1h 01m 44s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f8878f4507d0405e9fccada006f6c2cd : RETRY_LIMIT in 4m 33s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dedef08a79a94a909d765fd8c13fa355 : RETRY_LIMIT in 4m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b5d2889c92e64d7cb890996f8a6d43b3 : RETRY_LIMIT in 4m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec8c5ec07c9e48bfb0b676a8e100ed33 : FAILURE in 3m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7ae6a3a6b6b742a9b12428ab87ad9e32 : FAILURE in 3m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/087b4877009f43078bc3191296b0e0bd : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/acfe20d7042044f6be36e84648116a59 : SUCCESS in 13m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/df93b58eb96a4b3087bbddc647a0c314 : SUCCESS in 12m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f240ad745a3146229f632449572a30e1 : SUCCESS in 14m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a7a14cb64ce4246a7c1b91ddf76795f : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75dd93f8854c40548c4fe8d147c8aac5 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69389bb654b841f782c453a09428b77f : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e210d6f889754db89d3e6f3cd094e5d4 : SUCCESS in 15m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a25a04a793fa4e3ab3564522756af556 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d4f8ba02967f40d189bd63114b1b23d2 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f9f3072e4beb4d8f88179dab1e96d2ea : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3275c65cb2f84d32bb163be996b9761b : SUCCESS in 5m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d13700c91679486593781bfde3dcd9c5 : FAILURE in 10m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c62ea14ce1745999dff692b5d25482b : SUCCESS in 54m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/27b4ad9b4fad4d008336a7b3b2902ff4 : SUCCESS in 41m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c83acd5837bf4cbbbf8f4f5d00c2c3d3 : SUCCESS in 42m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f370ee299824b4886a4a7a77995447f : SUCCESS in 34m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/303044aded5445349f60e3856c725e8d : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f0310b74cf4b41932e7621f55c0f8c : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da3610a80c9c479fbfec945602c94ae5 : RETRY_LIMIT in 5m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e4fb30b056b40d5a26936d5bbf1437a : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1e2a35e6b376ef3f1843c04f5d9b71f2e945bfb0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-16 23:05:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99cd1b86f884c699b30c35d081151475a45b6581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 23:05:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00c227b1e35d6cfb114d728b053d41725e59fe4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-17 00:03:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb855f71eea04eaf82f690649714ec7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2ecd2944ab40b48a6c41cfa097731e : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/f6ed4b5299a740b39d4a8bf1f09ecd20 : SUCCESS in 46m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54fe47e51f7641dfadd8ac97cf5f6d88 : SUCCESS in 51m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfe1f9ff345f491a82d5316bf3e2236d : SUCCESS in 47m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b4d4353d6c004521bf9aadd8b5fbc3f8 : POST_FAILURE in 4m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8b515ff85800420f84eea165b69f028e : POST_FAILURE in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6cf4ce4919104fcc8ce5e338e0516657 : SUCCESS in 13m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09ef1d6acbf94b039b8efbefe8f6dac1 : SUCCESS in 13m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ab46ecf5e49471484617b0e812994f8 : SUCCESS in 12m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/46a4c5ed38ef4370b7a5344b550e5be1 : SUCCESS in 12m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff0084b613404075815a786c8c576afd : SUCCESS in 11m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c46fb8d7d45d4ef7a8f405479ff0b704 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e68eacf896504388900bc4da297fc48b : SUCCESS in 14m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70f5f8ff37fa47f2a5eb055dd61e3a79 : SUCCESS in 15m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1e8c7fff2deb48e9a59766b6cb0c898e : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e8b9ab1510bc4eed968884fcca38a795 : FAILURE in 17m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cdfa3f7abd764a41bfe27d477fc57ba0 : SUCCESS in 16m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9180b494b1f24c4397b8589c166206bc : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c649b925838a4b15948e29214e389060 : SUCCESS in 44m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/38a457d3d1334474b126d7f6929adedd : SUCCESS in 49m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25bd95f3bd9c4b578e54b141601cd28d : SUCCESS in 32m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d79a2f9cb1bd4767b054ad26fe080b77 : SUCCESS in 40m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7ac893dd3b1403790352b5f8490293d : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049f4eb0ed0c4336879a1af69a5462b3 : SUCCESS in 5m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8753b14d30d4d988ad5b3225ef8858c : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e2dad150663f695c1915074c5570156d623395cf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-19 20:08:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ceb3ebe4c6019f404f9bcc11ba5830041afdd3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 21:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b785994173a74da093b2cad63807c6aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e3489a63e34059847899933c36eca1 : SUCCESS in 7m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/005e824366f346d4b0b8277eb2b4783d : SUCCESS in 1h 01m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f331e967bc97463eafe9bec0f9e02ff2 : SUCCESS in 56m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e92ccb5de52b4ef2a4f335a7ebedde7a : SUCCESS in 1h 05m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4b287aa157a64c77bc62b27c4a18fe90 : FAILURE in 19m 18s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9d07045996104813962691731a351bf1 : FAILURE in 19m 49s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fb5a74f882c144998610f39eed27be3b : FAILURE in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1cafbbe8317e424fa0add80880b36cd7 : FAILURE in 2m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89ce97e5a97949d58a47fff7d4bb6673 : FAILURE in 2m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d5f07af601d474e9f4ee26b63ab39c2 : SUCCESS in 11m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18d323e1f777413bacd53169737c72ec : SUCCESS in 13m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b53458b7a7e4e9b8a78f34d9fa80fd9 : SUCCESS in 11m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7073d73a12414db7b4578b8832b50942 : SUCCESS in 11m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0cf497723a2c4258bf0ca279aa01b8cb : SUCCESS in 12m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f9ba2a2cb254609a8bc49de5a393a21 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d540818bada84e339b2a02f3e16ab7f8 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/74a520e7fc5040b8a5ae8cffbf69b69d : SUCCESS in 14m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2d9e1f87f1d94a1d829393f3e55a5e86 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1c8c9af7fbd4f1a8da7964addca5f17 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c05a68c0a2541a6918e4b81946b1f57 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd1ecd0e2f63445cb0956e2806392852 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0bcbf93f7fbb41bfbf63e34912598849 : FAILURE in 7m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59a4a485883248c18bf596e7cff4f736 : SUCCESS in 51m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b07d8a52e79f4f8a8dd0b612e9498414 : SUCCESS in 59m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d7f8c092f3940fca5610c2c3a869e7d : SUCCESS in 37m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22cae5a0f7934ed58bfb850ed11435c4 : SUCCESS in 28m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf08a8f7aa9d4624ad3d4a6c16d6cc4f : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc1e3af94fa4f8ca888390c4d1536d2 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3287c36869e642e798d33f452d42449a : RETRY_LIMIT in 5m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe99494670d9427facd617b6839dbf41 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"36edf86a198ddab6f57fefc1d1a0556f63b0600f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 21:22:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc93e8c6c34038f932518811c57925d083068b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 22:14:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c42eb0ac724fb19e0d9766ba9526b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5684e49f2fcd4f2abb95669b88153ba8 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/dbc8d9dd755a4068ba99a6b8a0040589 : SUCCESS in 51m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fb786af28fe44b38b23d1bf2c47721e : SUCCESS in 36m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f3baedff6c5432c83e1d6790aeb55ce : SUCCESS in 47m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4885ef439c3343a5b7fee235f74bdc87 : POST_FAILURE in 5m 04s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3ceb82617db743299a1fcaefc93ad998 : POST_FAILURE in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/beafb97f6586402b8f8e306d3a9287a6 : SUCCESS in 11m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0ebe47673f94fe09688757e032565cd : SUCCESS in 11m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/361af781ed0148f9bc8688ed20584130 : SUCCESS in 15m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e60485f45374ad3ba8ff9af6b78d9f9 : SUCCESS in 12m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b5cb5ea341443dc8f47ffcaf948a100 : SUCCESS in 12m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f38343e8c16408dae18aea1a1c4e307 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ebf74d67934647ada3ad6a430dff5e1b : SUCCESS in 13m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f92f786b74246b7a208f3b0c84f9c65 : SUCCESS in 15m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5a78913b00d54a4e9c6b2b6299440204 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/569722c1c82d41ec99af3877c838a84c : SUCCESS in 15m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/527a0fe811484aa7ab91c298d500ccfe : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83bd60c87af347e59cee7d3f032f867f : SUCCESS in 5m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/957562033bde4e77963dba27fc7b6356 : SUCCESS in 35m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a16d4de8c17422a874fff545697d065 : SUCCESS in 45m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1419889cb1d24164b28f7a5d22c509b6 : SUCCESS in 41m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/517ce95c235e4126a30889fda9be95c0 : SUCCESS in 28m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/735a0a51903c427f846f1fa6c83d50e1 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b43fdc133344427a363a2d1115fe5c1 : SUCCESS in 4m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c03da742101a460daac45f5edd7a35dd : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"189963874a06da51de7af116e5fd75b5ebb31ccf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 22:14:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"32a3ec089da663c4241be902a3244e9737739cd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 22:15:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2707ac2d8947c182fc803a2a5c4bbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e457af6bda664fa788c31ad7775387ed : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1581585f28564a36a62c228704b66481 : FAILURE in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/070d5ca5cd3b4255b9d7666259ed9355 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73c53dd77ac2d8c4a4e898ba8a15a4f373ac20da","revisions":{"73c53dd77ac2d8c4a4e898ba8a15a4f373ac20da":{"kind":"REWORK","_number":1,"created":"2022-12-06 14:01:26.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/54/866654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/866654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/866654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/866654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/866654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/866654/1"}}},"commit":{"parents":[{"commit":"6a1a8ce14be8b4625eaf4f2b4d18ddfbdbbb4289","subject":"Stop partial()ing hashlib.new","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a1a8ce14be8b4625eaf4f2b4d18ddfbdbbb4289"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-16 16:07:12.000000000","tz":0},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2022-12-06 14:01:26.000000000","tz":0},"subject":"py36: Fix syslog fallback to UDP","message":"py36: Fix syslog fallback to UDP\n\nWith py2 the SysLogHandler would raise an error if it could not\nconnect to the configured log_address socket. This prompted the\nutils.get_logger() method to fallback to using UDP. With py3 an error\nis not raised [1], so the fallback never happens. In that scenario,\nexceptions are raised when swift services are started and attempt to\nlog.\n\nThis patch modifies get_logger() to test if the log_address is a socket\nbefore instantiating a SysLogHandler and fallback to UDP if not a socket.\n\n[1] https://github.com/python/cpython/blob/master/Lib/logging/handlers.py#L833-L842\n\nChange-Id: Ic099fdc3925bf99f8528c7aa763b651c19df1322\n(cherry picked from commit c7d56fa9c57a18473da166254c01f4abc44c7ee3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c53dd77ac2d8c4a4e898ba8a15a4f373ac20da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c53dd77ac2d8c4a4e898ba8a15a4f373ac20da"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
