)]}'
{"id":"openstack%2Fswift~867121","triplet_id":"openstack%2Fswift~master~I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf","project":"openstack/swift","branch":"master","topic":"test_probe","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-01-05 04:14:51.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-12-14 11:35:15.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf","subject":"Sharder: add a new probe test for the case of slow parent sharding.","status":"MERGED","created":"2022-12-09 16:05:44.000000000","updated":"2023-01-05 04:16:06.000000000","submitted":"2023-01-05 04:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867121-test_probe","meta_rev_id":"9fabacecac4f17c32cc1216241f4b3b156b40a06","_number":867121,"virtual_id_number":867121,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-05 04:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2023-01-03 21:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-05 02:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-05 02:49:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 17:21:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 13:59:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-12-14 06:37:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-12-14 17:20:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-01-03 18:45:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-01-05 02:49:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"28986f9d83278b73dacffecc1355b8d551d8d8c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-09 16:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d2a0bfd0ce29e85c4164d320dc83b7f88371c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-09 16:15:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0f8e04ce719be6a817189821720cbe2249d1368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 17:21:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b8c19d83ad4591a2fb2604c4779e05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1eec0846bf44b5da41e5c2d73ce95c7 : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/152666044b5441d8ae2fd7834740b89e : SUCCESS in 58m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5743cc7018742bfb1a7560c4cbbecbe : SUCCESS in 1h 04m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca8732a3b04450990ea3ff50fd980d4 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"89942d1a533619780342f865ee22b2ab88cd0e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 17:43:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bb04c18627142ce97513d9fa12f7ead\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8a48f04805843dd8ffb9262680056bf : SUCCESS in 43m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e0e9d4fdea8450a9a55393eb913556c : SUCCESS in 54m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9cd66d97f504c55b0056e9e6521a118 : SUCCESS in 1h 26m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5160913512ff45b69f061c78021f5ea6 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8c5e335c02dce6077e7894717a91bc6159724385","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-12 13:59:41.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d11134f9a264ec7268097f8e0a686a55caf4f190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-14 06:34:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8cecbfcb7d8f4bac8d34ab29fbf3fc8413777635","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-14 06:37:57.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"129eff6aa94cf3f57252528230e6fd2ef2b9ff92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-14 06:38:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dd09a6b5f47da56607136c9dbbc96b9bfadf60f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:55:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83240f304a764f6287b03e7a820a6a89\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f7d4a0f0d6e4adc9821e590e3448ed5 : SUCCESS in 1h 14m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3059ae8c0686467facb02af80687f18b : SUCCESS in 47m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/41f5f32e3b0c452cb397ba3f7aed2e84 : SUCCESS in 59m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2f2bb9c0df946bbac98ac65c60efc91 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":4},{"id":"de1dd3847ebde0267ff66357f0e2cc5493d4f8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 08:11:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f93032de3f5949db8ebd247e97b677f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb35bb85cdb4c82a79e098c1ae54093 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1cd6bdcf55cd4ec98d47312c5a52ce3f : SUCCESS in 1h 13m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab288ac87ae7495eab533e607300f1eb : SUCCESS in 1h 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0b17d873a848deb0be4ac9a59a081c : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"850e570dbce3b38bd6b0da3d3311933b8fefe21b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-14 11:35:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11d3094826b9532debb36a0a518b8cd3d9ee1be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 11:35:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2864bb49b0cf118c80f0ba525b13677544b38a64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 13:23:39.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13c3957ac0784c1a907b4f0e57b42dfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad460dc491f24298873fea90880e2da9 : SUCCESS in 9m 21s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dc9cecc655a445fca85e3ae3bd97af15 : FAILURE in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b8bf29ed40434bcfab0c630e99fbee45 : SUCCESS in 17m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52925ff61dbd42faa2ad4e4b0f24d0d2 : SUCCESS in 13m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/89f29324264d45a8a7750d8e0ef8e7b9 : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21788f01b9dd4dbb8fb74d952f72804d : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/58ea04310c6d44479dae2e8cd6b9f55b : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c612126c5b1f4d5daacfb09520cca088 : SUCCESS in 15m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c72c9cbaa9904af9b89e24af00d90587 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/644e5aac87244c808c01e91c40c9dc92 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/058d15b52f034356afc694017ea6ba12 : SUCCESS in 18m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3dcc73b6825492cbab500021a9b8bbf : SUCCESS in 6m 31s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73997e4df65446ce898aef6920ee99d0 : SUCCESS in 8m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d85a9dd93ffc4fd996f0e76b641b8425 : SUCCESS in 1h 25m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e06263aaf1543dbb7e6710059b2d1ff : SUCCESS in 1h 26m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4618303b5141d99ad2b4c35d34b3c6 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":4},{"id":"cb5eaa2dda7d1023dde38de45c87fd830418d464","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-14 17:20:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf8472385dc98fa7ab6846e4cb063966ce1bd035","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 18:19:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b503bb71a2f476cbef553281a4c9c7e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a481d6b28b74e328210cefcf58f41a9 : SUCCESS in 57m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5515706cddf0447881d5547c3f04fafe : SUCCESS in 42m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3f506fb114e45d893e497650fb73b9e : SUCCESS in 42m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27afd62c4c3049919d78c2805f0805a8 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":4},{"id":"f400caaff5d398b93f4dfc9f0687a9b55306dc89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 18:51:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2dd11725650423bb10602c8b9be02f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deefdedc0d2747858fd7551537571567 : SUCCESS in 9m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35cf82723617490bbd5fda1eb021b7d6 : SUCCESS in 1h 12m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5eb9a6f513d54966aae254a366acf601 : FAILURE in 1h 26m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4729037d6f8489d86181dd16564eade : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"6c4bd5d420b27bd18738a0a214c1fa25be5e9e94","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-14 21:25:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e86baff4882604312af3bacb4fbd83ff90095ade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 22:29:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094f9ea4039a47ecac62e8d658b7fcf0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c87234497f34e869757cac2fb142f90 : SUCCESS in 1h 02m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e19616ded2d489dbd44fbb69398f0ac : SUCCESS in 48m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/09fc9a822e004546b1270521d2e9e665 : SUCCESS in 54m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c569eeafe004f9c89abebe87d846ef5 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":4},{"id":"001fa36c7d41a4e8cecf68853a5de8eec225b0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 22:46:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6346ac1d5b8c43a1974f190c1f90c7fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1303d9ee27e14dbe92a844b78c8f4542 : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51255df5193f4484a8ca7ac7381fb9f8 : SUCCESS in 1h 13m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c371fd4eaec94a2dbd92931eb4d5a9a4 : SUCCESS in 1h 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6db1565b5a741fbae037f0253c98762 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":4},{"id":"785ec84b346acfbf68e14f425778d52b9e5ba810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 22:47:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"626fbb1781193c75134051d25784f6b82a5bd632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 00:26:03.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3baf028d8fca4c7084215531cc21ea6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c11088d1714ebf8a77646eac1e4ff8 : SUCCESS in 10m 48s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1cc8feeddba54beea576c8a242eb6dd0 : SUCCESS in 17m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6db07a9d78d049a3bea460b65cd8b96b : SUCCESS in 16m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a48ce39890c40388e26e8064c4197f0 : SUCCESS in 16m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a5f64e4bb0e8421b8011c599356f4bb3 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a332679a39ad42669ab87476b56d0976 : SUCCESS in 20m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82c72c06d62646bdb20e9dedfaf3f22d : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffddb1aeefa1408fb937633aabfaee32 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d4f3edb4205641f7983b56c5bfe8bbcd : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7915674fb3cf4d139ca6cb53a4ff8cdd : SUCCESS in 12m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/182c45647e9342ae9d9411c4bae19cba : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7140956cba940e6b477ae3a575f2e01 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/501317a6deee4fc085d6dd919559806e : SUCCESS in 7m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/edd983052e614d2798cf9917bc59dbfa : SUCCESS in 1h 35m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/10893c3b64a849558508139a466c4b6c : FAILURE in 1h 06m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c56d07c2374957b5b922927855d45d : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":4},{"id":"a02b512b0bd88a95cfb84386db2102d6e7244bed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-15 09:50:40.000000000","message":"Patch Set 4: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"22cecf2b947e5150fa328e901bf33d303e274683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-20 07:41:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a3587725352aa850ab0af789dad3da3bfee6692","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-20 07:43:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71aa40bbe6d2b9b67622d18f1f372efd873eac4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 08:53:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0df170f04f548be9d29e2454e10504f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8877fbcd455d4301b6af612206eec9ea : SUCCESS in 1h 09m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8327d65857a94dceb2c41af29f049f30 : SUCCESS in 58m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/34d2e2261eef4e43a6a97973ce110152 : SUCCESS in 1h 03m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/20ba1a647a4b46fc80dc0e3f7184a094 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":5},{"id":"855db06a32d9a54ee4e37b316591bb4630b37353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 09:24:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5900bdf97d004787be0ac3dbadd1767e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70fe688cff048339491231a2be7169b : SUCCESS in 7m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eacf223571934ac08b429afad563d159 : FAILURE in 1h 19m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cefa7284eea4e42b089dca827b1efcd : SUCCESS in 1h 24m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451fca9dcd6445d0b6f592c0f827c438 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"a561300018303371436b47aa73af4fe0221249ef","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-20 10:02:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43687e42f2c0fc4ba8c56f4d7943a978fbcf95f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:35:50.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca1fc77ec1a40fcba08290b1f1d8d12\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/098424ebbcc944b995d989886072ca77 : SUCCESS in 1h 06m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c424f3d31e904e519fe950b78932dfd8 : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dacc2b3bec6a48b1aade6216c6beebd2 : SUCCESS in 1h 16m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd237ab628a543c98f6d2adf0e135f74 : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":5},{"id":"afa7e3e48918b950644c18e3c4f3821775de6aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:37:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e44328cf102847d69f0005ea6f610c05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7cbdc045c9e4cdfbb55f7b33335b228 : SUCCESS in 8m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6375a167bc04d67a81a758379f49e8c : SUCCESS in 1h 25m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/83f25212fd0442ffa76e5fb3776469f5 : SUCCESS in 1h 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1434740a2674adeb534f07b82613e68 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5b1b359d5f90fdf58d89058bcf27900a7be0796d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-20 18:17:56.000000000","message":"Patch Set 5: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4d417926b3b0e481f4a828e17758a963f795de93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-21 06:06:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9c4925b973be4e81330e9b6e8310a0e4ca23cd0e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-12-21 06:07:41.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2877de4017b08e7242b7a073b2ead4da40a0bcac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 07:32:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59f411dcc6a248ddabd1da7589f28cae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba9fc93e44b45afa765b2252c7eba8d : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d5776996c3fa4eb7aad75d6252728009 : SUCCESS in 1h 19m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5db1c67ff3b48638c578a7f40277358 : SUCCESS in 1h 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fdea4beb09c417f9f099542c5dd9a98 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6},{"id":"9cc9e676d1e0d368fe833485c8be939c948a08bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 07:50:01.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/093ba351c6204c53aff3cf981d9140e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1186f894b3a8457bb788ebca260eb521 : SUCCESS in 1h 27m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/464f54ab86d849bca0e08fc9df778c07 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7f77c628a1548a5bfd9eaea4dc83cc3 : SUCCESS in 1h 30m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/838e55604e5041df89242f34c04129de : SUCCESS in 1h 29m 42s","accounts_in_message":[],"_revision_number":6},{"id":"660350be6d2ff1cd7fb07deb9771784483cb8a76","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-03 18:45:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ae7384cb7dcc1a4ed90be03f0ff1aef496cfbbfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 18:45:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"306e6a9ac09d5e9e54daabc43a800a07107b2c12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 20:52:50.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4aa52fec7a64b858996f029dfa96f00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf94a4a10c254027b782ce6c9ea02a0f : FAILURE in 4m 25s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28f136efb6624fe6b2f38bc62dce02ea : SUCCESS in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a05502720acf44a0bf85dfa03cf1a563 : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/edab8e11839d460f97aea746d95250f9 : FAILURE in 4m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/23c849bb7b28406ba43e1d7dd2b69107 : FAILURE in 6m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c208329caf5c438e9bf32a11f887d1bd : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb97d39d0bd4480dba343bb25588937b : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0e9419f618fb4194a5c70c78f6b2467a : SUCCESS in 15m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0163fcdf1c0048ff8fd186e5be5c33e5 : FAILURE in 6m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ddc2b3fe3ea746848d831931a0e26314 : FAILURE in 3m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1232769ae334227bc206fd656271121 : FAILURE in 3m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bab2670951ef42fb824d8c324b1a5673 : SUCCESS in 7m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5fa83403f1904991b4bc4be1922101d4 : SUCCESS in 7m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3bbda2c773684f8c803c0589b24234cd : SUCCESS in 1h 14m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/32a0ffe5583743b28ab1d2ee001ae75b : TIMED_OUT in 2h 00m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6bc68a7fa340e0a6986af16667fd19 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":6},{"id":"0fcc25bcc9807393f63afde51da005722fa19f4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-03 21:03:01.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7d8981a0aef63e6255677ac7d52c42a3532a3d5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 21:11:51.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/521e2fadd3264619bafc152a8db93d7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc153da20923400d89dfa2200c64f370 : FAILURE in 5m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44c8ac405dbb4ebb874d0e567cf310d2 : FAILURE in 6m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/37aad989b35647d0987f6c71aa6f13b2 : FAILURE in 5m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/492299d320f74765b4b0c335ed7b289e : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":7},{"id":"01c5468a49aa954f0a919cdfc7b0355bff7d0935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 23:08:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4e32451237144b6a46e41445527551a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46cd8ad58174647870f4f04b9d21f5d : FAILURE in 4m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eb04fa87bfad40a1ab813c2d4dc8834d : SUCCESS in 1h 00m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c03cd87dfc654f0c8bc2333039fc033a : TIMED_OUT in 2h 00m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8fbbdd183dc48f0ac8f5d78c2dcc69d : FAILURE in 2m 57s","accounts_in_message":[],"_revision_number":7},{"id":"2b3196e5605e1a00a8e6e8b606992f786c527dfd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 00:23:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"533e68a0670df39b7a294d36f2847a85c599f558","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 01:44:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3347025142f4572af9a60020cc0e611\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cc67c5318504b56b9087d7c33aec0e9 : SUCCESS in 33m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/198717e2a65141bba495392688c0a728 : SUCCESS in 32m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c7e2acbca5641119a101645a549660b : SUCCESS in 1h 19m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c46c0a28dff4e439161bda18f5cccd0 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":7},{"id":"38f01fdc0acd3a459f778154568e5fc06dd8a1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 02:10:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/457d757d54714bdd85115860ccffb90c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7607656acdc4ce5a18992fd3e46bcb2 : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8fb51574a478468eb3847fe62b81dfa5 : SUCCESS in 1h 42m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/368f8622a6e64d59ac612d6159813ba6 : SUCCESS in 1h 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0612f97c8e1c45a0929ec54e7a09585d : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":7},{"id":"93e39c4efba11dcaa4edb420e4503b3e15c59376","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 02:49:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7366497bbd0eb6c2533182db6a6e8302fa48ebd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 02:49:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f2a0de7d0541eb62a56a32bcfe63fc4c12afcc2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 04:14:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2c1cfb289c54999865704182b5b61a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9745f708b88450583f2224b97cc781e : SUCCESS in 12m 06s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd50cb6003844535a7c19579e36be281 : SUCCESS in 31m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b2dceb79074950938d9b4f5758ab3a : SUCCESS in 33m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/18e84dc0820e45b9a141d23bec4a1bb4 : SUCCESS in 14m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8bfc694325f44ba79c005892d5bce52b : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f7d1ebe283fa479c8b33df92c5934522 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab3a3593d2584b279a3847f5d92aa4b1 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a66174df863445028fcfdb28fc6d9c9e : SUCCESS in 15m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3abdba580f784b5a8fb02e08305da64b : SUCCESS in 13m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d4c9ef3bcda4a47889f86e42faefdfe : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2b117b91091f43a0947f4a63946a5909 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dbeb5f0928214e7bb53e9ae69e74b65f : SUCCESS in 6m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d05b779b5a14652b96cc8d541c1c7e8 : SUCCESS in 9m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b151c2473d14cb88303389682dad247 : SUCCESS in 1h 17m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8dd7a443f4a849f3bd8dac24eb84d2d3 : SUCCESS in 1h 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a204a054dc2846b19f742a09bc9d82ad : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":7},{"id":"eac8664513564944e3877c7bf5a118a600189674","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 04:14:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"9fabacecac4f17c32cc1216241f4b3b156b40a06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 04:16:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0fa0c305764aafbc0378d46536793f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26d1a04c0274c1f820fa837ea1969de : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/348b8d6c056344ad853086c95a3d5c20 : FAILURE in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/746d9d6ba173496e8cf4550ab46259c0 : FAILURE in 40s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ec95047339bbf0edf8227a882c14cd8247bc2170","revisions":{"13ff986bd24cb969d6ca4bb06efce9248fc87447":{"kind":"REWORK","_number":1,"created":"2022-12-09 16:05:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/1"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 16:04:57.000000000","tz":-480},"subject":"A new probe test for sharder.","message":"A new probe test for sharder.\n\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13ff986bd24cb969d6ca4bb06efce9248fc87447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13ff986bd24cb969d6ca4bb06efce9248fc87447"}]},"branch":"refs/heads/master"},"38a90371ac746d329753026207a74991863793e4":{"kind":"REWORK","_number":2,"created":"2022-12-09 16:15:16.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/2"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 16:15:08.000000000","tz":-480},"subject":"A new probe test for sharder.","message":"A new probe test for sharder.\n\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a90371ac746d329753026207a74991863793e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a90371ac746d329753026207a74991863793e4"}]},"branch":"refs/heads/master"},"357749bb1341a041a8afb52c491f893405891cfb":{"kind":"REWORK","_number":3,"created":"2022-12-14 06:34:42.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/3"}}},"commit":{"parents":[{"commit":"9dc12a89029526f0ac8cd4aa30e121e4fdac1641","subject":"Merge \"CI: Add py310 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc12a89029526f0ac8cd4aa30e121e4fdac1641"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-14 06:33:53.000000000","tz":-480},"subject":"Sharder: add a new probe test for the case of slow parent sharding.","message":"Sharder: add a new probe test for the case of slow parent sharding.\n\nProbe test to produce a scenario where a parent container is stuck\nat sharding because of a gap in shard ranges. And the gap is caused\n by deleted child shard range which finishes sharding before its\nparent does.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/357749bb1341a041a8afb52c491f893405891cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/357749bb1341a041a8afb52c491f893405891cfb"}]},"branch":"refs/heads/master"},"a941e6b5a4191a3e7a4a19ee5c7e5cee2107df97":{"kind":"REWORK","_number":4,"created":"2022-12-14 06:38:48.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/4"}}},"commit":{"parents":[{"commit":"9dc12a89029526f0ac8cd4aa30e121e4fdac1641","subject":"Merge \"CI: Add py310 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc12a89029526f0ac8cd4aa30e121e4fdac1641"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-14 06:38:41.000000000","tz":-480},"subject":"Sharder: add a new probe test for the case of slow parent sharding.","message":"Sharder: add a new probe test for the case of slow parent sharding.\n\nProbe test to produce a scenario where a parent container is stuck\nat sharding because of a gap in shard ranges. And the gap is caused\n by deleted child shard range which finishes sharding before its\nparent does.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a941e6b5a4191a3e7a4a19ee5c7e5cee2107df97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a941e6b5a4191a3e7a4a19ee5c7e5cee2107df97"}]},"branch":"refs/heads/master"},"2a2d4f7c6d5727db7a8423d190967ae519817b1c":{"kind":"REWORK","_number":5,"created":"2022-12-20 07:41:39.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/5"}}},"commit":{"parents":[{"commit":"9dc12a89029526f0ac8cd4aa30e121e4fdac1641","subject":"Merge \"CI: Add py310 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc12a89029526f0ac8cd4aa30e121e4fdac1641"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-20 07:16:19.000000000","tz":-480},"subject":"Sharder: add a new probe test for the case of slow parent sharding.","message":"Sharder: add a new probe test for the case of slow parent sharding.\n\nProbe test to produce a scenario where a parent container is stuck\nat sharding because of a gap in shard ranges. And the gap is caused\n by deleted child shard range which finishes sharding before its\nparent does.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a2d4f7c6d5727db7a8423d190967ae519817b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a2d4f7c6d5727db7a8423d190967ae519817b1c"}]},"branch":"refs/heads/master"},"3c4b85a48ead5e2cf1f2e225eac12abae11f3ace":{"kind":"REWORK","_number":6,"created":"2022-12-21 06:06:54.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/6"}}},"commit":{"parents":[{"commit":"9dc12a89029526f0ac8cd4aa30e121e4fdac1641","subject":"Merge \"CI: Add py310 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc12a89029526f0ac8cd4aa30e121e4fdac1641"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-21 06:06:46.000000000","tz":-480},"subject":"Sharder: add a new probe test for the case of slow parent sharding.","message":"Sharder: add a new probe test for the case of slow parent sharding.\n\nProbe test to produce a scenario where a parent container is stuck\nat sharding because of a gap in shard ranges. And the gap is caused\n by deleted child shard range which finishes sharding before its\nparent does.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4b85a48ead5e2cf1f2e225eac12abae11f3ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4b85a48ead5e2cf1f2e225eac12abae11f3ace"}]},"branch":"refs/heads/master"},"ec95047339bbf0edf8227a882c14cd8247bc2170":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-01-03 21:03:01.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/21/867121/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/867121/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/867121/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/867121/7"}}},"commit":{"parents":[{"commit":"2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2","subject":"CI: Fix our usage of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-12-09 07:36:12.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-03 21:02:03.000000000","tz":-480},"subject":"Sharder: add a new probe test for the case of slow parent sharding.","message":"Sharder: add a new probe test for the case of slow parent sharding.\n\nProbe test to produce a scenario where a parent container is stuck\nat sharding because of a gap in shard ranges. And the gap is caused\n by deleted child shard range which finishes sharding before its\nparent does.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I73918776ed91b19ba3fd6deda2fe4ca2820f4dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec95047339bbf0edf8227a882c14cd8247bc2170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec95047339bbf0edf8227a882c14cd8247bc2170"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
