)]}'
{"id":"openstack%2Fswift~867146","triplet_id":"openstack%2Fswift~stable%2Fwallaby~If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","project":"openstack/swift","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-12-12 23:02:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34175":{"account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"last_update":"2022-12-12 08:47:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34175\u003e replied on the change","reason_account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}}},"hashtags":[],"change_id":"If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b","subject":"py2constraints: pin PasteDeploy version","status":"MERGED","created":"2022-12-09 21:52:07.000000000","updated":"2022-12-13 21:39:12.000000000","submitted":"2022-12-13 21:38:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867146","meta_rev_id":"9324d0e794f83d68c9adf8c8c6093571d8e43a01","_number":867146,"virtual_id_number":867146,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 21:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-12 08:47:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":2,"date":"2022-12-13 20:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},{"value":1,"date":"2022-12-13 20:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 21:52:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"state":"REVIEWER"},{"updated":"2022-12-09 23:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93f738e64b1c7fc01f318c3d58b9decbf4d2e478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-09 21:52:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b2ae11e3ed7491241f7cfac6c57b7c594e1043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 23:54:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4805664baa34a47ac13c411c080b257\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2674b568bd8a451e8550a38a66d0d0fa : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/613a537021004807a3abde4ad4374bd6 : SUCCESS in 1h 06m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d5a80c402fa47ee92a728c26ef38c8e : SUCCESS in 42m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd41e17be3cc4bc8a4f75670ceb4861a : SUCCESS in 1h 02m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/03340501d723441eb59de2ad631cb65d : RETRY_LIMIT in 5m 06s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f42ce11ae4d74ce2b2ee9a620c0c2be4 : RETRY_LIMIT in 4m 45s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/618201f3321c411bb427f79e55148eff : RETRY_LIMIT in 3m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/82f4120ca5d6446d943afabf76c57f3c : FAILURE in 3m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7478957a21d942028e6430ff04c58707 : FAILURE in 3m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71d842328ed1448bb0e170ca4da64baf : SUCCESS in 13m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5eeeedb83507485d869c57cb765b4201 : SUCCESS in 27m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1c33ca23bf5f47478373fc6bb92f54bd : SUCCESS in 28m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/89a8988a5ec14fcc90a7893002a58c80 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef570616d0646c681750f48fb798ba9 : SUCCESS in 31m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4dedad107cc043a4a667faac9a4ca247 : SUCCESS in 31m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33c75faecadc452f8f17bbceed8d1c90 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec04de4cda86437ea3f7e4710ffebe5a : SUCCESS in 36m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3ab2660c56c34bf39d5f1e663c24ee5d : SUCCESS in 33m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/967fb955b39e4dca94cc3255f8d0cf75 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5c97192836c7486686e3459cf0c53678 : SUCCESS in 36m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/200fb3a8b75a4bdc913c43e658650768 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46b8d423a40f45b4882a0590eb61c339 : FAILURE in 6m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88dbb689a22240f598a6623cfa9697b9 : SUCCESS in 51m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/668c66c8a996423cb93011e8092184db : TIMED_OUT in 2h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6ea3785817d4c0a8f36acaf4fbd6841 : SUCCESS in 44m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0cbf0c0ba16c40c4a40cf16e19449799 : SUCCESS in 44m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19bc5588ad5d4e42ac93a17be0ea2b18 : SUCCESS in 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7592cf7a30e4a5a87a88998c5abf929 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8970c1d0d56642d59064f4852eedfa2a : RETRY_LIMIT in 6m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc537f144e53466bb9fd9a6a26988a39 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"43254bfc7c2141c9867db3d3bc037455530b4fb0","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-12-12 08:47:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0be93f1a2c66e94e282efaa209a001717f72e79","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-12 23:02:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dce6cd57f73d79bf20b678c4bf4407d56752910f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 00:15:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a62d1830864d43bfded4dacda0c530\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a53fbdde634d17bab79aadcc0bec99 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c1612fa912ee415c8649ebbc545757b5 : SUCCESS in 1h 05m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23154bebef24486588bc2ac8a9304f0e : SUCCESS in 50m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e785d06a9c784fb78d6d1f78e2eec96c : SUCCESS in 1h 01m 29s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1c23ed2030c14fd0aff2e3dc7fa08ca1 : RETRY_LIMIT in 4m 04s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/649d25d4c9da442c97984168a3add8b5 : RETRY_LIMIT in 4m 10s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/033926f1568644129f1f199ef3e33b6e : RETRY_LIMIT in 4m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7b7030a5e6d40bf98044b3e0277bbca : FAILURE in 3m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ad5139705cf47019c0d8fc88816f553 : FAILURE in 3m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7001deb5dca4b5d9dbc2716645f6a90 : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9212eef44d1b4ddea42291c19c65476f : SUCCESS in 15m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2730ef7a6b1541399ceee95ea6803da0 : SUCCESS in 15m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9411e3c1a9946409fd0bc3217a797b3 : SUCCESS in 13m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/13a83c7d6c7b40589c68b35f890062b8 : SUCCESS in 13m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f29a6410bc2f4aa98d9af8547442d7a5 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c23143872e634c3fb23a65716c5bdb2a : SUCCESS in 15m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/903c919024c240e2a1b5be92a846a692 : SUCCESS in 17m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/588d114700594d649ad35ff675b5566f : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ea0dd67b61604505ad16af3fd94711ba : SUCCESS in 18m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/51b76e584135453486263e33a5b82774 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0ed58af91df4b3b88bb1ba557abc3a3 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24396d1821db4c52a44af6b02ef08ca6 : FAILURE in 6m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53a5d35e60b0489489f30dff73038372 : SUCCESS in 49m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/83a0c6a867ba4c7694c75a528f65be78 : SUCCESS in 51m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c6a4015f14044dba960f7b5b6fc0a87 : SUCCESS in 39m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33781c5fa12d47959ceefc2a6243bfba : SUCCESS in 41m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb987d756a247ba97f288d823d7c551 : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45bbccb5f8d43739347bc0edad9aad3 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a971ce21f26c4364a857bc038f580cdc : RETRY_LIMIT in 5m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dfa1c59d16f4f58a78dcee53eb0b87d : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d8364293efdbf90b71cfde7dfca1861d1de0dfb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-13 20:30:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d04a54507dcd1afe613b08a8620a55394516864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 20:31:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cb5a4d763dfe9a34c622db6691de3369ba65daf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 21:38:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d7836c22a54e3f8fa35f35c79a2b17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98ba304ea7f4788912e6a77bc1343b0 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/3fa9439f4cf046e9b2fbf86f6e778c20 : SUCCESS in 57m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/228d539ca52040699925ca707d91035f : SUCCESS in 41m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7181da19b11b43b8aef516eb1d88d08b : SUCCESS in 56m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/30b1a2eef274480880d1ef140bf96121 : POST_FAILURE in 5m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/857db684055840aaba37dca0bc56e995 : POST_FAILURE in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3559b9e324cf49f6910d1185d24481c8 : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7037d3910ce84a2d8939029b6bfe20d0 : SUCCESS in 11m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f41d74df520d4a3bb8109e18e6e8f4a2 : SUCCESS in 15m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/328696a985254779bec585654caf04f3 : SUCCESS in 13m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8abbcf626c84f2fa4ee2ebdb4314cc9 : SUCCESS in 11m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/502a5bf0a284463eb50262860d03954a : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a899df326294164b38dcd73da050c54 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0789422e10644dfc8b1ea5db077f0b6f : SUCCESS in 16m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a9e9b092677c4d6ebd0897cde9560128 : SUCCESS in 13m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/99fc828b8a9c48908015b99acf368ee8 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/839d853a726c47abb68621f4859428a2 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/28bffd6430554122a834a52eeeb80b71 : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/26ab0d2afbb240d28d5c324150152e96 : SUCCESS in 46m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d61d43942a34ca894faea0a3fa0b3d8 : SUCCESS in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1635953bdb4a4ea7aaecede3dbad2dae : SUCCESS in 26m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/213e7104872d4ca79fee8e5192548af5 : SUCCESS in 36m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0288a539c45b4fa7b85997f558f0680b : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08db9df36c374fce81acbafea0f3e17b : SUCCESS in 4m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89f3a472e93c440da96e945f9a27fe37 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e707ffc34baf8442286c44393de7b130718275fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 21:38:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9324d0e794f83d68c9adf8c8c6093571d8e43a01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 21:39:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d215016f5744adfa136d8a7434df510\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc50eefbe70f4dbdab5de748ad882edc : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/10e5b13b2aeb4607bdd1907e4b140b6a : FAILURE in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/90acfc7b31e441a480d2989f9fe7a7eb : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78533ac353d25714cc2008c2f5ac6d2b4393e01c","revisions":{"78533ac353d25714cc2008c2f5ac6d2b4393e01c":{"kind":"REWORK","_number":1,"created":"2022-12-09 21:52:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/867146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/867146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/867146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/867146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/867146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/867146/1"}}},"commit":{"parents":[{"commit":"6a1a8ce14be8b4625eaf4f2b4d18ddfbdbbb4289","subject":"Stop partial()ing hashlib.new","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a1a8ce14be8b4625eaf4f2b4d18ddfbdbbb4289"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-10-17 08:47:14.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-09 21:51:29.000000000","tz":-480},"subject":"py2constraints: pin PasteDeploy version","message":"py2constraints: pin PasteDeploy version\n\nPasteDeploy version 3.0 (2022-10-16) dropped support for python2 (as\nwell as \u003c3.7), which causes our py2 tests to fail. So cap the version\nhere at 2.1.1, the last which is compatible with our tests.\n\nEven doing this doesn\u0027t stop pip install swift pulling in a newer\nPasteDeploy in the\ntools/playbooks/saio_single_node_setup/make_rings.yaml playbook\n(causing the probes test on CentOS-7 to fail); so handle CentOS 7\nexplicitly.\n\nChange-Id: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n(cherry picked from commit cc033154ad4a4f345258457f3ceed9143fb3d46d)\n(cherry picked from commit d05ddb03648d70e4b8612243bb8cd2ddf8ae6baa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78533ac353d25714cc2008c2f5ac6d2b4393e01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78533ac353d25714cc2008c2f5ac6d2b4393e01c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
