)]}'
{"id":"openstack%2Fswift~867225","triplet_id":"openstack%2Fswift~master~Ie28dfffac8a511f30e7b753beb487e5b30849ae7","project":"openstack/swift","branch":"master","topic":"test_probe","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-01-30 09:05:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie28dfffac8a511f30e7b753beb487e5b30849ae7","subject":"sq: fix probe test","status":"ABANDONED","created":"2022-12-12 13:58:30.000000000","updated":"2023-01-30 09:05:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"37038efc68378023fdd1f18b3100efd6259553cc","_number":867225,"virtual_id_number":867225,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-12 15:02:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2022-12-12 13:58:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2022-12-12 13:58:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 14:55:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-12 15:02:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77555c0576d0896e61392aef7efebabdcb9e90fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-12 13:58:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b513c98a062bc624923f0c41c81bdc6e2e2eaecd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-12 13:58:53.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b30ff35a254bb501a91cd5bdae9d396848d26744","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 14:55:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91feeee3285943648cb912f34b887a3d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/272461f794944a7fb83d453b783229a8 : SUCCESS in 41m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/300105003ea24aeaa009cc36de4fef06 : SUCCESS in 51m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aadfbec6d3694623ac8df31d4306fdd1 : SUCCESS in 55m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/21bf60e3f24f4987a2f36d660246c2b9 : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"49b94a97bd640ec4cc23b6945965ef27564b7960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 15:02:59.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94a2572b9478402fa5c2a197e0ef3444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed7cff2cb22479a9d526942d074e441 : SUCCESS in 8m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab8c537dd257402c93e6d0c3d101ddcd : SUCCESS in 1h 01m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9eb62dfb81b54eed9c52087c64d18362 : SUCCESS in 53m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2882f56e4a624dbc9b8d1d79181852a2 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"37038efc68378023fdd1f18b3100efd6259553cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-30 09:05:35.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c661c083bb6a38144dcf76cacab192809b1396b4","revisions":{"c661c083bb6a38144dcf76cacab192809b1396b4":{"kind":"REWORK","_number":1,"created":"2022-12-12 13:58:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/867225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/867225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/867225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/867225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/867225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/867225/1"}}},"commit":{"parents":[{"commit":"38a90371ac746d329753026207a74991863793e4","subject":"A new probe test for sharder.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a90371ac746d329753026207a74991863793e4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-12-12 13:29:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-12-12 13:57:51.000000000","tz":0},"subject":"sq: fix probe test","message":"sq: fix probe test\n\nChange-Id: Ie28dfffac8a511f30e7b753beb487e5b30849ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c661c083bb6a38144dcf76cacab192809b1396b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c661c083bb6a38144dcf76cacab192809b1396b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
