)]}'
{"id":"openstack%2Fswift~868132","triplet_id":"openstack%2Fswift~stable%2Fzed~I35cade2c46eb6acb66c064cde75d78173f46864c","project":"openstack/swift","branch":"stable/zed","hashtags":[],"change_id":"I35cade2c46eb6acb66c064cde75d78173f46864c","subject":"Extract SwiftHttpProtocol to its own module","status":"MERGED","created":"2022-12-19 23:01:09.000000000","updated":"2022-12-20 01:48:53.000000000","submitted":"2022-12-20 01:47:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868132","meta_rev_id":"ff539bf2552cfc4d370782779c2d99e6f75f9066","_number":868132,"virtual_id_number":868132,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 01:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-20 00:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-20 00:11:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 00:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ce64f78ba3d04f5c6ec18d8bd861628d08d1bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-19 23:01:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ed308b52ba83c1040732f5859f0ec3caf32c08b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 00:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dce4481153f44d2aafed291fc560dad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6231a2d7ebca4c6a9e363b4aa7ec5876 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/2a6a9e6576ab4a8bbc2b65c4e761fe72 : SUCCESS in 1h 02m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9718a6e4f3d740b7a43a6480beec1bb5 : SUCCESS in 38m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87eb6db3da344c9faf0ca469db77f620 : SUCCESS in 58m 48s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f72152d2c9cc455b84710e6aa3402328 : FAILURE in 20m 04s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/26eecaa5aa144aada3d840eaa5e958c0 : FAILURE in 22m 09s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0231938406084c6ca28d5d29e52de4a6 : FAILURE in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f5a5769401948ad8aa5ed57e637873a : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6104d0b070cd49d29bdb1c4e201d8d60 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3f3c4192cdb46629522dbe97dd3630e : SUCCESS in 14m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/539a367c0f9a4638a50ce8cafcd7564e : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/23ef88dde7b54069b714887679b18eee : SUCCESS in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/622c3a66b8644410bd4cf19cb8ab0cf7 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/afa5e3a2152449a789f9710f94701c69 : SUCCESS in 21m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/09b0c45cb0d445e69626f18e38449bff : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e8c1ea6b2d924065924234e561609fc6 : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/486094c5479b49468d89d6d057ce7454 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/857c6468dbde4a6b8bc963ec98b7246c : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0ad334584264650b0644ed08847f043 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9e1860f63664f0c804a05739a1e97bc : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd99c870aa6b4330894ddc7d452b03de : SUCCESS in 6m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7496a4ec336a4a168abaa593d0cead02 : SUCCESS in 48m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/32073cfd78964c08bf60d1077559a0a9 : SUCCESS in 56m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfd7f602033e47ab99b7a6906e3143c3 : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad28d88cd37841c8bb6f521a198a0382 : SUCCESS in 39m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5c6d7d3953841a5874a59082e88cbee : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71148faa06248abb488919684fc9e96 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b37281a27724980b27091d210c1cf54 : FAILURE in 20m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b26ca95c6bc94f1e893768810ddd0540 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"076a1e97e67aba010ce9ce3502042777fa981648","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-20 00:11:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5feb8d93bcb7be2ee67afa6508bfa648cb5d228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 00:12:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bc8f0a3b8531c9fd1804522415d2399132f8cb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 00:18:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d6c0f3b615e40358a02e1469f24b241\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13fae992970a4e698ddade44b6c243c1 : SUCCESS in 53m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/889aafd56687408481531d7628bee671 : SUCCESS in 50m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfca5cbe260448d0baf3dcb8eb42ef18 : SUCCESS in 59m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b2b0d58fbfa4a2cbaa59c366993e924 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1509fca717c8ba161b41836ac3157b9263b0bf7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:47:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83ac58204cc4248b2a2038d7a649d4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ea8f3d2f36475087fedf3994a05746 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/faefcee2a062468cb7562de16fcc68d1 : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad94accd492e4f0cbd7a5cc8d719f000 : SUCCESS in 56m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6becb15165a46d89f5a12cce0777132 : SUCCESS in 57m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e76c79ab5b2c4893a187ea8714d5f153 : SUCCESS in 10m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/be0ef50d076d47c8832d17b167ead066 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d15f4b9504274ffab21cd680aad89844 : SUCCESS in 14m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e513d96142045c7a9976cf3b26861e1 : SUCCESS in 13m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31ebd37208624ff6a3bce5661356287d : SUCCESS in 14m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/94d5273ae3c24a34b0ef6c1b93e4d427 : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6772ca3e1e04fb1b88550fcc4c7c214 : SUCCESS in 20m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/094b46525d0448e184532c3efe58e68f : SUCCESS in 17m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc0850120cfd4c5e9ded4074e159a925 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f83269b367af4991819fcc189bff9c01 : SUCCESS in 13m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f869716576f5455ba26552e8b75bd85a : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bff4a9cda7fc4f9fb0c2a09840b9745d : SUCCESS in 7m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7af2e74c1ee744f7a87d823c8397d843 : SUCCESS in 6m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b9fb3dd3b25645d0a78e096835baa0bc : SUCCESS in 1h 04m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/667e69a8a2d44dd6b95690d7b0440abb : SUCCESS in 55m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31594f491828403f8f5c164164615b38 : SUCCESS in 28m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90fe8854fd5046e0a85f50859c8a17dc : SUCCESS in 38m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/689cb93d338e47c68a3b3e60b6f336f2 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7706de3e4aa461ea62e3cc6a32274c8 : SUCCESS in 5m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cff1ef0dd2349a6ab2ee021e4a1de37 : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c926939916e1a6e81c48510193f978e1b4b08c52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:47:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff539bf2552cfc4d370782779c2d99e6f75f9066","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:48:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4646a3e83a274950899a61699e569f01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a8daecbab645c18712bc0ffe96353f : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e14a4796b0f349e480d040d46781e82a : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2019796425e549aa8e274910a0a8c80c : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f37cf45f7082a8b2c3c015913b9302e41b3699ac","revisions":{"f37cf45f7082a8b2c3c015913b9302e41b3699ac":{"kind":"REWORK","_number":1,"created":"2022-12-19 23:01:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/868132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/868132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/868132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/868132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/868132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/868132/1"}}},"commit":{"parents":[{"commit":"c8662c6f0ef8b9f321234add40263d0947623c09","subject":"Mark rolling-upgrade job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8662c6f0ef8b9f321234add40263d0947623c09"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:15:53.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-19 22:58:07.000000000","tz":-480},"subject":"Extract SwiftHttpProtocol to its own module","message":"Extract SwiftHttpProtocol to its own module\n\nChange-Id: I35cade2c46eb6acb66c064cde75d78173f46864c\n(cherry picked from commit 597887dedcf1f2c855edfd1c591d4f30222be580)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f37cf45f7082a8b2c3c015913b9302e41b3699ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f37cf45f7082a8b2c3c015913b9302e41b3699ac"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
