)]}'
{"id":"openstack%2Fswift~868137","triplet_id":"openstack%2Fswift~stable%2Fyoga~Iae28097668213aa0734837ff21aef83251167d19","project":"openstack/swift","branch":"stable/yoga","hashtags":[],"change_id":"Iae28097668213aa0734837ff21aef83251167d19","subject":"Inline parse_request from cpython","status":"MERGED","created":"2022-12-20 00:13:25.000000000","updated":"2022-12-21 01:51:55.000000000","submitted":"2022-12-21 01:50:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868137","meta_rev_id":"ddfa5b75aae942ca7630ed9135bd79cad9390ee6","_number":868137,"virtual_id_number":868137,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-21 01:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-21 00:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-21 00:40:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 01:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9594607bda0fa5830c67737fa1bf8b4a496512be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-20 00:13:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b29f693f01373f5d7226189140e8e7f01cdf3539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:26:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5ee6eef3afd434f9c9ead0a6fe9995b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d16121f614b4e9e8d2d9e01dddd5e48 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/93122afbdbcd4c86bc8adbf00c2c297c : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/452c9b5d06324f309908932c1a5d9983 : SUCCESS in 1h 01m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/994bb73412204e2789881a2f2c5dc6e3 : SUCCESS in 59m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bc3e6ec2782b4681bc3198ef9a71f7a7 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/66869aa47014402f9856b797bcbdc653 : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f0cfe693040142c291d4d89e04d3329f : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8940185b7a1340b9bbc915fbfdbf97e5 : FAILURE in 3m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02810c8aadeb4863a768a8fb8536a80f : FAILURE in 3m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8f0da8b3abf42b8bbfd806ecefa2397 : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c15a11b33c2478cb132d0ea687e877f : SUCCESS in 16m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb3037e574754d5193d9c9deabe199c5 : SUCCESS in 14m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba8679c64e3c43aeb3ea659070da3fde : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11fb8e5f61dc41a7b562540cd78c869d : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a0908116caac47fe8b0ebfec00a470a7 : SUCCESS in 16m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b3dc1c784e5469087af9ae8c1047530 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/32ac6bc737af4cad9a120dcc2e485792 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4aef65a8006841e7bc0ebfa33ad974b7 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a039c9f3107b43f1b69bef33dfec8660 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bb96ec46f7e4473b0cb5ff0be482f60 : FAILURE in 7m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/637bd30e73cc479a892318b5f1cc740a : SUCCESS in 53m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/85aec9ae9d8e408faff22df6f822ee2b : SUCCESS in 57m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48cc5b78c6eb4ae986b8e04d0176eb39 : SUCCESS in 43m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91cbb46232cb4ca2be53baaa358d7cb7 : SUCCESS in 38m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46217a2e2e0e487daf95eb9c1d37b263 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2a5f017a92477dbe1d50efc55b8f2f : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bdfe5ce904b4525a05916adf34d3205 : FAILURE in 7m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c27af2171f1495cbb7222a1c952f1f1 : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":1},{"id":"29b757d1cf113a2897eaf538eb43d6ecaf70989a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 03:17:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e53f0ba9bf954d32872ca3b0f4e0a3ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3d86fad82ec485f872aa51d35464b15 : SUCCESS in 45m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9e3304a63ba4a86af6378b43c4060ae : SUCCESS in 41m 47s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3de8e49eeba24a8695ba170cd90335a1 : TIMED_OUT in 3h 01m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/80298327a6d1482591c11afaaa134411 : SUCCESS in 1h 05m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b99e7846903f42f686d51784036d7f5d : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1c16469d2c168d25000f9f220ae47412fe1855ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-21 00:40:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"678f6c905c1c2e36b8fc40615e6abfbcc4cffd7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 00:41:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b59863a3832d6a61f579d384b849dcd18ba1d9fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 01:50:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d052b22b5cf94203839fbd6bfd5ea373\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac804bd01ca440acb52eec9d77b13f83 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ad185cc234394a06ade884c6ac816dab : SUCCESS in 1h 03m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9660a6ac42f5466891227749c7b942ba : SUCCESS in 1h 00m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/182f763eedf440ed80c77bac39ab0c05 : SUCCESS in 1h 01m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3b80d905ef354f54ac5d87c74ec9b15d : POST_FAILURE in 5m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4715ac2e9cac4e34b157a37b2e1ed4d4 : POST_FAILURE in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89eb43913f364236a13ec73477e0a2c7 : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/954ad033cd75458386f0a1cb30f82f17 : SUCCESS in 15m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3500bc2a3eff4c39a459ded1e04a6f8f : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f1524ef548394591ba7834449ae2da03 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d86c899a1b844472a913b6abf37fdcae : SUCCESS in 14m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/702dc2213976457499bde274472ca4d2 : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8b7c19d26fd0437e8a1ce0eab6c4f993 : SUCCESS in 11m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9604b9c7b4c41c592abe63fd10e034f : SUCCESS in 11m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b76a598734d2401c943d7755c47c2e7a : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/659ecfc1cf104e5890dde62d8c7e8874 : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1b92d0275864866b0a89128b38a3c45 : SUCCESS in 55m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0417dfc762b645c18dbfae8d6afb6c66 : SUCCESS in 48m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62b8168e730c422baf2b011bb91ef8c7 : SUCCESS in 46m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27a89bdaed4e4f7bb59d67ffae45b6f9 : SUCCESS in 26m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18b72b052a4b418e9823a279a6fa35c2 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df4c8b6a9a14934807a0e93431a68cf : SUCCESS in 4m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65ade28b309e45d68f3b1a7755ca2c7c : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b1ad474d2fe9b431dc63617b7b4a059148cb6a95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 01:50:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ddfa5b75aae942ca7630ed9135bd79cad9390ee6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 01:51:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4128dff05d45eeaec666060730fa06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed03725d6a74d11b77948e6632b2509 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/40d534c49349499093b99a74b4d1bb09 : FAILURE in 25s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8f92bf3a944e4b3680dcb7c94658ed02 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6a85919a579bb194fc290fe719bf03269684de9","revisions":{"a6a85919a579bb194fc290fe719bf03269684de9":{"kind":"REWORK","_number":1,"created":"2022-12-20 00:13:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/868137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/868137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/868137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/868137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/868137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/868137/1"}}},"commit":{"parents":[{"commit":"4d6d6ba510187fd01297170ae50300fa2af75bf7","subject":"Extract SwiftHttpProtocol to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d6d6ba510187fd01297170ae50300fa2af75bf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:21:37.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-20 00:13:11.000000000","tz":-480},"subject":"Inline parse_request from cpython","message":"Inline parse_request from cpython\n\nApplied deltas:\n\n- Fix http.client references\n- Inline HTTPStatus codes\n- Address request line splitting (https://bugs.python.org/issue33973)\n- Special-case py2 header-parsing\n- Address multiple leading slashes in request path\n  (https://github.com/python/cpython/issues/99220)\n\nCloses-Bug: #1999278\nChange-Id: Iae28097668213aa0734837ff21aef83251167d19\n(cherry picked from commit 884f5538f8fb187b6ff18316249f1bd4b97b0952)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6a85919a579bb194fc290fe719bf03269684de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6a85919a579bb194fc290fe719bf03269684de9"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
