)]}'
{"id":"openstack%2Fswift~868138","triplet_id":"openstack%2Fswift~stable%2Fxena~I35cade2c46eb6acb66c064cde75d78173f46864c","project":"openstack/swift","branch":"stable/xena","hashtags":[],"change_id":"I35cade2c46eb6acb66c064cde75d78173f46864c","subject":"Extract SwiftHttpProtocol to its own module","status":"MERGED","created":"2022-12-20 00:14:30.000000000","updated":"2022-12-21 08:27:31.000000000","submitted":"2022-12-21 08:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868138","meta_rev_id":"c976a880da273e0278f6b8322598061ed0c1cad9","_number":868138,"virtual_id_number":868138,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-21 08:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-21 07:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-21 07:12:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 01:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3b191f211cc5f783c35acf4d002f3edfd907f13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-20 00:14:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e56bb6041cdd2d15aead098bd11bb1470f740672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:14:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c335e8424f4d50bfaa6e7dd032379c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea72816b8d7340ba8fae0968f91db3ae : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/0f2f988c377b4b1e875ecdce8fe5e3d7 : SUCCESS in 58m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7d0581c571d47fab5bcf7d1647d9fcb : SUCCESS in 42m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9388156d4b447ae9fcdd9bdb4e43245 : SUCCESS in 41m 09s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bca25e8fe4a5490eafc9bf106d187c11 : FAILURE in 19m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ff8060c73c79436789d0c5bf9c72f14d : FAILURE in 24m 20s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/461ac854e8d24d13bb256be94e9ebfd8 : FAILURE in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b53e0912bacf456385265f983ca49c42 : FAILURE in 3m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ce4af432da74ebcafd0b93eff27187b : FAILURE in 3m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0983b674342a45de8cb6f4fc1a82b7df : SUCCESS in 19m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b19112a61fa14938a92a2d4abba4102a : SUCCESS in 14m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b1bf3586c094f589eaf947eb669cae7 : SUCCESS in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b4f9b50a1cf64002a366c96fc79c1fac : SUCCESS in 16m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c897993967bd466e8d116bc252ca16e0 : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30d0c6dd0dde494a8ef0566c490d9862 : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/610dfff0238740708d59594b1daccd87 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2ea7eb92036457abafdcc95ece41f9e : SUCCESS in 21m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f334abd6dc084dbd9258fabc9dfa459a : SUCCESS in 20m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d8c3e1111944b70a6d67eb8fb0710ba : SUCCESS in 15m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/28f9087676334ffba08193e108951020 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/168d323897724630a05cecad4f4d0b32 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c8383dd3dcf47e48bb74ee26c7832fa : FAILURE in 8m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da8ea301d23544c69f8876e5f38970a9 : SUCCESS in 40m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d16831011624541a404b97063e263a8 : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04c65225144448ddafc5afad4dad357d : SUCCESS in 38m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ecd743b204b48b38519036af79fe3a7 : SUCCESS in 44m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aac0521b93a44483968f7770843dbba7 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/669c975b4bc14288af2367fa4965919b : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebf80434ef4d4242b9d6207f19552aa4 : FAILURE in 6m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/848afb9793f6489390d4fe11885a7396 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cd546cb419564f40e0a98e5b291aac3a21e1827c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 02:34:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b5e5f0c6d440a5bede56d8d95683fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7db1f8549f5b4b9bb790f54fc42fc880 : SUCCESS in 42m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a7f2da1759444968ac22a654e899db6 : SUCCESS in 46m 32s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3e2e2f72be564dfea64789ddcbf0e37e : TIMED_OUT in 2h 01m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9629783ce30343a4adaca18a759c6a66 : SUCCESS in 1h 05m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/201fe38b32cc4328bcb3e664dd5573ca : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"695461146b840cfdf35872aac9c848edf1080dca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-21 07:12:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68c89218dc8833c5d80e089d5ebba598ce2e672d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 07:13:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c8ed9552b565f8e3d2cd6e743476471a7c53c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:26:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/618f408bcc984c4895aed8ecc6d583d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc643b743724498c8fab703a26af080f : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c3e33d4d015b48499c1215c94f7c37d4 : SUCCESS in 1h 06m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7e376acaa2f4601a33e6b48962a97d0 : SUCCESS in 47m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6435a72a83541c3b0a4fba21cea228f : SUCCESS in 54m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8868be9018504739b2f936ba9f0868b7 : POST_FAILURE in 4m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/934d7885e55042f598a477cf6b354ef5 : POST_FAILURE in 4m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d1e3d40e26d4889ae22151d2fe85094 : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5608dbc8fff9466b95e78e8d90d898bb : SUCCESS in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/03276612053b4d42984a5611e910f496 : SUCCESS in 14m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7a947dcc278a4718a1706fe994da4307 : SUCCESS in 14m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e19e3d17df9b46e5a4be46a69da710dd : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7c211d50c1f435391a2e517cb640964 : SUCCESS in 13m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36cc9f6ab8cd4d8ba78b6fa33362c301 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd8a0c190fd7468f813685d12fcc911f : SUCCESS in 14m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/041116f3ce004eb5a359e2f1c03f5927 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59c467367b42455cb672e4a9d82b41b9 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d41033f3418444b7b04390100f417473 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b905a43e3b14b0d8c49de1f773d56bd : SUCCESS in 5m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/70e371e2e1414c76ad7893ef609d5148 : SUCCESS in 51m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a375591cbb09450f8614b1a03d2f171b : SUCCESS in 46m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12a32ca66c4d4462a45490df6313d0a1 : SUCCESS in 30m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71d682da6dd3483889c45c97bdc0c64e : SUCCESS in 36m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74aa494bf68d4a29993acd1b8dafd8c7 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34133a66700c4ab286f7c3136222b361 : SUCCESS in 4m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eced92a7f2404da79e03db7ea93ad632 : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"78f02e1f7efbd43abd3ad9ba9b4c04ea0a2ee460","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:26:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c976a880da273e0278f6b8322598061ed0c1cad9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4aa384d0a53467681caeb92d65f74a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920298b3591b4731928275c94045d84f : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7c91be86af054b388ecb23296ea66560 : FAILURE in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9c4d1f268a9446c193cbfe391d22d169 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75aae4c1b9b582ee0d447b40167fd0dee0c008c9","revisions":{"75aae4c1b9b582ee0d447b40167fd0dee0c008c9":{"kind":"REWORK","_number":1,"created":"2022-12-20 00:14:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/868138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/868138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/868138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/868138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/868138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/868138/1"}}},"commit":{"parents":[{"commit":"9a7562d60b67d786fe3e64151e0b877fd2cd9a3f","subject":"playbooks: replace ansible_ssh_user with ansible_user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a7562d60b67d786fe3e64151e0b877fd2cd9a3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:15:53.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-20 00:14:13.000000000","tz":-480},"subject":"Extract SwiftHttpProtocol to its own module","message":"Extract SwiftHttpProtocol to its own module\n\nChange-Id: I35cade2c46eb6acb66c064cde75d78173f46864c\n(cherry picked from commit 597887dedcf1f2c855edfd1c591d4f30222be580)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75aae4c1b9b582ee0d447b40167fd0dee0c008c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75aae4c1b9b582ee0d447b40167fd0dee0c008c9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
