)]}'
{"id":"openstack%2Fswift~868143","triplet_id":"openstack%2Fswift~stable%2Fvictoria~Iae28097668213aa0734837ff21aef83251167d19","project":"openstack/swift","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-07 00:22:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iae28097668213aa0734837ff21aef83251167d19","subject":"Inline parse_request from cpython","status":"MERGED","created":"2022-12-20 00:16:24.000000000","updated":"2023-01-07 02:36:35.000000000","submitted":"2023-01-07 02:34:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868143","meta_rev_id":"dfa1a4c17b21c3449f0131db1362abcb92ee79a9","_number":868143,"virtual_id_number":868143,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-07 02:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-27 21:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-27 21:25:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 01:34:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b00bedf913b512304bee13c362cf3f5e0ea6bc50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-20 00:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d5716c43b167dfa071e4a325e5b21a46d4a04a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:34:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/439fd86a8407447f812f26b022f1915a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cf4d958fbe430ebf04b473fcdfb449 : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/b8da49ea9afe4ea0a63a17cda0119e7f : SUCCESS in 1h 11m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef8d85deed044773b1dedbfb7e7090fc : SUCCESS in 46m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59c6c01af31c458ebfc3d96efb224f3e : SUCCESS in 55m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a4a9bec727049dfb404d072937fd9b2 : FAILURE in 3m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5c93a8bae591492b9b838a6ea85e5101 : FAILURE in 3m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9467c611d19f40a4b09c96ad346c7f1a : FAILURE in 3m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a89e319c30f14871a568ebb031db990f : SUCCESS in 13m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65964a9d7aa140c4b2bdf04ca01d1da7 : SUCCESS in 12m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1c66af99154a4098983007c76fa0e8 : SUCCESS in 12m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/981bb91b24e14cdb90970a7cd22f44e7 : FAILURE in 3m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eac500b24859424388277df2d0530fa6 : FAILURE in 3m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78aea9c2d33b49608c835164b758cd90 : FAILURE in 3m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5d54d9fca53b43cb93a99810fea0f1a6 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6fe918b1f3e44a01a47e9d1298a55d55 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3e0d14fdd5e348d3a4da9efbbb203f94 : SUCCESS in 17m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a7442e969dea4c92b697d91bd4eaa017 : RETRY_LIMIT in 4m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/395d71bd642a47259964ba55cb70d6bd : RETRY_LIMIT in 3m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3597b5d9e8f4e4f9381a94a6119a6a7 : FAILURE in 15m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a6fbd046860415391595227d86647d3 : FAILURE in 13m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa1132d797b647d08c9bdb0a3a54ada9 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fde844316d541de9a6952c8545ab4d1 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e2b9e9c5daa4cfbaf0401244eff951c : RETRY_LIMIT in 8m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff24ef5395a4eba90ccc5fbd792220d : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":1},{"id":"617e7c366fd07c09a3414f0de99cf30b27b88e3b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-21 23:59:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18c3275d7cb8353e686b81a07dcfe67dc31654ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 01:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/765912d576324af7a634ad01371affba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ad74063a164cd993b6b2981a5b75be : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a8b2c7ae923f478ea620c293ec044665 : SUCCESS in 1h 00m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcb8b549c16a4866b2d3d263b66ca6d9 : SUCCESS in 29m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef7a8a017478472f8248889ede674fd5 : SUCCESS in 41m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00227474afa54d629730fcf8b9f4913f : FAILURE in 3m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/83c56af665ca4e808b10fafb0f1f47ca : FAILURE in 3m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/610d015bfb3248ea93b97de4211ba0b0 : SUCCESS in 11m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/325f3ce9a57c4cb1a4f44f13765a7704 : SUCCESS in 9m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d990ab561c00419eb425c102c9da31bd : SUCCESS in 11m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8515a67ab187456c88ed441feaef705f : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/419b90c5511f49f48259670964b64a2f : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2e7d2ec6c054f6f83e49dcef67c9c42 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1273cb95ae1947379d906dd19521ebb8 : SUCCESS in 14m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a9d79b9420ea4671ad99a62eade917f2 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/12d45d68abbe4788a76dc91a0e01f681 : SUCCESS in 13m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7a665b86181440d7abb6e3bb9728bed0 : SUCCESS in 16m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c3a292dc74546d49113234b17a2f378 : FAILURE in 5m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/833f385ffc314e0d9da6e01cb7aaf8cf : SUCCESS in 38m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6720c3c709ed46308e8fc11a90a804f3 : SUCCESS in 25m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/408a7a82b89b437ea9714a6b43983ab2 : SUCCESS in 27m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72e9dc328e154bffa34c4597716edd73 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9366246cf4314bf49f5fcaeff47d3877 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80c95f3122bd41cab774767a2bbbfdee : RETRY_LIMIT in 5m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/307e79c143b4433ba45282feffcd283f : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6fb56a9fb05b1e6801476f053e3825c22836bd30","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-27 21:25:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b38117f51f5e3f7eb1f1d38997c52b832386e75d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 21:25:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f48bc19fb75699fdc5bf6ef23eb401e52f21c09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 22:38:19.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/868142 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"334a32215497cb85b00440d4268c777ed9876810","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 21:55:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32293cd0df2a15cfb133d7c90b590aeb9d0c5b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:03:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf03070986b9447d8aa80b0c8476a4c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a553995443424dd5b0b5a8af2611c50f : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/bf6542b870924089b97885b7de90f1fd : SUCCESS in 58m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3970613d85c44e64850908318816375b : SUCCESS in 35m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bc1d1a80b554909a613e11b0e2dc9e2 : SUCCESS in 58m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46c5a865f349434aa055f1f1e7da814b : FAILURE in 4m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d01d16ee01034a839c4c2ab23b4d2fa5 : FAILURE in 4m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a409bf20179b40df9f281d99a38536fc : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f8077c4e4bce495791e002f0a96d916c : SUCCESS in 12m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3593942ff434dea98b6b0dfd6e7904e : SUCCESS in 12m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/344d323c1da04098ad7e50f2794630e4 : SUCCESS in 13m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00b8199c79954cb88c5fe04ac3a3e25a : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa36f8ba27bf49a9ac752b5da559d2cc : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/92a28b609a1e45dda0472499c2a28a5f : SUCCESS in 16m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f87fd8352e354e918914ba9690fb6dce : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7c50541d8ecf4468a00ecb7fc44a9411 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b673001ce0040ce8c08eaec0092deee : SUCCESS in 19m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/724a818cb47f43a7ba2feedb41c32bae : FAILURE in 6m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0baaf6fbffe465a86aa52659a3d4531 : SUCCESS in 47m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfcc9ca0a84740d4a61e5b8f05ac9ecb : SUCCESS in 44m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f18e40085f424f31876a79b826d1b2bf : SUCCESS in 45m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8fce6fb92eb42dcbfc4ca159d0962a5 : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ad62b365ed4f759fcdf26892f7699b : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b235cb8e3294fb7bff8a7f9e8984682 : RETRY_LIMIT in 6m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d46511541d4b469db8383e52776afc : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"41ba4dddbcdbb026e372f74fb5ffa3eb2e579f20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:03:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09b549845e58e3ae754d5c5a868494912fd4c9a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 00:26:32.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/868142 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"346c29ceb1c399c2baee2d5325ad8d941e8ff395","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 17:36:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d261d2361edc2088b3b5f57a264fd7a4d2fff68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 18:45:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca8e64ba1c0848308e31cd58aa5bf4f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92346b426b794fb791b9e7177edbd9d5 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/3d0ae2391d1a4ac0a8893c80abce15b3 : SUCCESS in 56m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2c6c3d8c6824e07975c17477a1b6083 : SUCCESS in 29m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5686c397d8524670bcc8094cced5b6cd : SUCCESS in 52m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5693e7d59b734f5a9163ccaef82fa1c6 : FAILURE in 3m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6270444319ac49c98b004d7540cd5ba1 : FAILURE in 3m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d56b55bdc838433ca080914ae42fd984 : SUCCESS in 11m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2021f403138941489fcde7a3def77831 : SUCCESS in 12m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f20dfd60974e4fc8a462ae8d3ad94f47 : SUCCESS in 12m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c3dd12da6024406283e43c98a9f61da8 : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f2df01e2f624eb69f9e541dcb94a8e1 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/370fd49f1d724aa58cca228265d90b81 : FAILURE in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a917bc3dab674714b57e2bfd523ce5bf : SUCCESS in 17m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/015e025de6614b708d327345b77b639e : FAILURE in 13m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/59a2faeee3c543a8a61fc37e9774d0ea : SUCCESS in 19m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/28534bdad5484e9ab24321e0806855d7 : SUCCESS in 18m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65e77ebbbb2a4f24ae142eae4188a311 : FAILURE in 7m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad279fe5716f40978a1159ca069f0463 : SUCCESS in 44m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bda67e4ba45460fac1990bee5a9b732 : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ad4baaff65b4c85bd87a9ac63af3873 : SUCCESS in 37m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f66e33a15d1b4f78bf7df6f22aa00251 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f55090a20d4cc59af7d3994e884321 : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad6b77b8c6cd4cadb3446ac92a0bf496 : RETRY_LIMIT in 6m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46561e4363424d7a813a63afbf71e5d0 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":1},{"id":"778ed4b7d57824f89fbe0aa329e4df034cb2ae6b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 18:58:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e0ec268321165fc32119324207651e2af429614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 20:38:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/784ef856bcf54b3c956f0f1c14ca970b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202f04d06b2d4fb9b6b48ef03d4c1318 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/48bd988f288745a39299ce309a77a7e7 : SUCCESS in 1h 04m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7abe4cc3f3be4309981f4a6797864c6f : SUCCESS in 53m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7aa27dae0154862826ed2b73322d034 : SUCCESS in 1h 31m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f9a0449e35744b7a124141a5cad80c0 : FAILURE in 3m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edeb69ab510d4e5ca240eeeec2d3aad6 : FAILURE in 3m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/53c59e84198947aa8f582b444f42ed57 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5176de834ad14b75b5cdda4346887290 : SUCCESS in 14m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1528fa1dc1d94ad694eedfeeadd853b7 : SUCCESS in 15m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dac4b127066746bb9aa7ade823574e2e : SUCCESS in 14m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/09c369ee9b1941de8f5749d41cc43e42 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b55b0981b31d4334bab050aaf77ee083 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/698738a7a5a14833a21b042cce103afc : SUCCESS in 18m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/57bbb27628cc413ab8f7226412d93bfb : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d409118e941f4bc0a46414669ba905f7 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e0ebb81724d44a2d9a382732b2592bd6 : SUCCESS in 20m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09ca0b6cc9e844ebbb2279c5029d0a26 : FAILURE in 6m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e755539c81214fef9947b0020de53124 : SUCCESS in 47m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e7fb68b4562408a9a965916a97852c2 : SUCCESS in 39m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/443796c3ae7941de88197faeadbe45f5 : SUCCESS in 39m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8f314cc49d445397f3b8a32a8f79bc : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f5e3a6a4950466d81c396149bc1d1c8 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12e410f1edc94f05a48af326f102d337 : RETRY_LIMIT in 6m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba8bb9866a5640569b66b415d54f6d3e : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7c640b7fc0601c8daa2868aa4103ed0fbc268906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 20:39:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dee2ae4697232e3945bfca25fa678dfc2ba2a61e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 21:54:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07a54515e57341a4aead0ff8ed72d51b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c99dcbc3b6f4df4a614bef3b89dde48 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/f42001bfc1bf4776aa386aa93496cebd : SUCCESS in 1h 11m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/deca19344f704b37a820874ee730a2f1 : SUCCESS in 35m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/381063404a8647d19457b5d4c39e8d32 : SUCCESS in 1h 08m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cd2b823863094cbf96744b8c265ac56f : POST_FAILURE in 6m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/283ef0014fb344779a8c03ce9ff0bbfc : POST_FAILURE in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2fa5895b632f4e45aea76d57cdc87ccb : SUCCESS in 13m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5608e4daf254f7c8df934d9cff568ca : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/74dea74d94d94322826bd81117a18896 : SUCCESS in 13m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7d549a5860b54f93a372c22c71745e64 : SUCCESS in 12m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f890a6a68c47468d9694740b803dec1e : FAILURE in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49a1df409cc243a9af4ca1ef0f4a8eec : FAILURE in 16m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d154fed4f47478a8e4b43363d0b25a8 : SUCCESS in 19m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f370c817cc154157a5048a00ccc066b4 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/649232927aa84ae499756241cf14d77e : SUCCESS in 17m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/196ac5b3b0b9466395ee54ad6ded6abe : SUCCESS in 40m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0b080700b7a4496b142345366a4cdf3 : SUCCESS in 47m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d942a13bbd1346da8089eabdfdaba851 : SUCCESS in 37m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d25a61321de64b10bd6bf8a5721bf18d : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76cfc73fb12d442fba0c474fee9b7f18 : SUCCESS in 5m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5958e78089bb4f21a7df6739e7979f13 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"862451df309e1ba131c33fc6e1d535069184afe3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 21:57:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29613f0085dc1892b95ab1b8fa867f4fb750168f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 23:01:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bba44526aa434d8f8de2fc0cd48c75f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22466d47fa4b42a38b9485aecbcfe7e4 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/2f81bfbcd89f4fcea5b4e43051ff6612 : SUCCESS in 45m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7536a5df3bcd482d9d48fa87b747d7af : SUCCESS in 40m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdbb3605825342adb8ce34f67eac21c1 : SUCCESS in 46m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ae4eead3d5341fa98379a6b3898f7d8 : FAILURE in 3m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/25f1a76855624de8bd0cc4099b88a15f : FAILURE in 3m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c093de797c14afb80acc495f9bc123f : SUCCESS in 17m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00b973a1e23b4e5186693f2b078e514a : SUCCESS in 15m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e077d0238ee4530a58b3915ea509bc2 : SUCCESS in 15m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9931f226d37542e2807ade1c73c59b54 : SUCCESS in 17m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1bd7b1a5ac5e4574b2ed79c4ceddebc4 : SUCCESS in 21m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b2f48c581b946f6b69e6f82e8acea34 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03e2d167d7584a8194dc9fa81dd0bd34 : SUCCESS in 22m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9195a309a73a4e2bac75ae9f6305ba9c : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88961cc3f7c84d16b98d5e0aa0825f25 : SUCCESS in 22m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9858046e83b2476a9f0643120d2f3b7f : SUCCESS in 17m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b992d4b70d814478ac9c59cbff7ff884 : FAILURE in 7m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4dc57c014fd7484ba2d865cb997a7c3c : SUCCESS in 49m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed85d3c0a26d4480a948a5a90106eab0 : SUCCESS in 44m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9a39088f9a941c0bbf99cfcfad52989 : SUCCESS in 32m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22c13571040341419969cf0fc6141b78 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2031c0d2ab40f2b3ad322f49d3166a : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29c1006d993c49059d27e244e074e16a : RETRY_LIMIT in 5m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d8340b1171e40b292bc6f56bbed2edd : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0a74e07c8bc1ec9b29df72505ea31da5c44fbf92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 23:02:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f1d43968ee08b607c07a9fb798e81dd201f932a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:21:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fdb7022afef4f8a84cd85e201d39325\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3971b0802190432e9f278730e1c4749a : SUCCESS in 6m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/cb63b8e752964084836fa6a94bac21d0 : SUCCESS in 1h 13m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bf328d7fd4d435eb8aaaa66b4e894cf : SUCCESS in 55m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92c94256542b4ff2b9109c1268607e7c : SUCCESS in 1h 01m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a0632ba734d74fc8871e2e2636b5a2c4 : POST_FAILURE in 5m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/941f25a975e642dfb86e89b948323bd5 : POST_FAILURE in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7aa293cb7df146898e1981ebf2ea5970 : SUCCESS in 11m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac12374a19544f281690a67ef0c65db : SUCCESS in 11m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d34ca8f1dbda4b9b9d06724526710811 : SUCCESS in 13m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ae815c58296d4fc198d4039e22e24501 : SUCCESS in 11m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78e76a3b397843a4a252755d520dec76 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ead872a34e954c72a7af48f16f292bd2 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/016c3a54da904102b92cd3e6668db338 : SUCCESS in 19m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3860c7bfe7ad425ab07bada4cba80513 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cda21fa6c0454f10b7f4194b57e9b261 : SUCCESS in 19m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f2abd6c7fe8e4371aae6e6d137774ad0 : FAILURE in 48m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3d6eed50f5941fbbeff5943f1888d83 : SUCCESS in 42m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e864fe9224d14a78bac273141f1f7678 : SUCCESS in 42m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6083aefcfb4f4e8b9e9867b2458ef61e : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe0a0280b0543b490e8824cb5d8d0d0 : SUCCESS in 6m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d929513b48f4d37972ed48e018d702f : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":1},{"id":"853ba450597e6c98a82b3070c3af7a2f68add338","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 00:22:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cb84124b7e1ca5ad5071d68788a0e4a736cf169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:23:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e08c1be45f424a7986add14a88cc9a17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4f0d45ed874cf785403f5307ba9b50 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/a3477be76501405098c2af56fa701512 : SUCCESS in 51m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53e921819cca4bd28618b95832fe5a35 : SUCCESS in 39m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22fb8fd5e6bd4f18b9a6f961a0e52cbb : SUCCESS in 52m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e12a06c57564150b563b9ff097132d2 : FAILURE in 4m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/39b5ee18c67b417e9b336af159444cf5 : FAILURE in 4m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2280b50ba81f4222be5915f640e3e9ff : SUCCESS in 13m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c578ce0fbf6427c894c9021bf269aaa : SUCCESS in 13m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9581c1833f764936821c9118b08d2f0e : SUCCESS in 13m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fe22960c00e447cca45ce69077a7fabf : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71926d48fe044dec8596754ab1bf0e2e : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f5d5834e7b849d1a92d207d6548844e : SUCCESS in 14m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40d10764b6d74ddbb6a9c5fb6c5e15db : SUCCESS in 15m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cafac97e495244c6aef4866bf51d4ff4 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a699c29a86274d19bf6efdef9b2d938c : SUCCESS in 17m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4998100a510f48549cab7e5df01b84dd : SUCCESS in 18m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c50eb65a32244d5bea47999e7ad8dcb : FAILURE in 8m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/50dab7a9ab0c49e8b7db6bf3baf099e9 : SUCCESS in 41m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e5968a2ddb54129a8186de55f829c43 : SUCCESS in 38m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6afc9c37a2484fa8bfa5b3ad298c5530 : SUCCESS in 39m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcca97097be041ea9a2f89a8cf972c18 : SUCCESS in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af805f6873e8495a86b68c6ceed17c01 : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ea4d000a9c4428bba71a02578e9f7a1 : RETRY_LIMIT in 5m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1588d66b2504bb89576df323c0decc1 : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":1},{"id":"085f9f85b46cf617a51a117ab9ed15a1b3b3c2e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:23:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cb04f1e9748ab76ff2ee9f38d534fff7ebfb591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:34:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120fb46e88ee4f0fa3744bfbee0758f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9be04860a4747c3b56edca39dcc85d4 : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5cb511ada9ed4ab59c42059551dbf570 : SUCCESS in 1h 10m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1969bfd0c58f42ebb0255cc96d120367 : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f40acf6c7e594216a574c93575bcd44c : SUCCESS in 52m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d44d32c81dbd4c6590435cad12602b5a : POST_FAILURE in 5m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ef774227540a4eb08c1ed292d8bd6bd3 : POST_FAILURE in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/010784f1789e41e3acaba41ad6e2c74f : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03f4e138acb34148a6e1970e5a3ff596 : SUCCESS in 12m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff2c6d2e16c547b4a3e9943f50f61202 : SUCCESS in 10m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43146f6f53384cf086555549bd42282b : SUCCESS in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fad4b7013ba94bea80f6fd80e61423cb : SUCCESS in 13m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/234eb7754db34c8b96eea9234bb88e48 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f4b196f82134cfa80f29eaa59c38cdd : SUCCESS in 16m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a1d1a1b703144b08a06a62207dd95f11 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/01a344697bda4e40b9916fce9eee7487 : SUCCESS in 16m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a58dddedf91a43a38966a29d8e9f87dc : SUCCESS in 49m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8d2f589621c43c0b8d7cf3258854f55 : SUCCESS in 37m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5bb921609ac54df8b207fda4158d8637 : SUCCESS in 47m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9eefd218da864a79ae8d04c145728cd5 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338912518ac940619471bac49a00ee12 : SUCCESS in 6m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb97e7376f1a48139263e9fa10d9d83c : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ad1ca9da4deb135ade6250c3b48908d8fc3248a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:34:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfa1a4c17b21c3449f0131db1362abcb92ee79a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:36:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679a4a21ad4a44489bdfbebd34a6ede0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159677c42aec48d382970ee8717dfa9d : SUCCESS in 1m 21s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a9edd0fc60d0489f9752b853f13de21d : FAILURE in 1m 01s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/297f09fa895b46489ae5cf3ab9d3cd62 : FAILURE in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bd18dc20db079dd9312e6c3b7da68aa6b7cc2c7","revisions":{"4bd18dc20db079dd9312e6c3b7da68aa6b7cc2c7":{"kind":"REWORK","_number":1,"created":"2022-12-20 00:16:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/868143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/868143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/868143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/868143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/868143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/868143/1"}}},"commit":{"parents":[{"commit":"40429a6ae75b8bbb8b20709ee69fd4b17e0541b6","subject":"Extract SwiftHttpProtocol to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40429a6ae75b8bbb8b20709ee69fd4b17e0541b6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:21:37.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-20 00:16:15.000000000","tz":-480},"subject":"Inline parse_request from cpython","message":"Inline parse_request from cpython\n\nApplied deltas:\n\n- Fix http.client references\n- Inline HTTPStatus codes\n- Address request line splitting (https://bugs.python.org/issue33973)\n- Special-case py2 header-parsing\n- Address multiple leading slashes in request path\n  (https://github.com/python/cpython/issues/99220)\n\nCloses-Bug: #1999278\nChange-Id: Iae28097668213aa0734837ff21aef83251167d19\n(cherry picked from commit 884f5538f8fb187b6ff18316249f1bd4b97b0952)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bd18dc20db079dd9312e6c3b7da68aa6b7cc2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bd18dc20db079dd9312e6c3b7da68aa6b7cc2c7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
